)]}'
{"id":"openstack%2Fkolla~703075","triplet_id":"openstack%2Fkolla~master~I5f55c892b1ae9eba9b6af8a45ef7dfad515c7b05","project":"openstack/kolla","branch":"master","topic":"add-cinderlib","hashtags":[],"change_id":"I5f55c892b1ae9eba9b6af8a45ef7dfad515c7b05","subject":"Add cinderlib RPM to cinder-volume","status":"MERGED","created":"2020-01-17 13:20:45.000000000","updated":"2020-01-29 09:46:20.000000000","submitted":"2020-01-29 09:42:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"703075-1580290932677-845097a9","meta_rev_id":"de45059e8346165c0c1b15e5ebf855489879954f","_number":703075,"virtual_id_number":703075,"owner":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-29 09:42:12.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-01-23 14:15:17.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"date":"2020-01-23 13:49:42.000000000","_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2020-01-29 08:02:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2020-01-18 19:07:55.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2020-01-29 08:02:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-18 19:07:55.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-01-23 14:15:17.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-01-29 08:02:28.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-01-29 09:42:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7fa11de422974ace88502394bfcf5fa04f7814b5","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2020-01-17 13:20:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e63fd530a96e429d1333d50e97aec91c40655782","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-17 13:45:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/b1564f35b9dd4917bb01b2bbdb6b3b54 : SUCCESS in 24m 04s","accounts_in_message":[],"_revision_number":1},{"id":"4288e4fbe5060a2a60d8f389c89d9ecbcf692885","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 14:20:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/635c7ce102f440b9bcb8a3d090da6ffb : SUCCESS in 59m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"682a5a5f1ca27aa3f38019307341c9b7a9857c44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 16:05:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/04bd65dc26124b908dbf3f6a98f288da : SUCCESS in 34m 08s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/b2c0ff2ba12d4d4abe6111e8cb3581e1 : SUCCESS in 39m 01s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/f9f88bb52c12489998a4da557997a235 : SUCCESS in 7m 29s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/07c205e43ab44f57938fa2594c892d7a : SUCCESS in 8m 04s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/e7e657fffff84c0d9ece5f36ae6177dd : SUCCESS in 1h 31m 48s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/43ead9e7055e46fd8914ffc11b22fb23 : SUCCESS in 1h 34m 28s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/d026a68178fc4c8b9c38635912a3134c : SUCCESS in 1h 46m 43s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/0d850cf6cce74b1290d57240377c8ede : SUCCESS in 1h 00m 06s\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/43d7546373a04558afab401aaac69549 : FAILURE in 1h 44m 16s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/822d52e4a2db475f8219f0604c5bef6b : TIMED_OUT in 2h 40m 11s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/82a92078ab43440f969c5d7cfe3de1bc : SUCCESS in 1h 54m 16s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6406ec42ab394000ba9109ff67f9f50d : SUCCESS in 33m 26s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/8f9ca23e9acf468a885d6a4f99f2e1ab : SUCCESS in 19m 09s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/86c6ace9eafd46fca1acf59e0d865555 : SUCCESS in 54m 09s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d816a52a82734346b6ec651c9c1d7c30 : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d5dff88e2fd48eba982c781109a8b73 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e8687faa351141f8ad2cccd4f3baa795 : SUCCESS in 6m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2da7f887acde496f82f986e0905c899e : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58b497747302470d9745005354f577d6 : SUCCESS in 5m 11s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2902c3251274424281d9f6c2de5ea967 : SUCCESS in 21m 49s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d0ef1f87ba7e493b846361c4eb42a042 : FAILURE in 35m 27s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6be4e288bd384a9889d436ba27b0f8ab : SUCCESS in 54m 06s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f7c05120c28c40d3a98fe6f7a029da18 : SUCCESS in 55m 12s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/bd6e78ebea1a4ead9dc1e554215939bc : SUCCESS in 1h 07m 03s\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/babc735e5958456592f50d3b1278685e : SUCCESS in 1h 20m 27s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/fe8eb796b06e4da0b93b3da620b0ce9e : SUCCESS in 1h 41m 16s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/3c5cd3acc051426887aada8f2d0ba605 : SUCCESS in 1h 20m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"45d12a84849690182897733d48e2a4994a7717c8","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-18 19:07:55.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f3f44de1ae9927f4ca5c2ee5295cd1eb075b7f64","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-18 19:36:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/022691d3cb7b4fc2bc8090577c0a2b70 : SUCCESS in 27m 20s","accounts_in_message":[],"_revision_number":1},{"id":"76f89f303e230f1db7fee504c4ed105dc6540be5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-18 20:12:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/dc91d35cbc24495ba5959210d10a9dce : SUCCESS in 1h 03m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"89aab1ef6edc334c4a33bbc12b6a5a3757ec5a05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-18 21:50:49.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/afafa9e250f94458b476b8fd6c19252d : SUCCESS in 57m 35s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/f6181674b4ba44228446f16faf947f2f : SUCCESS in 1h 04m 23s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/576410d1bea04c2693ac05b212184134 : SUCCESS in 6m 17s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/027f795045084181be6c42913f828148 : SUCCESS in 6m 21s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/9f8ad724ab7c4cec99c00fc7f6b57145 : SUCCESS in 1h 37m 12s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/ed3624aa75454b708cfc99013910a3c9 : SUCCESS in 1h 39m 40s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/73ed413187ab4b69b5cf8dc57b298485 : SUCCESS in 1h 51m 35s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/9edd2d0febd147d38385af060e4d8074 : SUCCESS in 47m 06s\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/ca3902767db8411e8c16f683fd203b38 : FAILURE in 1h 19m 27s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/3805078fd0914ac8814715736525afa0 : SUCCESS in 2h 33m 52s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/748a787c90d9422eaa1396ec530d9372 : TIMED_OUT in 2h 03m 54s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/fabff1d9917743c2b1cbea42c866e3cd : SUCCESS in 51m 31s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/50c48623df2248e79e0434cfada21e84 : SUCCESS in 24m 08s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c061e26ffd65418a9ca0475a17109bf8 : SUCCESS in 1h 00m 06s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5c735aa786344d51a0d7588d2ecc4da2 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2bd1e15df564a048e76371aa8e23cd5 : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c891fca1dd6447ab984d48ec7c33e31 : SUCCESS in 3m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1b773aafa96d46f9a4b4be58b8905bd1 : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36cb2810fbf94d2487a9bf66701bdce6 : SUCCESS in 6m 41s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5ab6a359265042c0b558cb6692bcc3ec : SUCCESS in 32m 27s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f63d98209a8d42c28fc899124139ba68 : FAILURE in 33m 21s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/428752a666f848c39cf33bc7caf8c5d9 : SUCCESS in 58m 02s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2c22e9648c164f75aadd21771da9f2b1 : SUCCESS in 53m 47s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/642daed3dfae48f1ab2f7d0d0bfb556d : FAILURE in 1h 58m 34s\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/18b8a7421d2d4b079bf9f9c400caec51 : SUCCESS in 2h 15m 22s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/acc055e60eaa49f7906fab3244ba2736 : SUCCESS in 1h 57m 24s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/34e8a281229749b9bd456f21350b6fac : TIMED_OUT in 2h 03m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"663d6f021481adcfcbd892caa1f5b7e528bebcab","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2020-01-23 13:49:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6818679ee2242d1abd6bfb9d05b251a5c7286ad7","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-23 14:15:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/ea9b4f7b03cf4a60a8554ff99540d5a9 : SUCCESS in 24m 45s","accounts_in_message":[],"_revision_number":1},{"id":"36bc523f53e0b71ba147e7ccc846c6b0697cd515","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 15:14:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9d5cf702d7644842b9a3304d69976cf3 : SUCCESS in 1h 23m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"30874a9f2515851c24f8a228d425f9dfb96bf1ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 16:24:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/be3966ed3525408592c602685c1e2cbe : SUCCESS in 47m 35s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/9b0f8a955ed44b619e251593c26a5f7b : SUCCESS in 45m 20s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/c013a9c5adc8462b93a63f485dd69c34 : SUCCESS in 6m 41s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/82e39012e1f34e958a583b2718faa144 : SUCCESS in 5m 55s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/d331fa41e37d4869b660ce0b9ee203a3 : SUCCESS in 1h 27m 34s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/b40b305b23de4e8993020743e0c9f96c : SUCCESS in 1h 22m 29s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/6890f748dae54a37a427d737431e4b12 : SUCCESS in 1h 45m 38s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/719c188066ec4f8892451560335cc00a : SUCCESS in 52m 34s\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/5dae45d2f3494842a1ab5745bf406a5b : FAILURE in 2h 25m 42s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/e0e2d9ff049c4f2382755d672ccd8ce0 : SUCCESS in 2h 10m 24s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/f8ecf3d4cea54889a136d943646f4c2b : SUCCESS in 1h 44m 02s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/795f87785ed543b99b7d0e942c7f538f : SUCCESS in 25m 00s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/bd6ef6ca47974a9cbc7c24594479d8d1 : SUCCESS in 22m 29s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/669f28e8426243afbe4d5b792b5a94b1 : SUCCESS in 1h 01m 09s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dc6ecb3a03e84d33b5a3fe3150f3ac63 : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68c1af87a58f4280abd56a8e3740ab33 : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/302a3e2daf0d4133b985f3e7db9f7615 : SUCCESS in 6m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/156b02dc691f4f9a82029e3b7d3bee9e : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/550bf7537bf445648c10475dbd619702 : SUCCESS in 5m 19s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4309c31385764db8b9819d75204f0bb0 : SUCCESS in 22m 15s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/49361403d9ae4bffa4a065bdb0ca8218 : SUCCESS in 28m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/48a266ee331c497193ee0afef45593b3 : SUCCESS in 1h 04m 03s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/57fcae8fde0e432e9124e19118849836 : SUCCESS in 1h 05m 53s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/78e665d40f054a37b1daf900cc67d330 : SUCCESS in 1h 00m 24s\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/342e28de2e7d4e3f9876e23ed103a69e : SUCCESS in 1h 41m 14s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/85c877d0aa6b4ac48207a804d74428de : SUCCESS in 1h 40m 10s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/1e4a00dd85fb4c618e22a046f3d68dde : SUCCESS in 1h 37m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6a9a0f8c5ce97b9ee113b2fa6b287e972724b21d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-01-29 08:02:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f572cdf0e5e6ea6bf67692450618d26c4fa67ee9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 08:02:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f1cc714c73a1fff9d7629b3bcbcf8d09c8cf05b3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 09:42:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"155cb2e1c3c5e6974444ad89c169e23f7a27aac4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 09:42:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/6a54a05a03144745bea69be6441c8b03 : SUCCESS in 44m 19s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/d2f23ae4c72c4031a75388d38d23ddf8 : SUCCESS in 6m 26s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/121ba0834a0843a39ba87885636ea7c7 : SUCCESS in 1h 35m 59s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/3ab6d8d9e70b473fb8537683b9213255 : SUCCESS in 1h 27m 12s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/608dcdd1d0c9475ab0758f648005a436 : SUCCESS in 51m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ed18e61d35ad4290a94de6dfd5c1fd0b : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59994dbca52f43829de8182d94350086 : SUCCESS in 6m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e11ab6a583544df94b77f12eb8a7c3c : SUCCESS in 5m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2bf4929324d3482e88a00098fa88a2a3 : SUCCESS in 6m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72c2ed9bd745429e9a8522b7bb0298a0 : SUCCESS in 6m 28s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f5243c3958714340bb1d7ba4c273dcd9 : SUCCESS in 27m 37s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/423bb9b48aae4589acf13fc6cedd1bc4 : SUCCESS in 55m 28s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/73323457dad74458ad5a93652ac16295 : SUCCESS in 1h 13m 22s","accounts_in_message":[],"_revision_number":1},{"id":"b43520d1d4216ab96f2384f35aaf34827d85664c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 09:46:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9933db5767954f9a97ca0ae7231a4e44 : SUCCESS in 1m 30s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a1b74b76977f8300938c6a6a9adf67af6cd50086","revisions":{"a1b74b76977f8300938c6a6a9adf67af6cd50086":{"kind":"REWORK","_number":1,"created":"2020-01-17 13:20:45.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/75/703075/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/75/703075/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/703075/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/703075/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/703075/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/75/703075/1"}}},"commit":{"parents":[{"commit":"c10f9d7acf2fd3be2ad09c6c148d429d6b4c4268","subject":"Merge \"Revert \"Add AArch64 publisher for Debian/source\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c10f9d7acf2fd3be2ad09c6c148d429d6b4c4268"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2020-01-17 13:15:17.000000000","tz":60},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2020-01-17 13:15:17.000000000","tz":60},"subject":"Add cinderlib RPM to cinder-volume","message":"Add cinderlib RPM to cinder-volume\n\nFinding the root cause of attach/detach operation on volumes can be very\ncomplicated.\n\nThis patch adds the cinderlib RPM to the cinder-volume container, which\ncan, in many scenarios, really help debug these issues on live nodes\n(controller and compute) without affecting our running services.\n\nChange-Id: I5f55c892b1ae9eba9b6af8a45ef7dfad515c7b05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a1b74b76977f8300938c6a6a9adf67af6cd50086"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a1b74b76977f8300938c6a6a9adf67af6cd50086"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
