)]}'
{"id":"openstack%2Fkolla~703078","triplet_id":"openstack%2Fkolla~stable%2Fstein~I77fdbcb905b1a7f2d535c3a37b2899118d1442c5","project":"openstack/kolla","branch":"stable/stein","hashtags":[],"change_id":"I77fdbcb905b1a7f2d535c3a37b2899118d1442c5","subject":"config: make kolla work with oslo.config 7.0.0+","status":"MERGED","created":"2020-01-17 13:34:11.000000000","updated":"2020-01-20 13:34:18.000000000","submitted":"2020-01-20 13:32:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"703078-1579527130166-17819b5d","meta_rev_id":"84b027d143fdeea99b2e45588ee1506bb6e727d0","_number":703078,"virtual_id_number":703078,"owner":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-20 13:32:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2020-01-20 11:02:24.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":-1,"date":"2020-01-20 11:02:24.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-01-20 11:02:24.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-01-20 11:02:29.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-17 13:36:16.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-01-17 14:18:59.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-01-20 11:02:29.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-01-20 13:32:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"204467939805814097fd22f2cc4a453f1f82a76d","author":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"date":"2020-01-17 13:34:11.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"24c6c554b3cb74b6bd15b6c4a33053f6c35a23aa","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-01-17 13:36:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5971e67d11539e63c85ffcaf8c516c4f4435a570","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-17 14:18:59.000000000","message":"Patch Set 1: Code-Review-1\n\nThis should not be necessary - train upper constraints has oslo.config 6.x\n\nhttps://opendev.org/openstack/requirements/raw/branch/stable/train/upper-constraints.txt","accounts_in_message":[],"_revision_number":1},{"id":"e3adc3812280244f79deb3de293e387f65f71abf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-17 14:57:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/b644c058dd4142689a969ad1ca3b103d : SUCCESS in 38m 29s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/622bba56ba6349cc8c0aea1fe87cac8a : SUCCESS in 50m 58s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/7ef3b541650948399630a202a32bb7ed : SUCCESS in 1h 12m 35s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/3ab2103d1ef449188ef61dca55e644ca : SUCCESS in 1h 11m 02s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/85d7a037dfdb49e2be49a3ef246e2fb2 : SUCCESS in 1h 22m 28s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/1e3d154fd6a943dcafb5ca61c7593aba : SUCCESS in 59m 51s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/a5472aab0fd64d549f719bc5e911d82d : FAILURE in 49m 00s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/a2a0ea0b23d74e579a71739f0d6a3aef : SUCCESS in 44m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4e3d76802bb34157980f2f7734b27418 : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e90c7ec739e4ac7a487fb29b255b049 : SUCCESS in 3m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7584759ae2e44f119b30642c689eeb2e : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/32baac8658ba41599f856bf0726d9a5d : SUCCESS in 3m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c5c288390ab043b3ae8cfd26df78f53c : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea391149940042b78136a57c3567bf33 : SUCCESS in 4m 01s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5902ca2179f0461e9170deb7a63bf24e : SUCCESS in 22m 47s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0c8038f683534a23ac2f0579cc210820 : SUCCESS in 26m 45s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/66717f3acf924d54bd0b03f3558c8d92 : SUCCESS in 55m 57s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4c840a5d6b7c41f1896c6ae1d5404c8f : SUCCESS in 48m 57s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/bc9292704e484eb18d1163b789982b07 : SUCCESS in 1h 04m 00s","accounts_in_message":[],"_revision_number":1},{"id":"037836bfdd24300dcdd09be8500c9acd0d2bff2f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-18 19:06:52.000000000","message":"Patch Set 1: Code-Review-1\n\nseems unnecessary","accounts_in_message":[],"_revision_number":1},{"id":"29409a780833cf600833cd3ad5812dc0ecf09a0a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-20 07:41:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nvery necessary, trivialfix","accounts_in_message":[],"_revision_number":1},{"id":"bf63c8b0e13690eb8f30e1e6728bd60e97baf3fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 07:42:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2abe5ec1372b9e7fa6ba4833f19e956f0d63fef5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 09:30:45.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/2a8f28d914cb420e969008969204aef3 : FAILURE in 53m 21s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/52eb9914a11c4c3dbafdcceb869e4a07 : FAILURE in 1h 03m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/66460cc2d45d4cf79408290beca0b16d : SUCCESS in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f00f1360d7db4b919e0b19823d62c704 : SUCCESS in 4m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9c9cd553990d4712958e8e207b5a8efa : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e80f60bbb57049e7a632f6a5c94a4156 : SUCCESS in 6m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/226f56b29d35497ca27640f7d9f0a96a : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/202ec668cc504531bf260849882e8168 : SUCCESS in 4m 30s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/60a4fec0eb94492590e679e6f85d5f54 : FAILURE in 40m 45s","accounts_in_message":[],"_revision_number":1},{"id":"96f2e39b1bb88cbdbc75069b691b089bca5abec1","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-20 09:35:30.000000000","message":"Patch Set 1:\n\nINFO:kolla.common.utils.neutron-base:\u001b[91mERROR: Package \u0027vmware-nsx\u0027 requires a different Python: 2.7.17 not in \u0027\u003e\u003d3.6\u0027","accounts_in_message":[],"_revision_number":1},{"id":"5ff6232bd6b94ef31e9b25ef2f1282c00919dc10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 10:59:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/294985fb6c6b4dffb7bcf56de8d48bff : SUCCESS in 48m 34s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/f0f6e891d866424bbfc637b8721a50cb : FAILURE in 58m 05s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/71ef2b229eaa4b269a10bc9470885fee : FAILURE in 44m 40s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/f89172b5c9894246860a92522bb57358 : SUCCESS in 1h 16m 47s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/f244613832014b2bb09f65a34f6bb43f : FAILURE in 32m 28s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/c761c63ac1c54713946f6d669f41cccd : SUCCESS in 58m 08s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/ee50cf899e7e451c9b4aa19741c59640 : FAILURE in 50m 47s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/3ba22a06019a413ebe26fc81c5661bf6 : FAILURE in 57m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b4ed812b5a254845b8e503bd29b246d3 : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c89857960c024e23b87a485b90ffe68a : SUCCESS in 4m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/52506663e62e40e1bc58174dd2643001 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a980d5a099b40daa36ea549a6674825 : SUCCESS in 3m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7c19911323e941a98c5458dfd1925ebf : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5030cbe91cbb456eb2162cd10cddf202 : SUCCESS in 4m 01s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e518f05c8b0c4e1cbab29207f46f145a : SUCCESS in 29m 41s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/32831003f9064566b61ac175645bc265 : FAILURE in 28m 30s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/789b8c09b5464fe7bc1a56d11b087ce4 : FAILURE in 25m 32s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/06abcc7624614df7af3a24c7a68be01a : SUCCESS in 50m 56s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8993b24f0f594545a18d44f9f5c6a6ac : FAILURE in 27m 38s","accounts_in_message":[],"_revision_number":1},{"id":"87c249d1fd7e7853b03e37e59181319c0ebf8460","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-20 11:02:24.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"52146b87f2e8fa9609c0f44c7b4e043290429c7a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-20 11:02:29.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"610c5655b3a593ac2429528485bae88549d458a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 12:27:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/a5377a17a8d94c3da93da8f287cf768e : SUCCESS in 56m 56s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/9d50e7538b7f45058e2a01cf352028db : SUCCESS in 46m 10s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/336d3d8ec3bc41ca863f541b53087ed8 : SUCCESS in 1h 12m 48s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/015228fd4c4247fd80cb29df3fd14dba : SUCCESS in 1h 14m 07s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/725f8ca05f2b4e0a8f0aadf1caba2722 : SUCCESS in 1h 19m 55s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/167c2370a9024ffc9950ec016277a5c0 : SUCCESS in 53m 20s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/56135c35adb24d8198bef079098afa61 : FAILURE in 50m 46s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/3a333cc335574733933c3eae4a7e13ac : SUCCESS in 57m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/45e628c689f24989a76736294018d6e2 : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74285c4117c74e9783f4adba331b420a : SUCCESS in 5m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/336f52a131b441e78ba3985a68ae6046 : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eb4cf96b9b894109a68199c4cb0e0742 : SUCCESS in 5m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bd2f4db0aa88449496c322659e0eaacf : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/208bd0eca7b143e0b469fe82e058f6f0 : SUCCESS in 5m 51s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/060730387d9943748045bf2719a710e2 : SUCCESS in 34m 59s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/08fc228cc0114fd19a30723cb4814454 : SUCCESS in 40m 39s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/951e4d790b6f47f092e9db3c72b19479 : SUCCESS in 1h 00m 06s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1316d234fa304314bf1a7d6101e2703a : SUCCESS in 58m 24s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a524bfc282f847bba4dd174f8f315e4a : SUCCESS in 1h 06m 48s","accounts_in_message":[],"_revision_number":2},{"id":"0e6c1482aed6bef7e46b6b1d321b1c6a7e2b3a55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 12:28:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cee22c2ceb92a2449994dab4d3cbb689ef1f3589","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 13:32:09.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/82fbacc894e44e73880964e660e83c17 : SUCCESS in 40m 50s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/d163bdf0cb3a4ce49c003af9d7305e74 : SUCCESS in 58m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/31e12fe1ab244d21a6c73af0ff3ec30a : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/715d4db5230044448f9f1e3fdc0184bc : SUCCESS in 4m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7b7cd53d026d4754a97e5b2799902e2e : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef68c52a612940f28789b8c5180f83c5 : SUCCESS in 4m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/33f4f1f32eb94da9b6049d94ebe48bfd : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e02479c2020441b90b4e261547e3083 : SUCCESS in 3m 25s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/079fa5c9bed640ea8acd73b329800cc7 : SUCCESS in 27m 33s","accounts_in_message":[],"_revision_number":2},{"id":"f9594db6896fa51ba3ec283c92b485920b789f62","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 13:32:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"84b027d143fdeea99b2e45588ee1506bb6e727d0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 13:34:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/717ac142011144d2907b469366e2794f : SUCCESS in 1m 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"45497837cdb1e239b3a7a69db632aa3b8b6b4224","revisions":{"6452b60c41d85b05d20b19ae5b3a696a5d922ee8":{"kind":"REWORK","_number":1,"created":"2020-01-17 13:34:11.000000000","uploader":{"_account_id":28374,"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","username":"viktormichalek"},"ref":"refs/changes/78/703078/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/78/703078/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/703078/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/703078/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/703078/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/78/703078/1"}}},"commit":{"parents":[{"commit":"ddc5fea2c685072ec2245b0180889f99ffa4eb83","subject":"Merge \"CI: Fix symlinks for failed build logs\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ddc5fea2c685072ec2245b0180889f99ffa4eb83"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-01-10 12:19:00.000000000","tz":60},"committer":{"name":"Viktor Michalek","email":"viktor.michalek@ultimum.io","date":"2020-01-17 13:34:11.000000000","tz":0},"subject":"config: make kolla work with oslo.config 7.0.0+","message":"config: make kolla work with oslo.config 7.0.0+\n\nFrom oslo.config 7.0.0 release notes:\n\n\u003e Positional options are now required by default, to match argparse’s\n\u003e default behavior. To revert this behavior (and maintain optional\n\u003e positional arguments), you need to explicitly specify positional\u003dTrue,\n\u003e required\u003dFalse as part of the options definition.\n\nSo let\u0027s follow.\n\nChange-Id: I77fdbcb905b1a7f2d535c3a37b2899118d1442c5\n(cherry picked from commit aaa472bfa257aea53abcb3642ad4beeaed1b99a8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6452b60c41d85b05d20b19ae5b3a696a5d922ee8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6452b60c41d85b05d20b19ae5b3a696a5d922ee8"}]},"branch":"refs/heads/stable/stein"},"45497837cdb1e239b3a7a69db632aa3b8b6b4224":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-01-20 11:02:24.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/78/703078/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/78/703078/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/703078/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/703078/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/703078/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/78/703078/2"}}},"commit":{"parents":[{"commit":"8bc3b193642b68c7070a32f6887dfb070c451011","subject":"[stein] Bump versions and pin vmware-nsx","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8bc3b193642b68c7070a32f6887dfb070c451011"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-01-10 12:19:00.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-01-20 11:02:24.000000000","tz":0},"subject":"config: make kolla work with oslo.config 7.0.0+","message":"config: make kolla work with oslo.config 7.0.0+\n\nFrom oslo.config 7.0.0 release notes:\n\n\u003e Positional options are now required by default, to match argparse’s\n\u003e default behavior. To revert this behavior (and maintain optional\n\u003e positional arguments), you need to explicitly specify positional\u003dTrue,\n\u003e required\u003dFalse as part of the options definition.\n\nSo let\u0027s follow.\n\nChange-Id: I77fdbcb905b1a7f2d535c3a37b2899118d1442c5\n(cherry picked from commit aaa472bfa257aea53abcb3642ad4beeaed1b99a8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/45497837cdb1e239b3a7a69db632aa3b8b6b4224"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/45497837cdb1e239b3a7a69db632aa3b8b6b4224"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
