)]}'
{"id":"openstack%2Fkolla~703252","triplet_id":"openstack%2Fkolla~master~I02932857454bdb09b72f539038d394760457109d","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I02932857454bdb09b72f539038d394760457109d","subject":"Revert \"Debian/source: do not force tag in build jobs\"","status":"MERGED","created":"2020-01-18 19:09:03.000000000","updated":"2020-01-20 15:18:41.000000000","submitted":"2020-01-20 15:15:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"703252-1579533343084-9115562a","meta_rev_id":"b19ffa47c8b56528bcc6229535dff6995f79a17a","_number":703252,"virtual_id_number":703252,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-20 15:15:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2020-01-20 09:17:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2020-01-20 08:17:54.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2020-01-20 09:17:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"date":"2020-01-20 08:17:54.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":2,"date":"2020-01-20 08:17:54.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-18 19:09:03.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-01-18 19:16:35.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"},{"updated":"2020-01-18 21:27:08.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-01-20 09:17:41.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-01-20 15:15:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0aa091b29c0edab782cb9f9229cb3594523453a0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-18 19:09:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cda377776c6157247cf610ba359a068db1c68a1e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-18 19:15:09.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"222b2b1b8569fb4e8a559af2781caa49462ca6fb","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-18 19:15:22.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"f56198915a5233bfdea064be69a5d25d7fbba550","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-18 19:17:19.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\ndebian is non-voting so this is the lesser problem - it is blocked on ubuntu bifrost","accounts_in_message":[],"_revision_number":3},{"id":"ab4517dbf16d0e698bd107f8a8ecb9e18465dfec","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-18 19:17:41.000000000","message":"Patch Set 3: -Code-Review -Workflow Review-Priority+2\n\nactions oops","accounts_in_message":[],"_revision_number":3},{"id":"050b5d21c9cd162269e1d6e691922b3f33738326","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-18 20:36:25.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/aff05bd8d3fa429e93fd5288e23ef282 : SUCCESS in 34m 20s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/7fae99b2a86844ccb9b8e37eeb616cbf : SUCCESS in 40m 07s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/23b03bc8ffa74360b837fd60c5ab4b46 : SUCCESS in 4m 42s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/c449d9f34dfb4fd7a9bd1a0a2837feba : SUCCESS in 5m 22s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/993fb663eaba483c855af6958a64aa71 : SUCCESS in 1h 18m 44s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/94cf70ba1f8a480e9a9ccae9861ee0e0 : SUCCESS in 1h 09m 29s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/84ba9563b56e4cb7857da6a2cd2dae6f : SUCCESS in 1h 20m 10s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/29526db825a4491ba987a42167d4335a : SUCCESS in 39m 11s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e8021010797a41b197bc28b8e70a3334 : SUCCESS in 38m 48s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/bc781abb3f08496d9b718762412e5699 : SUCCESS in 59m 58s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5651822042fa45469ef206b5a3ed4998 : SUCCESS in 3m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c28913ef28174d40a368d5c57f705028 : SUCCESS in 3m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e819d92080054eb1a8a0afaa731d8744 : SUCCESS in 2m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bd7409f7d784457c94c66f739c7d3a95 : SUCCESS in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1311cfc879934eaa8aebf0495b148311 : SUCCESS in 6m 27s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1c6d4c3d26cd49bcbd204901045f8dc6 : SUCCESS in 42m 23s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6d71369fe9a64b34af6eaf0195396e59 : FAILURE in 38m 15s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4f8f13379aec463493ae976033f62ce8 : SUCCESS in 55m 54s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1de3fcfacfef42c4ac8103d04e63411c : SUCCESS in 59m 03s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/229691132bad484e866db9ea933a88ab : SUCCESS in 1h 02m 44s","accounts_in_message":[],"_revision_number":3},{"id":"c8ad8747891e9427d146bab844a12e35fce983a5","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-01-18 21:27:08.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"1417f7957bfdcdec2d7bc5dab86801b451fa6d44","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-18 21:28:17.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/040f49e083a6426db7488e8ed523fb04 : SUCCESS in 2h 12m 25s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4e66b626089394e1e6e29996259c9342baad0a5c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-20 08:17:54.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"67208fd98f822b602164d8bd0d86acf9b1b79686","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-01-20 09:17:41.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"bf06470c9ad9ac6cd93d05a0789fb0ad2eab6c39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 10:01:04.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/cb1f9a30b4154ae6b58a9723e5c0f5db : SUCCESS in 31m 30s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/a556763b57e2466bbc995df4e37c58fa : FAILURE in 46m 13s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/26eaf7b8c0584bbeaf8e87789a395b67 : SUCCESS in 6m 25s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/f4fd94422be74239850eff7653c08d23 : SUCCESS in 6m 21s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/b2b5aefe18294e89963d2423baab732d : SUCCESS in 1h 33m 58s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/76eb40499cf04871ad15e006f09fe49c : SUCCESS in 1h 23m 07s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/acfb15c21b02445490eb70060b2f8699 : FAILURE in 1h 39m 13s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/0e27cb4e0f414c6a86180cca9221f7bd : FAILURE in 26m 57s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/f0f6c77397e44d51b2ef4018e8d5a57c : FAILURE in 20m 41s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/62451c851cf74678bc379b16bb4de389 : SUCCESS in 57m 33s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/61e3326d644e459b9d3ab1890369a5ac : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c147edd254b34fd5beb92902e3ab9747 : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd8a4d08b73e4c9c847eae09e7c010cf : SUCCESS in 4m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d9e4fb58017f4f328673363fb5e4ec7d : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1986bc045355462d85e510fadd3c3e40 : SUCCESS in 5m 12s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a9b65fb9d2424f2f908337be557627ee : SUCCESS in 27m 21s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/00aa64cf0c0f42b3beee4c2a09e3ea90 : SUCCESS in 26m 44s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8df895205ec24596b90781c16b094c87 : SUCCESS in 56m 27s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/88a84e28d46e4c19a1d375ade40650d9 : SUCCESS in 54m 42s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/21b3445939814b5e90239fe078bc209d : SUCCESS in 1h 04m 28s","accounts_in_message":[],"_revision_number":4},{"id":"f1a48ffca250668d41dfacb9214ddf04587ae193","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 10:24:17.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d85ba3e36c1a49788c89f1e35093f8bf : FAILURE in 2h 05m 42s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"fb547154caf5238d486f783203917736854c0fff","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-20 10:26:02.000000000","message":"Patch Set 4:\n\nrehceck\n\nrandom issue","accounts_in_message":[],"_revision_number":4},{"id":"ec5661417fe83e43b6ff04eb500ca3a82ee929c5","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-20 11:36:34.000000000","message":"Patch Set 4:\n\nrecheck\n\n(also random typo in recheck)","accounts_in_message":[],"_revision_number":4},{"id":"22f6ee35e694f6e14e3e1868444f4d8178dd0c72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 13:38:17.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/3ca619ce5ef54a47a6d0c145a0753609 : SUCCESS in 34m 09s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/0ba6aa10945d482b8cd29723fe27b099 : SUCCESS in 52m 33s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/bb9280323d884f568b99ffff3c6ca555 : SUCCESS in 7m 03s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/5265916739c147e1a13bec08a1448b3e : SUCCESS in 7m 16s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/14c5176ea29f415990cff4e4c672a3d2 : SUCCESS in 1h 21m 34s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/deddfd278e3c48c5be41144b9ef7d0ad : SUCCESS in 1h 47m 22s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/10e79953f94a432c8c4ade99465fd03f : SUCCESS in 1h 59m 50s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/3ea07864e27346ab8897445f1d738e64 : FAILURE in 35m 15s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6f63ce10758e420eba4986c35b44bead : FAILURE in 33m 22s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/06947ff001334be1bf203474856accc1 : SUCCESS in 1h 14m 40s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e2f94c71075f4c4c87453083b45a7a6c : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ca41a46df37463d8d8fec1f692c9441 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/468e346354a2466a92903b3bee78702c : SUCCESS in 4m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1d7bbeaa2e0c417085e7c859798bef4a : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5de6911484a6479aa06776e0e9d07ed9 : SUCCESS in 6m 20s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4054d69e3ba140c88a535656ab7a98ea : SUCCESS in 27m 47s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3509463b87184eb38dedadfd6e8dcdc0 : SUCCESS in 29m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7537adc375a94cb692e26596a840e5da : SUCCESS in 56m 22s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8ce8f9cdedac421e9b04a485c941aa2e : SUCCESS in 1h 05m 59s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e42df2f677fa49e78ff1432267eebe60 : SUCCESS in 1h 18m 39s","accounts_in_message":[],"_revision_number":4},{"id":"c5a1cbba0503a126b8a546ab4ef1df39f1e8df69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 13:38:29.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"f4753965b2b78bbec4d59f557ef871fe4308df4e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 14:56:52.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2a35f50257d04b85985057e06d7226f8 : TIMED_OUT in 3h 04m 59s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f4cfcf6b42b5fc01e01d1879644a59cfc51acab5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 15:15:42.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/a03d7971f978488a9f3b51d22f6a7ad7 : SUCCESS in 39m 20s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/349e90243d9b49f089a0eadfadaf10f5 : SUCCESS in 5m 38s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/f8e71c90e87f47f584c7239d4fabe7a7 : SUCCESS in 1h 20m 41s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/9acdadaddbe44e1f84767e2821b057dc : SUCCESS in 1h 31m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/44b0b65c486b4c20960d585031eb8354 : SUCCESS in 6m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59cbb37b528b4322870214ac51fa1066 : SUCCESS in 7m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/078a1e0cfa984da49083c81c9c9ca8bd : SUCCESS in 4m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2a46dc6b1dbd40a88fe716f1744a8465 : SUCCESS in 7m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd6f4e79a93e49d695d93dc4fdd9e8b0 : SUCCESS in 5m 38s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9478c0c0ec9e44c094ba49efdfc3c15c : SUCCESS in 33m 50s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c80013fcd85444dcb8d7e087c5b132ae : SUCCESS in 1h 13m 00s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/eda82a6648ca4c259853e0387c00858b : SUCCESS in 1h 06m 11s","accounts_in_message":[],"_revision_number":4},{"id":"63248992fa45a93d8abe631bac51d75653557e00","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 15:15:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"b19ffa47c8b56528bcc6229535dff6995f79a17a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 15:18:41.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca1e4a3818304ad791c251c643708938 : SUCCESS in 2m 17s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"1cb4760b055bcc38380e6741273337c33c0b563c","revisions":{"eeb7a57630d6b5359f2205bcc56f77d0723daed2":{"kind":"REWORK","_number":1,"created":"2020-01-18 19:09:03.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/52/703252/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/52/703252/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/703252/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/703252/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/703252/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/52/703252/1"}}},"commit":{"parents":[{"commit":"ca4da16836e4b3ef1d84059a63378c3f5bb2517e","subject":"Debian/source: do not force tag in build jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ca4da16836e4b3ef1d84059a63378c3f5bb2517e"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-01-18 19:09:03.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-01-18 19:09:03.000000000","tz":0},"subject":"Revert \"Debian/source: do not force tag in build jobs\"","message":"Revert \"Debian/source: do not force tag in build jobs\"\n\nThis reverts commit ca4da16836e4b3ef1d84059a63378c3f5bb2517e.\n\nChange-Id: I02932857454bdb09b72f539038d394760457109d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eeb7a57630d6b5359f2205bcc56f77d0723daed2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eeb7a57630d6b5359f2205bcc56f77d0723daed2"}]},"branch":"refs/heads/master"},"78b41f2aaa2332a9e783e4ac6efd6a1c28091c14":{"kind":"REWORK","_number":2,"created":"2020-01-18 19:15:09.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/52/703252/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/52/703252/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/703252/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/703252/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/703252/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/52/703252/2"}}},"commit":{"parents":[{"commit":"ca4da16836e4b3ef1d84059a63378c3f5bb2517e","subject":"Debian/source: do not force tag in build jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ca4da16836e4b3ef1d84059a63378c3f5bb2517e"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-01-18 19:09:03.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-01-18 19:15:05.000000000","tz":0},"subject":"Revert \"Debian/source: do not force tag in build jobs\"","message":"Revert \"Debian/source: do not force tag in build jobs\"\n\nThis reverts commit ca4da16836e4b3ef1d84059a63378c3f5bb2517e.\n\nThis is adapted to revert only the missed part of\naarch64 publisher which now blocks publishing for debian.\n\nChange-Id: I02932857454bdb09b72f539038d394760457109d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/78b41f2aaa2332a9e783e4ac6efd6a1c28091c14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/78b41f2aaa2332a9e783e4ac6efd6a1c28091c14"}]},"branch":"refs/heads/master"},"4de9f1a12a1eec18ffaeb0602499a84e3a563255":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-01-18 19:15:22.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/52/703252/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/52/703252/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/703252/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/703252/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/703252/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/52/703252/3"}}},"commit":{"parents":[{"commit":"c10f9d7acf2fd3be2ad09c6c148d429d6b4c4268","subject":"Merge \"Revert \"Add AArch64 publisher for Debian/source\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c10f9d7acf2fd3be2ad09c6c148d429d6b4c4268"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-01-18 19:09:03.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-01-18 19:15:22.000000000","tz":0},"subject":"Revert \"Debian/source: do not force tag in build jobs\"","message":"Revert \"Debian/source: do not force tag in build jobs\"\n\nThis reverts commit ca4da16836e4b3ef1d84059a63378c3f5bb2517e.\n\nThis is adapted to revert only the missed part of\naarch64 publisher which now blocks publishing for debian.\n\nChange-Id: I02932857454bdb09b72f539038d394760457109d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4de9f1a12a1eec18ffaeb0602499a84e3a563255"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4de9f1a12a1eec18ffaeb0602499a84e3a563255"}]},"branch":"refs/heads/master"},"1cb4760b055bcc38380e6741273337c33c0b563c":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-01-20 08:17:54.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/52/703252/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/52/703252/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/703252/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/703252/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/703252/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/52/703252/4"}}},"commit":{"parents":[{"commit":"347bd599e597f382b0b6b8ddebedad5a90a67f55","subject":"To-revert: Temporarily disable Ubuntu bifrost image building","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/347bd599e597f382b0b6b8ddebedad5a90a67f55"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-01-18 19:09:03.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-01-20 08:17:54.000000000","tz":0},"subject":"Revert \"Debian/source: do not force tag in build jobs\"","message":"Revert \"Debian/source: do not force tag in build jobs\"\n\nThis reverts commit ca4da16836e4b3ef1d84059a63378c3f5bb2517e.\n\nThis is adapted to revert only the missed part of\naarch64 publisher which now blocks publishing for debian.\n\nChange-Id: I02932857454bdb09b72f539038d394760457109d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1cb4760b055bcc38380e6741273337c33c0b563c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1cb4760b055bcc38380e6741273337c33c0b563c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
