)]}'
{"id":"openstack%2Fkolla~703351","triplet_id":"openstack%2Fkolla~stable%2Fstein~Iacd75261d14dc7d9fe89d0db7502831976a6b894","project":"openstack/kolla","branch":"stable/stein","topic":"bump-versions-and-pin-vmware-nsx","hashtags":[],"change_id":"Iacd75261d14dc7d9fe89d0db7502831976a6b894","subject":"[stein] Bump versions and pin vmware-nsx","status":"MERGED","created":"2020-01-20 09:47:02.000000000","updated":"2020-01-20 12:22:21.000000000","submitted":"2020-01-20 12:20:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"703351-1579522832786-a981c286","meta_rev_id":"4ed394c38979d077ea01bfe0ee7339bd3ef60881","_number":703351,"virtual_id_number":703351,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-20 12:20:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2020-01-20 11:21:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2020-01-20 09:47:57.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2020-01-20 11:21:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-01-20 09:47:14.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"date":"2020-01-20 09:47:22.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-20 09:47:22.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"},{"updated":"2020-01-20 09:47:57.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-01-20 11:21:43.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-01-20 12:20:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7b642c01b67147720a47855fd9113efc395e2e5f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-20 09:47:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0e6e7d240ab7ca0ceeeb1f814b0d20e3ddd67b7f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-20 09:47:14.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"e516729128e57b65eb38860575caa1083121659c","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-01-20 09:47:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"634b74d037367de67fa2800de8ef9b065fd84c06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 11:18:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/f9991c06303c43d1ada1f453f34dd896 : SUCCESS in 41m 16s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/66655a7c323c41fb94608b745e3cdd41 : SUCCESS in 1h 00m 27s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/14910082944443e1b17b91682723507e : SUCCESS in 1h 09m 20s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/8749c60d86544c49b8e5af75c10faa4c : SUCCESS in 1h 12m 00s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/632eb654c1974a4aba3d4039441e7f84 : SUCCESS in 1h 25m 34s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/ae63036505fc489881132c9788056f0d : SUCCESS in 58m 40s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/c50cef13df7942a48cc52b5a881dbfb2 : FAILURE in 45m 21s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/7b43c88ae279410c8aa659476159bbb6 : SUCCESS in 54m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/acb0a9b4c12c45d192da0a694edf5ab9 : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab6b4f00813345b586367adad2eafe31 : SUCCESS in 3m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/77670850bfff4a8cbce9e155d650e850 : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7759a61074d04da0aed115981430ef4a : SUCCESS in 4m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9d8ef68a421146aa9717cec110cda846 : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73d8cbbf16d1491695ab2131154f84ad : SUCCESS in 6m 15s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/05238921110640b79e2e39ceecbd78e3 : SUCCESS in 38m 55s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/42c02c32af5649bf88a0cc09188fc0f2 : SUCCESS in 47m 28s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4d4b15f47eec4836b1bd1fa14cda739c : SUCCESS in 58m 26s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8a5c2df7dce448c0a6d5f73c541e1772 : SUCCESS in 46m 51s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/246e29846aca43238dfdddfee070a513 : SUCCESS in 1h 05m 38s","accounts_in_message":[],"_revision_number":1},{"id":"eb481dfccca0478289bee2e16365d62fdbf26053","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-01-20 11:21:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"50fa326fe812b900a1784d477c387123d2cbc0ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 11:22:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"318a88601f5b49d3abd36436c49fa80f62c9ba0b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 12:20:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/18195685fc38476280450de6d1a94ddc : SUCCESS in 57m 35s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/a72203334e0f42a499cab2337779daa3 : SUCCESS in 46m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b064b946b5044aa79fbf6bbddbd03685 : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/635c60a948e34b1ca2c2f5af920623ab : SUCCESS in 4m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e883559d06d14d959c7eaf15eedb9477 : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b042733167d24a1f80d2ad2b5e7c9ce6 : SUCCESS in 4m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/98aa19af17f34926a886aaf58e1f3528 : SUCCESS in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7257fdb948d7437aa619bf6c3ca4c6db : SUCCESS in 7m 02s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/05c857c139f44b1dbc0332d47a46e3a0 : SUCCESS in 36m 16s","accounts_in_message":[],"_revision_number":1},{"id":"464707e5ec677790faabbaabee44a32b69dcc1ad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 12:20:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"86c3f403c8293b890d80b356fe10c51d1318c927","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 12:22:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac784dff7d6342d19873dfdf7bed1ac5 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8bc3b193642b68c7070a32f6887dfb070c451011","revisions":{"8bc3b193642b68c7070a32f6887dfb070c451011":{"kind":"REWORK","_number":1,"created":"2020-01-20 09:47:02.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/51/703351/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/51/703351/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/703351/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/703351/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/51/703351/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/51/703351/1"}}},"commit":{"parents":[{"commit":"ddc5fea2c685072ec2245b0180889f99ffa4eb83","subject":"Merge \"CI: Fix symlinks for failed build logs\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ddc5fea2c685072ec2245b0180889f99ffa4eb83"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-01-20 09:45:12.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-01-20 09:45:12.000000000","tz":60},"subject":"[stein] Bump versions and pin vmware-nsx","message":"[stein] Bump versions and pin vmware-nsx\n\nFor some reason vmware-nsx releases outside openstack/releases\nkeeping the neutron version in sync.\nSomehow we missed the pin in stein (it is pinned in rocky and train).\n\nChange-Id: Iacd75261d14dc7d9fe89d0db7502831976a6b894\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8bc3b193642b68c7070a32f6887dfb070c451011"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8bc3b193642b68c7070a32f6887dfb070c451011"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
