)]}'
{"id":"openstack%2Fkolla~703357","triplet_id":"openstack%2Fkolla~master~Ie136940c883b15a082e0f538a1c46134edbb10cc","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"Ie136940c883b15a082e0f538a1c46134edbb10cc","subject":"Add note about vmware-nsx to version-check","status":"MERGED","created":"2020-01-20 09:58:35.000000000","updated":"2020-01-21 08:19:57.000000000","submitted":"2020-01-21 08:18:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"703357-1579594701381-c94174cc","meta_rev_id":"c780239d6b7a63417f673b9fec47529cb2a4e527","_number":703357,"virtual_id_number":703357,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-21 08:18:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2020-01-20 11:51:08.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-01-21 06:33:29.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2020-01-21 06:33:29.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-20 11:29:09.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-01-21 06:33:29.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"},{"updated":"2020-01-21 08:18:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"02f0275490e560518454100bb7ceca61cee0b324","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-20 09:58:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d52d852ef48bb346dd3ec4b505d7734a89b5e5db","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-01-20 11:29:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bf26b50623ae4f1d5a46d24dc8cdfe27ec911ff5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 11:42:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/7e240f5ff28c4c2a9ce26dbb22f45cda : SUCCESS in 29m 32s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/e8ee3af7380940f9819c637c406378fe : SUCCESS in 41m 42s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/107a65a505be469888c0690abcd5754c : SUCCESS in 7m 14s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/f71c441a55c541aea20455c9f5e534be : SUCCESS in 6m 33s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/64003952c7e64afdbe5574b39d4aa695 : SUCCESS in 1h 41m 32s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/96e3f804f3b14f1fbf83554a7a30c847 : SUCCESS in 1h 19m 41s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/12c9e18632bd4ded976644a6dabcfe8f : SUCCESS in 1h 35m 08s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d9aeaa9734b94d39aa926ec7e98aee88 : FAILURE in 38m 04s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/830fe7e07c91420283d66c1becb58f8d : FAILURE in 21m 01s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e69c80517ab34aae8adf11524ebe05b5 : SUCCESS in 1h 14m 20s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a04bc5f0437b48919663ae658f405f83 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ead3cd047d34c298f317840f908edf0 : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f45d6b60fdd4991a028b307f7fc7dfb : SUCCESS in 4m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b1e657e5d1fa4f0bb4e480d5aa3d33db : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cce41f04027747f1b7c2f6607b22febf : SUCCESS in 9m 17s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f84a592a52b24562920c6de3f80889cb : SUCCESS in 28m 31s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f2774807a2e04a3aa58d3e2b5ba41239 : FAILURE in 41m 42s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/18523cf204ae45809340c29d71059e05 : SUCCESS in 1h 00m 13s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bb4dc9f02c3f47a2a712ffd2c0d47bda : SUCCESS in 1h 01m 28s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/41d594cd227d4ec1aad4ea808f1de5e4 : SUCCESS in 1h 03m 37s","accounts_in_message":[],"_revision_number":1},{"id":"014854cf04ffbdc7a5293a087de4fefa49d9cfbe","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-20 11:51:08.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"55a4846c1fe9d3d33ad01b6b62e125c329545070","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 13:28:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/b6a89f82a6114f71acd1b178ce094b9e : SUCCESS in 41m 46s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/afec9ef5c0b146d49a673b529178e9fc : SUCCESS in 48m 34s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/c2643d38dd3f49fd8d59ff82a1d56b48 : SUCCESS in 8m 13s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/e4bfb80169354bbda66714770dd47010 : SUCCESS in 5m 33s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/8f93b91e8e2047b68bd72455c5e112f5 : SUCCESS in 1h 23m 25s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/60658dc52a4c4aec856a981eb48c7a34 : SUCCESS in 1h 18m 04s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/ad9fc70bde034582ade4f84e4e995f2b : SUCCESS in 1h 32m 22s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/48011ed515c947e8b924a82b9317d0ae : FAILURE in 33m 20s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/c82408f3a8124dad8619afd25e3472de : FAILURE in 36m 30s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/328901a95f664817826cff88fb8109bd : SUCCESS in 1h 16m 07s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f9d7bb9b6a4d489eb1e802ff47bb6943 : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30d1bf6ccbe1473fb2a71b1afd213ba3 : SUCCESS in 6m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ffbcc1426de43d0b7590f86a1e01664 : SUCCESS in 6m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3c2f3ba3912a4fd38557926a1fe284d4 : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/948cf00f78f5415883bfa6931e7c61f3 : SUCCESS in 6m 49s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7d0e7a86eb8c444bafc66654388582ac : SUCCESS in 50m 00s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/efd83604e7ad421bb81feaf149aaf3f8 : SUCCESS in 52m 17s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/96e60d7fb4c34c1596da61f0930992e9 : SUCCESS in 1h 22m 47s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/475657751b8548f89556dd6b72f5513b : SUCCESS in 1h 07m 23s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/cac6b19105c7474fa3ca9cb936e578a1 : SUCCESS in 1h 08m 51s","accounts_in_message":[],"_revision_number":2},{"id":"40a83e63bae734a84b30835acd7949a2147a4878","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 15:06:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0d35975d76f64540b59d7dfee0aba2d2 : TIMED_OUT in 3h 04m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cf7204a1a74f23b78ebb78a8d14cbbc8d7510ba9","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-01-21 06:33:29.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"973998a92294db3dc40b649a1c6aa76480894fad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 06:33:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"caf02ac2520e9a2b2c863375b18da9640ce46283","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 08:18:21.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/c60799bcbbfa4f6b8fc07e9cd3c9b648 : SUCCESS in 41m 35s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/2b6225b103604f138caf5f673a469ff4 : SUCCESS in 7m 41s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/f7fc7affd5a1403ba9baf563aafab93f : SUCCESS in 1h 42m 11s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/d77b49d2ee0240159bdf0af9cf7a732c : SUCCESS in 1h 39m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2e672592ffa546748c92d2c7c166d6b8 : SUCCESS in 6m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e13a38bf7624542a65c8493f9f4ebc7 : SUCCESS in 7m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc794bf045e44a4895ed103ddaad5d7a : SUCCESS in 6m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e6654317d86840fead20c5d10d106cbf : SUCCESS in 7m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a7781929b3441f7bbad18660f8421e8 : SUCCESS in 8m 46s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/823fb70736d04556859b88907efbefd8 : SUCCESS in 32m 39s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c53a1ae254bd454699be928f84fd3507 : SUCCESS in 57m 44s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1b2668aba7db4094895dfeaefb85e76a : SUCCESS in 1h 14m 15s","accounts_in_message":[],"_revision_number":2},{"id":"2897966c2f0777a86154c08f8851071b8a96b48d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 08:18:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"c780239d6b7a63417f673b9fec47529cb2a4e527","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 08:19:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51bb5e845d31429abac65d5dd6dc9fca : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d9fa85629a8c50ec84c6a2e6268c085a64791fba","revisions":{"5ad134718269a0f0e90939733fec9f4bc802b67b":{"kind":"REWORK","_number":1,"created":"2020-01-20 09:58:35.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/57/703357/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/703357/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/703357/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/703357/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/703357/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/703357/1"}}},"commit":{"parents":[{"commit":"c10f9d7acf2fd3be2ad09c6c148d429d6b4c4268","subject":"Merge \"Revert \"Add AArch64 publisher for Debian/source\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c10f9d7acf2fd3be2ad09c6c148d429d6b4c4268"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-01-20 09:57:57.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-01-20 09:57:57.000000000","tz":60},"subject":"Add note about vmware-nsx to version-check","message":"Add note about vmware-nsx to version-check\n\nChange-Id: Ie136940c883b15a082e0f538a1c46134edbb10cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5ad134718269a0f0e90939733fec9f4bc802b67b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5ad134718269a0f0e90939733fec9f4bc802b67b"}]},"branch":"refs/heads/master"},"d9fa85629a8c50ec84c6a2e6268c085a64791fba":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-01-20 11:51:08.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/57/703357/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/703357/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/703357/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/703357/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/703357/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/703357/2"}}},"commit":{"parents":[{"commit":"347bd599e597f382b0b6b8ddebedad5a90a67f55","subject":"To-revert: Temporarily disable Ubuntu bifrost image building","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/347bd599e597f382b0b6b8ddebedad5a90a67f55"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-01-20 09:57:57.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-01-20 11:51:08.000000000","tz":0},"subject":"Add note about vmware-nsx to version-check","message":"Add note about vmware-nsx to version-check\n\nChange-Id: Ie136940c883b15a082e0f538a1c46134edbb10cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d9fa85629a8c50ec84c6a2e6268c085a64791fba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d9fa85629a8c50ec84c6a2e6268c085a64791fba"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
