)]}'
{"id":"openstack%2Fkolla~703486","triplet_id":"openstack%2Fkolla~stable%2Fstein~Idddf51730821644948f51577807f5beaa18c6262","project":"openstack/kolla","branch":"stable/stein","topic":"ci-unb0rk-failed-logs-stable/train-stable/stein","hashtags":[],"change_id":"Idddf51730821644948f51577807f5beaa18c6262","subject":"CI: Fix symlinks to failed logs","status":"MERGED","created":"2020-01-20 19:22:30.000000000","updated":"2020-01-22 15:38:52.000000000","submitted":"2020-01-22 15:36:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"703486-1579707419326-641ca554","meta_rev_id":"a6384fc909ecbc3f20d49f003ac670809733235e","_number":703486,"virtual_id_number":703486,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-22 15:36:59.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2020-01-22 13:17:45.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-01-21 05:28:40.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2020-01-22 13:17:45.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-21 05:28:40.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"},{"updated":"2020-01-22 13:17:45.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-01-22 15:36:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e863af664e00ad3e027e77a1eb35e3f4618a1132","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-20 19:22:30.000000000","message":"Patch Set 1: Cherry Picked from branch stable/train.","accounts_in_message":[],"_revision_number":1},{"id":"185864ff81e288ed773c7839478e5ab0ade1710e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 21:00:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/c409c1b7930e4bf2bd2a1bd2475bd1de : SUCCESS in 1h 00m 16s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/646cb45c0cf144b6818701655fdea951 : SUCCESS in 49m 53s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/dc1658493cfa4dc3855a9b98476acdcc : SUCCESS in 1h 19m 56s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/4b119ce8583940c4a308c8ff71f9082d : SUCCESS in 1h 08m 36s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/c126f1fd11514ea8b9e9588a912cdcd3 : SUCCESS in 1h 25m 44s\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/1b3c9d9e32ad487d9ddf49b29dc0568c : FAILURE in 54m 34s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/2d408ed1b2574c1881bdd75e64886f90 : SUCCESS in 47m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d44c7cd224d441e3938e3747cb5c1e38 : SUCCESS in 6m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc38061a1dc6470999e8b4938c7f08bb : SUCCESS in 3m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/58b9f70ca54544f1a7240570993b2ab7 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0cc8df9da37f4dac85d6515af77ca0af : SUCCESS in 5m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4bb632ed512943898b54374c2dfbbbb0 : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df54d2d7e31a46718a12812039780e41 : SUCCESS in 4m 33s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3ab0c717d51142afb805b913f2990027 : SUCCESS in 24m 13s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c3dc2e35d3c342ef931ee0b42956bf80 : SUCCESS in 35m 38s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e03c13a17d6748a5a01ea6397cb86eb7 : SUCCESS in 55m 09s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7987b3f1bcc14ef3b9c85feb270c5044 : SUCCESS in 49m 47s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3fe85e5e3d294de280e5733df1b46630 : SUCCESS in 1h 03m 54s","accounts_in_message":[],"_revision_number":1},{"id":"9c7599d92dab1b8079ff8d75a8e46662ef9792d7","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-01-21 05:28:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3331338d33b18333d9c4132122d515c879b84cd7","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-01-22 13:17:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"86144251cfa9da4ffdd93f2c172ffe725bf626b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 13:18:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0d419ad1c45fec2d8ba11b67fea230159bc295fa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 15:36:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"68fd47e490922838985b6362247b0968dd2305b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 15:36:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/0408e4cf6a36461f94b9bc2019e5ce2c : SUCCESS in 50m 53s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/be8fd913247d4a5eac4e2f3d428865b9 : SUCCESS in 44m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bb2e9dc9a9e640bebbf43174eec7cbd2 : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0339554180c45a8a84aea25047eec3a : SUCCESS in 4m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/394ca04c810f4a71a9fb2f48dd44a923 : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/75683ae957594c07a21f3254ec6b301f : SUCCESS in 5m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7d21adbb24364c6681c4eb8b66851196 : SUCCESS in 8m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d5b378f1be8413f8000ed456aa55d70 : SUCCESS in 5m 18s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2b84c79551b546ff9381dc32d797e524 : SUCCESS in 29m 26s","accounts_in_message":[],"_revision_number":1},{"id":"300ff42823cc3d70f73a701987c2c7f842d5b833","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 15:38:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f55db94affbc435d81709de44c2f65cf : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"067770667cf62195580a8406dda1219d6a709f57","revisions":{"067770667cf62195580a8406dda1219d6a709f57":{"kind":"REWORK","_number":1,"created":"2020-01-20 19:22:30.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/86/703486/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/86/703486/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/703486/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/703486/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/703486/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/86/703486/1"}}},"commit":{"parents":[{"commit":"45497837cdb1e239b3a7a69db632aa3b8b6b4224","subject":"config: make kolla work with oslo.config 7.0.0+","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/45497837cdb1e239b3a7a69db632aa3b8b6b4224"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-01-15 09:54:18.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-01-20 19:22:30.000000000","tz":0},"subject":"CI: Fix symlinks to failed logs","message":"CI: Fix symlinks to failed logs\n\nStop renaming files. Swift should be able to serve text/plain files\nbased on content, not extension.\n\nFailed logs symlinks were broken because we stopped gzipping log\nfiles and improperly rewrote the logic there.\n\nChange-Id: Idddf51730821644948f51577807f5beaa18c6262\n(cherry picked from commit 4a0182f15acbabc88df0aa660abce81d61f528da)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/067770667cf62195580a8406dda1219d6a709f57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/067770667cf62195580a8406dda1219d6a709f57"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
