)]}'
{"id":"openstack%2Fkolla~703492","triplet_id":"openstack%2Fkolla~stable%2Ftrain~I8fa3fcad88b35743e042d8a62ccee722b08af673","project":"openstack/kolla","branch":"stable/train","topic":"bug/1859607-stable/train","hashtags":[],"change_id":"I8fa3fcad88b35743e042d8a62ccee722b08af673","subject":"Fixes swift-object-expirer for Debian and Ubuntu binary images.","status":"MERGED","created":"2020-01-20 20:13:05.000000000","updated":"2020-01-22 15:12:44.000000000","submitted":"2020-01-22 15:09:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"703492-1579705756498-641ca554","meta_rev_id":"0977da2bc523036a5d6127355f7ffc1ac5f7ab1d","_number":703492,"virtual_id_number":703492,"owner":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-22 15:09:16.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2020-01-22 13:17:08.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-01-21 07:16:52.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2020-01-22 13:17:05.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-21 07:16:52.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-01-22 13:17:05.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-01-22 15:09:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"16afba0332e5d34e3acf979b5997fd43c275a82c","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-01-20 20:13:05.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"7c7b62598b8e2e91573675956e3a31bcc2ba93fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 21:40:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/18260f5dab2c48c4803c39a30f768147 : SUCCESS in 45m 42s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/1d8c37493bb84b1aa9b24be0576eca07 : SUCCESS in 55m 55s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/d80297c3bf3d487896bbb989f4c18e11 : SUCCESS in 1h 21m 22s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/c61924fd53ee45a8a050ec0f9774f391 : SUCCESS in 1h 21m 53s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/de3a4dcba18542f8bcda3ddf251143eb : SUCCESS in 1h 26m 00s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/001fd5a2cdea41a58b9f5e4c94aaebe5 : SUCCESS in 53m 34s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/bb8c1db9a36047df913931b3af537f7a : SUCCESS in 38m 18s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a5d586dfefa04ba49277def8c9357abc : SUCCESS in 35m 20s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/37470d1520814d579dc3e2b902ae99d1 : SUCCESS in 1h 02m 41s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5a9941a4d6de4bafab13cae090b4ec8e : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5150d21c1c884b5d8fbbd090e2893b45 : SUCCESS in 4m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b628aa351d9c4d589ad9d590d36f0a7d : SUCCESS in 3m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4637976001d24933bdb8aa6726fa3e1c : SUCCESS in 4m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a70e2acc324845548e65af563815d5a0 : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16c539bf65294fb99e6c58b133591b29 : SUCCESS in 6m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/62aa241ce6854746b60374be4eed3ed2 : SUCCESS in 5m 00s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/88da983f704f4c26b96f6d7d6bbf2c99 : SUCCESS in 40m 31s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/614731bc5cca4d2d999330cfe1bd7f63 : SUCCESS in 41m 36s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/182d45b8a6404fd7a7750f0d1b59b0c8 : SUCCESS in 57m 31s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d43231ff25da426ab5272c765c9c88e8 : SUCCESS in 1h 22m 10s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5c137a34996e45bea349fe45383484a7 : SUCCESS in 1h 13m 09s","accounts_in_message":[],"_revision_number":1},{"id":"5c683a4f7f3cc7a6d47bd94c6e20eab884c77d5f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-21 07:16:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9412b5d0088f821fae52cd1006fdc03ac7d723b2","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-01-22 13:17:05.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a2855bbbec3d7d1041103d1c7aae0421ade2e5b6","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-01-22 13:17:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d90d72f01423a95afa39403e139ec21a1e9cc25b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 13:17:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fdf85c7c3c4730872e44b514f8e786bce3d2883b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 15:09:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ebb429f79a8d45d89d334a05a6edfbe782ae0436","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 15:09:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/4b70f4f4249b4438a2bcc0ed61645861 : SUCCESS in 48m 24s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/d034c6e003264190afda0ff2ec52c704 : SUCCESS in 1h 20m 21s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/d14b7ef2ad4447c6a17d6357d5a02199 : SUCCESS in 1h 44m 38s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/53cc6ba3094e4a9fa2042794860de134 : SUCCESS in 1h 00m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/20a0acb5c8654ef3842e74741b9d51b8 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99e6ae4ade424ee9b2ed2a833ea79bd5 : SUCCESS in 4m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ce40c88f299c49b3a1eed5e34252a217 : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bee021b348ca49e2ad65548145ef4d84 : SUCCESS in 3m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0b520383343e472899bf408f3ff3eb25 : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cab5e47b76e4a748d664ee5006a128e : SUCCESS in 6m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/73ef71e55924478980fbabd300e3cacb : SUCCESS in 5m 12s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6d89700c6522403fbab21942110adee8 : SUCCESS in 33m 22s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f2bddd7164c546e6a53979a6af66c52a : SUCCESS in 58m 14s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/431809a6a1f3471c901c4e630026cf74 : SUCCESS in 1h 02m 49s","accounts_in_message":[],"_revision_number":1},{"id":"6fbf564cb6a2bce2e07d06662e6af14eb7f2f299","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 15:12:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14e7c7c49d2643e2a6a4a15f1856418f : SUCCESS in 2m 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1e2aca8849a24e448a8c46e98cd05b7c : SUCCESS in 2m 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6632c5cf52d10e714468198b188dd9975dcba36f","revisions":{"6632c5cf52d10e714468198b188dd9975dcba36f":{"kind":"REWORK","_number":1,"created":"2020-01-20 20:13:05.000000000","uploader":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"ref":"refs/changes/92/703492/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/92/703492/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/703492/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/703492/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/703492/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/92/703492/1"}}},"commit":{"parents":[{"commit":"9adfb291184ec30a0aeaf2f4b792735e02980b00","subject":"config: make kolla work with oslo.config 7.0.0+","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9adfb291184ec30a0aeaf2f4b792735e02980b00"}]}],"author":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2020-01-19 19:47:10.000000000","tz":180},"committer":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2020-01-20 20:13:05.000000000","tz":0},"subject":"Fixes swift-object-expirer for Debian and Ubuntu binary images.","message":"Fixes swift-object-expirer for Debian and Ubuntu binary images.\n\nReplaces \u0027swift-object\u0027 package with \u0027swift-object-expirer\u0027 package.\n\nChange-Id: I8fa3fcad88b35743e042d8a62ccee722b08af673\nCloses-Bug: #1859607\n(cherry picked from commit 190a6f7a264c1ea8c0c565d2b75fb73fdc29d500)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6632c5cf52d10e714468198b188dd9975dcba36f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6632c5cf52d10e714468198b188dd9975dcba36f"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
