)]}'
{"id":"openstack%2Fkolla~703493","triplet_id":"openstack%2Fkolla~stable%2Fstein~I8fa3fcad88b35743e042d8a62ccee722b08af673","project":"openstack/kolla","branch":"stable/stein","topic":"bug/1859607-stable/stein","hashtags":[],"change_id":"I8fa3fcad88b35743e042d8a62ccee722b08af673","subject":"Fixes swift-object-expirer for Debian and Ubuntu binary images.","status":"MERGED","created":"2020-01-20 20:13:23.000000000","updated":"2020-01-22 15:11:52.000000000","submitted":"2020-01-22 15:09:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"703493-1579705757665-04bc3352","meta_rev_id":"0cc34896521e5878f5df37011008ca66003b517f","_number":703493,"virtual_id_number":703493,"owner":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-22 15:09:17.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2020-01-22 13:17:15.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-01-21 07:16:48.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2020-01-22 13:17:15.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-21 07:16:48.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-01-22 13:17:15.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-01-22 15:09:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2d89c7bcfafc0507ee471d95a11249621cbd699c","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-01-20 20:13:23.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"e4c13569cc5411ab000a63bbd96e81dae9b20d35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 21:53:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/86c8e4a358364a1db8664377d3a33243 : SUCCESS in 39m 20s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/62fdb9e3012643eb9c63a79b6bf77633 : SUCCESS in 53m 38s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/1914f48eb5d349c5969d23208a7f26d9 : SUCCESS in 1h 39m 18s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/698f7e547acc47339d99d68722a194d2 : SUCCESS in 1h 33m 19s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/a937994f5421437289fe8773e14b37c4 : SUCCESS in 1h 22m 32s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/5c32d951ad8949eeb6780af0a9a67f47 : SUCCESS in 1h 01m 39s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/c1b07bdedac8477d81bd375860216d9d : FAILURE in 49m 31s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/746c3c39164b46dcb499a1ac4ae9041d : SUCCESS in 54m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bf6259a21509419eb1ce0399e68c7d7b : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b550026130e042728be30f3f4dc226b3 : SUCCESS in 3m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bfd2121d45064a35924d674030b1f238 : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4bea2a474613474c9e8393b482541d67 : SUCCESS in 4m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2f737246958b4f87907fa3b2c59e1218 : SUCCESS in 3m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eed232631f13493a8e183fdb17956101 : SUCCESS in 6m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6af263a2f1b0451aae235be8c571c974 : SUCCESS in 5m 00s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/811c4c7f302649c1b45e9c71c9a890ca : SUCCESS in 24m 12s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/aa1ba3f689a64c94bf27ae85bdccf047 : SUCCESS in 31m 29s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/de73257302674a328e95ca0f6765151a : SUCCESS in 54m 44s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/775991de1703450d93c09540cc9e0d71 : SUCCESS in 54m 39s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9f37abbad1fb4f2e8815fb7366b1490b : SUCCESS in 1h 00m 29s","accounts_in_message":[],"_revision_number":1},{"id":"185bdfa00c26e1c7a99344cd559ffdfd73b023ea","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-21 07:16:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"819a9c0276c9030d831d0bcb6132ffd37318a3f7","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-01-22 13:17:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"263ffafb5277abb4c2534ac137c86a58af47e4fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 13:17:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"20c7e9a76b2d050e0977d99c37c9a7bf2cb3ed23","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 15:09:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"bfd48a5b96c3371f03464f646193839ab04553fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 15:09:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/5b0efe4dd0bf4b5a9360da0470cd5847 : SUCCESS in 45m 05s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/167b2519c01f4b2487a70aa063189ddb : SUCCESS in 51m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4122854312a746c5b02793d55d3e98ef : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f78e0d90141437bb360317da55487ca : SUCCESS in 3m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/480ecd016dca48008e39883a1aee60d3 : SUCCESS in 3m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f080dd748f12427ea03274b9b829f7f8 : SUCCESS in 3m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/acdbe04742c542089c102efcdef55c46 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/caa3561574254bbf8e464e700d4e1458 : SUCCESS in 4m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ef855cf7f034fae93c5ada9d79c7e52 : SUCCESS in 4m 59s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/48eec0c3bb3d45eba1678a727e61e06c : SUCCESS in 38m 59s","accounts_in_message":[],"_revision_number":1},{"id":"21badcc1437e4dac1f9a5fd5d77223c6c9ec067f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 15:11:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/969dce8e255741119131d20f1cda8e33 : SUCCESS in 2m 07s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/44ab6c8a9b874a9a8f1bbebf51e4f65b : SUCCESS in 1m 27s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d9c63d7b81848677c8aa7bc4a6c9c3c0189ae8fd","revisions":{"d9c63d7b81848677c8aa7bc4a6c9c3c0189ae8fd":{"kind":"REWORK","_number":1,"created":"2020-01-20 20:13:23.000000000","uploader":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"ref":"refs/changes/93/703493/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/93/703493/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/703493/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/703493/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/93/703493/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/93/703493/1"}}},"commit":{"parents":[{"commit":"45497837cdb1e239b3a7a69db632aa3b8b6b4224","subject":"config: make kolla work with oslo.config 7.0.0+","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/45497837cdb1e239b3a7a69db632aa3b8b6b4224"}]}],"author":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2020-01-19 19:47:10.000000000","tz":180},"committer":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2020-01-20 20:13:23.000000000","tz":0},"subject":"Fixes swift-object-expirer for Debian and Ubuntu binary images.","message":"Fixes swift-object-expirer for Debian and Ubuntu binary images.\n\nReplaces \u0027swift-object\u0027 package with \u0027swift-object-expirer\u0027 package.\n\nChange-Id: I8fa3fcad88b35743e042d8a62ccee722b08af673\nCloses-Bug: #1859607\n(cherry picked from commit 190a6f7a264c1ea8c0c565d2b75fb73fdc29d500)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d9c63d7b81848677c8aa7bc4a6c9c3c0189ae8fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d9c63d7b81848677c8aa7bc4a6c9c3c0189ae8fd"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
