)]}'
{"id":"openstack%2Fkolla~703610","triplet_id":"openstack%2Fkolla~stable%2Ftrain~Iaeb34f0c66dd3b25b0fa300d8e838ddc9704acf2","project":"openstack/kolla","branch":"stable/train","hashtags":[],"change_id":"Iaeb34f0c66dd3b25b0fa300d8e838ddc9704acf2","subject":"[train] Bump versions","status":"MERGED","created":"2020-01-21 13:21:41.000000000","updated":"2020-01-22 12:47:28.000000000","submitted":"2020-01-22 12:45:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"703610-1579697138257-090dce9f","meta_rev_id":"f6668cc1b054dec785409d36b844516fa49b5d74","_number":703610,"virtual_id_number":703610,"owner":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-22 12:45:38.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"all":[{"value":2,"date":"2020-01-22 10:55:45.000000000","_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":2,"date":"2020-01-21 17:59:21.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"all":[{"value":1,"date":"2020-01-22 10:55:45.000000000","_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-21 17:59:21.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-01-22 10:55:45.000000000","updated_by":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"reviewer":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"state":"REVIEWER"},{"updated":"2020-01-22 12:45:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"39c8e2002a896223a330d8d5ea7ed914c9e2d62a","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-01-21 13:21:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b27d52e57512a1af18938c3260fe42bb1c84dc52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 14:56:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/45a55c84f7504115b4b57a966cce01d3 : SUCCESS in 45m 56s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/a970119117d547678dd16110936969e6 : SUCCESS in 44m 34s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/90f2caf88c21442eb682506abd037afe : SUCCESS in 1h 18m 41s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/098f32f27d65469d9cdafc0c3085f059 : SUCCESS in 1h 17m 47s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/4ce8d7d08f3d4450a66940952ee9d9fb : SUCCESS in 1h 28m 14s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/dace0a9d92494ca1b0c3f720a1638de2 : SUCCESS in 43m 07s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/c701e6c3a6c1412a9e4d8d25f7c20459 : SUCCESS in 30m 49s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/1b8b924d4bf8466a96b359f1b2532872 : SUCCESS in 25m 57s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7f237fe0eabe49d8a4dff4b0ee8fede9 : SUCCESS in 1h 03m 55s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/35e0a33a0f08480eaf8206dbcf32dc5b : SUCCESS in 6m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3fd9038084343d9ac50dd8b26a48cf7 : SUCCESS in 5m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8f8b56de525e4f15b6c06310bc2d5e7b : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b3f5789b04f649eeb0b78457204a07bd : SUCCESS in 5m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/af051c7a05c944488696d6005c7e2a8f : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33ea0159bd7447eb8ee3e1ca010dbcd1 : SUCCESS in 10m 20s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d702f524324b47c6b1fa97854b6ae516 : SUCCESS in 28m 40s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6a8be1a6bd1e4fe8b0bf481b1683d50c : SUCCESS in 32m 50s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/cc971848f9b34786bd5e754a524414b4 : SUCCESS in 1h 03m 13s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cf54c061ee86498bb307b086e0dc4333 : SUCCESS in 57m 51s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d0bf80d8e14941e79469d43d220cb99e : SUCCESS in 1h 05m 42s","accounts_in_message":[],"_revision_number":1},{"id":"1d0e7c2ec55c1b8d68d4cfb9e6c72b4ec59246f7","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-21 17:59:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6aaba51cb924988d415d4f59e102755b0caf1501","author":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"date":"2020-01-22 10:55:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dec7b537811db8532b2ac8b2812982d002b3962a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 10:55:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"11ef60334cae8531390779bfa91a9a26d2c64c1f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 12:45:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/eb648d6a960f4c29a079b0d9e7c511ae : SUCCESS in 1h 03m 50s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/06a364abc06f4d409499f9902de667bf : SUCCESS in 1h 36m 06s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/09a6953fadcb44e19a29b19d00262bef : SUCCESS in 1h 45m 21s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/05ac325415684740a8047bc304315e08 : SUCCESS in 57m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a04898f19b664ef69deda66aa9d1764e : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7c51c0b5a9b4dfaa0d10cb1a9622d62 : SUCCESS in 6m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0686f088082f4c92b6203899035752f9 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a60533ac75845f7a1b4a0c31c90c607 : SUCCESS in 4m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a51b30ef0786470cadc55d0b64431c28 : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afac1482c15b43c58ca8e3081448b315 : SUCCESS in 6m 08s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/41b3efe5d4e345dd8ad7ff468194a4fb : SUCCESS in 28m 44s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0dd1f8f9ddfb4c73b635d50d95729072 : SUCCESS in 1h 01m 29s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f2077451836149c185618370fe498a74 : SUCCESS in 1h 03m 09s","accounts_in_message":[],"_revision_number":1},{"id":"4d87dc4f51047d7c1cd728aecf425059fc0f103d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 12:45:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f6668cc1b054dec785409d36b844516fa49b5d74","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 12:47:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b5e6e30b239408f8dc36818e740e634 : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cfe3780322905af79c03a185e30af4f496d02602","revisions":{"cfe3780322905af79c03a185e30af4f496d02602":{"kind":"REWORK","_number":1,"created":"2020-01-21 13:21:41.000000000","uploader":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"ref":"refs/changes/10/703610/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/703610/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/703610/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/703610/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/703610/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/703610/1"}}},"commit":{"parents":[{"commit":"9adfb291184ec30a0aeaf2f4b792735e02980b00","subject":"config: make kolla work with oslo.config 7.0.0+","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9adfb291184ec30a0aeaf2f4b792735e02980b00"}]}],"author":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2020-01-21 13:20:41.000000000","tz":180},"committer":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2020-01-21 13:20:41.000000000","tz":180},"subject":"[train] Bump versions","message":"[train] Bump versions\n\nChange-Id: Iaeb34f0c66dd3b25b0fa300d8e838ddc9704acf2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cfe3780322905af79c03a185e30af4f496d02602"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cfe3780322905af79c03a185e30af4f496d02602"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
