)]}'
{"id":"openstack%2Fkolla~703611","triplet_id":"openstack%2Fkolla~stable%2Fstein~I0a524bf7147446f7604ec3555fe926e7f251d736","project":"openstack/kolla","branch":"stable/stein","hashtags":[],"change_id":"I0a524bf7147446f7604ec3555fe926e7f251d736","subject":"[stein] Bump versions","status":"MERGED","created":"2020-01-21 13:23:08.000000000","updated":"2020-01-22 12:37:57.000000000","submitted":"2020-01-22 12:36:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"703611-1579696576964-090dce9f","meta_rev_id":"96ac0f85fa1483388adb2b05d8fdc50073f01b33","_number":703611,"virtual_id_number":703611,"owner":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"date":"2020-01-21 16:31:35.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-22 12:36:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"all":[{"value":2,"date":"2020-01-22 10:55:24.000000000","_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":2,"date":"2020-01-21 18:00:29.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"all":[{"value":1,"date":"2020-01-22 10:55:24.000000000","_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-21 18:00:29.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-01-22 10:55:24.000000000","updated_by":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"reviewer":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"state":"REVIEWER"},{"updated":"2020-01-22 12:36:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2dd4633df2b85c5f905822249f545b1e0176288d","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-01-21 13:23:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"937be125131d5a7c411f5c1ea9c9b104f9843932","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 15:00:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/6889e214d6dd469ea36b40fea842f71c : SUCCESS in 45m 55s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/2c0adfa088ef4bc2b4efe8f7b062608b : SUCCESS in 1h 00m 03s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/dcf49297a5bc4ef59411559dfa3c708c : FAILURE in 51m 33s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/2ad8e022f03048cf9d4fd09a075fa01b : SUCCESS in 1h 24m 56s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/81a990bacaae47a1bb57278862191767 : SUCCESS in 1h 19m 19s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/4f3f518a66e0464f8a0d197fe5d459b8 : SUCCESS in 58m 26s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/919301edadee4db486a34d6552e95e4b : FAILURE in 54m 17s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/e254828a466647eab7fc3d8d25d6d918 : SUCCESS in 53m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/81680adbdc00481c8fbc301a5f8cca0f : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88e39dde93ac422ab97a196596f329eb : SUCCESS in 4m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ee923f574c6c4674a70f2bf542f16c05 : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e341fd41b6f345dd9fee6d558b1d2d66 : SUCCESS in 4m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6ade3d9e5bc94e6c82382425e2a7b6b8 : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b886feedfe64a229c9f7872241180e0 : SUCCESS in 4m 00s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e171be0186b442639912ba70cfa1f7bd : SUCCESS in 38m 46s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9f2539a241fb4bb7af9831ba67d3a60c : SUCCESS in 42m 27s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/711af644bda74441b8df34e388088e9a : SUCCESS in 55m 41s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/906c53b0de6543ed8ee3399c22465d5c : SUCCESS in 48m 31s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0aa99a5f6bd14bdabeebbd94cfcfc3d8 : SUCCESS in 1h 02m 49s","accounts_in_message":[],"_revision_number":1},{"id":"801bf6855d568543fc11b0dc54be095671c764e8","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-01-21 16:31:35.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"19cc2c910ef7b20fda2bc40e83ee3e30a989e828","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-21 18:00:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"96bd73894b719abf2497dd899106de4cb0d9be2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 18:22:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/2edd6a13c53a461c9563e1f260d40606 : SUCCESS in 58m 39s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/aa176fc85b094a6bbd1b8e24bead6010 : SUCCESS in 57m 06s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/5cac7e3a273f44cba2d77056089446f7 : SUCCESS in 1h 14m 21s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/b052198913614c30b453e834bd28ee41 : SUCCESS in 1h 23m 54s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/d337d12342b449aebff9e560df594fe0 : SUCCESS in 1h 46m 53s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/275ec6dd02a8448ca4098ad22ecacbae : SUCCESS in 53m 52s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/1050a2ddc94c492e88bc4adaf33fec80 : FAILURE in 53m 11s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/1b21fd0538594a9b82ceb742d28cc680 : SUCCESS in 51m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/30b7e151d1ec4ae08c74e337bb32f899 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa48b35d9474451bbb1981cf0778696b : SUCCESS in 6m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e0f1405347574167b4cc6abce7fde4dc : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f7b99970479b425ca6b97b0265a5f2a9 : SUCCESS in 4m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3d69236845624d4d8d7289ff8829a6fc : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/630b50bd7f17465abbdc538df36b422e : SUCCESS in 5m 38s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c0e48a4e7a3b40b5bcffa24849e9cc8f : SUCCESS in 27m 01s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/711e89d1893f42f387be09bfcdd78cb3 : SUCCESS in 36m 36s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/991a8f35428f499e81acdc52c4f445a5 : SUCCESS in 51m 38s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/08f4d41c28084b8788e7a1ab2ea7e243 : SUCCESS in 49m 06s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/07f95aa034454649aebe503be053090d : SUCCESS in 1h 11m 43s","accounts_in_message":[],"_revision_number":1},{"id":"a49955234abc1a38fe18e5ef6c4b1a7771932ffe","author":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"date":"2020-01-22 10:55:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"10352f966223c6a64c9962b23e9e9d1c653d1317","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 10:55:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3611eb03c0b3212aae4db1ac968d7c9a533da40a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 12:36:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/e5cb94c6ac3f40b4869bb04750f8459e : SUCCESS in 1h 06m 00s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/4b526ed2a91046568f572c837b4b53d0 : SUCCESS in 1h 01m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/41ff08a4898a4eb18526c475df1bd076 : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad4dbc4fc77c4b8c8d6a5854f040c87c : SUCCESS in 4m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5c2a2ef019724d739630d6fad4b42c8a : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aeeab3d82f8a410487a002c2f79f91fd : SUCCESS in 4m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/676b0108624b4283b19cea8b0b064168 : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a10cd9bb10046ee8ba55b3b7280df7c : SUCCESS in 4m 39s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/83d8d1a03f184f1d9bb1117f8c2637a8 : SUCCESS in 30m 55s","accounts_in_message":[],"_revision_number":1},{"id":"9568cf4e50698ded8801fe1450e46b7bdd68fb1b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 12:36:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"96ac0f85fa1483388adb2b05d8fdc50073f01b33","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 12:37:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a0f526427ed49349e7bf8fb1c26e301 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0de137f86684122fd6d78a4349877f9437fc7f2d","revisions":{"0de137f86684122fd6d78a4349877f9437fc7f2d":{"kind":"REWORK","_number":1,"created":"2020-01-21 13:23:08.000000000","uploader":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"ref":"refs/changes/11/703611/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/11/703611/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/703611/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/703611/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/703611/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/11/703611/1"}}},"commit":{"parents":[{"commit":"45497837cdb1e239b3a7a69db632aa3b8b6b4224","subject":"config: make kolla work with oslo.config 7.0.0+","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/45497837cdb1e239b3a7a69db632aa3b8b6b4224"}]}],"author":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2020-01-21 13:22:44.000000000","tz":180},"committer":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2020-01-21 13:22:44.000000000","tz":180},"subject":"[stein] Bump versions","message":"[stein] Bump versions\n\nChange-Id: I0a524bf7147446f7604ec3555fe926e7f251d736\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0de137f86684122fd6d78a4349877f9437fc7f2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0de137f86684122fd6d78a4349877f9437fc7f2d"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
