)]}'
{"id":"openstack%2Fkolla~703640","triplet_id":"openstack%2Fkolla~master~I7fae5c78719597da58d43529868c7880136c5ccd","project":"openstack/kolla","branch":"master","topic":"centos8-containers","hashtags":[],"change_id":"I7fae5c78719597da58d43529868c7880136c5ccd","subject":"[WIP] centos8: remove rabbitmq pinning","status":"ABANDONED","created":"2020-01-21 15:22:45.000000000","updated":"2020-04-07 07:14:57.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1cbf4d6660277ec85858f7f2a89e1c922e607815","_number":703640,"virtual_id_number":703640,"owner":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-02-06 22:38:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-02-06 22:51:10.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},{"value":-2,"date":"2020-04-04 07:38:57.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-21 18:05:01.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2020-01-21 18:16:09.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-01-22 09:15:03.000000000","updated_by":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"reviewer":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"state":"REVIEWER"},{"updated":"2020-01-28 14:42:51.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-02-06 22:38:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-06 22:51:10.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-04 07:38:57.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"}],"messages":[{"id":"7a385fa0edcfa07a75f17128d276531f9894dde5","author":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"date":"2020-01-21 15:22:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f092ad5d1bd9682100dd68f1b6a88f424f854b9a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-21 15:31:43.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/805846df694e458580c61ccf4fde9195 : FAILURE in 6m 08s","accounts_in_message":[],"_revision_number":1},{"id":"ba91c3234a6eb750dec8f12fe5fa96dc291303a6","author":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"date":"2020-01-21 16:39:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a814824b17e8168ec0c076d7c9134e406892822f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-21 16:55:40.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/ddc134621fb246b28979e756537484c9 : FAILURE in 7m 04s","accounts_in_message":[],"_revision_number":2},{"id":"d4216fb4bce24b6244255abfa61dafd81cc30a4f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-21 17:55:37.000000000","message":"Patch Set 2:\n\nmixes two things; btw, how urgent is this? We might prefer to merge https://review.opendev.org/#/c/694799 and https://review.opendev.org/#/c/700877","accounts_in_message":[],"_revision_number":2},{"id":"0725046a49ef943db8b118f62f83d4d027693dd5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 18:00:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/150a71ca5e37498096b5d1561a3ce09e : SUCCESS in 1h 18m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c6342f477816e2701ce87d19ff7cb86e5b76504c","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-01-21 18:05:01.000000000","message":"Patch Set 2:\n\n@Radoslow we can continue w/o merging this and wait for the other two patches to land.","accounts_in_message":[],"_revision_number":2},{"id":"0152ff6720be1f7f551916d1907bd18788ecf231","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-21 18:16:09.000000000","message":"Patch Set 2:\n\nack, thanks, Wes","accounts_in_message":[],"_revision_number":2},{"id":"81fabf6955e9d37cc34def44bd6b07770cfa083d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 19:24:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/99d42c083e47496abb531a0771f8eb15 : SUCCESS in 38m 22s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/0acd8a5ffd37405ca464ff76c1c2398a : SUCCESS in 56m 25s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/d8e5cb0130614fe4ab76445987841fb0 : SUCCESS in 20m 40s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/d497d10d799d48b9a2f93a296d6eac21 : SUCCESS in 31m 25s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/01ebf49255eb4bc39f2e5356af419ef3 : SUCCESS in 1h 36m 43s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/be4197c8c00942fe8921d77915d4ee2f : SUCCESS in 1h 30m 22s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/e10e8d623d714932bb8b22de9abf4ebb : SUCCESS in 1h 40m 48s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/482eb03e330446e99513dd4d6b058874 : SUCCESS in 46m 17s\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/8c18ed5e48c345f899a3850ed2860c75 : SUCCESS in 1h 56m 57s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/091d659fe24645d59aad433c20d0c8df : TIMED_OUT in 2h 33m 34s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/04c65d10dab740a2890bc37013735d7a : SUCCESS in 29m 49s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/20458dcdaa3a41baad2eff947cf8fee0 : SUCCESS in 41m 02s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e9cb288f860c4dd8b7e3e7f9a1d5efb9 : SUCCESS in 54m 53s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7618e7afae0e4a728bb5f5764d7b1dbd : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17d6f8c0ac9241cc9c660e3e4611ef96 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ab065fdb17dd4a56a9b05b5ac5e16c54 : SUCCESS in 6m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/61c6929261824bdca0b696b65b29e695 : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9602bd60e924653bcf91de0d0008738 : SUCCESS in 6m 21s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1001e32ac45a4ae69e240a5aac638c35 : SUCCESS in 25m 23s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8100fd7d63014251ae5547f199626a7b : SUCCESS in 26m 50s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/fbe577a259cd48f5b4bcbc0adf5e2cec : SUCCESS in 57m 51s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b733c23d85aa4bf0b73009955ecbb578 : SUCCESS in 53m 39s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ab43bab8da2849d88b01947098ecda48 : SUCCESS in 1h 05m 06s\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/7e88812a36504b03b826d4f7687ce43c : SUCCESS in 1h 35m 13s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/9f408a7a8f934b869527dd6e87b2d042 : FAILURE in 1h 24m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bcd5f1cfc8a85b57ec355e0a723e5140825046b1","author":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"date":"2020-01-22 09:15:03.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"bdf9cbbc48ccb4da737173c3b297924d1422c3dc","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-01-28 14:42:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"03002664dcc6c4326b424c21c5e72ae9c6291190","author":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"date":"2020-02-06 20:31:32.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2dbaaeb7730315deaeaa45ef23497592e7c7ce85","author":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"date":"2020-02-06 20:32:52.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"73beaeecd7031028c1d7b5ec0f023a52ce49ae54","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 20:33:06.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"6d47fe1c9acd6f33ec372e9dbed4fefd68edf5da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 20:33:07.000000000","message":"Patch Set 3: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"e1fac74d48942a2b5350cb630fa680ad4520dcbf","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-06 20:33:08.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"d3fa034486875b8b390c4bb47bf2b0d50a7f58fe","author":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"date":"2020-02-06 20:33:15.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"2bcc7581780ea77d280a63c6ba8ab948daa78e88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 22:38:48.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/26a061a28c594a3f9c492d16a958555d : SUCCESS in 42m 01s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/91441817db574be6bfab308061d695f2 : SUCCESS in 46m 30s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/59dacc70363540748676e25f4445c9a1 : SUCCESS in 32m 22s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/4ba7b6ac2de94ce883353cbc886da137 : SUCCESS in 33m 02s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/de0936191c3b48a38a55db45ae2d161e : SUCCESS in 1h 24m 22s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/7b556e399dc64d04ae839a0cce96c06f : SUCCESS in 1h 24m 31s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/2ee88f7503f7429d92f39c9eb1a94c75 : SUCCESS in 1h 35m 45s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/4e03824c77be4feeb8eb9634b20a06f6 : SUCCESS in 1h 01m 24s\n- kolla-ansible-centos-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/e86b1ee471ae450ca196c5de5c7e3a63 : SUCCESS in 1h 49m 04s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/1f61b4595d34488c95ebefc92b07b5ba : SUCCESS in 2h 02m 23s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/82dd0377de0f459b9da7c04a0c3e397f : SUCCESS in 26m 55s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/c9461aaf009d4be3b8f36b89f0c1360b : SUCCESS in 23m 01s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/6f50ba9d12084d568eb9b1ca090beb90 : SUCCESS in 58m 47s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0d5a5069506842c997b062e47d014f6e : SUCCESS in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/904548e3db674c66abd2f8f914f68b70 : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a338741058c34bf1b1e681cd8dc673df : SUCCESS in 5m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/64ef80c20d7142a1ad4032e209e21da1 : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c91ab2a9e5de4c8588d9f235f421c05c : SUCCESS in 3m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5d4648c96d744eaa794155d848040a7 : SUCCESS in 5m 34s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/dd75de27d1d6421d8ae3a573609a590e : SUCCESS in 24m 41s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e1fd51cb8ed94b45892967f23362ff37 : SUCCESS in 26m 45s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/09d404ee230c4009b5d10c7100a4b2a2 : FAILURE in 50m 22s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/14b28206eee444a3bca4333eff179681 : SUCCESS in 57m 03s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/37bc131983d243d1a89f38a888b9c73b : SUCCESS in 1h 42m 02s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/6df11bba6a8649438007a9a76886c968 : SUCCESS in 1h 25m 23s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/d5d5ec5f893940cbae97884f0ac91585 : SUCCESS in 1h 36m 47s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a8d6cdceb1c1f306a831e5281c7372725823e9f0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-06 22:51:10.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/a521e3d3051143c684d7193ba22579f2 : SUCCESS in 26m 35s","accounts_in_message":[],"_revision_number":4},{"id":"11915ae815a8e420ae9c120eff6c2a335acec242","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 23:16:14.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f693d6bfb8c44989aac2ff0714788b27 : SUCCESS in 2h 31m 37s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c887e8d18be7a88d4b8f64e6bcc1c6d17f6e7e71","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-04-04 07:38:57.000000000","message":"Patch Set 4: Code-Review-2","accounts_in_message":[],"_revision_number":4},{"id":"484f13cb231a2b1ee97d67fb06860c0c9dcd0d3a","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-04-07 07:14:57.000000000","message":"Abandoned\n\nhttps://review.opendev.org/#/c/694799/ has been merged.","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"03ba99f038062c1f6d91bb26123272360e58c20e","revisions":{"e94a91f032b5c92787251c9e261858bfcee7526f":{"kind":"REWORK","_number":1,"created":"2020-01-21 15:22:45.000000000","uploader":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"ref":"refs/changes/40/703640/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/40/703640/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/703640/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/703640/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/703640/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/40/703640/1"}}},"commit":{"parents":[{"commit":"d1304acb318454ddf0410d3f41d2c4f2d00b2c3d","subject":"CentOS 8: Update packages in images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d1304acb318454ddf0410d3f41d2c4f2d00b2c3d"}]}],"author":{"name":"Rafael Folco","email":"rfolco@redhat.com","date":"2020-01-21 15:19:55.000000000","tz":-180},"committer":{"name":"Rafael Folco","email":"rfolco@redhat.com","date":"2020-01-21 15:19:55.000000000","tz":-180},"subject":"Remove rabbitmq pinning for CentOS8","message":"Remove rabbitmq pinning for CentOS8\n\nCentOS8 provides rabbitmq 3.8+\n\nChange-Id: I7fae5c78719597da58d43529868c7880136c5ccd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e94a91f032b5c92787251c9e261858bfcee7526f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e94a91f032b5c92787251c9e261858bfcee7526f"}]},"branch":"refs/heads/master"},"14cbc1d3de2fb36d28449f308ae9b43894c2ed20":{"kind":"REWORK","_number":2,"created":"2020-01-21 16:39:26.000000000","uploader":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"ref":"refs/changes/40/703640/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/40/703640/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/703640/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/703640/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/703640/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/40/703640/2"}}},"commit":{"parents":[{"commit":"d1304acb318454ddf0410d3f41d2c4f2d00b2c3d","subject":"CentOS 8: Update packages in images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d1304acb318454ddf0410d3f41d2c4f2d00b2c3d"}]}],"author":{"name":"Rafael Folco","email":"rfolco@redhat.com","date":"2020-01-21 15:19:55.000000000","tz":-180},"committer":{"name":"Rafael Folco","email":"rfolco@redhat.com","date":"2020-01-21 16:38:51.000000000","tz":-180},"subject":"centos8: fix rabbitmq and qdrouterd","message":"centos8: fix rabbitmq and qdrouterd\n\nIgnore epel/opstools repo disabling for qdrouterd\nRemove rabbitmq pinning for CentOS8\n\nChange-Id: I7fae5c78719597da58d43529868c7880136c5ccd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/14cbc1d3de2fb36d28449f308ae9b43894c2ed20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/14cbc1d3de2fb36d28449f308ae9b43894c2ed20"}]},"branch":"refs/heads/master"},"8ee721ea63b20b94cd9f8351eb6a076081449e13":{"kind":"REWORK","_number":3,"created":"2020-02-06 20:32:52.000000000","uploader":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"ref":"refs/changes/40/703640/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/40/703640/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/703640/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/703640/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/703640/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/40/703640/3"}}},"commit":{"parents":[{"commit":"d1304acb318454ddf0410d3f41d2c4f2d00b2c3d","subject":"CentOS 8: Update packages in images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d1304acb318454ddf0410d3f41d2c4f2d00b2c3d"}]}],"author":{"name":"Rafael Folco","email":"rfolco@redhat.com","date":"2020-01-21 15:19:55.000000000","tz":-180},"committer":{"name":"Rafael Folco","email":"rfolco@redhat.com","date":"2020-02-06 20:32:41.000000000","tz":0},"subject":"[WIP] centos8: remove rabbitmq pinning","message":"[WIP] centos8: remove rabbitmq pinning\n\nRemove rabbitmq pinning for CentOS8\n\nChange-Id: I7fae5c78719597da58d43529868c7880136c5ccd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8ee721ea63b20b94cd9f8351eb6a076081449e13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8ee721ea63b20b94cd9f8351eb6a076081449e13"}]},"branch":"refs/heads/master"},"03ba99f038062c1f6d91bb26123272360e58c20e":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-02-06 20:33:15.000000000","uploader":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"ref":"refs/changes/40/703640/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/40/703640/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/703640/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/703640/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/703640/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/40/703640/4"}}},"commit":{"parents":[{"commit":"ccd734b319ff3250444fb25f64d73e72dc70b94b","subject":"Merge \"CI: Replace ceph jobs with ceph-ansible\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ccd734b319ff3250444fb25f64d73e72dc70b94b"}]}],"author":{"name":"Rafael Folco","email":"rfolco@redhat.com","date":"2020-01-21 15:19:55.000000000","tz":-180},"committer":{"name":"Rafael Folco","email":"rfolco@redhat.com","date":"2020-02-06 20:33:15.000000000","tz":0},"subject":"[WIP] centos8: remove rabbitmq pinning","message":"[WIP] centos8: remove rabbitmq pinning\n\nRemove rabbitmq pinning for CentOS8\n\nChange-Id: I7fae5c78719597da58d43529868c7880136c5ccd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/03ba99f038062c1f6d91bb26123272360e58c20e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/03ba99f038062c1f6d91bb26123272360e58c20e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
