)]}'
{"id":"openstack%2Fkolla~704090","triplet_id":"openstack%2Fkolla~stable%2Frocky~I36e167629ead1d337f2b824d2cd6dbed0cf51001","project":"openstack/kolla","branch":"stable/rocky","hashtags":[],"change_id":"I36e167629ead1d337f2b824d2cd6dbed0cf51001","subject":"[rocky] Bump versions","status":"MERGED","created":"2020-01-23 23:37:38.000000000","updated":"2020-01-27 08:44:31.000000000","submitted":"2020-01-27 08:44:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"704090-1580114671129-46a14a57","meta_rev_id":"3fb3f08b5481fa5e6a0b162d653d26837835b2eb","_number":704090,"virtual_id_number":704090,"owner":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"date":"2020-01-27 04:56:46.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-27 08:44:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-01-24 09:44:09.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-01-24 09:53:35.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-01-27 07:30:59.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-24 09:44:09.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-01-27 07:30:59.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-01-27 08:44:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8dcfe1d2fc9606bb9187d30b1f4d68dc103a579f","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-01-23 23:37:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6928a8b64fb1c3260d3e5cd777a0aeb04fe57a29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-24 01:38:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/16a6e18120ae416e9f2be24053c9202f : SUCCESS in 1h 00m 32s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/2859cedc5ac04de8b138eb66b0979ea4 : SUCCESS in 1h 15m 31s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/53fe6a376dd84172a80f0579a9765ca1 : SUCCESS in 1h 48m 25s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/2170aa73e9524430a9da6e4beb2b1473 : SUCCESS in 1h 42m 41s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/da4a336b258b47f7a1275f4c1d95469e : SUCCESS in 1h 07m 43s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/2ec86532250f4e0aaf0fd4f36cb275ad : SUCCESS in 1h 10m 40s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/0edffb13c38f4eb7be487e4071767fbe : SUCCESS in 1h 32m 06s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/ba62c82fe8fb499d8314c7ea7c53db9e : SUCCESS in 1h 22m 34s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fc1897ea91b841dc9c9e8240b1063819 : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8e4d370f298481fb90c12b23b3094d9 : SUCCESS in 6m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4019f23d19ca405aa99a41b638dba890 : SUCCESS in 4m 38s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/91eda61424944fe58698db081733acf3 : SUCCESS in 6m 01s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/fb41d26bac9e4edba309406f5286eb49 : SUCCESS in 4m 38s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d014c4a037e94e379885364a4d35f0f2 : POST_FAILURE in 1h 13m 40s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e088585c4bb4417bab1d7e0ee2ae6ac4 : POST_FAILURE in 32m 16s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/82c08ad3541541fb9f00194555042591 : FAILURE in 37m 05s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7b3dba25058a402f85137221749218be : FAILURE in 1h 19m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1ff59cd01d92f130f6ffeb3688e2aa6ef395a4c3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-24 09:44:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bcba77b89a6343d7fb89fd69e08e3c7fc9ed949a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-24 09:53:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3b99b22a79bd46ab28871f360b396508a51ecec5","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-01-27 04:56:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fb437a36d43f78d5fb4d51ed0713179f62fa0c3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 06:45:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/82d7b301fdf741ef8addcd5b170ef6f1 : SUCCESS in 38m 49s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/7dad69e24f6f4c689e647fdfafbddff1 : SUCCESS in 49m 16s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/31c38b69f45b404eb0b83e6a7d98c97b : SUCCESS in 1h 24m 47s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/17e38b1216bf4f27be115b861b27f50e : SUCCESS in 1h 04m 18s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/bf984ca00b484777b8a18ed0b5c56044 : SUCCESS in 41m 04s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/aae734535f5c4b7c9b0f67c708bc9fe7 : SUCCESS in 52m 24s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/573375f47a24406da03781bf3a4e5428 : SUCCESS in 1h 12m 11s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/3932157da6bd4cfa8a9b80f447c1ee0e : SUCCESS in 1h 17m 26s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c1524ba9acec4e80b5031c31fd56b897 : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88f2d66c43774c2a85e5fcd7ee24671b : SUCCESS in 4m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/871dc187beb94a1893ff2e2c010e2547 : SUCCESS in 4m 18s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d265c495844f4916840ded8a512bff73 : SUCCESS in 3m 59s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/e23dde20d27141c1aac1a2d6b2b2e165 : SUCCESS in 4m 05s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/82a67953f8624075bb4588b7f4f08c0b : SUCCESS in 1h 26m 00s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6107b033b8da45ff95f68ade7d128bb6 : SUCCESS in 28m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b9b36db0810d40929857fc92d7c2358e : SUCCESS in 1h 00m 31s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d603e0442838400ba91b476e9ac5f38a : SUCCESS in 1h 41m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1699d1d346ebc60a8bc3c2920d6354b1b182631b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-27 07:30:59.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"034e9a741670154d3cc010e5e49d1d157a3330d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 07:31:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1380b2536bdc7ae5beac117765fa32b5ddf3ece7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 08:44:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/1fc6247fac5a4687be3c904b0fe7f70f : SUCCESS in 48m 04s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/96d3fab25dc943168c685e6925938ccb : SUCCESS in 47m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/46c8f5db4146421bb458943f801be735 : SUCCESS in 6m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d1e08cb847d435a99e912d7d26bac12 : SUCCESS in 4m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/eb20c3e47087456cbc1c92c8f4158c35 : SUCCESS in 3m 26s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7ba3f26baa3c4665bf881ffb917d3a5b : SUCCESS in 3m 52s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/aefb002f9d2941bcb982eea2f5cde90a : SUCCESS in 3m 54s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d0147a006ac44744a4ec0a4aa2e2b62c : SUCCESS in 33m 38s","accounts_in_message":[],"_revision_number":1},{"id":"3fb3f08b5481fa5e6a0b162d653d26837835b2eb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 08:44:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"58998f8026b6c02427f856cf2c2b520dbfa5f987","revisions":{"58998f8026b6c02427f856cf2c2b520dbfa5f987":{"kind":"REWORK","_number":1,"created":"2020-01-23 23:37:38.000000000","uploader":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"ref":"refs/changes/90/704090/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/704090/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/704090/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/704090/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/704090/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/704090/1"}}},"commit":{"parents":[{"commit":"0ff85f9bb89b25d62082e86534f6bc6b0695de83","subject":"Merge \"[rocky] Bump versions\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0ff85f9bb89b25d62082e86534f6bc6b0695de83"}]}],"author":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2020-01-23 23:36:22.000000000","tz":180},"committer":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2020-01-23 23:36:22.000000000","tz":180},"subject":"[rocky] Bump versions","message":"[rocky] Bump versions\n\nChange-Id: I36e167629ead1d337f2b824d2cd6dbed0cf51001\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/58998f8026b6c02427f856cf2c2b520dbfa5f987"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/58998f8026b6c02427f856cf2c2b520dbfa5f987"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
