)]}'
{"id":"openstack%2Fkolla~704092","triplet_id":"openstack%2Fkolla~stable%2Ftrain~I7846a797a0d8bd83c7406e1746556f0d5554b925","project":"openstack/kolla","branch":"stable/train","hashtags":[],"change_id":"I7846a797a0d8bd83c7406e1746556f0d5554b925","subject":"[train] Bump versions","status":"MERGED","created":"2020-01-23 23:42:43.000000000","updated":"2020-01-24 11:42:52.000000000","submitted":"2020-01-24 11:41:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"704092-1579866065739-641ca554","meta_rev_id":"258dd8f891dd623e8d950e5021e88e345bf43afb","_number":704092,"virtual_id_number":704092,"owner":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-24 11:41:05.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-01-24 09:44:01.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-01-24 09:53:47.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-01-24 09:53:47.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-24 09:44:01.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-01-24 09:53:47.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-01-24 11:41:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"43ecd1d090dd6a2ef9fa662bc541099583d87846","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-01-23 23:42:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"531594802eccaa2b1803b9f53a090316c737fa2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-24 01:24:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/f831090097424f5aa7b7af66c39b6826 : SUCCESS in 55m 08s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/c16a4c4d7efd4fb78485e8393dbcbafc : SUCCESS in 55m 12s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/2f08eb1d81bc4d0da344929147cba3f6 : SUCCESS in 1h 25m 35s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/9bba2f23b6974c4fa462e5086d884183 : SUCCESS in 1h 30m 03s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/9f502bdfd2eb47b39efe17f3978df85b : SUCCESS in 1h 36m 02s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/6eb468c2a77f4e72a1b2982d52afc3b0 : SUCCESS in 1h 02m 58s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e16e243925ee4f0381f23a37938d8ef8 : SUCCESS in 26m 18s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/94fe01f2babc4cc0961f0d5e9607faec : SUCCESS in 18m 43s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f9d340fff4bd417c988c9f464fecf33d : SUCCESS in 1h 02m 08s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5a4cb1d9a29f42d2ab463f21fb68ffb4 : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8482cb8a26a44fea6d8a60717aca99f : SUCCESS in 5m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/950d15a897c24bb8812fe1ede143dad8 : SUCCESS in 3m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3066cde7c8e84031bf34724d46a41d61 : SUCCESS in 3m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/aff8cf401fc64509b258d4032d0d46f3 : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bc9928f082b4316bb835652d7e5ac21 : SUCCESS in 5m 16s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4ded7a2e411544ea86aad1255171e861 : SUCCESS in 22m 38s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a36944cf7ead4cd289870beeb8817f6e : SUCCESS in 35m 14s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e5799c9b3ba54a9a9cfe99f372fefcae : SUCCESS in 57m 16s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7c0826c5af1b43f2982ab4e5ee801d58 : SUCCESS in 54m 21s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2690dc6c689c418abfa5790ec85ff256 : SUCCESS in 1h 12m 52s","accounts_in_message":[],"_revision_number":1},{"id":"18e8df99d6a0c109e42112a70825b91d6b207608","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-24 09:44:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9f0d64890cc733e50179b885e6605844c06eeed0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-24 09:53:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1f125e36c604643c3749e22873da09c26b2e4678","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-24 09:53:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"12ca817e96968b262bfb11e38b8d196cab944c30","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-24 11:41:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d15bb02cb68cc06282117c784fd0d64ddf78a8b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-24 11:41:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/79dbd218527043f9b6dc0a411b52061b : SUCCESS in 55m 11s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/5c9b5a67c2ca40cdbc37357d24615927 : SUCCESS in 1h 36m 08s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/31403e3926394c61ba4a462a11d4f077 : SUCCESS in 1h 44m 38s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/e9ac88a6a58c4f46a4d8937e2471dd18 : SUCCESS in 58m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f8de352e68914c3f95c16109cd1fa7f3 : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/446424be8a07496c83a6492b2f73dc3d : SUCCESS in 4m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f40b5d95fcef451cb65b69c29e441d48 : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/08895c31517448738ab5fff2ccb3263e : SUCCESS in 3m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a30725c11dc144a8ae308f7042e9344f : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b7ffefff7164701b53a56e833388920 : SUCCESS in 7m 11s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/419e7b518b664b43ad6e80a81159a11f : SUCCESS in 29m 03s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/094f6939a94b43eaaf5b645788f9f545 : SUCCESS in 58m 22s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3f54d3f3db114638b8db4d3e7d0146ee : SUCCESS in 1h 05m 29s","accounts_in_message":[],"_revision_number":1},{"id":"258dd8f891dd623e8d950e5021e88e345bf43afb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-24 11:42:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edb8d0bf427c458abf454498507820f6 : SUCCESS in 1m 23s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9b8dbab77f2990036a09ff6146042b4c18dba3a8","revisions":{"9b8dbab77f2990036a09ff6146042b4c18dba3a8":{"kind":"REWORK","_number":1,"created":"2020-01-23 23:42:43.000000000","uploader":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"ref":"refs/changes/92/704092/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/92/704092/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/704092/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/704092/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/92/704092/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/92/704092/1"}}},"commit":{"parents":[{"commit":"e4b41156eb0a4c0df364b0c5d1119ef5811a7410","subject":"Merge \"CI: Fix symlinks to failed logs\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e4b41156eb0a4c0df364b0c5d1119ef5811a7410"}]}],"author":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2020-01-23 23:42:16.000000000","tz":180},"committer":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2020-01-23 23:42:16.000000000","tz":180},"subject":"[train] Bump versions","message":"[train] Bump versions\n\nChange-Id: I7846a797a0d8bd83c7406e1746556f0d5554b925\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9b8dbab77f2990036a09ff6146042b4c18dba3a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9b8dbab77f2990036a09ff6146042b4c18dba3a8"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
