)]}'
{"id":"openstack%2Fkolla~704311","triplet_id":"openstack%2Fkolla~master~I20312b9719c6f5e3dca0b4198607ddae183152f1","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I20312b9719c6f5e3dca0b4198607ddae183152f1","subject":"Add new option to enable unbuildable images","status":"MERGED","created":"2020-01-27 11:47:03.000000000","updated":"2020-01-28 18:12:44.000000000","submitted":"2020-01-28 18:09:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"704311-1580234971450-87017a1f","meta_rev_id":"739c6cc9951c0d1a3cee027036a93ed61b06d801","_number":704311,"virtual_id_number":704311,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-28 18:09:31.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-01-28 10:55:26.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"all":[{"value":2,"date":"2020-01-28 16:31:45.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":2,"date":"2020-01-28 10:12:14.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"all":[{"value":1,"date":"2020-01-28 16:31:45.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-28 10:12:14.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-01-28 10:55:26.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-01-28 16:31:45.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"},{"updated":"2020-01-28 18:09:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4f84702d0bb8bbb7ddf73f585122865814781d1f","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-01-27 11:47:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"43728af3b0502df52e2158cd763e0e504b774836","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-27 12:16:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/293ff2aac9ff465ba90599fd2235c215 : SUCCESS in 28m 05s","accounts_in_message":[],"_revision_number":1},{"id":"33904dfc5c447811a4366adeaaead10a715e1824","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 13:32:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/947636d938774ee8966aa2c7bee00c86 : SUCCESS in 1h 45m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6465d9c35fc7f118c9c0844aae7aee7e5f1eb55a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 14:02:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/8be8c39f172143318786e0c6a117c452 : SUCCESS in 31m 34s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/c8333a58c0044a1daa7dfff3580354b2 : SUCCESS in 49m 25s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/8b5f0942065b44118299c46d7fa24edd : SUCCESS in 5m 27s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/3262fa20329c4349ac4cf0a3e0743542 : SUCCESS in 5m 25s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/783271b6dbca44338c74184a6f394b29 : SUCCESS in 1h 42m 02s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/16815c59cb16452b847d5a06b1495fd9 : SUCCESS in 1h 40m 21s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/e407204cfbae48c483b94f6a2d3362bf : SUCCESS in 2h 07m 36s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/8ac8965dea224f92a6ce325a33009fa1 : SUCCESS in 50m 47s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e22a688de54b46e4946d97da65f0f950 : SUCCESS in 34m 37s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/1d37b11d01ad43eabe3a894da1aeeedf : SUCCESS in 18m 58s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/1b0caf80c3fa458286dcd7e6bd745158 : SUCCESS in 1h 30m 26s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c06d2997f2a242b38c1a14bf8204ab07 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/822686798c764d878150a2a527416695 : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/61939ec781e74138aa221840e02ab5cb : SUCCESS in 3m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8113bdd77a314f5896adb5758ad9161e : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/535c97ab3eb84d28ae66335ab3d19138 : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2dbc3a025a547c896810b0fec6fb639 : SUCCESS in 5m 52s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f298a1119d6e4292bccb0385810aa6a2 : SUCCESS in 25m 03s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e3047a28244b4a9b9fdfd685078a10e7 : SUCCESS in 26m 08s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b09eb3adc50a4bfca04d8f427f2adf7f : SUCCESS in 56m 12s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/933a2b4ed77f4190b14421f1d50962d8 : SUCCESS in 52m 04s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/dafaa93984074c8fa29e3260d584087c : SUCCESS in 1h 03m 37s","accounts_in_message":[],"_revision_number":1},{"id":"f61af48dd92e21aa2d5fa270bf6aa7cffda4922e","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-01-28 03:24:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4aa5389c3e1b90cc415df4c79fe7b937c4ae6e9e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-28 08:03:04.000000000","message":"Patch Set 1: Code-Review-1\n\nneeds a rn (also for its predecessor which changed behavior related to templating but I did not want to hold that)","accounts_in_message":[],"_revision_number":1},{"id":"29d29d9dea9a4bfc59e40a979bd6e8e808a3c698","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-01-28 10:04:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"af8384281f1595199a95425c8c78bd800b0381b0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-28 10:08:52.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"4a33dd0ffc45cd22a6464fe258b1cf6971a3c7d5","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-01-28 10:10:27.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"41c70e34ab3f34a2ab8ac578ec5b141c4c85f5fc","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-28 10:12:14.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"44e4233e3eaec455e772fa712a2e29c95e548f69","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-28 10:55:26.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/c36b05b2fca541b88cf0d66c221e1564 : SUCCESS in 28m 38s","accounts_in_message":[],"_revision_number":3},{"id":"67545bfcc52b84e35e1cf024bd2f6fbd729bb2aa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 13:15:10.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ff1f56dec0754a8a9e600ed6f9690e42 : TIMED_OUT in 3h 03m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a38f44bfa65e54cc3605bc033609b5107c6a58cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 13:45:09.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/8db61eae6f724048ba687e11ff7d1092 : SUCCESS in 39m 34s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/4278d912d45d4679bd580eb521abd1bc : SUCCESS in 47m 55s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/58b3aa00dc9e41dbb030f2d7826a36d2 : SUCCESS in 7m 55s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/f48fa5ed4c4940ffa4c837334e9eee31 : SUCCESS in 6m 34s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/e5042d05e5144e9ab5005f5788bf5a2a : SUCCESS in 1h 31m 17s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/511baaa614374b4e8fd2b1d8062bafdf : SUCCESS in 1h 17m 15s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/00dbbb6a93c54d0aa1b574168a285dfa : SUCCESS in 2h 13m 26s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/cb94cf89005b4ee79a414bcac5ffcf27 : SUCCESS in 59m 04s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/7717409705824d9f9e3beeffb7ae30a1 : TIMED_OUT in 3h 02m 09s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/580753400bbe45a0a21bf0fb89d3445f : SUCCESS in 22m 15s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/75de4bef55d94acd93453e320a821278 : SUCCESS in 57m 46s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ee211a60169146c8836013f152ca7d2d : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a9a945e17a74336ba72e8509a3cd93d : SUCCESS in 7m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1909dc5f7b3a41e9a5720ca327798916 : SUCCESS in 6m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d6912dc07cea49769bf3720c269786eb : SUCCESS in 7m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97ea9bac312f45beb8e65012ef2e41cd : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a44db2265824198884c4f1737d815b5 : SUCCESS in 8m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dc297d621e214bc7a5bc5e17e224a375 : SUCCESS in 6m 05s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e8e7feef7c984842bf83d7a97b83330a : SUCCESS in 23m 18s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/00ce7eaf7a2145d8874dd2419ff34f38 : SUCCESS in 24m 05s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d17c2d07120a4a3d9d088399c639e05d : SUCCESS in 53m 44s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/149551d5389b4f169bc8722114537c2a : SUCCESS in 53m 06s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6aea010313934e67b92ec6b7981248b9 : SUCCESS in 1h 11m 01s","accounts_in_message":[],"_revision_number":3},{"id":"fbb1fc97626d7808b512ab904d0b4c9ccec6ebba","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-01-28 16:31:45.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"24f06c694e80145b57a6f41fc92cafc4fc635fce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 16:32:02.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c6fbafaa5f06576f0626c5dd539dad836ba7db10","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 18:09:31.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/a4776f325fec4a56aad238ac3830ac6c : SUCCESS in 45m 04s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/03ed85f7584c4805b0abbfb159e4f23c : SUCCESS in 5m 29s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/20953a466ad94805a8d00ba5c4927c3f : SUCCESS in 1h 32m 52s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/df0f44735f854370b5ea13a4e0d64396 : SUCCESS in 1h 33m 10s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/c5c972e911f6481c81f9efa05cc80c8d : SUCCESS in 52m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/549383b4f1744e0b86b0a6e988b742a4 : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea4097044de94171bedc17d4473c9a76 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dbdd9f2ef3154899aecd5fe3e78a48c1 : SUCCESS in 4m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1f359765e4ce4f5bbb8cc88c4d8cca38 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1eda94c898ac4413a1a56969e533806a : SUCCESS in 5m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/74a36df64d70497ea054a384eabba78a : SUCCESS in 5m 33s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4dceef99e8334340beaf0fbd9c597189 : SUCCESS in 38m 10s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1c47ebf88c294736894ce45c82b2f166 : SUCCESS in 55m 47s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/860f1110cec244b5993b6195c6b2be94 : SUCCESS in 1h 04m 21s","accounts_in_message":[],"_revision_number":3},{"id":"2d112e4ceb2eb8662b80240141b4382561b1515c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 18:09:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"739c6cc9951c0d1a3cee027036a93ed61b06d801","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 18:12:44.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2dc223ef77aa417e8ecaa31c6f8487bd : SUCCESS in 1m 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/458542a271b44f2ebc33ddcd2fe879d5 : SUCCESS in 1m 28s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e22c529166f78e79760aeb0cc3b4ab33dd3b6f9c","revisions":{"d2e706f7b1c4c16c7e0bd642bf08db4490b06577":{"kind":"REWORK","_number":1,"created":"2020-01-27 11:47:03.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/11/704311/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/11/704311/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/704311/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/704311/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/704311/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/11/704311/1"}}},"commit":{"parents":[{"commit":"fa2e1304362d3b81b9903b77d180aa2eb3f1af8b","subject":"build: list only buildable templates for --template-only","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fa2e1304362d3b81b9903b77d180aa2eb3f1af8b"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-01-27 11:43:14.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-01-27 11:46:52.000000000","tz":60},"subject":"Add new option to enable unbuildable images","message":"Add new option to enable unbuildable images\n\nI7a92b762f6a783f92ebca61e5595e1422a92df4f added checking for unbuildable\nimages when someone wants just templates (like TripleO for example).\n\nBut what if someone wants all images despite their buildability status?\nOr wants to build for some new distro/arch combo?\n\n--enable-unbuildable just ignores internal UNBUILDABLE_IMAGES list and\ntries all images.\n\nChange-Id: I20312b9719c6f5e3dca0b4198607ddae183152f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d2e706f7b1c4c16c7e0bd642bf08db4490b06577"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d2e706f7b1c4c16c7e0bd642bf08db4490b06577"}]},"branch":"refs/heads/master"},"7fe72231ce7b85000d0de35dc15f4aed8cc0f76a":{"kind":"REWORK","_number":2,"created":"2020-01-28 10:04:37.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/11/704311/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/11/704311/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/704311/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/704311/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/704311/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/11/704311/2"}}},"commit":{"parents":[{"commit":"fa2e1304362d3b81b9903b77d180aa2eb3f1af8b","subject":"build: list only buildable templates for --template-only","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fa2e1304362d3b81b9903b77d180aa2eb3f1af8b"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-01-27 11:43:14.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-01-28 10:04:22.000000000","tz":60},"subject":"Add new option to enable unbuildable images","message":"Add new option to enable unbuildable images\n\nI7a92b762f6a783f92ebca61e5595e1422a92df4f added checking for unbuildable\nimages when someone wants just templates (like TripleO for example).\n\nBut what if someone wants all images despite their buildability status?\nOr wants to build for some new distro/arch combo?\n\n--enable-unbuildable just ignores internal UNBUILDABLE_IMAGES list and\ntries all images.\n\nChange-Id: I20312b9719c6f5e3dca0b4198607ddae183152f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7fe72231ce7b85000d0de35dc15f4aed8cc0f76a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7fe72231ce7b85000d0de35dc15f4aed8cc0f76a"}]},"branch":"refs/heads/master"},"e22c529166f78e79760aeb0cc3b4ab33dd3b6f9c":{"kind":"REWORK","_number":3,"created":"2020-01-28 10:10:27.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/11/704311/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/11/704311/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/704311/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/704311/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/704311/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/11/704311/3"}}},"commit":{"parents":[{"commit":"fa2e1304362d3b81b9903b77d180aa2eb3f1af8b","subject":"build: list only buildable templates for --template-only","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fa2e1304362d3b81b9903b77d180aa2eb3f1af8b"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-01-27 11:43:14.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-01-28 10:10:22.000000000","tz":0},"subject":"Add new option to enable unbuildable images","message":"Add new option to enable unbuildable images\n\nI7a92b762f6a783f92ebca61e5595e1422a92df4f added checking for unbuildable\nimages when someone wants just templates (like TripleO for example).\n\nBut what if someone wants all images despite their buildability status?\nOr wants to build for some new distro/arch combo?\n\n--enable-unbuildable just ignores internal UNBUILDABLE_IMAGES list and\ntries all images.\n\nChange-Id: I20312b9719c6f5e3dca0b4198607ddae183152f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e22c529166f78e79760aeb0cc3b4ab33dd3b6f9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e22c529166f78e79760aeb0cc3b4ab33dd3b6f9c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
