)]}'
{"id":"openstack%2Fkolla~704560","triplet_id":"openstack%2Fkolla~master~If5044346a90580c4051d6138a84ce10238807460","project":"openstack/kolla","branch":"master","topic":"collectd-pmu","hashtags":[],"change_id":"If5044346a90580c4051d6138a84ce10238807460","subject":"Add collectd-pmu package for Intel PMU (C7 only)","status":"MERGED","created":"2020-01-28 12:55:11.000000000","updated":"2020-02-07 11:43:42.000000000","submitted":"2020-02-07 11:41:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"704560-1581075706898-5f1f21a0","meta_rev_id":"01950c08317c1bccbc8964e136f1b1661a810b44","_number":704560,"virtual_id_number":704560,"owner":{"_account_id":31539,"name":"Piotr Rabiega","email":"piotrx.rabiega@intel.com","username":"prabiegx"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31539,"name":"Piotr Rabiega","email":"piotrx.rabiega@intel.com","username":"prabiegx"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-07 11:41:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-02-07 09:02:45.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":31539,"name":"Piotr Rabiega","email":"piotrx.rabiega@intel.com","username":"prabiegx"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2020-02-07 09:19:47.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":2,"date":"2020-02-07 08:32:46.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":31539,"name":"Piotr Rabiega","email":"piotrx.rabiega@intel.com","username":"prabiegx"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":1,"date":"2020-02-07 10:05:48.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31539,"name":"Piotr Rabiega","email":"piotrx.rabiega@intel.com","username":"prabiegx"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":31539,"name":"Piotr Rabiega","email":"piotrx.rabiega@intel.com","username":"prabiegx"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"date":"2020-02-07 08:40:00.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":31539,"name":"Piotr Rabiega","email":"piotrx.rabiega@intel.com","username":"prabiegx"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-28 12:55:22.000000000","updated_by":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"reviewer":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"state":"REVIEWER"},{"updated":"2020-02-06 15:34:55.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2020-02-07 08:40:00.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-02-07 09:02:45.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-07 09:19:47.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-02-07 10:05:48.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-02-07 11:41:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0b7ef7d0e9eed63e5e5957dd5fe4f74dd1e14a83","author":{"_account_id":31539,"name":"Piotr Rabiega","email":"piotrx.rabiega@intel.com","username":"prabiegx"},"date":"2020-01-28 12:55:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"88aa2c1ca7f71fbe6d11d5cae169ccbb9ec6c8f3","author":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"date":"2020-01-28 12:55:22.000000000","message":"Patch Set 1:\n\nThank you for your first contribution to OpenStack.\n\nYour patch will now be tested automatically by OpenStack testing frameworks\nand once the automatic tests pass, it will be reviewed by other friendly\ndevelopers. They will give you feedback and may require you to refine it.\n\nPeople seldom get their patch approved on the first try, so don\u0027t be\nconcerned if requested to make corrections. Feel free to modify your patch\nand resubmit a new change-set.\n\nPatches usually take 3 to 7 days to be reviewed so be patient and be\navailable on IRC to ask and answer questions about your work. Also it\ntakes generally at least a couple of weeks for cores to get around to\nreviewing code. The more you participate in the community the more\nrewarding it is for you. You may also notice that the more you get to know\npeople and get to be known, the faster your patches will be reviewed and\neventually approved. Get to know others and become known by doing code\nreviews: anybody can do it, and it\u0027s a great way to learn the code base.\n\nThanks again for supporting OpenStack, we look forward to working with you.\n\nIRC: https://wiki.openstack.org/wiki/IRC\nWorkflow: https://docs.openstack.org/infra/manual/developers.html\nCommit Messages: https://wiki.openstack.org/wiki/GitCommitMessages","accounts_in_message":[],"_revision_number":1},{"id":"8bfd24b2961bc1db0bac04e3c6724f4ee10b5df0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-28 13:13:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f82c5c6b6a019d87e054b1c701c4b23e16e28c04","author":{"_account_id":31539,"name":"Piotr Rabiega","email":"piotrx.rabiega@intel.com","username":"prabiegx"},"date":"2020-01-28 13:18:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c75d25df290dbcf133b19c94ffdb77a0806abcb5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-28 13:22:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/e120e8d67dc746b5bf5c162c3efe1318 : SUCCESS in 26m 17s","accounts_in_message":[],"_revision_number":1},{"id":"0f315453816143d57b5e61a7a1a11f73d1ff6229","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-28 13:25:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"65cc92eee2431a519d06a64648bdc1258049f176","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 14:51:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/66adea6fe40f4d28b5a46c26af4a3c3c : FAILURE in 38m 12s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/19eeb509770f4beb944a569a0cfe0a89 : FAILURE in 45m 50s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/d94638dc639b4c9e82be6eb20c46400a : SUCCESS in 9m 18s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/9df5767743294ac6ada19909bc730786 : SUCCESS in 9m 31s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/319d8f33066b49b8b03e961c09ed7622 : SUCCESS in 1h 25m 11s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/a4f765df8b8d4f3ebfe9080f895f2556 : SUCCESS in 1h 14m 58s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/374e329af36048d9942023bd197f5c03 : SUCCESS in 1h 52m 28s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/3d6959209374422589aa3b2f0dfde449 : FAILURE in 59m 23s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/11f87dd9064c413ea4868338c1b9c3fb : SUCCESS in 29m 11s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/83be4f89aca240f08bc04963399d4eae : SUCCESS in 21m 18s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/711c71a8b1104013ae7a46860dd37d61 : SUCCESS in 55m 47s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/302135c373d348d5889fcf77511985d6 : SUCCESS in 6m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd787a2f17ea4dc6b039dd98281f0ab9 : SUCCESS in 7m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/89e40225850440f9a0ba6031d90783fc : SUCCESS in 7m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bc542c800995449b908a1d9e26ca025b : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6a536f1e8177473bac9b13f8772d8772 : SUCCESS in 3m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/321a388761af432cb7df224fb0cccec4 : SUCCESS in 5m 48s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0737b4da36fa4730943df1c12f84e3ee : SUCCESS in 27m 08s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/98ac71b2a7e345c79600a0fc460649f9 : SUCCESS in 24m 00s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/cd780328c4ab42f9b5683863a9c9d7ae : SUCCESS in 57m 56s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ad94867d4acc415aa0deb6776561615b : SUCCESS in 57m 45s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/28302602b0474f55ba5998b985c8b33d : SUCCESS in 1h 16m 40s","accounts_in_message":[],"_revision_number":1},{"id":"e0bf10a577abffacba0945361173b69ee4cc54e7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 15:51:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/adf5eef232c842dfade93a069f98451a : SUCCESS in 2h 40m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2e8305e48ac885516aff2d396a6d6a6d48afa2e9","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2020-01-29 08:37:49.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nThank you for this contribution. For the failure, theer may be multiple reasons:\n- different package name\n- unexpected package source\n\nI\u0027d remove the python2/3 switch and clean that up. Not sure if that is relevant to this current patch, but it\u0027s a technical debt to be solved here sooner or later.","accounts_in_message":[],"_revision_number":1},{"id":"ca0b4d41a6afe3bda310567c35af368ef80ddf08","author":{"_account_id":31539,"name":"Piotr Rabiega","email":"piotrx.rabiega@intel.com","username":"prabiegx"},"date":"2020-02-05 08:44:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0fb11c04ab9208cce601488f58c2132e8a5dd648","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-05 09:12:36.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/61c1d1f8045b42e2a339a4436eaf730e : FAILURE in 27m 10s","accounts_in_message":[],"_revision_number":2},{"id":"4f893552d805c067533d65914ea1c2e9273cf6e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 10:24:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/a9c93e4533b44f6b90dbcb30f98c1b5c : SUCCESS in 43m 50s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/3094059cbaa3464a853cc05f8df78013 : SUCCESS in 50m 54s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/eb2804c54de449099bf6e31d10632ba3 : SUCCESS in 23m 45s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/569a4ec2bbc241019cfcabaa3fee5188 : SUCCESS in 30m 05s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/3842c28b995c456b8b1373dd422fe34e : SUCCESS in 1h 31m 27s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/dfcb3b77b1414bf5b71195117273b8ab : SUCCESS in 1h 36m 08s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/ae5ba36304d0495b89992ced9061d408 : SUCCESS in 1h 29m 59s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/83a818ea74e841309ecdf9c5761637ea : SUCCESS in 1h 01m 05s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e12a0d87c57c47c5bc3072b4f4716419 : SUCCESS in 27m 35s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e73a38a773894d20a1eba3c8e687c9c4 : SUCCESS in 26m 08s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f7bbeab8c54e4d3f8d2f1a1d90e8e397 : SUCCESS in 1h 01m 34s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/74e499f50c1c4d968037934b5291e777 : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1400f089717461cb3c6dac3d71812a6 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2286d6b23aca4018a5cbcd8a88fa50a5 : SUCCESS in 3m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/da264e42e08d48b2a9711bbb172f0806 : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8e453ce1b4f8455faf922e4b68822248 : SUCCESS in 3m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0791acd02c924e7a99fa236b9b2f3302 : SUCCESS in 5m 03s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/83337458d70a46cc923843755d120ec5 : SUCCESS in 24m 29s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/09f093a702dc44b89e60ed81f94f300a : SUCCESS in 27m 56s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/970b9725af124817b5fadb0fbc74be0e : SUCCESS in 57m 15s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c8c39047fba04fe8b0a63bdaa371e46d : SUCCESS in 57m 51s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4b4557c5dfdf4a8c953c5f2402a123df : SUCCESS in 1h 05m 47s","accounts_in_message":[],"_revision_number":2},{"id":"8d59cd193d59be40e10e3df829579ac967838bf3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 11:01:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8a4756d06a7d454ab68715a0176de60b : SUCCESS in 2h 16m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c9e5414bba6e6ca79101fd022803f0be11f6acc7","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2020-02-05 11:32:22.000000000","message":"Patch Set 2:\n\n(1 comment)\n\ncongrats, looks good to me!\n\nbuild for centos-opstools on centos8 to come.","accounts_in_message":[],"_revision_number":2},{"id":"dd40ce68e3a12d15cb26c585ae75c46eebcd5a5c","author":{"_account_id":31539,"name":"Piotr Rabiega","email":"piotrx.rabiega@intel.com","username":"prabiegx"},"date":"2020-02-05 11:39:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"907d63164550eced606260fead83ae88ce77ed34","author":{"_account_id":31539,"name":"Piotr Rabiega","email":"piotrx.rabiega@intel.com","username":"prabiegx"},"date":"2020-02-06 07:57:42.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"325962e79d6948667aa8c1e36c3bc2e5ebc9ddc6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-06 08:23:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/20aec5df5a5e4a6d9bbe727dd799c792 : SUCCESS in 24m 38s","accounts_in_message":[],"_revision_number":3},{"id":"e251da8b8db84bb37b7563b75ac9bac55a306f99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 09:34:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/0198d2c94fdb43a98c32fb325da1d7df : SUCCESS in 35m 06s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/833ec541d5e842999ccd666fcc942578 : SUCCESS in 42m 33s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/39ccdfa3c77040d2884b4a94118c2714 : SUCCESS in 22m 06s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/aaad0365b276494080d59ed836ec492b : SUCCESS in 29m 38s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/68f3a15e640a4a8eb660326bad81e8a0 : SUCCESS in 1h 32m 15s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/d682c76e0ff64ed0a61937509611b6ed : SUCCESS in 1h 18m 43s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/bc250eda0ea6460f9fcb2ccd47ecd946 : SUCCESS in 1h 29m 32s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/69cbb11dffb74f66b92ebfa5b4fae0f0 : SUCCESS in 58m 59s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6f8b2d592ed1490c97c8dbdbf686e45e : SUCCESS in 20m 12s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/8c1767d1e5624bbfbbf9142cdec9de0d : SUCCESS in 18m 47s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0f448b8925354e93ab6701c351a94c88 : SUCCESS in 52m 36s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e341b48a76974596a80435d5af84f973 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/933a4be288714a18919070b1c56078cb : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4bdd79edcbcd4fdba6fd0e34288c356e : SUCCESS in 3m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8cb0af8217574fc7b314f9aa08b79f32 : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a2c4c31ff2b4e6ca887826ffeeadfff : SUCCESS in 4m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac581e0eeaf648eca884c57ec937d21e : SUCCESS in 5m 12s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cee8e10702584b489b115531cc04e5c6 : SUCCESS in 22m 27s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/89d800b89ea04ba39416e89183a9f918 : SUCCESS in 31m 08s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/adcc6d66ecc14c4bbe4fc9af21d8b59b : SUCCESS in 53m 37s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/af58c4ca09cb42e99252cdaff874c408 : SUCCESS in 55m 10s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0d0964f2b9304a6aa4d5cb9dc8c6a943 : SUCCESS in 1h 04m 18s","accounts_in_message":[],"_revision_number":3},{"id":"f314a27f2be46c88ae3b6e3023cf31f9b93369e5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 10:13:54.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5b4b10178cee4442beb9fac3167dadb5 : SUCCESS in 2h 14m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"37c65e35d5914b258cc9219106773f437ed1a69c","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2020-02-06 15:34:55.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"88e7405459ef0708b97f3c5057fbbe8c932b345a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-07 08:27:09.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"57afb7af00a76fcb01d1075c9130140935d95989","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-07 08:27:16.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"2f514ba52adb7cb72d2aacbf04d2915bde4e3176","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-07 08:27:54.000000000","message":"Topic changed from add_collectd_intel_pmu to collectd-pmu","accounts_in_message":[],"_revision_number":4},{"id":"01a2cbecde39c020c3b44b294a34e98583a4e4eb","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-07 08:29:12.000000000","message":"Patch Set 4: -Code-Review Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"fb09046214a439be11be94b37e767c28dc8f69cb","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-07 08:31:50.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"dc0e1cce9db13142c1d21d22580882b1c3c6b075","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-07 08:32:46.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"11897aabaa83aa1ee7ead059091bd8e543289c96","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-07 09:02:45.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/a7e3189972cd4679b54d26e81124faa8 : SUCCESS in 29m 47s","accounts_in_message":[],"_revision_number":5},{"id":"288ca1899da24d5d8c19a94e95de9326c8e9aca7","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-02-07 09:19:47.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"a8002874cb866c39b5df3a813789cdc44b178851","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 10:04:34.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/b9271231bc3145298c49fce9fe7d2b50 : SUCCESS in 32m 53s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/9ba548811646454db9afd174c7027438 : SUCCESS in 41m 13s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/1e396dc26e0342428d4d45105cff27f8 : SUCCESS in 27m 38s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/238656e3740c477ea902938293e3dceb : SUCCESS in 24m 32s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/c52773bdf9b24867996f64f5a7cc5b84 : SUCCESS in 1h 17m 41s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/bc21d76dfa8843bc8c86c6e638b2b70d : SUCCESS in 1h 18m 37s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/a0bfa6b6f4f84004af8818450e21e06b : SUCCESS in 1h 31m 04s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/c097bfec014e4ef7add56f7c257c554e : SUCCESS in 45m 00s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/93235d2da107433f910a94fbfd5c95c7 : SUCCESS in 25m 04s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/8df4691be5404fe2b5332080957681dd : SUCCESS in 18m 49s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/46fc97e0517340bc912ad106da9b2aed : SUCCESS in 55m 41s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0a61853f79cb497fb4bd8dbdefe1ceb5 : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cb5da58e5924ff19bd1847f772f114d : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc7b808c6b634d749a6eeeca1a6cc143 : SUCCESS in 4m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3e363b048d4f40399c1f93b120285aed : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/68015c3527e441fe9a7b9a129151ae01 : SUCCESS in 4m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d16b2c37ff7740ec9741ad84435db151 : SUCCESS in 5m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fcab249f8b16454699b958af375c3750 : SUCCESS in 4m 35s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/364378b8a10d4781a6716133fa62a4fa : SUCCESS in 38m 44s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f159ec1821ab4d83a587dcc6937aaa9d : SUCCESS in 38m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7890aa393e49493ab319e48d51404930 : SUCCESS in 56m 40s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6cfc8dfcbd894b13b3f52da09218acb1 : SUCCESS in 58m 35s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d995081f841948fc987864140ff1b6b3 : SUCCESS in 1h 12m 37s","accounts_in_message":[],"_revision_number":5},{"id":"df22b725fc4ea50f1a9183e04e96f1bc0b3129cc","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-07 10:05:48.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"1bfd0499d979fedfa7810b99eb5c8d3f7d0104d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 10:06:02.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"50fcece2fbf642442cce875d609fe6b53728f109","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-07 10:06:07.000000000","message":"Patch Set 5:\n\nMatthias confirmed no backport pending","accounts_in_message":[],"_revision_number":5},{"id":"4da2b5ca9c5a5bbb1d7c060e50ab1221482bebdc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 11:08:03.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/38781d1ce97743f9b6e0af7370253bef : SUCCESS in 2h 32m 18s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c5da06bfc0ff548218d94a245bbdaa04f0e386a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 11:41:46.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/1690c9366e994db3a247a008c7f9f0a0 : SUCCESS in 44m 23s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/ef1c9e665f2c456ea33b3e14fcdd2072 : SUCCESS in 36m 50s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/a93ea6d3c7704d42a5aee5c91202c1e8 : SUCCESS in 1h 21m 01s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/e2c4ccaf66d24dbb88a7058997dd906e : SUCCESS in 1h 33m 51s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/2dc936ba905c46538cb4d6985a87d972 : SUCCESS in 1h 00m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/02188bd3b4de4103a1661e2c90036341 : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4f4c438f49344048e511761f420123d : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ec16ef0408684a14901265208bb083c2 : SUCCESS in 4m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/630c0807369b47eeb9c2cbe3f8a8302c : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f76cc940f0bf468fb0a3ef54109fef9b : SUCCESS in 4m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a0ed5077eb044004b3add0cd86dbcbaf : SUCCESS in 5m 06s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b74b061258fe4063b2392531f159a73c : SUCCESS in 25m 43s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b5a74cbac22f4dfba63d774a7c40c13d : SUCCESS in 1h 09m 33s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/853df05261e74043b673040e65b593db : SUCCESS in 1h 20m 14s","accounts_in_message":[],"_revision_number":5},{"id":"ec1729dd80411fe9309edd7ffb38cafb62c0ef7f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 11:41:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"01950c08317c1bccbc8964e136f1b1661a810b44","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 11:43:42.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9457e9f7eba54aa98dbda465a40c12d9 : SUCCESS in 1m 14s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/108b0f2ac06341e2bf47d3af8b1a5e11 : SUCCESS in 1m 27s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"e6e76ac249a551b58c892f5121b1583b2882ffe8","revisions":{"1cae846b00daa9b918eb72799165e214722e72e1":{"kind":"REWORK","_number":1,"created":"2020-01-28 12:55:11.000000000","uploader":{"_account_id":31539,"name":"Piotr Rabiega","email":"piotrx.rabiega@intel.com","username":"prabiegx"},"ref":"refs/changes/60/704560/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/60/704560/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/704560/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/704560/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/704560/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/60/704560/1"}}},"commit":{"parents":[{"commit":"fa2e1304362d3b81b9903b77d180aa2eb3f1af8b","subject":"build: list only buildable templates for --template-only","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fa2e1304362d3b81b9903b77d180aa2eb3f1af8b"}]}],"author":{"name":"Piotr Rabiega","email":"piotrx.rabiega@intel.com","date":"2020-01-28 11:25:22.000000000","tz":60},"committer":{"name":"Piotr Rabiega","email":"piotrx.rabiega@intel.com","date":"2020-01-28 12:52:12.000000000","tz":60},"subject":"Add collectd-intel_pmu plugin","message":"Add collectd-intel_pmu plugin\n\nThe intel_pmu plugin collects performance counters data on Intel CPUs\nusing Linux perf interface.\n\nChange-Id: If5044346a90580c4051d6138a84ce10238807460\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1cae846b00daa9b918eb72799165e214722e72e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1cae846b00daa9b918eb72799165e214722e72e1"}]},"branch":"refs/heads/master"},"4ce116d6fdd62c4922ce7f0af00a162883a89539":{"kind":"REWORK","_number":2,"created":"2020-02-05 08:44:31.000000000","uploader":{"_account_id":31539,"name":"Piotr Rabiega","email":"piotrx.rabiega@intel.com","username":"prabiegx"},"ref":"refs/changes/60/704560/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/60/704560/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/704560/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/704560/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/704560/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/60/704560/2"}}},"commit":{"parents":[{"commit":"fa2e1304362d3b81b9903b77d180aa2eb3f1af8b","subject":"build: list only buildable templates for --template-only","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fa2e1304362d3b81b9903b77d180aa2eb3f1af8b"}]}],"author":{"name":"Piotr Rabiega","email":"piotrx.rabiega@intel.com","date":"2020-01-28 11:25:22.000000000","tz":60},"committer":{"name":"Piotr Rabiega","email":"piotrx.rabiega@intel.com","date":"2020-02-05 08:42:40.000000000","tz":60},"subject":"Add collectd-intel_pmu plugin","message":"Add collectd-intel_pmu plugin\n\nThe intel_pmu plugin collects performance counters data on Intel CPUs\nusing Linux perf interface.\n\nChange-Id: If5044346a90580c4051d6138a84ce10238807460\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4ce116d6fdd62c4922ce7f0af00a162883a89539"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4ce116d6fdd62c4922ce7f0af00a162883a89539"}]},"branch":"refs/heads/master"},"f3de86dd22b477f371818fbfaf5795644cebb472":{"kind":"REWORK","_number":3,"created":"2020-02-06 07:57:42.000000000","uploader":{"_account_id":31539,"name":"Piotr Rabiega","email":"piotrx.rabiega@intel.com","username":"prabiegx"},"ref":"refs/changes/60/704560/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/60/704560/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/704560/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/704560/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/704560/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/60/704560/3"}}},"commit":{"parents":[{"commit":"fa2e1304362d3b81b9903b77d180aa2eb3f1af8b","subject":"build: list only buildable templates for --template-only","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fa2e1304362d3b81b9903b77d180aa2eb3f1af8b"}]}],"author":{"name":"Piotr Rabiega","email":"piotrx.rabiega@intel.com","date":"2020-01-28 11:25:22.000000000","tz":60},"committer":{"name":"Piotr Rabiega","email":"piotrx.rabiega@intel.com","date":"2020-02-06 07:54:29.000000000","tz":60},"subject":"Add collectd-intel_pmu plugin","message":"Add collectd-intel_pmu plugin\n\nThe intel_pmu plugin collects performance counters data on Intel CPUs\nusing Linux perf interface.\n\nChange-Id: If5044346a90580c4051d6138a84ce10238807460\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f3de86dd22b477f371818fbfaf5795644cebb472"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f3de86dd22b477f371818fbfaf5795644cebb472"}]},"branch":"refs/heads/master"},"10ca9217736fc4c595806e146023cf0a7c45eb69":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-02-07 08:27:09.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/60/704560/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/60/704560/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/704560/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/704560/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/704560/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/60/704560/4"}}},"commit":{"parents":[{"commit":"fa2e1304362d3b81b9903b77d180aa2eb3f1af8b","subject":"build: list only buildable templates for --template-only","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fa2e1304362d3b81b9903b77d180aa2eb3f1af8b"}]}],"author":{"name":"Piotr Rabiega","email":"piotrx.rabiega@intel.com","date":"2020-01-28 11:25:22.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-02-07 08:26:41.000000000","tz":0},"subject":"Add collectd-pmu package for Intel PMU (C7 only)","message":"Add collectd-pmu package for Intel PMU (C7 only)\n\nThe intel_pmu plugin collects performance counters data\non Intel CPUs using Linux perf interface.\n\nC7 only - C8 packaging still pending.\n\nChange-Id: If5044346a90580c4051d6138a84ce10238807460\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/10ca9217736fc4c595806e146023cf0a7c45eb69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/10ca9217736fc4c595806e146023cf0a7c45eb69"}]},"branch":"refs/heads/master"},"e6e76ac249a551b58c892f5121b1583b2882ffe8":{"kind":"REWORK","_number":5,"created":"2020-02-07 08:31:50.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/60/704560/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/60/704560/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/704560/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/704560/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/60/704560/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/60/704560/5"}}},"commit":{"parents":[{"commit":"fa2e1304362d3b81b9903b77d180aa2eb3f1af8b","subject":"build: list only buildable templates for --template-only","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fa2e1304362d3b81b9903b77d180aa2eb3f1af8b"}]}],"author":{"name":"Piotr Rabiega","email":"piotrx.rabiega@intel.com","date":"2020-01-28 11:25:22.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-02-07 08:31:41.000000000","tz":60},"subject":"Add collectd-pmu package for Intel PMU (C7 only)","message":"Add collectd-pmu package for Intel PMU (C7 only)\n\nThe intel_pmu plugin collects performance counters data\non Intel CPUs using Linux perf interface.\n\nC7 only - C8 packaging still pending.\n\nChange-Id: If5044346a90580c4051d6138a84ce10238807460\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e6e76ac249a551b58c892f5121b1583b2882ffe8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e6e76ac249a551b58c892f5121b1583b2882ffe8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
