)]}'
{"id":"openstack%2Fkolla~704723","triplet_id":"openstack%2Fkolla~master~Idaaee886ceb6cf29494369e771f8ec453550e3f9","project":"openstack/kolla","branch":"master","topic":"netcontrold_rpm","hashtags":[],"change_id":"Idaaee886ceb6cf29494369e771f8ec453550e3f9","subject":"Add support to build netcontrold binary image","status":"ABANDONED","created":"2020-01-29 04:40:35.000000000","updated":"2020-04-20 04:57:35.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4fb63f766dce79bd17eeea32831a6abe1f61645b","_number":704723,"virtual_id_number":704723,"owner":{"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":31108,"name":"Gowrishankar","email":"gmuthukr@redhat.com","username":"gmuthukr"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-01-30 12:08:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-01-30 10:45:37.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"date":"2020-01-30 10:36:55.000000000","_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31108,"name":"Gowrishankar","email":"gmuthukr@redhat.com","username":"gmuthukr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":-1,"date":"2020-01-30 10:11:00.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":31108,"name":"Gowrishankar","email":"gmuthukr@redhat.com","username":"gmuthukr"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":31108,"name":"Gowrishankar","email":"gmuthukr@redhat.com","username":"gmuthukr"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":31108,"name":"Gowrishankar","email":"gmuthukr@redhat.com","username":"gmuthukr"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":31108,"name":"Gowrishankar","email":"gmuthukr@redhat.com","username":"gmuthukr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-29 14:22:21.000000000","updated_by":{"_account_id":31108,"name":"Gowrishankar","email":"gmuthukr@redhat.com","username":"gmuthukr"},"reviewer":{"_account_id":31108,"name":"Gowrishankar","email":"gmuthukr@redhat.com","username":"gmuthukr"},"state":"REVIEWER"},{"updated":"2020-01-30 08:36:06.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-01-30 10:11:00.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-01-30 10:45:37.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-01-30 12:08:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"aee48333df0810041b02ceda14fae41d3b00b741","author":{"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"},"date":"2020-01-29 04:40:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8608877228145847fec276a6b7ad776b8ee8b432","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-29 05:05:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/88847eed7dd94189be9cf5e672a9cce2 : SUCCESS in 24m 01s","accounts_in_message":[],"_revision_number":1},{"id":"f69f9072a4fe788201749876b6384d876ace0953","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 05:47:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/bb0e676d0275480d9c4f9afd7f5ddb4a : SUCCESS in 1h 05m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"06b2115f10a6c8fa9f49bc807bdf860367685a68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 06:28:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/b2de9b6de9a8419a97a3e966904a4ee3 : SUCCESS in 35m 03s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/2e1c2ae91e024e09bd9bac97645642bd : SUCCESS in 39m 29s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/c75fb14c402d4173b4ebf43596895d56 : SUCCESS in 5m 02s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/4852789a6fb04b54a9327208edad6b41 : SUCCESS in 5m 11s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/ec9facc1986043b7a13dd03d053945a2 : SUCCESS in 1h 31m 45s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/2f21a31e6a8a4b9a9849246e01f2f6e5 : SUCCESS in 1h 42m 21s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/c0ad15a9dd34487fa27d3bc202612f4e : FAILURE in 1h 18m 29s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/2406ecb19c8b46f5a62fc824b74a34f1 : SUCCESS in 55m 36s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/3025bfeb80594817aff0cc7c0e3d706a : SUCCESS in 27m 18s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/efba70fce3814401b2567465bd695720 : SUCCESS in 32m 13s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5d07e5cead0340f0acfbd834b51772c6 : SUCCESS in 1h 05m 06s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b2958f50b5654be580c2a8aa5bf799ac : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60bbb9623dd64d20b04b7eb370cd3dfe : SUCCESS in 6m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3b1a0a87d8bc48ea9bd0be2a22223bb9 : SUCCESS in 3m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/50e5461d5929426a8156a650d8394c81 : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/390299e563ed44be920ae0e2b143fd10 : SUCCESS in 4m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba3bbe9b8a76431ca79156d93675b707 : SUCCESS in 7m 53s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3289151a8f4c400d900e9d93e6a1107b : SUCCESS in 44m 18s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/fac5abfca68942aab6713d7b1ca39de8 : SUCCESS in 26m 45s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1caf9cddc9334ff4bacf35523e39103e : SUCCESS in 53m 36s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e1c4371ab5f04b329f6f39ebbf7ce493 : SUCCESS in 51m 30s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6fc99b3aeb634170ba767539fc3ddc68 : SUCCESS in 1h 01m 38s","accounts_in_message":[],"_revision_number":1},{"id":"b64701621f732ece1a418c8806c8ce3a0fb99839","author":{"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"},"date":"2020-01-29 07:37:22.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cc6c416e2b279910eab4bc330d1c9741fa941d2c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-29 08:03:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/dcb8e9a7398f4c099e11ec167db9d2d6 : SUCCESS in 24m 53s","accounts_in_message":[],"_revision_number":1},{"id":"4865084bf64018d9d1c69cd94d1d29963934c876","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 09:22:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/ba644a4a3730430a990f5f7cd8c0eec5 : SUCCESS in 38m 08s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/74fb1ee21c034998b7368cc358658b15 : SUCCESS in 41m 30s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/e254d2ce50b34f708195c5d000cc0fa8 : SUCCESS in 6m 04s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/cd49e696c629461dabd5441c7395c474 : SUCCESS in 5m 47s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/a99b8ca15fa5451ebaf4905a20e55fdb : SUCCESS in 1h 26m 59s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/8106aa1fd0ee49a4b26c7df7f5f403eb : SUCCESS in 1h 26m 14s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/277edf533b1544b097d22abeae5573ea : SUCCESS in 1h 40m 39s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/60ba82899532475db7d73cee41c38e7a : SUCCESS in 51m 00s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/bea917e5d3124842a6876d2c35380df7 : SUCCESS in 23m 24s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/7621eaf8285641eeaeecb3cfee7d6380 : SUCCESS in 20m 06s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/20c26e6b5c8a4044ba13508eed6586d2 : SUCCESS in 53m 56s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0df6317d4cc845a6a33c1c4f462caa4a : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b6efdf2cfac44bdb1308ad73bf9010a : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/803825cb6b9a4cdbb97f372d34b8edca : SUCCESS in 4m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/20840987a3cd45a0ac0f2ff966d6bfa6 : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8b631beb692f473db702d517079db4ec : SUCCESS in 3m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0a35236a5cf4a618fb1bff9b60b72e0 : SUCCESS in 5m 39s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/38eca6529d9e45109ad82456a34cdee1 : SUCCESS in 42m 05s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/548abf0fd5694139a48a5f6646b6c340 : SUCCESS in 27m 37s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/08ed243756d24f6cb761844cef5bb633 : SUCCESS in 1h 05m 57s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/87f2cacf18ce4e77a0d96139d44956f5 : SUCCESS in 1h 06m 13s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/51b6d30d0a41421bbaf77b22e824fa91 : SUCCESS in 1h 31m 36s","accounts_in_message":[],"_revision_number":1},{"id":"652f6fe87e6617e8460d3f70d29880bb09bdea97","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 10:09:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/61c06125de5046a0ab94cefbced7921c : SUCCESS in 2h 31m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"58f2190c71ea965d37078f3c9e62e4c4e6797e95","author":{"_account_id":31108,"name":"Gowrishankar","email":"gmuthukr@redhat.com","username":"gmuthukr"},"date":"2020-01-29 14:22:21.000000000","message":"Patch Set 1: Code-Review+1\n\nAgreed to support binary build if package available.","accounts_in_message":[],"_revision_number":1},{"id":"f78e6e881003004631e03000e0d50217610166ce","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-30 08:41:20.000000000","message":"Patch Set 1: Code-Review-1\n\nwe support binary only from official repos","accounts_in_message":[],"_revision_number":1},{"id":"6c26a3ccd4bb7e7519f9b995cbbb212d5031cc6a","author":{"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"},"date":"2020-01-30 09:25:45.000000000","message":"Patch Set 1:\n\n\u003e we support binary only from official repos\n\nThanks for your comments. As this tool is associated with OVS, we will work on it to align with official repo. Till it is ready, if we need to build the binary image, do you suggest any other alternative? Is it acceptable to build with pip for binary image too?","accounts_in_message":[],"_revision_number":1},{"id":"56210e4de35aa36744fdde4011957515a931632e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-30 09:28:13.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nwell, locally you can do whichever works best","accounts_in_message":[],"_revision_number":1},{"id":"731fdf3d0a99aa5248c87f3517db1a9e3de85eab","author":{"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"},"date":"2020-01-30 10:07:49.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"18f63a6d5458c711091386352ac9da021c1a23d9","author":{"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"},"date":"2020-01-30 10:09:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"845a1ff6678642a322cb0516e5d64360da1ce3fd","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-30 10:11:00.000000000","message":"Patch Set 2: Code-Review-1\n\ndoes not RDO want to build these packages?","accounts_in_message":[],"_revision_number":2},{"id":"fc3db6c05d204cefe67429f43ee36f9c4df41734","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 10:14:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f043425a910da4f85a428de854e9d7a81e7c763f","author":{"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"},"date":"2020-01-30 10:36:55.000000000","message":"Patch Set 2:\n\n\u003e does not RDO want to build these packages?\n\nnetcontrold is not specific to openstack. It can be used in other deployments with ovs-dpdk, like kubernetes with ovs-dpdk. Ideally, it is preferrable to place it as part of platform packages. There is no conclusion on it yet, still in the process of finding the right home for it. \n\nMeanwhile, we wanted to enable openstack ovs-dpdk deployments with netcontrold so that multi-queue can be enabled to enhance the packet throughput performance.","accounts_in_message":[],"_revision_number":2},{"id":"2078eebe0e462e123928dc4961fc61da7ffb9269","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-30 10:45:37.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/ea155f2dc34a4c92a0681f602a3e8cce : POST_FAILURE in 28m 53s","accounts_in_message":[],"_revision_number":2},{"id":"4d486d08f2750bbd0d4602b2b567a1d6da8d3ad7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 12:08:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/c86600ae3db9491fad79ffbb24495bd3 : SUCCESS in 30m 38s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/bb28fffef1ef44bab184e35ead1c83a9 : SUCCESS in 50m 02s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/6513fe15e40f495089aec0084dc8a92f : SUCCESS in 8m 09s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/b511d45f0d194b229cb33b9e0b14dc6c : SUCCESS in 11m 54s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/91d7864ad9184214a74d136b83bff773 : SUCCESS in 1h 35m 16s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/66349ede8c5f41e198df9d0470e32dae : SUCCESS in 1h 10m 31s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/c4718623f5c7402e999f3648a63b30c6 : SUCCESS in 1h 57m 47s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/cc670971b0994b24ba2185283ccab3dc : SUCCESS in 53m 33s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/a792cb9384294f709ed81c60052e2b89 : SUCCESS in 24m 18s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/5d2de1e760d24ca7b60d93cf2db35a0f : SUCCESS in 40m 17s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/dd92458dec5a4895867537bf80dd122a : SUCCESS in 58m 24s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/114b60a1c3b04fbc9d69a0ebaab32ddc : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7465a8a551f74139a48737f3e14d6ac5 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/08094880f337496ebe497b24e989c417 : SUCCESS in 4m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a5821338719a403aa3a127fdbc336b79 : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b764d1e802824ace848cef3f6b708f70 : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe898d2561f64bebabfc21d1666e9f26 : SUCCESS in 8m 14s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9b831b8f6ee24e0bb41705ce49960643 : SUCCESS in 45m 45s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ec68dcb04e83426d98783beafc6983d0 : SUCCESS in 26m 43s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4189ca20caf648a698711c0e89acf63f : SUCCESS in 1h 06m 40s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6491128a1ba844a9be03ae5760dee93f : SUCCESS in 1h 06m 10s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/854705c29e524f54a1a5bf79a86a8761 : SUCCESS in 1h 29m 10s","accounts_in_message":[],"_revision_number":2},{"id":"b72babb27766d9e730e6fefb174f04ee12e938ac","author":{"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"},"date":"2020-04-20 04:57:35.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"aae8d41937211047adc9d35f061fd3249bc4c7ea","revisions":{"f6c3b25d6af960f7c21f35518df41a68e63773fd":{"kind":"REWORK","_number":1,"created":"2020-01-29 04:40:35.000000000","uploader":{"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"},"ref":"refs/changes/23/704723/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/704723/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/704723/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/704723/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/704723/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/704723/1"}}},"commit":{"parents":[{"commit":"dcd3669e379991c73169fcee0f6649d14387957a","subject":"Merge \"Revert \"To-revert: Temporarily disable Ubuntu bifrost image building\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dcd3669e379991c73169fcee0f6649d14387957a"}]}],"author":{"name":"Saravanan KR","email":"skramaja@redhat.com","date":"2020-01-28 15:50:49.000000000","tz":330},"committer":{"name":"Saravanan KR","email":"skramaja@redhat.com","date":"2020-01-29 04:40:31.000000000","tz":330},"subject":"Add support to build netcontrold binary image","message":"Add support to build netcontrold binary image\n\nnetcontrold release supports rpm binary version.\nAdded support to build container with netcontrold rpm.\n\nChange-Id: Idaaee886ceb6cf29494369e771f8ec453550e3f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f6c3b25d6af960f7c21f35518df41a68e63773fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f6c3b25d6af960f7c21f35518df41a68e63773fd"}]},"branch":"refs/heads/master"},"aae8d41937211047adc9d35f061fd3249bc4c7ea":{"kind":"REWORK","_number":2,"created":"2020-01-30 10:09:20.000000000","uploader":{"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"},"ref":"refs/changes/23/704723/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/704723/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/704723/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/704723/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/704723/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/704723/2"}}},"commit":{"parents":[{"commit":"dcd3669e379991c73169fcee0f6649d14387957a","subject":"Merge \"Revert \"To-revert: Temporarily disable Ubuntu bifrost image building\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dcd3669e379991c73169fcee0f6649d14387957a"}]}],"author":{"name":"Saravanan KR","email":"skramaja@redhat.com","date":"2020-01-28 15:50:49.000000000","tz":330},"committer":{"name":"Saravanan KR","email":"skramaja@redhat.com","date":"2020-01-30 10:08:04.000000000","tz":330},"subject":"Add support to build netcontrold binary image","message":"Add support to build netcontrold binary image\n\nnetcontrold release supports rpm binary version.\nAdded support to build container with netcontrold rpm.\n\nChange-Id: Idaaee886ceb6cf29494369e771f8ec453550e3f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aae8d41937211047adc9d35f061fd3249bc4c7ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aae8d41937211047adc9d35f061fd3249bc4c7ea"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
