)]}'
{"id":"openstack%2Fkolla~704848","triplet_id":"openstack%2Fkolla~master~I59dc2c4207af6812501b6c6acdb34e51a3e848c4","project":"openstack/kolla","branch":"master","topic":"bp/centos-rhel-8","hashtags":[],"change_id":"I59dc2c4207af6812501b6c6acdb34e51a3e848c4","subject":"CentOS 8: Allow SSH access to keystone_ssh and nova_ssh","status":"MERGED","created":"2020-01-29 17:06:19.000000000","updated":"2020-02-01 15:45:28.000000000","submitted":"2020-02-01 15:43:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"704848-1580571818998-9ea6cd4b","meta_rev_id":"c9c4fe61ee9bdd1b6c1e0b3485a4342ca1cdb5ce","_number":704848,"virtual_id_number":704848,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"date":"2020-01-31 21:45:01.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-01 15:43:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-02-01 12:30:17.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2020-01-31 07:17:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-01-30 11:32:37.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2020-01-31 07:17:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-30 10:42:18.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-01-30 11:32:37.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-01-31 07:17:42.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-02-01 12:30:17.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-01 15:43:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"372c101ed812a6ce918e228bab9714ddb6c6c47b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-29 17:06:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"025889f579bb9177dce0ad049660dd408dff1506","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 18:15:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/dfc6ac9343284965be627dfb13a0addf : SUCCESS in 1h 06m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6eabd64b93fb0c90ab8432ddb89fedc3265ab4d0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-29 18:16:12.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/564fbb8049fc4508a85a08d1d17b4c0f : FAILURE in 5m 45s","accounts_in_message":[],"_revision_number":1},{"id":"22e5a544d284d0e310c7de4273647c8e91e61bce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 18:57:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/54538450ac2947b09b43ef9162450bbc : SUCCESS in 39m 36s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/9dc3034d6a7d4f22b7e9db9cb607e39b : SUCCESS in 46m 49s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/4b97816e884245a99442f142f2f08b82 : SUCCESS in 34m 34s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/490923f8df6f4b4585f5d8e1c908d882 : SUCCESS in 32m 48s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/9e3b730254ae45458cc2764e08ad48c6 : SUCCESS in 1h 44m 39s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/f284ed79457143c98b70a975bc01f86a : SUCCESS in 1h 23m 18s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/52cb95b1ee8b4b908b9547a4dd29a2e9 : SUCCESS in 1h 45m 14s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/df0c5a798e454490b81d96482da6f871 : SUCCESS in 1h 01m 16s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/7e96e62fec2c415b9f65dc40273ac9b0 : SUCCESS in 35m 16s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/9d067c10106e4d04bd3ad3facbe50513 : SUCCESS in 34m 51s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b9ed2a9fc1e54965a44a79227a2e7cc1 : SUCCESS in 1h 03m 18s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ab0c570ba8644169a4d2c83400809d82 : SUCCESS in 6m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5a62d72911d499cb1e18fa5b545d880 : SUCCESS in 6m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aa1de25ec41c41bb832512f57773784b : SUCCESS in 5m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3100d3a64abd49b981d25b98bb4fb0a9 : SUCCESS in 6m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/539f4c6b13e54c6e8d61cd655c9dacb7 : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/812545a4724840b692719191189299c7 : SUCCESS in 8m 43s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7f4de29dbdc14e5d908c3c9d22aae847 : SUCCESS in 34m 24s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b5d5fef72e974721a7c0fd2a0c12a2bc : SUCCESS in 39m 10s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/436282fa74f542c197b9c3620136f3f7 : SUCCESS in 56m 52s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/20a32bb50f024627bb5df836c9bd2ea1 : SUCCESS in 1h 01m 23s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b74fde55b51a445b81b1df090b69b149 : SUCCESS in 1h 07m 19s","accounts_in_message":[],"_revision_number":1},{"id":"638c73c74f6fb40c37bafd0182e30217975f2525","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-01-30 10:07:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"91355a9680425b779764aa7bfef04cae83f31b0f","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-01-30 10:08:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"de3c13ec5d655ed508d1e22e21a62d1fc391fe1b","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-01-30 10:08:27.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5d996143ec8206e7621ef3948281353f74f26c3d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-30 10:09:23.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0f92411c0e8dbcd046c2eafb7848564c42933214","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-30 10:13:02.000000000","message":"Patch Set 1:\n\nIn a queue :)","accounts_in_message":[],"_revision_number":1},{"id":"656bf181920e28dde2c01f69d87d6ca57311059b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-30 10:27:57.000000000","message":"Patch Set 1: Code-Review-2 Workflow-1\n\nnot helping, to check","accounts_in_message":[],"_revision_number":1},{"id":"d448401c811caa187f64c1073fb3041cc592cfa8","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-01-30 10:31:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2b13025bd04505a96baea634ef448036a9055dc3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-30 10:33:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0d82f0e2096460a1ee869f8b542bc023b0af6649","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-30 10:37:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0b676b4d2a2f867def5bbc47fb17c4b6b4d8d326","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 10:39:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4f3c672dbcb68e25025d6aebe2b63bafbbfe2aba","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-30 10:41:20.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b84a86523dcf8b18212aa649564cf942f670e67e","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-01-30 10:42:18.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"cb1c786594dc3a679e8ed4a44fc972b85849607f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-30 11:10:48.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/efde3828cb0f4f4e998ef15f97c0b169 : FAILURE in 6m 40s","accounts_in_message":[],"_revision_number":2},{"id":"f96ca24b76e4ed554ef706fc82a65416b55532ef","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-30 11:27:52.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2f6ca7b5dfc30246eee368a2b6139e59425bbb00","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-30 11:32:37.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"195cee4ec1a5ef7de592499379b92c75d4a38328","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 11:41:37.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d6619c0d2525ba6f026378471aa4cd44c8430765","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-30 11:51:44.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/fa78a81ce69043cba24e38dc5657bee8 : POST_FAILURE in 5m 54s","accounts_in_message":[],"_revision_number":3},{"id":"4c97c0d342612dd97aea934791fe397be8385144","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 13:29:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/a63a789043944de8b3a666c17bbc62f6 : SUCCESS in 33m 08s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/712249262e1d44f0bd1ec5763e4ffeb5 : SUCCESS in 44m 16s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/0a624c479a3d4f2c9b843a4b2cd16af0 : SUCCESS in 26m 27s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/0611d3d7274641a5a321bfb0731e1c06 : SUCCESS in 30m 43s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/c8cc6320547b45589d1c0b22de8db7bd : SUCCESS in 1h 44m 24s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/975e12dcb97948ac8cd175af26326a82 : SUCCESS in 1h 23m 37s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/fb23af9e365b41019a93bd4463dd8825 : SUCCESS in 1h 43m 12s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/7773d8d0154946cca98fe62c59c000db : SUCCESS in 55m 37s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9609f9a4b0204d0ebb67665847596da6 : SUCCESS in 29m 38s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/c12a5a71cdc04caf9356e39e4acaa7dd : SUCCESS in 21m 29s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/36178615e7434f469fc7718118b46dc6 : SUCCESS in 55m 50s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a0fceb328d974aab85b9bb9300ab21e0 : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0af242457b404de7adcaad3cc7445f35 : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/06c203d8bc644a06a0e0a321fda99735 : SUCCESS in 6m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1daa302814fe4168b6dbb8c92bdf451c : SUCCESS in 6m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a98a535306c44054829a7f0528e43e3c : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdee1024d8814a71abc08beeba103f14 : SUCCESS in 8m 27s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a2846b5a594f4b4dab30b7450f24deaf : SUCCESS in 34m 03s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/48c27c5ab998465ba36df09be08af17a : SUCCESS in 42m 43s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2a2c46b307d8498c9a2d6f89c33273d9 : SUCCESS in 58m 48s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/adf26fb097924b999097fa53a33891ef : SUCCESS in 58m 03s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/992637beb2f0488bb3c2e8544885f7ab : SUCCESS in 1h 05m 30s","accounts_in_message":[],"_revision_number":3},{"id":"bad33a41e6b8a7fb7b241173402671cd50eeffd8","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-01-31 07:17:42.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"9e77efb6b76d44f1da0900f01cf2fbd4c8ec79c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 13:20:48.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"bc2d7489e79abf8aed36f90b465ecb233078a71e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 15:09:51.000000000","message":"Patch Set 3: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"e9118e8e3c5585bd531445dbbb95a3e7ec54a07c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-31 15:19:35.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"425d1a2845179b36f5bca78644cd3004cb1b87a3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 15:22:37.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7d86bf87e0830be9764692eb32a0561a1ce4d682","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-31 15:29:39.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/f6cc9241e034429bb54ff32975dfd28f : FAILURE in 6m 59s","accounts_in_message":[],"_revision_number":3},{"id":"b267731489a10c0068e82e08e95301ebba3a02d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 17:03:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/9d712bb764e04d879c81dba2c7aca580 : SUCCESS in 32m 18s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/bc9809a194f44f69a3f7f7e2d0fc01ec : SUCCESS in 51m 22s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/3ef91fa8fc804aa59274c9e97de16aac : SUCCESS in 32m 27s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/92e9a059eb1d4c3d93408e8fdb187c13 : SUCCESS in 32m 41s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/fbf5c9e5568f445482b548c2fdd69651 : SUCCESS in 1h 35m 57s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/5cf4e0362fbf475b8e5b0b1949153592 : SUCCESS in 1h 18m 39s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/a281d1dd4c3949669e068fe33268254c : SUCCESS in 1h 39m 07s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/3ae9c35aa74a47b19fcb1c412de4764a : SUCCESS in 1h 16m 15s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/bdebe8bb981b4bd9b46cbacebed9426a : SUCCESS in 34m 13s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b429bdfdb0654053ab14683a28204199 : SUCCESS in 30m 58s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/49d386b101534b4b8db027d6eabfffea : SUCCESS in 1h 00m 21s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/988dc7438705417c80359d89ee07f9ea : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d8a8e44d23e4154a898c0e8b48ddc50 : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ce8d2645f8e404abc1752017d1e2f41 : SUCCESS in 4m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a4e5419bef374d7d857993a5298e0eb7 : SUCCESS in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/64293e54a8544af39dacc1b8039e6b23 : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fc8beba2c9c43a9b5afc403413cab79 : SUCCESS in 6m 55s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/aa311db8aef34720afa9799296565d54 : SUCCESS in 27m 06s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9f31305c6b6745e2bcd869e791f6fe1b : SUCCESS in 49m 05s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/edf93cb8edea4ce0921b95f3be251f8d : SUCCESS in 1h 05m 33s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a2437a567ff043c0ac48d3c7629a19b9 : SUCCESS in 1h 41m 30s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a985068ed4af4772a79bcec6e7aab2d4 : SUCCESS in 1h 37m 51s","accounts_in_message":[],"_revision_number":3},{"id":"6908bd668b5422982c8457fc03213a1630c0e602","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 17:42:20.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5936e10863cfc3388a23cc070406b087e14ab4e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 19:26:03.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/3a74759ab0424fc9b00a78027895fde1 : SUCCESS in 45m 00s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/51a2b80d78a740a6a063a2fc77be0b31 : SUCCESS in 33m 56s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/b88beb5a986f41cbaa1b2561ffe02b3a : FAILURE in 58m 04s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/d124c74cc97a4b2198dacd0370fc47de : SUCCESS in 1h 25m 18s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/469680b7bf694030be1c1c03aa0ec21f : SUCCESS in 1h 01m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/870d96df569141c69fe807bedf80bfd4 : SUCCESS in 8m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e449e8414fc34400a54478a5b502aa6f : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6be0fe8c5ce249dfa7bfb23fc8a5f393 : SUCCESS in 6m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/55914ff1f5d642b59d44de689ebd88cc : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87622514b40243b59c62913b8b541c7a : SUCCESS in 16m 24s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a097eb399c3c4f1085f687f12bb895b7 : SUCCESS in 34m 02s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/505664c3a0df41d4bfe002674807ade3 : SUCCESS in 52m 40s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c3dd9bfddd024416a4d3f5120216107b : SUCCESS in 1h 07m 11s","accounts_in_message":[],"_revision_number":3},{"id":"f4ef3abf74749ad0ae946d9eee60f527bb9377e3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-31 21:45:01.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"f605f6cd844637a136efcf3c7c352355757bce88","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 21:46:38.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f5581743861972a2875e730882f74f3ab9239fab","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-31 22:13:23.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/bcd23c7f3a374464b449daa1b92cb369 : SUCCESS in 25m 25s","accounts_in_message":[],"_revision_number":3},{"id":"817decf32b05730182fdc83104ea9c94e2d463db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-01 00:18:28.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/0bd9def96c8f477585f8538e3f4a3e41 : SUCCESS in 32m 11s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/f319007f0533475fb3ba60265fd23cfa : SUCCESS in 52m 24s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/e48c3d7bc2c041e48e9eabad11a804ab : SUCCESS in 45m 59s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/2f3ccaab8f02488580c0de8938174b4d : SUCCESS in 27m 15s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/02db565f90284c5b843a271f0e8228dc : TIMED_OUT in 2h 02m 35s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/bb55f7f50a1d4392bb3d3520f505a979 : TIMED_OUT in 2h 06m 28s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/86c3c5a9c2a94dd5b988654e801b5b2e : SUCCESS in 2h 31m 22s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/45f7409618db4e47841a9a2a73ca8006 : SUCCESS in 52m 11s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/4429d2d2e179433a962cf66dbcaa53ba : SUCCESS in 25m 46s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/84928c98663246c79e4781b490cf34a3 : SUCCESS in 19m 41s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/6cc4af7d57c84e85bf8616c82c13dd25 : SUCCESS in 1h 17m 51s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1cbf65f46ae2463198353e63ea9ee8ec : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b643f103a3b14ae0ba5929eac65ddeb6 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/98ad0d813ed54c509102b808f8f5e700 : SUCCESS in 3m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5c96507d5fb847cab51cebccbf415154 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f312d13bce694e9bab12e0a10906d58c : SUCCESS in 3m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c32346f24424aeda0b66243f39e6d55 : SUCCESS in 5m 11s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/071e9fd05a0844e0b0f8ed7bb3b12bdf : SUCCESS in 23m 00s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3541d661e57b4a50bf24402792d7a46f : SUCCESS in 24m 43s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5bb1789ff5d346d48c034114a45e2fcb : SUCCESS in 52m 21s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b268678cd2fb49fb874916a65bce85a1 : SUCCESS in 56m 02s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c7fea4ab70ee4db188ecdd5e6bd89c51 : SUCCESS in 1h 06m 46s","accounts_in_message":[],"_revision_number":3},{"id":"73b4e85b73e1ec453d6572130e7cdbb1434326a6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-01 08:24:37.000000000","message":"Patch Set 3:\n\nrecheck\n\npoor c7, soon you be gone ;D","accounts_in_message":[],"_revision_number":3},{"id":"d63a62710e018fd452d4ebccbe33f71decaefe7c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-01 08:25:28.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8b0c46475b7c17963bc83703ad3e7ef81fdbeb9b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-01 08:51:32.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/3b0ac990cf6b499dae3b04d140cde10b : SUCCESS in 25m 53s","accounts_in_message":[],"_revision_number":3},{"id":"4ffd6f7d49df9a7bb8781a5cb04288ed8448cd31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-01 10:04:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/acf74b3ae1a341939db460bd301ddc50 : SUCCESS in 32m 58s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/bcfcb2c6af5d49c59e95f56e1925e4ec : SUCCESS in 37m 50s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/b5b5754fbc6d47e8af2ad16a4e237a60 : SUCCESS in 21m 10s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/0f0769726e75401daae0c4269bf9ef92 : SUCCESS in 24m 48s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/4477fe23ed5245e8a276cd2747dbf164 : SUCCESS in 1h 24m 40s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/817ead4cb5c84018941ee57ab2f791f1 : SUCCESS in 1h 10m 34s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/ae7dff87048349f69ccaf4ef16bf81bb : SUCCESS in 1h 35m 37s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/1fb215d3a53c4f1ca2205469a2e5f038 : SUCCESS in 51m 29s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9af340151a4a49e8b351c2d6573d9064 : SUCCESS in 21m 53s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b23a9e7676b640b48b67b603dbac2864 : SUCCESS in 18m 53s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/97d2fb7d95f14098b7f5615016f0f088 : SUCCESS in 1h 03m 31s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8fe40965116f4be0af0adb074de67494 : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40280453cd2c4e74a091bbd8100dbe78 : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef3e0331e48849e095a9f669e087db2f : SUCCESS in 3m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/adbbf3feacb1453bb75c5e6db3358dc9 : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba71c27831a74fdda659be3888400568 : SUCCESS in 3m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9118bc62573f440cb889f24cc5e3bce8 : SUCCESS in 5m 41s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9215ddcfcfc044b5972e01dace672a31 : SUCCESS in 20m 53s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6d83104a955741c9b049956072f0abf9 : SUCCESS in 23m 59s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d462e00f9d644de088576f50cc733a44 : SUCCESS in 1h 08m 49s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3c7353e4f7de4f61abe4331a67f307ed : SUCCESS in 48m 18s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/90cbdded565b454c85d85609f3119825 : SUCCESS in 1h 00m 37s","accounts_in_message":[],"_revision_number":3},{"id":"35f4a25477ee8c725577a9dfb8ab893a469b06c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-01 10:04:39.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f4620f0511589321abc1c9b0b863d7805a4a5693","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-01 11:30:57.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/51dd3c260b8440f1a97659f4121dec68 : SUCCESS in 36m 49s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/5f4621d6cc0444c393faf5ddd14897b9 : SUCCESS in 28m 22s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/fa326a0a861d4813952d989abab7d6e8 : SUCCESS in 1h 17m 02s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/a83bc9858ed0474e832c735ebb0eeeb4 : SUCCESS in 1h 25m 21s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/915b3331d288491ab827a3c3439196ba : SUCCESS in 41m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e4007090b2e8498db6af44349c6baa2b : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8ec3272f40a4942b7d02b243fc72dec : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b37639bdc0cb4ea496a7487ff87080a9 : SUCCESS in 3m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/68884f77e2fc4fcab5f28ec61d4fcd58 : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32579b39a6ac46a0a70b3336a2acc16a : SUCCESS in 6m 06s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4e198db848884589abe76927d501f829 : SUCCESS in 25m 35s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7ab63b79d8d14a34a8797f4d85388ef9 : SUCCESS in 55m 00s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c58bc45029994d99b804133074561d8e : FAILURE in 13m 09s","accounts_in_message":[],"_revision_number":3},{"id":"221d4ad87115af9c8c9664f627a586d3387098fe","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-01 12:06:26.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"f66c0f748cde416e5502f0ecaca991e1859e8250","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-01 12:07:16.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3de8553611bdfae3fbc14553685d161c70c08de7","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-01 12:30:17.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/ce88236358e1491582bd989a7dc066c8 : SUCCESS in 22m 59s","accounts_in_message":[],"_revision_number":3},{"id":"8ded90813e147b114369aa41d0cb26b5a39b7527","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-01 13:41:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/bf23f825f7684c6f8822be70d14318a1 : SUCCESS in 44m 15s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/8534e10912c143288b18686afcd6772f : SUCCESS in 39m 33s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/2b8e61fb023b4636ae8cb5673fd0427e : SUCCESS in 32m 06s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/2c8fba9de2834cfb81c7ce773e2ca003 : SUCCESS in 35m 19s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/b0835b0929e54532b3b41ab4531b04d4 : SUCCESS in 1h 23m 39s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/3e4330ac36d249e8b8b9feaeb0251430 : SUCCESS in 1h 32m 08s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/584834ed014e4e0facedd8674b9af14e : SUCCESS in 1h 25m 52s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/939197b4a8554649853f57c0215e1fc3 : SUCCESS in 58m 49s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9aff7ce13b9544fd843b249d25621423 : SUCCESS in 27m 55s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/63930f91a1454d95aa70e801cf506ad4 : SUCCESS in 27m 30s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/424b4150ec8247dfa308a4cc12e31fb0 : SUCCESS in 59m 29s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/021b28946de3412c9370ef90c04b6f0c : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80573a0ad035488d9246885189d252b9 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f74bce8c948c4b3aa464b86e50dacb0a : SUCCESS in 2m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ba06c566ee7341af91102d61839f07eb : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4dae00c6d0ee4f1eb9d5f88afaec672e : SUCCESS in 3m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf818c2236d4420985bdbebaa235da6a : SUCCESS in 8m 27s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1d32548f0e4f4ba4b06cdc338ce474eb : SUCCESS in 26m 56s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/426ad31ec1b94d15914298d8651f41f2 : SUCCESS in 26m 28s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/25cf1b6171da4f6298d2732e9de03de9 : SUCCESS in 1h 16m 57s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/59114b786b1b483086a62b21b80b4953 : SUCCESS in 50m 33s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/391798c4111b4232881fc7ba0dff3261 : SUCCESS in 1h 29m 21s","accounts_in_message":[],"_revision_number":3},{"id":"0521d87c7c0a61df4a5880e09eebb222a1867f40","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-01 13:41:40.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"28fe208d72639cbe1fdfde9698bbd4cbd4d35a32","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-01 15:43:38.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/4fa5798d731e4f7da26242fa7b663c1f : SUCCESS in 1h 04m 04s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/3e86c60f6a8d4c6d857182ad4cb900ec : SUCCESS in 42m 49s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/b46c291881ba47479adff292790e0a7d : SUCCESS in 1h 47m 38s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/0416158137e24711b834612a06028453 : SUCCESS in 2h 01m 06s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/9999aeeefbc54e07be8c2a14f5d9068a : SUCCESS in 1h 04m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d1fb2d63e1e941d6a4c1d0aac41bdd59 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dde95f090c5b4194b2c92b3ddda0b55a : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc98e01430fa4e1199ca796eb04a335d : SUCCESS in 4m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/37e2c0160f6b44909c1bd4b454efe83f : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c049f147d704e78b63808140ac6af07 : SUCCESS in 6m 58s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9eb7c23be1d34ca08bd64535b3b9af52 : SUCCESS in 46m 50s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4b73569fc76b4f138d94840c2689db8a : SUCCESS in 1h 20m 56s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7489c280707c4d1eb31d299c2ad0a6f5 : SUCCESS in 1h 28m 58s","accounts_in_message":[],"_revision_number":3},{"id":"69113d45f1518cbefed5e190424da3ff2b9c05e3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-01 15:43:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"e764f9db475438179d98004a8f3cfeeb8abc207c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-01 15:45:28.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f08fafa7bfa5453a853a87e2f873d1ce : SUCCESS in 1m 28s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"892ae92d971fd7a2c8d39e416512bb69f7fd833f","revisions":{"9490b517b9e04a1d8ca7f78812d5e687f1c2f401":{"kind":"REWORK","_number":1,"created":"2020-01-29 17:06:19.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/48/704848/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/704848/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/704848/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/704848/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/704848/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/704848/1"}}},"commit":{"parents":[{"commit":"aefd25fe58d4d9cdd4dbe6867a104fc46f0ebb89","subject":"CentOS 8: Use upstream Ceph/master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aefd25fe58d4d9cdd4dbe6867a104fc46f0ebb89"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-01-29 17:04:43.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-01-29 17:05:27.000000000","tz":0},"subject":"CentOS 8: Allow SSH access to keystone_ssh container","message":"CentOS 8: Allow SSH access to keystone_ssh container\n\nThe centos:8 image contains a /run/nologin file, which prevents SSH\naccess to it. Remove this file in the keystone_ssh container to allow\nlogin via SSH.\n\nChange-Id: I59dc2c4207af6812501b6c6acdb34e51a3e848c4\nPartially-Implements: blueprint centos-rhel-8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9490b517b9e04a1d8ca7f78812d5e687f1c2f401"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9490b517b9e04a1d8ca7f78812d5e687f1c2f401"}]},"branch":"refs/heads/master"},"c147d635f2d63ca2062d66bd0de8702659d05c2f":{"kind":"REWORK","_number":2,"created":"2020-01-30 10:37:23.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/48/704848/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/704848/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/704848/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/704848/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/704848/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/704848/2"}}},"commit":{"parents":[{"commit":"70423f19596f92ed01c89e7d62f8c46a6dba6a23","subject":"CentOS 8: Use upstream Ceph/master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/70423f19596f92ed01c89e7d62f8c46a6dba6a23"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-01-29 17:04:43.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-01-30 10:37:07.000000000","tz":0},"subject":"CentOS 8: Allow SSH access to keystone_ssh container","message":"CentOS 8: Allow SSH access to keystone_ssh container\n\nThe centos:8 image contains a /run/nologin file, which prevents SSH\naccess to it. Remove this file in the keystone_ssh container to allow\nlogin via SSH.\n\nChange-Id: I59dc2c4207af6812501b6c6acdb34e51a3e848c4\nPartially-Implements: blueprint centos-rhel-8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c147d635f2d63ca2062d66bd0de8702659d05c2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c147d635f2d63ca2062d66bd0de8702659d05c2f"}]},"branch":"refs/heads/master"},"892ae92d971fd7a2c8d39e416512bb69f7fd833f":{"kind":"REWORK","_number":3,"created":"2020-01-30 11:27:52.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/48/704848/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/704848/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/704848/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/704848/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/704848/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/704848/3"}}},"commit":{"parents":[{"commit":"70423f19596f92ed01c89e7d62f8c46a6dba6a23","subject":"CentOS 8: Use upstream Ceph/master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/70423f19596f92ed01c89e7d62f8c46a6dba6a23"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-01-29 17:04:43.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-01-30 11:23:48.000000000","tz":0},"subject":"CentOS 8: Allow SSH access to keystone_ssh and nova_ssh","message":"CentOS 8: Allow SSH access to keystone_ssh and nova_ssh\n\nThe centos:8 image contains a /run/nologin file, which prevents SSH\naccess to it. Remove this file in the keystone_ssh and nova_ssh images\nto allow login via SSH.\n\nChange-Id: I59dc2c4207af6812501b6c6acdb34e51a3e848c4\nPartially-Implements: blueprint centos-rhel-8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/892ae92d971fd7a2c8d39e416512bb69f7fd833f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/892ae92d971fd7a2c8d39e416512bb69f7fd833f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
