)]}'
{"id":"openstack%2Fkolla~704965","triplet_id":"openstack%2Fkolla~master~I6d409e56b424a039f5aff9af86ca1b026c2c6286","project":"openstack/kolla","branch":"master","topic":"bp/centos-rhel-8","hashtags":[],"change_id":"I6d409e56b424a039f5aff9af86ca1b026c2c6286","subject":"CentOS 8: Relax ironic iPXE EFI bootloader location","status":"MERGED","created":"2020-01-30 11:27:52.000000000","updated":"2020-02-01 15:26:42.000000000","submitted":"2020-02-01 15:25:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"704965-1580570717357-87017a1f","meta_rev_id":"fab2b57a8f9f31273e7c5818f591d6e06b1360fc","_number":704965,"virtual_id_number":704965,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-01 15:25:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-02-01 12:31:56.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2020-02-01 12:01:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-02-01 12:01:11.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-02-01 12:01:17.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2020-02-01 12:01:11.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-31 09:26:22.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-02-01 12:01:17.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-02-01 12:31:56.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-01 15:25:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ba292a833aa4ef99969bd85e66ac514ee0acc09d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-30 11:27:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"23e8906d3f18df8d2597bbd2d08d6a13d09387d8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 11:44:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"476493ebb024a64d58b49c629941f8ab7fbb6ef1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-30 11:54:53.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/f368ec3b222f4b7e9599bf3233ed90b8 : POST_FAILURE in 7m 22s","accounts_in_message":[],"_revision_number":1},{"id":"ccf796f749d24b93a8a44de2abcc6fd83b134560","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 13:34:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/2bacc563c12b4532b8b499f20f9d5195 : SUCCESS in 46m 11s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/38a5b94dc0c94c9ebd4ca8e40466c19f : SUCCESS in 55m 13s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/57bd544439a0451b9c1dd7fe70766400 : SUCCESS in 29m 47s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/e17f624f79b34d68abe4df5c3b28f20d : SUCCESS in 35m 31s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/fce11a6ff288414abc697710e237d09c : SUCCESS in 1h 28m 08s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/77de41b62b534254972eaa359bf3c34f : SUCCESS in 1h 17m 26s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/70c11ef8ea544cada52fd7ba787ab7b4 : SUCCESS in 1h 38m 47s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/67eb9fc3c2a44c7999bd9a531232f4fe : SUCCESS in 1h 02m 46s\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/d1e1f984dc6644f6b860deefe7f73a0b : SUCCESS in 1h 36m 11s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/767391141c6b4208ba18aaa30042b193 : SUCCESS in 25m 18s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/3bfda7e420874356a588229da5811c5d : SUCCESS in 21m 29s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/6b93a29ff2124ed79917d254907b029c : SUCCESS in 1h 00m 35s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a721cab9b2ec4df1ba4b528af51786ce : SUCCESS in 6m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53275ca27f294b748b6099ee308d9c44 : SUCCESS in 8m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/80b93cce21a94ffe81dc4b614bd345de : SUCCESS in 7m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/96176b4aa1244991a8b575dc4fbfe87b : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6be59a0abb03407b870ed5e223ebbdc5 : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad8ea6f6922a4a3fad999cceee5c4a16 : SUCCESS in 5m 40s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9704c1d511e24c2caacaa17be46f6271 : SUCCESS in 24m 02s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1d676e1169064c88a67cc932534026dc : SUCCESS in 28m 22s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/98c68f68c1cb4f159c172f41a62dc350 : SUCCESS in 1h 06m 43s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/877784ab86f34c66ba1f1afd8d06e7fe : SUCCESS in 1h 09m 45s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3e18a80bcf9c42298d9635aa4dc44eaf : SUCCESS in 1h 10m 44s\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/621e54d07a0d4729b9615509726aed5c : SUCCESS in 1h 04m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"22afb494ed57c94152ad1ff650ac9cc07a81b804","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-31 09:09:39.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"ca414eeb755670036ac77d583b97c2acb6aea136","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-31 09:12:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"50ac0b04afcfc4cb975313e34e188de6bef9e74b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-01-31 09:26:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0dd270e3d4e50631ce34b35b7b291dd370ac6a7b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 13:20:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8bb0d9b3125e3c2c76d287c7b76668847e575829","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 15:09:52.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"9101331d7b20c14fb4a6cbfb8a2688bb645fba97","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-31 15:19:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"15ac87f090864ff7d2fcf07029bcdea743f9ed63","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-31 15:19:44.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5d540ec06414640278f643a67977604b563333ef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 15:22:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c190b37e6a980f0210a28d1747c06cd4cbaea069","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-31 15:29:54.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/9ec3aff8f0dc41ad94c6c285b130520f : FAILURE in 8m 04s","accounts_in_message":[],"_revision_number":1},{"id":"cc685510ece14d31083936cf1b07a15f5e7db7e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 16:59:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/4bfbaa0f2969409f9758320781bd6127 : SUCCESS in 1h 00m 31s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/9ce066c00b7e4f15865c8cb526be5981 : SUCCESS in 1h 04m 10s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/721e1cb95a8b444d8e99093fbba220e4 : SUCCESS in 35m 17s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/541b984d6c3e4492897e694de590f97e : SUCCESS in 33m 24s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/123bbac1306647adbf598d8b6d7fab2c : SUCCESS in 1h 20m 31s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/b8a646a397fa4714a11b877af4cfe4b8 : SUCCESS in 1h 38m 33s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/3255d5da8fd448d38ac7f03fb8783c08 : SUCCESS in 1h 25m 28s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/3ef4e46014fa4826aaa70a953b8e1593 : SUCCESS in 55m 09s\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/131cbce616e84ace8723173d1c43e015 : SUCCESS in 1h 31m 27s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/33a6cd6a23334c0bb00d7d7bd08730d0 : SUCCESS in 45m 34s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/d8f49cdc009240f79da3b8a8a654625f : SUCCESS in 23m 28s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/aaf02661f3b245faadf413735f02c71c : SUCCESS in 55m 06s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/389a5443f9834e9c8fdf1b666f0f14c9 : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/730c2110fead4606a0c5fcf4c3406592 : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/53eb62ce5f69439b9b0365b623e89fd9 : SUCCESS in 4m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c5d8936c9f7c43f29ed963591e6eedef : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d9203b5d70e54638a42141753b332626 : SUCCESS in 4m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22add7f2b9654b2e83955ef5e8b966b5 : SUCCESS in 7m 37s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4db08d9797ef4ac6a10661a3f296b909 : SUCCESS in 49m 46s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c6abc964d1054b218debda45d8345543 : SUCCESS in 30m 15s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f53d3b6d8d70466e81ed671fe0f06bb7 : SUCCESS in 1h 00m 29s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/056235dffa9e4ca4953a2372c6f75ac2 : SUCCESS in 1h 02m 39s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d3f2666e9ba4472b8bd5d535fbbef85c : SUCCESS in 1h 07m 06s\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/5eb141949824404e839691e27be3d98b : SUCCESS in 1h 17m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"398d9e868ae4790dbb990757ddb9b9734eb880cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 17:42:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d12db05235446e2f9c7c80a5fd5dbfc1be17c6c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 19:26:04.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"7d3b28af0bd677d8cc91f10a6f90a53b718b9d57","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-31 21:45:14.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d6e03ea781af75a2a312aaab8c1487fd190e0065","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 21:46:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e47db13480f15f08ce6dc7440d063c6e064aa263","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-31 22:16:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/65f5031980ed4c84a9b2d9d5528ea570 : SUCCESS in 28m 04s","accounts_in_message":[],"_revision_number":1},{"id":"7699999e2985e1be7e39a667060826ab73126205","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 23:32:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/69a6aeba025e47d2a23be810b231af66 : SUCCESS in 29m 24s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/c4bc2ffa130c4c7884b409eecd97c56b : SUCCESS in 58m 17s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/0cce8c446ad847fdb3eaef438da1ca37 : SUCCESS in 1h 26m 52s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/2dd07f9b021b425f82b95b1f0030ef36 : SUCCESS in 27m 00s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/f88d95658fd946fca370cf435ca2d4fb : SUCCESS in 1h 28m 14s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/ae64169f4a944447a6ee2bcdb54e61ec : SUCCESS in 1h 14m 53s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/a2989457cd1847239c864a3b6e264db1 : SUCCESS in 1h 28m 47s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/a3bfb39f01bc4dae989df6e03d32e6ce : SUCCESS in 54m 30s\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/ef88c86c477e4026853c41036db5d3da : SUCCESS in 1h 41m 55s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6e010ef7bed540edbe90a9c0e13b1aac : SUCCESS in 20m 34s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/eb5620a7994e46b18624b43d11275770 : SUCCESS in 19m 03s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/03d12ea44ccf47b5ab0bdf1fc7fe162d : SUCCESS in 55m 23s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9743075e2e4c4d138e4c5dfd7267d1f6 : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d992e9cf8675432d818bbc24b777a77a : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d69142db6e9d4e6abd1840a6da8331a3 : SUCCESS in 3m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6e24fe508d49486f96d8a5b5abf5bfad : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5157730a278a4a67a3c4734031ca9298 : SUCCESS in 4m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d836e21a57c4827829593007ae7f792 : SUCCESS in 5m 09s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/abb9ae9808b44b1a9a975591b3991449 : SUCCESS in 36m 01s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7a51e5a47ab84dcdad0ae0dddf847637 : SUCCESS in 25m 22s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/067cea86dabc4025a29fc4612ab28469 : SUCCESS in 54m 24s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3e1aca9e2e414c39bef2523dc5614790 : SUCCESS in 54m 24s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7a6bb5dd1f764c1bbb1cb5a8a5032370 : SUCCESS in 1h 01m 01s\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/c1c3fe1fdf554c50b32e880040e749c9 : SUCCESS in 58m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"213e8e4b92582c36ce18f49efcceae45d3cee885","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-01 10:04:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9f222e9806bdddba131de2b35ea7e567f14103d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-01 11:30:58.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"8f694cac043b7f4933a18be2c6c4733601e32d79","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-01 12:01:11.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"1328f2caa91e247e79d2f2edd2808ced9836f7d9","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-01 12:01:17.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d34149f353fa6fe0210c7fbce94cf9f25579600a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-01 12:02:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5f0243743a4b8344307f71ea79f73e53c81a0c08","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-01 12:31:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/1c869c6467644c06a1bdd60ae88ed185 : SUCCESS in 29m 49s","accounts_in_message":[],"_revision_number":2},{"id":"bd5c23f6e8fe341cc485f32cb9e6859b00c5e308","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-01 13:38:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/59aa0706eb94469fb3e5685aebe2fd82 : SUCCESS in 28m 45s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/75586713fa5d4c4d8267066f20419225 : SUCCESS in 39m 33s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/75b797e9101b4d4e988ff675eb5a1db3 : SUCCESS in 22m 11s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/e9e19da1121e4b6c9766d1b8cd30812f : SUCCESS in 34m 20s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/71476465cadf4db19bc75b66596db7f7 : SUCCESS in 1h 27m 24s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/db10ec2009ef4039a1a129042bb6bf32 : SUCCESS in 1h 05m 51s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/17a7d5e63b794a85b028ba722726c867 : SUCCESS in 1h 21m 28s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/48d0bc4b56624b28be25a4e0791f6951 : SUCCESS in 54m 10s\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/240693dafcf245ab8e766f70da77efdf : SUCCESS in 1h 32m 30s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9b1638fc08524aa18e18066d2f1d3528 : SUCCESS in 31m 15s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/884f19ea86bf42bba2020846b42296d2 : SUCCESS in 18m 25s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/eed78f1a1ba640d9a9b1c3f5ead52ba6 : SUCCESS in 1h 09m 16s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fccffccfc5964e1ab583e2ff2d0f5906 : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d27921fd22640ceb24c1358606fa4d7 : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/383187001bc24bfe950224631b69c5fc : SUCCESS in 3m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/099037acc477490a95cb01210d9f26b1 : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dfe6b23827244bf38c9aba34fe927731 : SUCCESS in 3m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6710fee05eb4505b95451aa85f003cf : SUCCESS in 4m 34s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2531bdae6c294268adba1d7f8225d4df : SUCCESS in 31m 39s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/61d8e27c247845ef9b600a3298ff9514 : SUCCESS in 23m 15s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a6344b5868e040d293b2671bdb080a5f : SUCCESS in 59m 43s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6ef6808e8a8749bca21a21896f9e7b84 : SUCCESS in 1h 06m 55s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6fe41a306f5045d38ef4aaeacbf6644c : SUCCESS in 1h 06m 41s\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/11d68b477f804feabf253d9db4fe83cc : SUCCESS in 54m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9c8d141edc361b37fc4d5ca348d26d1ad2ef1bce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-01 13:38:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ac5c242099667a2054f0f0135f65b1ad4bbfddab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-01 15:25:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"3811631b04d90f79789fe2139155f17cae4dbc01","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-01 15:25:17.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/a40e09934c0d4d8c8d6f8f958630b875 : SUCCESS in 1h 01m 20s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/5d84c83ba5e945f68beb6ff59f102020 : SUCCESS in 42m 27s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/106f1627cb9a4c92a39e5892c290c5bc : SUCCESS in 1h 17m 52s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/c56e8607ca504e8a89081442a6bf6549 : SUCCESS in 1h 45m 19s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/003ccf80eed74f4695f33b40e248a802 : SUCCESS in 51m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a2aad2333a1848f89080d4ec8876304f : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59c6c1d4254d4209a2d738786a20b2a3 : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b903d524d5344c8aa442fe0038f0fc86 : SUCCESS in 4m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/65eee09160ae4572944c885e7adbe841 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af30e7264b25491192c9d09d83b32a77 : SUCCESS in 8m 42s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1379e49449004425af371033f902b8bb : SUCCESS in 42m 15s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/37c4920f596b4d0481c2865f35462633 : SUCCESS in 1h 15m 46s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9ec8114913e34c33aff156f5fb71d0cf : SUCCESS in 1h 21m 11s","accounts_in_message":[],"_revision_number":2},{"id":"ec5feedbd64b29313f61bd157636e1cecbef0bb3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-01 15:26:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9c6fc0bf21046ac8e13663dcf86c7bc : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"adcdef7d85b4d1a89592f01ae4902b43b768f0d3","revisions":{"42c64996c9e6c1ea5da6adae51f9588f3140a998":{"kind":"REWORK","_number":1,"created":"2020-01-30 11:27:52.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/65/704965/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/65/704965/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/65/704965/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/65/704965/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/65/704965/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/65/704965/1"}}},"commit":{"parents":[{"commit":"892ae92d971fd7a2c8d39e416512bb69f7fd833f","subject":"CentOS 8: Allow SSH access to keystone_ssh and nova_ssh","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/892ae92d971fd7a2c8d39e416512bb69f7fd833f"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-01-30 11:24:18.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-01-30 11:26:22.000000000","tz":0},"subject":"CentOS 8: Relax ironic iPXE EFI bootloader location","message":"CentOS 8: Relax ironic iPXE EFI bootloader location\n\nCentOS 7 had ipxe.efi. CentOS 8 has ipxe-i386.efi and ipxe-x86_64.efi.\nUse a wildcard to copy any of these that are present.\n\nChange-Id: I6d409e56b424a039f5aff9af86ca1b026c2c6286\nPartially-Implements: blueprint centos-rhel-8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/42c64996c9e6c1ea5da6adae51f9588f3140a998"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/42c64996c9e6c1ea5da6adae51f9588f3140a998"}]},"branch":"refs/heads/master"},"adcdef7d85b4d1a89592f01ae4902b43b768f0d3":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-02-01 12:01:11.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/65/704965/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/65/704965/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/65/704965/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/65/704965/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/65/704965/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/65/704965/2"}}},"commit":{"parents":[{"commit":"d00fa3a6dda3914e66ed414ef758bf2aeda2906c","subject":"Merge \"CentOS 8: Use upstream Ceph/master\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d00fa3a6dda3914e66ed414ef758bf2aeda2906c"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-01-30 11:24:18.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-02-01 12:01:11.000000000","tz":0},"subject":"CentOS 8: Relax ironic iPXE EFI bootloader location","message":"CentOS 8: Relax ironic iPXE EFI bootloader location\n\nCentOS 7 had ipxe.efi. CentOS 8 has ipxe-i386.efi and ipxe-x86_64.efi.\nUse a wildcard to copy any of these that are present.\n\nChange-Id: I6d409e56b424a039f5aff9af86ca1b026c2c6286\nPartially-Implements: blueprint centos-rhel-8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/adcdef7d85b4d1a89592f01ae4902b43b768f0d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/adcdef7d85b4d1a89592f01ae4902b43b768f0d3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
