)]}'
{"id":"openstack%2Fkolla~705040","triplet_id":"openstack%2Fkolla~master~I92be17ef1ce0f81ab3f49433a0d1e688874ba64e","project":"openstack/kolla","branch":"master","topic":"bp/centos-rhel-8","hashtags":[],"change_id":"I92be17ef1ce0f81ab3f49433a0d1e688874ba64e","subject":"nova: use proper uefi package names under CentOS/RHEL 8","status":"MERGED","created":"2020-01-30 15:28:57.000000000","updated":"2020-02-24 13:48:18.000000000","submitted":"2020-02-04 15:26:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"705040-1580829984436-5b512e01","meta_rev_id":"154afb08d7ccf4becbdba887c404418b673b2b9d","_number":705040,"virtual_id_number":705040,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-04 15:26:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-01-30 17:20:01.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2020-02-04 13:27:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2020-01-31 08:50:46.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2020-02-04 13:27:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-30 17:20:01.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-01-31 08:50:46.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-02-04 13:27:56.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-02-04 15:26:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b535a81bbded79032a86ba79d7ca4e6d7557f337","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-01-30 15:28:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1063f659d53245f744e15ac3f4ce7be1f594c509","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 15:57:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"564b6cf1a0307b3e2fb122998fd4bdbb4b6cc21b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 17:11:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/7e906fa9417f4230bee15cd4e39615f9 : SUCCESS in 34m 55s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/c45bb11091d6445a9e3494e43f94867f : SUCCESS in 1h 20m 32s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/6b32851446e744488ebe61598b1db8bd : SUCCESS in 10m 34s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/f5d1a37cabe84bd69223e659613137fd : SUCCESS in 12m 03s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/f20679279d87461c86db4016dd1011f8 : SUCCESS in 1h 20m 13s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/e1ad25182f0040e89038900d3d083ca5 : SUCCESS in 1h 28m 17s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/827f30e759a9449cbc21fde830f737dd : SUCCESS in 1h 36m 49s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/fb14c034087d462db18d57c1fc1709e2 : SUCCESS in 1h 01m 56s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/42079729b4fd4cffb50277f5f327ed6e : SUCCESS in 29m 22s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/5af3e294874c42199a7c434dc6899997 : SUCCESS in 27m 42s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f6191bff1a38469ca7908f7704f8a38e : SUCCESS in 58m 45s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f5955d71a17e44028b9fd61e8012b8d8 : SUCCESS in 6m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ea342bd14704570be3fd602cfa110a9 : SUCCESS in 6m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6cea5e2151ad4e75a5e6b697616892da : SUCCESS in 6m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e0cd2ce01ac44803b1e87504674d9df9 : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/43d1f60d951b49d09d4140f972d8888a : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7f1c042905f49e99fd17e9c8453e911 : SUCCESS in 5m 19s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ad6105fc443747eaa317b2cf1dd88cad : SUCCESS in 28m 20s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/52a5a7c0022e431380720236c92591cb : SUCCESS in 30m 33s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/92ffae3f244a40cda8e8b23a1eb049a2 : SUCCESS in 55m 36s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/77f663e9210d4305a796140ffbbc18a3 : SUCCESS in 57m 12s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a48ddde732ff4bc4a964d533fb51daf9 : SUCCESS in 1h 11m 17s","accounts_in_message":[],"_revision_number":1},{"id":"8b1116978d0e6ed5cf715ed6b9db4a195c3d0881","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-30 17:20:01.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/2f6668ef77b54a66b9c8412a1b7dc785 : POST_FAILURE in 25m 10s","accounts_in_message":[],"_revision_number":1},{"id":"dd06631052bb392d41ffa15c2c012cf765e9ab60","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-30 18:23:36.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d58a26f9c2d3b977233003b3aa72cdd27aa9e24f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-31 08:50:46.000000000","message":"Patch Set 1: Code-Review+2\n\nok, this change is nice, no need for rn since it falls into c8 general goal","accounts_in_message":[],"_revision_number":1},{"id":"12e725acaf8510aaedfe96f43295a83b828af45f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-31 08:59:59.000000000","message":"Topic set to bp/centos-rhel-8","accounts_in_message":[],"_revision_number":1},{"id":"7ed9709d358e064ddc9cc8ba3eece75dc5a0721d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-02-04 13:27:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f76ad178e95ddcc0200188ade0019d117a4e68e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-04 13:28:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4ac1caf68b103ed5f77b803c0f07f159db58e960","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-04 15:26:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b8345cdbff195b90625adff05b62f1311e7e5ce6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-04 15:26:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/c44bb5e8dedd4d32a357cfc2f06afbff : SUCCESS in 50m 13s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/cb9d494696594fd7bf3627410c6977a2 : SUCCESS in 39m 16s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/1e8b67805bb042a0a32e127e66266260 : SUCCESS in 1h 19m 53s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/e846e1de6ab4469598190dad0eae6b69 : SUCCESS in 1h 27m 43s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/d97009e9a43a4c0a8807605bbc5b2871 : SUCCESS in 59m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d00be293474244c4a40202f17d9635c5 : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d4f77592edd44b39e503e22c8f481da : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b6942260f21d47319f7fbc7d3ebad627 : SUCCESS in 4m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/19a72d23946c407784b62808490babf3 : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/973dd5b543f040a0bfbc93f0c33422a2 : SUCCESS in 7m 09s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/fab1e4205ddc45b8a3324c02c6ce63dc : SUCCESS in 30m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/bcb6fe5911c64640a1e62087c1e06c12 : SUCCESS in 52m 36s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/429c7b96352e49d680a7342b8871ac4b : SUCCESS in 1h 03m 16s","accounts_in_message":[],"_revision_number":1},{"id":"5a84aba89aea92d8c8472c8249b8685bad7c6590","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-04 15:31:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e1fc6882b304df3b98ff139bb0fcb7f : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1},{"id":"154afb08d7ccf4becbdba887c404418b673b2b9d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-24 13:48:18.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit ad00caaf9341bb58fd73969782a3f89b8189fb94","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a04963a6abdf34a49ecb815d97de394d8c7a152b","revisions":{"a04963a6abdf34a49ecb815d97de394d8c7a152b":{"kind":"REWORK","_number":1,"created":"2020-01-30 15:28:57.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/40/705040/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/40/705040/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/705040/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/705040/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/705040/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/40/705040/1"}}},"commit":{"parents":[{"commit":"a68223412e6e3289173adca1af760157b25121c2","subject":"Deprecate OpenDaylight","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a68223412e6e3289173adca1af760157b25121c2"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-01-30 15:26:08.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-01-30 15:28:48.000000000","tz":60},"subject":"nova: use proper uefi package names under CentOS/RHEL 8","message":"nova: use proper uefi package names under CentOS/RHEL 8\n\nChange-Id: I92be17ef1ce0f81ab3f49433a0d1e688874ba64e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a04963a6abdf34a49ecb815d97de394d8c7a152b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a04963a6abdf34a49ecb815d97de394d8c7a152b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
