)]}'
{"id":"openstack%2Fkolla~706042","triplet_id":"openstack%2Fkolla~master~I9790cd6c9d142a4a3ced6d5c9a9af621c3892eb0","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I9790cd6c9d142a4a3ced6d5c9a9af621c3892eb0","subject":"Install Fluentd Prometheus plugin out of the box","status":"MERGED","created":"2020-02-05 16:58:07.000000000","updated":"2020-06-17 08:02:18.000000000","submitted":"2020-06-16 23:28:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"706042-1592350082448-035ca625","meta_rev_id":"6b5164e284aa7ad1823ee32bfe7d9ea930d07aaa","_number":706042,"virtual_id_number":706042,"owner":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-16 23:28:02.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-15 21:22:55.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2020-06-15 16:35:54.000000000","_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-06-16 08:39:15.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-06-16 08:55:54.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":1,"date":"2020-06-16 21:48:42.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-06-16 09:02:27.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-25 18:00:54.000000000","updated_by":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"reviewer":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"state":"REVIEWER"},{"updated":"2020-02-26 09:58:18.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-06-15 21:22:55.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-16 08:39:15.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-06-16 08:55:54.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-06-16 21:48:42.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-06-16 23:28:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"862f35813107c2dfb7ecd079458b769bf8d5f62a","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-02-05 16:58:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0e594af8ab84bf9b579eb515747a2ad5dc4a1893","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-05 17:51:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/6316a46e4a224a5db4949f6b097f728e : SUCCESS in 29m 12s","accounts_in_message":[],"_revision_number":1},{"id":"61065481e85d5ab864d8c817b074c5881ed29413","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 18:32:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/cf9ad24b713c4dc39dc0a82695476dbe : SUCCESS in 47m 55s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/e8c982b8a1694e1fa247633b218277a7 : SUCCESS in 45m 50s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/cca77a48884e4c4bb61ca736369161c9 : SUCCESS in 29m 22s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/b0a6e2d61b784100a25cc8c94a42bc87 : SUCCESS in 33m 09s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/61a4c1487aaf460b86ca935c69e1e27f : SUCCESS in 1h 18m 19s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/8862b198d10b4769be09b3c6216702a5 : SUCCESS in 1h 20m 58s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/efb81bf6ad174a6b927b96c7096d39f9 : SUCCESS in 1h 25m 56s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/3c3e41da8f14456d8a2d84a752e32b85 : SUCCESS in 52m 46s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/a454a9c3f4b6491c864c6bf5c1b1c1d3 : SUCCESS in 22m 28s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/59b684e542b4444aaff628873b8fae52 : SUCCESS in 21m 38s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8cc00f5c99d74b7c8b9cecc4f70c9ad0 : SUCCESS in 59m 06s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9de0470b12c840c69b2d8cb5676e9d41 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4bb03a6a6d442fb9cbad20adc6de209 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ff5ac1f34ad45748dce8a3b36a1439f : SUCCESS in 4m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/24bfc36cd7ac40af98186e143ec1b529 : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f4121d0496146ce82fa2755fd1945a0 : SUCCESS in 5m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/203e116a5e82421e98e426274e04b738 : SUCCESS in 7m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4a74fd5d313440eba541901b521c58f9 : SUCCESS in 6m 41s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/03d78e76b5b3406a8d9ccea512e1160c : SUCCESS in 34m 16s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3421b60766064373b1c9fc40621fdfb2 : SUCCESS in 34m 11s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a505fbee1bbd4806bc9bb9f6078eb968 : SUCCESS in 1h 01m 34s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/92008398b3f645a68e0caf467993bf78 : SUCCESS in 1h 01m 38s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/965154294d7f402d9a89b92aa9022a8a : SUCCESS in 1h 10m 14s","accounts_in_message":[],"_revision_number":1},{"id":"c0c6e85288b8a510865c1b2e177eac9b7d47cf7b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 18:38:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"117a9e8fa4efbbf9e54746e1e8040abf4e041fc4","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-02-06 12:05:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"aa7caa36db2eb27718ad7a325b46b4c13aec612f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-25 14:43:22.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"131af5e05aa54d39db6dcce25ccdc930d9f79a19","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2020-02-25 16:25:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4f8d2539a5310cc8dd6b4fe6351dc3d3748d7ae1","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-02-25 16:29:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5f577f50fa71aba0f0e71465942882bb610ab381","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2020-02-25 18:00:54.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"5df8867214bef08911734a54ccf58f045584bd60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 18:28:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/f2fbc4f19cdd4473ab203d4124f7f056 : SUCCESS in 40m 19s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/116c85d83b524a1baced6e651e0c834a : SUCCESS in 44m 30s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/f9731a77b8ae45238aba2f2998630e20 : SUCCESS in 30m 38s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/5d011bb504ad4fd79bcf0ce754b96837 : SUCCESS in 34m 21s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/4b2a19c9a6714def96b40f4db250ded8 : SUCCESS in 1h 19m 51s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/3069c15f584a4e19a4f6d75f4237f360 : SUCCESS in 1h 08m 58s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/e51cbfcfb6cc4335933e463134b02fa6 : SUCCESS in 1h 17m 51s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/9138f4423a844fc5b9905eb9e9bb58c2 : SUCCESS in 1h 15m 59s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/20440b8674e54702956ab56ae6e7e535 : SUCCESS in 1h 46m 30s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/474702f144ac433ca6eac43aad89389a : SUCCESS in 54m 43s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/a643fbf691264fb6b1472d264e530175 : SUCCESS in 25m 52s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6e792d97d42b49a1a15b4090c5ea18e3 : SUCCESS in 23m 30s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a72e6c36140a4b5dab822d7bdc9f6425 : SUCCESS in 1h 01m 53s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b30223e6083e436d8e84f9f4aeb94cd3 : SUCCESS in 7m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15804d98807b4c7c8729dfb6e0cdd8d2 : SUCCESS in 7m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b845dd1eb3ad485a8ede713fc37e236c : SUCCESS in 6m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e6921af6cb7c45beadf96a1ec1c95197 : SUCCESS in 6m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/69b14b12e06243f6b23471ba4c50a5e6 : SUCCESS in 6m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e7b7c70696d47bf9d907e4da4bed2c4 : SUCCESS in 7m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f91f7815c0d342b38eac63783cf4a443 : SUCCESS in 7m 20s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/12188e5335b14057a34c8a526d498c56 : SUCCESS in 30m 40s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/cd062c84b566419e9713b8f0cfe5001b : SUCCESS in 34m 08s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c2f3d7180ca44429838d6a54299fd467 : SUCCESS in 59m 26s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c2a6d0ab0cf64107b31691f8e20717fe : SUCCESS in 1h 06m 27s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1cd3b687b79e4d2eb41525293180d087 : SUCCESS in 1h 14m 42s","accounts_in_message":[],"_revision_number":2},{"id":"c3a963b8671bb44e81b537c97bfa1d760d0da405","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-25 19:26:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/bbb5f389056a451e996719d6e70aa315 : SUCCESS in 38m 17s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/7e80f20be2084851a37f4365db4eb384 : SUCCESS in 43m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2c708ce4333ee6785339021cb073a83e11ced5cc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 19:35:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3b722056fce44890849d211fd5dbc5a0 : TIMED_OUT in 3h 01m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f651b58b4029aab49afd6e2702250477ac4df91b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-02-26 09:58:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f48615b658cfb696c96ef3c14d24468e1cf9c2aa","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-04 09:49:53.000000000","message":"Patch Set 2:\n\nrecheck\n\nARM","accounts_in_message":[],"_revision_number":2},{"id":"992798bd71ff3add866a70e39e844654d8ab6eca","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-03-04 10:27:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/d369a53579d4415ca5078d9ef3f3180e : SUCCESS in 26m 04s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/53379631c5ff4eecb31b8fcb9c3f58a4 : SUCCESS in 36m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ff4857bdd72c2879d2be31604a47aeb712bcad2d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-04 10:52:00.000000000","message":"Patch Set 2:\n\nwould need rebase to check on arm","accounts_in_message":[],"_revision_number":2},{"id":"a83f8a406b6313b10c063c7f891046c22b04c18e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 11:27:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/5f4b4701359444b382b7df621870bd3f : SUCCESS in 29m 59s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/ae638c72f32f425cb7e46e4a96b82fa1 : FAILURE in 44m 16s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/61e55ce7aaa04d87bc7b43c275d3867a : SUCCESS in 23m 50s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/4c44c918c7744e84a15909e72092cd94 : SUCCESS in 26m 56s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/c93c9daf3ed54426a82a665a85043b63 : SUCCESS in 1h 21m 35s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/fba98dbce762411dbd90f04c10f5c0a3 : FAILURE in 34m 35s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/d7c895dcf3004bc695f1e53203a563b8 : SUCCESS in 1h 22m 26s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/51e278d3a65047de961406556b7722ee : SUCCESS in 1h 15m 26s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/dfc1d43b5cea4c5aa66a510358abef87 : SUCCESS in 1h 35m 01s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/803143f94d26463dafceb62857fac580 : SUCCESS in 52m 57s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/df0116d7d1b040939fcd74f68c22b6dd : SUCCESS in 21m 52s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0eaf594343d14437a5061cd29ecfda59 : SUCCESS in 21m 48s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/be95e02c975a44cb8ae73261e4498d8d : FAILURE in 32m 02s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/71c701a53b644c38b13e01fb9a7d4fb3 : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6b6774fac754cf18ea7775bb2acc9f4 : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b68a134bffa4b05a1d1aaa10e1be7b7 : SUCCESS in 4m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cd48abe4240a4f599eb67c69b1fa47a4 : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b0ec380ac8574cb083d89161c8229f23 : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41b558678a50435ab719066d411a6795 : SUCCESS in 6m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/631a907cbe72496fb2ebadc743a1fdce : SUCCESS in 5m 52s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8e77226c1a1042cd9a1f740b8f0c2a4f : SUCCESS in 25m 09s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/40e0ab849c884048b8a6609a69d42d31 : SUCCESS in 27m 10s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f008bfaa079b4b729803c1368da46da3 : FAILURE in 28m 47s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4d37f6e773de4ba3b5cff6f5f1ae2aec : SUCCESS in 1h 01m 02s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/67d639303e43407eb84436796101f828 : FAILURE in 1h 03m 46s","accounts_in_message":[],"_revision_number":2},{"id":"2ac479f93504aac2354bbfe2c002589e6b199b6e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 12:52:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5d19c797969e4a9a995c61ae9e2be708 : TIMED_OUT in 3h 01m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bfc3175f030c431a000500e6602531ea415d95a1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-15 16:35:54.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"d00dd07229fbee3bed732c9f249962fa8c4dc8bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 18:51:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/09dd4fc607744fac8cb253b12f7a341a : SUCCESS in 25m 00s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/cad0a3b9917740d28d192b0872bbeb9f : SUCCESS in 32m 30s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/2b34d77d98d243c78824922eb5fb3d2e : SUCCESS in 1h 21m 04s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/eddc8d83c5c14031b85ca9a4ba07c279 : SUCCESS in 1h 20m 28s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/2b4fa3ebf55d4071be4e90a11954b796 : SUCCESS in 2h 05m 43s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/e28768f7c3e04ffa8d8a7122e506892f : SUCCESS in 1h 05m 50s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b5133a97cff3402fbb41e746f9ff085d : SUCCESS in 32m 25s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a15e5539eabd48c4af2d89b593a2b0cb : SUCCESS in 23m 09s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/31b8fcb35fe048daa861af51d5278c95 : SUCCESS in 1h 10m 15s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/32161ebcd998410187253ece0f6d2cb7 : SUCCESS in 7m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26786c1a15964a2ea296b39101f56c5f : SUCCESS in 9m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c9113f31088e4acbad899c0aa20fdb92 : SUCCESS in 7m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/333734447a2340d9bcaf491f135a0459 : SUCCESS in 7m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/925684a634ba4190b70a333d0a2e96a7 : SUCCESS in 9m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7bbae601a9f848dda3848203d3fc1b5f : SUCCESS in 7m 07s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3513ff722ef941c6917497d1e1e9dbe6 : SUCCESS in 30m 34s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/681aad527f994a73a8f4e16654e36907 : SUCCESS in 30m 26s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4727faf2432845eaa050e8572b3bd0cc : SUCCESS in 1h 09m 50s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5c939a07d8bd4a0dadae08a114c97248 : SUCCESS in 1h 09m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7117b8dc5e9441c999967104069b1230 : SUCCESS in 1h 11m 31s","accounts_in_message":[],"_revision_number":3},{"id":"20fb9b519d2d135d169e40e7d198e9a10615e330","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 19:24:42.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/f4f9fa612db84be7b4ee045a325afd6b : SUCCESS in 1h 30m 07s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/58e75d2e7c27442082bd75553ef20493 : SUCCESS in 1h 32m 04s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ca1fc68a900249b49fa7ca7ef2291961 : SUCCESS in 2h 37m 14s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/d804399a25d14a61b085bf676a1239e0 : SUCCESS in 1h 36m 21s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9e2652d24d2a106a05ec5d6e361fc87b35d2a987","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-15 21:22:55.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/a31ffc4775ea483aa1e5f865ec7ea0e4 : SUCCESS in 27m 21s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a64de4b85720011ad13c99f6ac3978acd0bd8ff3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-16 08:39:15.000000000","message":"Patch Set 3: Code-Review+2\n\nWait for CI fix","accounts_in_message":[],"_revision_number":3},{"id":"df3195b1f51d58f5d0f47f30102ef8c0d69ece73","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-16 08:55:54.000000000","message":"Patch Set 3: Code-Review+2\n\nbackport to ussuri?","accounts_in_message":[],"_revision_number":3},{"id":"f32091f09d13b2f66a6afbb08da42ac89e881dc5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-16 08:58:30.000000000","message":"Patch Set 3:\n\n\u003e backport to ussuri?\n\n+1","accounts_in_message":[],"_revision_number":3},{"id":"c790d63f1b86900a1c904e51eb4218c03f21b642","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-16 09:02:27.000000000","message":"Patch Set 3: Backport-Candidate+1","accounts_in_message":[],"_revision_number":3},{"id":"9eb18a2e1aeb5f5be0647fff87f2e580853afdd8","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-06-16 21:48:42.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ad22e7a4c0e5ca4afc1c32847f328bf6eead148e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 21:48:56.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"379bae0069603480ace6130e4c7b27bc339aa433","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 23:28:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"8c0ec159250ce67df221de2aea81e9f5268018be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 23:28:02.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/7470aecf53884e9288f11c0896d635f0 : SUCCESS in 29m 35s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/25402381588440769c21b2be115ef178 : SUCCESS in 1h 17m 49s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/10a4f7cd9fd64d90ba57c3f09062b903 : SUCCESS in 1h 37m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1edb69724c5f42ed9dadc9d35f50d2db : SUCCESS in 9m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/459e649a3d7b41b996cd55a3e56a8ad4 : SUCCESS in 9m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a47bdf6cc2a1438c9884b30bc6caaffd : SUCCESS in 9m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/303b9651d2c546819d13fab8c3c35136 : SUCCESS in 9m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/789895556ec24fb4afbbedbcd8c0b066 : SUCCESS in 8m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fc1ba8a46b2f4349a9321efe97d5e6e0 : SUCCESS in 7m 29s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d8d208d997fe4388a3b685988bda131f : SUCCESS in 27m 01s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/90a9f266e86441c7847125d0c7c87bbc : SUCCESS in 55m 58s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3550940e06024006ac77f94230844d54 : SUCCESS in 1h 20m 47s","accounts_in_message":[],"_revision_number":3},{"id":"6e74c3bbe76545961ec5373b27c84c9908d332a8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 23:29:53.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d9185e3d1084fa2b73a122fb434330e : SUCCESS in 1m 12s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/86c026d432f44b0ca28e73435c4a50fe : SUCCESS in 1m 31s","accounts_in_message":[],"_revision_number":3},{"id":"6b5164e284aa7ad1823ee32bfe7d9ea930d07aaa","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-17 08:02:18.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit b040d72ac167493aa11f63455c01f6c26193f05b","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e20fe5eaa5eb6c46f815ce7295ecdcf67a4cd042","revisions":{"d6ee512a4156ea6ecb67ce6a910f5d5efad815d1":{"kind":"REWORK","_number":1,"created":"2020-02-05 16:58:07.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/42/706042/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/42/706042/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/706042/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/706042/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/706042/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/42/706042/1"}}},"commit":{"parents":[{"commit":"a0058693f192d866a54a21137424b9a67ae8ef4c","subject":"Merge \"CentOS 8: Allow SSH access to keystone_ssh and nova_ssh\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a0058693f192d866a54a21137424b9a67ae8ef4c"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2020-02-05 10:20:14.000000000","tz":0},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2020-02-05 16:56:52.000000000","tz":0},"subject":"Install Fluentd Prometheus plugin out of the box","message":"Install Fluentd Prometheus plugin out of the box\n\nThis plugin [1,2] supports a number of use cases which are likely\nto be useful out of the box to users enabling either Monasca or\nPrometheus:\n\n1. As a system admin I want to monitor the status of Fluentd\n   to ensure that it is functioning normally.\n2. As a system admin I want to know the time to response for\n   calls made to endpoints defined in HAProxy.\n\n[1] https://docs.fluentd.org/deployment/monitoring-prometheus\n[2] https://github.com/fluent/fluent-plugin-prometheus\n\nChange-Id: I9790cd6c9d142a4a3ced6d5c9a9af621c3892eb0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d6ee512a4156ea6ecb67ce6a910f5d5efad815d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d6ee512a4156ea6ecb67ce6a910f5d5efad815d1"}]},"branch":"refs/heads/master"},"1254a6734bab79f634af4b4b38489c81ade0605b":{"kind":"REWORK","_number":2,"created":"2020-02-25 16:29:31.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/42/706042/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/42/706042/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/706042/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/706042/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/706042/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/42/706042/2"}}},"commit":{"parents":[{"commit":"a0058693f192d866a54a21137424b9a67ae8ef4c","subject":"Merge \"CentOS 8: Allow SSH access to keystone_ssh and nova_ssh\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a0058693f192d866a54a21137424b9a67ae8ef4c"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2020-02-05 10:20:14.000000000","tz":0},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2020-02-25 16:29:20.000000000","tz":0},"subject":"Install Fluentd Prometheus plugin out of the box","message":"Install Fluentd Prometheus plugin out of the box\n\nThis plugin [1,2] supports a number of use cases which are likely\nto be useful out of the box to users enabling either Monasca or\nPrometheus:\n\n1. As a system admin I want to monitor the status of Fluentd\n   to ensure that it is functioning normally.\n2. As a system admin I want to know the time to response for\n   calls made to endpoints defined in HAProxy.\n\n[1] https://docs.fluentd.org/deployment/monitoring-prometheus\n[2] https://github.com/fluent/fluent-plugin-prometheus\n\nChange-Id: I9790cd6c9d142a4a3ced6d5c9a9af621c3892eb0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1254a6734bab79f634af4b4b38489c81ade0605b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1254a6734bab79f634af4b4b38489c81ade0605b"}]},"branch":"refs/heads/master"},"e20fe5eaa5eb6c46f815ce7295ecdcf67a4cd042":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-06-15 16:35:54.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/42/706042/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/42/706042/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/706042/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/706042/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/706042/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/42/706042/3"}}},"commit":{"parents":[{"commit":"c07409bfabdd8135316609160e6477d202b46b95","subject":"Merge \"Revert \"[to-revert] Pin ovsdbapp in neutron to 1.1.0\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c07409bfabdd8135316609160e6477d202b46b95"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2020-02-05 10:20:14.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-15 16:35:54.000000000","tz":0},"subject":"Install Fluentd Prometheus plugin out of the box","message":"Install Fluentd Prometheus plugin out of the box\n\nThis plugin [1,2] supports a number of use cases which are likely\nto be useful out of the box to users enabling either Monasca or\nPrometheus:\n\n1. As a system admin I want to monitor the status of Fluentd\n   to ensure that it is functioning normally.\n2. As a system admin I want to know the time to response for\n   calls made to endpoints defined in HAProxy.\n\n[1] https://docs.fluentd.org/deployment/monitoring-prometheus\n[2] https://github.com/fluent/fluent-plugin-prometheus\n\nChange-Id: I9790cd6c9d142a4a3ced6d5c9a9af621c3892eb0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e20fe5eaa5eb6c46f815ce7295ecdcf67a4cd042"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e20fe5eaa5eb6c46f815ce7295ecdcf67a4cd042"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
