)]}'
{"id":"openstack%2Fkolla~706059","triplet_id":"openstack%2Fkolla~master~Ie095590b82d6e1b3922d99178d3a4019626ab630","project":"openstack/kolla","branch":"master","topic":"ci-replace-cinder-lvm-with-zun","hashtags":[],"change_id":"Ie095590b82d6e1b3922d99178d3a4019626ab630","subject":"CI: Replace cinder-lvm scenario with zun scenario","status":"MERGED","created":"2020-02-05 17:34:03.000000000","updated":"2020-02-07 13:24:15.000000000","submitted":"2020-02-07 13:21:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"706059-1581081701109-b4850ae3","meta_rev_id":"f23acaa07d43e0a6daec5c6ef83c88371c4e5420","_number":706059,"virtual_id_number":706059,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-07 13:21:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},{"value":2,"date":"2020-02-07 10:19:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-02-07 10:23:50.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2020-02-07 10:23:50.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"date":"2020-02-06 19:53:27.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-02-06 19:42:41.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-06 16:46:07.000000000","updated_by":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"reviewer":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"state":"REVIEWER"},{"updated":"2020-02-06 18:52:18.000000000","updated_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"reviewer":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"state":"REVIEWER"},{"updated":"2020-02-06 19:53:27.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-02-07 10:19:28.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-02-07 10:23:50.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-02-07 13:21:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"340d6568b705f9e0defebe1b1fc3010f667994bb","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-05 17:34:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7030a0508cd92cb19a35362bcc74644e91dc1f1b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 18:29:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9c5f3dc0a8ddd2337cd2951549944b5145dc9e13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 19:37:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/7d74e3df5c7f462fa7b9845cb78bafd5 : SUCCESS in 42m 33s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/cdb51ee8c68c49b1bfbb51901173e668 : SUCCESS in 41m 23s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/f238471cef4c4a48b39022d319c27887 : SUCCESS in 24m 36s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/ad83517c20254d50a7e895f0d9da5540 : SUCCESS in 39m 25s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/7d22b133929d49f9bc1dd5e66ed10414 : SUCCESS in 58m 50s\n- kolla-ansible-centos-source-zun https://zuul.opendev.org/t/openstack/build/4be050b4aa994d75874ae356a01de780 : SUCCESS in 1h 57m 24s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/5cd7e002bf21482cbcfa3610b6c0df45 : SUCCESS in 22m 21s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ac361b3dc7434ebd8e8d76558a79e202 : SUCCESS in 22m 01s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0f4cba3c2f6a47578fc46619afdbe19a : SUCCESS in 7m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1b70bf66584499fa3ddd93876122ee0 : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4dd8854d86c44c38a58670eb95356e28 : SUCCESS in 5m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f42e78eef05e41b19d14812ef716df58 : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a15210fb7d14593a2d3d1e5e69b3e57 : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c54e348acbce4cf89c7c8fdcc34bf122 : SUCCESS in 6m 19s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/758d5b8fb3e44b678e65ff24978fb660 : SUCCESS in 25m 33s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/fea8ceb0ff0a45288fa597cd48635825 : SUCCESS in 27m 22s","accounts_in_message":[],"_revision_number":1},{"id":"db4b82976e4dc91e68220451f8acec37c65f97a4","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-06 10:48:03.000000000","message":"Patch Set 1:\n\nmerge conflict, eh ;/","accounts_in_message":[],"_revision_number":1},{"id":"80647150556225d87ba841c057b0e50b998844d8","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-06 10:49:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e5fe17677c87f33c0ffc03e810b5bccbd507135a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-06 10:50:09.000000000","message":"Patch Set 2: Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"20ee3d164efbfa61b148be72e2e5d3405cf241c7","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-02-06 10:52:04.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c870253c98557886b75ccd66a9e02f6c65128ae9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-06 10:55:38.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d901bfd1a67bf7be1fdb0b36e73b843583f2075c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 12:38:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/f59b2496907f404ca09dd9d49948ed1c : SUCCESS in 36m 07s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/a5ea71f0ecec4361a4af9b0ff338f033 : SUCCESS in 40m 13s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/0b43fc02be9a455891445d9ed481bc68 : SUCCESS in 27m 35s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/b7d7891ca7cf4d0e9e7c59d3fcba6b6b : SUCCESS in 30m 30s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/6230b1cf556f4cd0abb98622e21bf70c : SUCCESS in 1h 04m 46s\n- kolla-ansible-centos-source-zun https://zuul.opendev.org/t/openstack/build/ad5616b86f424a809bbc1e7cb49db61b : SUCCESS in 1h 45m 22s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ad981804abc74f079226928b9e441ad2 : FAILURE in 26m 54s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/7554c7a77e2848a6b81c7b687e533727 : FAILURE in 24m 58s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/86a4842b68964baab598a83205653921 : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9578dd66b88458a85b0abf2ae473b0c : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d1a475f4837448d93347445f2a4f6c0 : SUCCESS in 4m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8e2f1b2c24a84d42919f4d176a929fb1 : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/835a2ace0b6840bcbf1cb3b7709c5b01 : SUCCESS in 7m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cbee837bcdd46bc9e662572ec604407 : SUCCESS in 5m 19s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/436bd80a626546f4b3c008237ffe9fd8 : SUCCESS in 50m 11s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d9085c611e2d4b9195ead17b0d18fa12 : SUCCESS in 50m 11s","accounts_in_message":[],"_revision_number":2},{"id":"b8d8912dea40d866142305512540a5c52b05e271","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-06 12:44:36.000000000","message":"Patch Set 2:\n\ndebian is fried;\n\nfor some reason it did not check ubuntu zun - infra on it","accounts_in_message":[],"_revision_number":2},{"id":"23fc7e32967b091265ad0f24073706e4cf60a30f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 13:45:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/383013c30bb6477bad092d33b621a094 : SUCCESS in 2h 46m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"01bdc19841dd0ad9f92f9528e9835e082bdcb752","author":{"_account_id":4162,"name":"Paul Belanger","email":"pabelanger@redhat.com","username":"pabelanger"},"date":"2020-02-06 16:46:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6dde1266963e71e27ecff6011007dd84ba21c44b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-06 17:03:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8f8bcbeef1bb6e131c2da8e790e62172130175df","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-06 17:16:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0dca9eed60efed568ca02deb2c8545c1818e8ae5","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-06 17:25:08.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"363d309c52968253c295ed7c6fa547678e9e0fa5","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-06 17:25:21.000000000","message":"Patch Set 3: Review-Priority+1","accounts_in_message":[],"_revision_number":3},{"id":"d7d9d83f54f83087096af686eb349944896eef7d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-06 17:26:40.000000000","message":"Patch Set 3:\n\nremoved duplicate - no change in zuul...","accounts_in_message":[],"_revision_number":3},{"id":"d90143cace5bd76f1fb6ea8dfa2eab47c82d3baa","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-06 17:32:06.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"23353df9ef0cdc715b66603b24d0ce3a802a544f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-06 17:32:15.000000000","message":"Patch Set 4: Workflow-1\n\ntesting... for infra","accounts_in_message":[],"_revision_number":4},{"id":"4a012a1f4fa178959f792b03a6ff0e41d9619a2b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-06 17:40:01.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"19bc6ddbacefb0ad443dc01d61f0a232605da6a3","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-06 17:40:09.000000000","message":"Patch Set 5: Workflow-1","accounts_in_message":[],"_revision_number":5},{"id":"2ee8dc02bd372dc26daad2d50718bb8d6b0486f5","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-06 17:42:15.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"1a0761db04c2af43e3f7b732814e900a736f49ee","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-06 17:42:22.000000000","message":"Patch Set 6: Workflow-1\n\ndisabled centos...","accounts_in_message":[],"_revision_number":6},{"id":"cdffde3fb73cfb52594a3f65498ba5f5682c3ea0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-06 17:47:11.000000000","message":"Patch Set 6:\n\nnote-to-self: restore ps3","accounts_in_message":[],"_revision_number":6},{"id":"84b52560e9084e1642d011a4673911707fb6ceab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 18:42:21.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/93e8c243e411427d8a2814dd813fe1e7 : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/903c3eb575cc4886944d01bdf8ffc92e : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7251c6f8928e4d30a9e2471230e08863 : SUCCESS in 3m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d2f1e9729f6c446ab05a79de6a43206f : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b2f21c0988de451a9167be2104e513a5 : SUCCESS in 4m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c02229e141f42a4bb59a340ccaa4f83 : SUCCESS in 5m 10s\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/85e6f7d87a7d4e75b1df20418d1e21ab : SUCCESS in 33m 15s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/aea4db48d35f4e9980314f581f31548f : SUCCESS in 39m 25s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/6e96e83097274638a037f75f56a838ab : SUCCESS in 25m 58s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/ec29f06431f747588ba3393929c2b8e6 : SUCCESS in 26m 59s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/b2d1ea7bb88449078d3db9db5dc44939 : SUCCESS in 57m 08s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1ea1b00aa3a94c30be955ce0274b7192 : SUCCESS in 22m 39s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/518623e3d4804b9f9b245f95b812feae : SUCCESS in 30m 03s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ed2ea99ac3be454595f6dadb869504c3 : SUCCESS in 21m 44s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/8f6a7d3573ca4ccd8c82491b38240074 : SUCCESS in 22m 07s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"a8637a00bfb4459bed139fb1bb5d9b393c7300a0","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2020-02-06 18:52:18.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"14a987f74966197443768e00e37cca7dcf349f0c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-06 19:42:41.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"3568be4551efbefaf1397f4180e582ad3a26d809","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-06 19:52:09.000000000","message":"Topic changed from ci-replace-cinder-lvm-with-zun-patch3 to ci-replace-cinder-lvm-with-zun","accounts_in_message":[],"_revision_number":7},{"id":"b267ec45c8cd1ec01f51ebe35e31c295636a51ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 21:10:25.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/49a0e8cc957c45158b46a68f3fa7685b : SUCCESS in 1h 07m 38s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/1a79ccf9dc88440a83b291e2920df793 : SUCCESS in 1h 16m 23s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/131ed89ecc154f73922b51a736d700f5 : SUCCESS in 1h 06m 06s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/cfa3b6b796cd4591836e734265f01c28 : SUCCESS in 57m 47s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/3c12c7f937444dc9b42ff545b008ce38 : SUCCESS in 58m 22s\n- kolla-ansible-centos-source-zun https://zuul.opendev.org/t/openstack/build/5d6b6f2e0ece4b9eb328a6e6b5aefd1b : FAILURE in 1h 26m 08s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/94c35bcee7b34f5b9f2efb7c76173ddf : SUCCESS in 35m 24s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e2c92f6bb06b4bd59e8553c50a6da636 : SUCCESS in 57m 27s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4f1ff916eaf848c58a6c92a67a93541a : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d610195ec4e44209a92f78057bbbf20 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f3ae6a3fd7b4ccd8450284916b2f46e : SUCCESS in 4m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ccf8c6a75c90454b9f6f49c82ce1f8fb : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e8d235e2eca434ca782fcbe06260a91 : SUCCESS in 5m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4012115d524e4374b6e60af523bba337 : SUCCESS in 10m 37s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/52396817db874bc9a6ee46dc336ee6bb : SUCCESS in 43m 24s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/15b9b06d831b4a969731e7186a466d28 : SUCCESS in 25m 24s","accounts_in_message":[],"_revision_number":7},{"id":"9058c00420c92372abc0f44a8b51a4da27f2bd77","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 22:18:21.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e048dc55b16e40849672f1b3b778aa1b : SUCCESS in 2h 33m 08s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"4d9bb69c6f6246717aa2e0d287b2902b10693d56","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-07 08:49:14.000000000","message":"Patch Set 7:\n\nBadRequestException: 400: Client Error for url: http://192.0.2.10:9696/v2.0/subnets, Invalid input for operation: Requested subnet with cidr: 10.0.0.0/24 for network: e7c1bd4a-a9bc-40bf-beaa-8a27f6ead1b0 overlaps with another subnet.","accounts_in_message":[],"_revision_number":7},{"id":"3e943bdfbab5ecaa6629d5c58e657c91d5c1b5d6","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-02-07 10:19:28.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"e21e76ef10008f07a8e8e25e2b2f381a17a7a3b1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-07 10:23:50.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"50a240687a3cc0d32d85ca921ff1af4d92412f18","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 10:24:08.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"11dcb32c829abd0e62f6601805522fb88acf2133","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 13:21:40.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/ea41b7cfa26e46f1bc229a3ba4c6e3fe : SUCCESS in 40m 59s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/3ba5dec5f0d247da85701154ba27ba69 : SUCCESS in 30m 38s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/3accfdef3d37452eba5ec64c9c7f70d6 : SUCCESS in 54m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/329e58321f3e421a82577f91ffbfb79e : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cc10687213245b298ce92ebdb996b06 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fed6615b03ce4958946df43a7d62f435 : SUCCESS in 3m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/77cec0c81a204dd4951477620427f187 : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1401dc3426804ef1a57ad0eae84dea87 : SUCCESS in 5m 14s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4a3cfd0ef3434b0798440b5415dfa2b3 : SUCCESS in 32m 28s","accounts_in_message":[],"_revision_number":7},{"id":"a30cdf04ba88b2bf1edb0d30326b57b981db6d74","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 13:21:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"f23acaa07d43e0a6daec5c6ef83c88371c4e5420","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 13:24:15.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efe3de9fed854645a86eb5306a8cf82e : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"a7c2e4b34c91bbe947490940da23bc0f6ea4867c","revisions":{"aa0f5da07ee6474d33bf9d2641d305a507f93ee5":{"kind":"REWORK","_number":1,"created":"2020-02-05 17:34:03.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/59/706059/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/706059/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/706059/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/706059/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/706059/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/706059/1"}}},"commit":{"parents":[{"commit":"49445b68f8e195f4db7aebe972415d1d620665dc","subject":"Merge \"nova: use proper uefi package names under CentOS/RHEL 8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/49445b68f8e195f4db7aebe972415d1d620665dc"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-02-05 17:33:32.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-02-05 17:33:32.000000000","tz":60},"subject":"CI: Replace cinder-lvm scenario with zun scenario","message":"CI: Replace cinder-lvm scenario with zun scenario\n\nAs discussed during the meeting. [1]\n\n[1] http://eavesdrop.openstack.org/meetings/kolla/2020/kolla.2020-02-05-15.00.html\n\nChange-Id: Ie095590b82d6e1b3922d99178d3a4019626ab630\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aa0f5da07ee6474d33bf9d2641d305a507f93ee5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aa0f5da07ee6474d33bf9d2641d305a507f93ee5"}]},"branch":"refs/heads/master"},"e524d01c7c982e9f7276e17dc3e99b3e54dc77d4":{"kind":"REWORK","_number":2,"created":"2020-02-06 10:49:57.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/59/706059/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/706059/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/706059/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/706059/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/706059/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/706059/2"}}},"commit":{"parents":[{"commit":"ccd734b319ff3250444fb25f64d73e72dc70b94b","subject":"Merge \"CI: Replace ceph jobs with ceph-ansible\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ccd734b319ff3250444fb25f64d73e72dc70b94b"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-02-05 17:33:32.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-02-06 10:49:46.000000000","tz":60},"subject":"CI: Replace cinder-lvm scenario with zun scenario","message":"CI: Replace cinder-lvm scenario with zun scenario\n\nAs discussed during the meeting. [1]\n\n[1] http://eavesdrop.openstack.org/meetings/kolla/2020/kolla.2020-02-05-15.00.html\n\nChange-Id: Ie095590b82d6e1b3922d99178d3a4019626ab630\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e524d01c7c982e9f7276e17dc3e99b3e54dc77d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e524d01c7c982e9f7276e17dc3e99b3e54dc77d4"}]},"branch":"refs/heads/master"},"4740341ab2f0057607669a2b64e6297d1766d9d1":{"kind":"REWORK","_number":3,"created":"2020-02-06 17:25:08.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/59/706059/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/706059/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/706059/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/706059/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/706059/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/706059/3"}}},"commit":{"parents":[{"commit":"ccd734b319ff3250444fb25f64d73e72dc70b94b","subject":"Merge \"CI: Replace ceph jobs with ceph-ansible\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ccd734b319ff3250444fb25f64d73e72dc70b94b"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-02-05 17:33:32.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-02-06 17:16:30.000000000","tz":0},"subject":"CI: Replace cinder-lvm scenario with zun scenario","message":"CI: Replace cinder-lvm scenario with zun scenario\n\nAs discussed during the meeting. [1]\n\n[1] http://eavesdrop.openstack.org/meetings/kolla/2020/kolla.2020-02-05-15.00.html\n\nChange-Id: Ie095590b82d6e1b3922d99178d3a4019626ab630\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4740341ab2f0057607669a2b64e6297d1766d9d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4740341ab2f0057607669a2b64e6297d1766d9d1"}]},"branch":"refs/heads/master"},"440c40967032158a7a5de6ca8086e24467e51421":{"kind":"REWORK","_number":4,"created":"2020-02-06 17:32:06.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/59/706059/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/706059/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/706059/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/706059/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/706059/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/706059/4"}}},"commit":{"parents":[{"commit":"ccd734b319ff3250444fb25f64d73e72dc70b94b","subject":"Merge \"CI: Replace ceph jobs with ceph-ansible\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ccd734b319ff3250444fb25f64d73e72dc70b94b"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-02-05 17:33:32.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-02-06 17:31:55.000000000","tz":0},"subject":"CI: Replace cinder-lvm scenario with zun scenario","message":"CI: Replace cinder-lvm scenario with zun scenario\n\nAs discussed during the meeting. [1]\n\n[1] http://eavesdrop.openstack.org/meetings/kolla/2020/kolla.2020-02-05-15.00.html\n\nChange-Id: Ie095590b82d6e1b3922d99178d3a4019626ab630\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/440c40967032158a7a5de6ca8086e24467e51421"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/440c40967032158a7a5de6ca8086e24467e51421"}]},"branch":"refs/heads/master"},"817b21b05289febbf762d1af7b3dadf6c7f6c0c6":{"kind":"REWORK","_number":5,"created":"2020-02-06 17:40:01.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/59/706059/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/706059/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/706059/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/706059/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/706059/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/706059/5"}}},"commit":{"parents":[{"commit":"ccd734b319ff3250444fb25f64d73e72dc70b94b","subject":"Merge \"CI: Replace ceph jobs with ceph-ansible\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ccd734b319ff3250444fb25f64d73e72dc70b94b"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-02-05 17:33:32.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-02-06 17:39:53.000000000","tz":60},"subject":"CI: Replace cinder-lvm scenario with zun scenario","message":"CI: Replace cinder-lvm scenario with zun scenario\n\nAs discussed during the meeting. [1]\n\n[1] http://eavesdrop.openstack.org/meetings/kolla/2020/kolla.2020-02-05-15.00.html\n\nChange-Id: Ie095590b82d6e1b3922d99178d3a4019626ab630\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/817b21b05289febbf762d1af7b3dadf6c7f6c0c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/817b21b05289febbf762d1af7b3dadf6c7f6c0c6"}]},"branch":"refs/heads/master"},"3cad10cda5eaff4af7499fb54f70ed4a6d72ac50":{"kind":"REWORK","_number":6,"created":"2020-02-06 17:42:15.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/59/706059/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/706059/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/706059/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/706059/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/706059/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/706059/6"}}},"commit":{"parents":[{"commit":"ccd734b319ff3250444fb25f64d73e72dc70b94b","subject":"Merge \"CI: Replace ceph jobs with ceph-ansible\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ccd734b319ff3250444fb25f64d73e72dc70b94b"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-02-05 17:33:32.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-02-06 17:42:07.000000000","tz":0},"subject":"CI: Replace cinder-lvm scenario with zun scenario","message":"CI: Replace cinder-lvm scenario with zun scenario\n\nAs discussed during the meeting. [1]\n\n[1] http://eavesdrop.openstack.org/meetings/kolla/2020/kolla.2020-02-05-15.00.html\n\nChange-Id: Ie095590b82d6e1b3922d99178d3a4019626ab630\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3cad10cda5eaff4af7499fb54f70ed4a6d72ac50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3cad10cda5eaff4af7499fb54f70ed4a6d72ac50"}]},"branch":"refs/heads/master"},"a7c2e4b34c91bbe947490940da23bc0f6ea4867c":{"kind":"REWORK","_number":7,"created":"2020-02-06 19:42:41.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/59/706059/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/706059/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/706059/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/706059/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/706059/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/706059/7"}}},"commit":{"parents":[{"commit":"ccd734b319ff3250444fb25f64d73e72dc70b94b","subject":"Merge \"CI: Replace ceph jobs with ceph-ansible\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ccd734b319ff3250444fb25f64d73e72dc70b94b"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-02-05 17:33:32.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-02-06 19:42:32.000000000","tz":60},"subject":"CI: Replace cinder-lvm scenario with zun scenario","message":"CI: Replace cinder-lvm scenario with zun scenario\n\nAs discussed during the meeting. [1]\n\n[1] http://eavesdrop.openstack.org/meetings/kolla/2020/kolla.2020-02-05-15.00.html\n\nChange-Id: Ie095590b82d6e1b3922d99178d3a4019626ab630\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a7c2e4b34c91bbe947490940da23bc0f6ea4867c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a7c2e4b34c91bbe947490940da23bc0f6ea4867c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
