)]}'
{"id":"openstack%2Fkolla~706061","triplet_id":"openstack%2Fkolla~stable%2Fstein~I1d5cd3d9af98444acac5bedd7daeaa6c6673dcd6","project":"openstack/kolla","branch":"stable/stein","topic":"bug/1814552-stable/train-stable/stein","hashtags":[],"change_id":"I1d5cd3d9af98444acac5bedd7daeaa6c6673dcd6","subject":"nova-libvirt: add UEFI packages to support UEFI instances","status":"MERGED","created":"2020-02-05 17:36:48.000000000","updated":"2020-02-06 14:43:53.000000000","submitted":"2020-02-06 14:41:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"706061-1581000093757-71032318","meta_rev_id":"8d0e6b929af1b0915f8aaaa4a251ac8fa36b988b","_number":706061,"virtual_id_number":706061,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"date":"2020-02-05 20:07:58.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-06 14:41:33.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":2,"date":"2020-02-06 13:04:53.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-02-06 13:40:37.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2020-02-06 13:40:37.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-05 17:36:48.000000000","updated_by":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"reviewer":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"state":"REVIEWER"},{"updated":"2020-02-06 13:04:53.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-02-06 13:40:37.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-02-06 14:41:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"08dd197c4e48a4928a904f81702d743190ec54dd","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-05 17:36:48.000000000","message":"Patch Set 1: Cherry Picked from branch stable/train.","accounts_in_message":[],"_revision_number":1},{"id":"ae6cc9df70f16c124ed3ac622eebcdaa2027d9d5","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-05 17:37:30.000000000","message":"Patch Set 1: Workflow-1\n\nshould have no c8 support","accounts_in_message":[],"_revision_number":1},{"id":"9f2cd3474a7b21b1a0f2a53d5c29b0a5fdf9a4e8","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-05 17:41:47.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"eb0cb7f4b3a477f3032c34c43ffaecea5fc24c95","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-05 17:42:42.000000000","message":"Patch Set 2:\n\nhmm, rocky does not like a simple merge","accounts_in_message":[],"_revision_number":2},{"id":"e0eabb1e33d97126076a708c797dfb6e9fc6d276","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 19:14:24.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/9f70731ee3b34672a03c8b5952281146 : SUCCESS in 46m 33s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/8f94410dcb8749ba9cf48db48c374c29 : SUCCESS in 1h 05m 44s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/b1746d7de9ca4465bc25c55b2cdee84e : SUCCESS in 1h 27m 20s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/64e188b38ffd4888ab69db7f597b6758 : SUCCESS in 1h 17m 28s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/dc4a522d0a21479f905de371d6bfdce5 : SUCCESS in 1h 19m 21s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/a762c3ad6fe544f9ae16619f06817f2c : SUCCESS in 50m 13s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/4abdee14d84649a1830595cf2d65fc0f : FAILURE in 1h 05m 56s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/31b68b36e6d045cc8205e3420405a403 : SUCCESS in 1h 00m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/be1b045b2d13479c83b948360df2986e : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0b6fc3fe7f14792905738b23c8c2d8b : SUCCESS in 5m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bfa2dc58df28442cab05565d5f5fc4ed : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cbdc4040b9534b7c84444231e7afb1cb : SUCCESS in 4m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/594c9930ceec4612bd894571d1b7b00a : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c3d9fe2156146b08d433c9b711f31b2 : SUCCESS in 5m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c2c0120b51ad44e8affb8e389a3f3633 : SUCCESS in 5m 07s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5d7dc51d87f14fa3beacf69ee33a5c9f : FAILURE in 28m 11s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/aa721469b72740d29034036e6fe713ce : FAILURE in 41m 21s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/bcea33556c16485bb4e02d52a905c2b4 : FAILURE in 32m 34s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/58e8614a57c24a7fb65af59cb8421647 : FAILURE in 26m 12s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4e64c8ae75ca4745b0022c3ec2464e03 : FAILURE in 31m 47s","accounts_in_message":[],"_revision_number":2},{"id":"ca848e92e51916fb51f8dc0e78110b3cb2111e62","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-05 20:07:58.000000000","message":"Patch Set 2:\n\nrecheck\n\nlooks like mirror update in progress","accounts_in_message":[],"_revision_number":2},{"id":"1402fdff3397ff1a765f5787f8699e50d3a5e664","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 21:53:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/cae059646df44e94a47fe16d043bd21f : SUCCESS in 39m 42s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/ece3edd0560542a0a7903581ec14b230 : SUCCESS in 45m 27s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/00556be037d145c6adffda9382dada4c : SUCCESS in 1h 16m 13s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/c9b1b9c708df4e31b918ad89cb379718 : SUCCESS in 1h 13m 47s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/7b6ee368669a434db048fd2796626312 : SUCCESS in 1h 39m 24s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/d2948202b5db4ca1ab459292994d38ba : SUCCESS in 55m 04s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/fa5d5df9ce5e49a8adb15ccaedce38f2 : FAILURE in 50m 30s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/0075861183a04c47be4b7d09461870e5 : SUCCESS in 46m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bff6195782a94386975034e144f73238 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/380d8e5cd52e4cb69f890dfd77d1f48b : SUCCESS in 4m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0c8138b724094a29a83a1f8a16264b2a : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c929cc1cf2534719a97a966f88aeceb6 : SUCCESS in 4m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2a2835a34ffe4aa5b39bf32d468c27fb : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/871653d96c2c417a94811dde3d4309c1 : SUCCESS in 4m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/437fd11b890642d1a2b816ac295e96b1 : SUCCESS in 5m 39s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/75dfc665eefa4b15a9d591fcac463354 : SUCCESS in 24m 20s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1b5199d507c14b908c17a1de4541ebb9 : SUCCESS in 31m 12s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a3b6dcf3bb6440658500c75176d85f73 : SUCCESS in 53m 19s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fab285eaa52c494e8969c5a41fc99a30 : FAILURE in 33m 10s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2ed3cddf94c64663969b75621e9f2a08 : SUCCESS in 1h 02m 02s","accounts_in_message":[],"_revision_number":2},{"id":"68c018301c4768c5993d4e035ce66ce937c82b12","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-02-06 13:04:53.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"724cf6941a49371e6f9c52f59b6f786eb7a8a98f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-06 13:40:37.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"fccbdee263a43c99a1fc4a6d6ed7c87774becc1d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 13:40:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"39bb09a232f28e0978147dc124885e5c12523f84","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 14:41:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"df4b09362cc4742d14f3e172fd19569bdd806997","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 14:41:33.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/5731cb83f0344881b9e4613b3573c8dc : SUCCESS in 49m 30s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/1e9f03837260470f9eb02bb155b6b40f : SUCCESS in 56m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4117a426cf02440693248a42c3d44181 : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/723d3a1c881240b3a3339f191bd84761 : SUCCESS in 3m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/afd14ab018f447afa1d2662729e2714f : SUCCESS in 3m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/40de15b74243488b94a74af50ca16371 : SUCCESS in 3m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a6731e3a43fb4c19b580084f73f5033f : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b8831b3912f4f5fb11a67854a6a5821 : SUCCESS in 4m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/27677a93777a4a2d843c0d7309937c61 : SUCCESS in 4m 57s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/dd4665b72e894db294b9fc66813a1079 : SUCCESS in 33m 01s","accounts_in_message":[],"_revision_number":2},{"id":"96649a19d60a7de7e0661270e05f7ba94a1d84b4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 14:43:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3804e03959d2446e83986216e9f1a178 : SUCCESS in 1m 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ea83e1500ac9499db36ca8a539ca01fa : SUCCESS in 1m 53s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"757fe57563ed54e53c14c8cfae780b71fb49b7ba","revisions":{"16a743ad52285dfefc8e1a45478f3b8e791688a6":{"kind":"REWORK","_number":1,"created":"2020-02-05 17:36:48.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/61/706061/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/61/706061/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/706061/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/706061/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/706061/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/61/706061/1"}}},"commit":{"parents":[{"commit":"983bee4f95bf524fee9c0b9f87ec105d005eca96","subject":"Change fetch_fernet_tokens.py behaviour to check for minimum number of tokens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/983bee4f95bf524fee9c0b9f87ec105d005eca96"}]}],"author":{"name":"chenxing","email":"chason.chan@foxmail.com","date":"2019-09-16 06:46:41.000000000","tz":480},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-02-05 17:36:48.000000000","tz":0},"subject":"nova-libvirt: add UEFI packages to support UEFI instances","message":"nova-libvirt: add UEFI packages to support UEFI instances\n\nFix inability to run UEFI-based images/instances by installing UEFI\npackages also in nova-libvirt image which is not based on nova-base.\n\nIncludes support for C8.\nBackport below Train w/o C8.\n\nCloses-Bug: #1814552\nCo-authored-by: Marcin Juszkiewicz \u003cmarcin.juszkiewicz@linaro.org\u003e\nCo-authored-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\nChange-Id: I1d5cd3d9af98444acac5bedd7daeaa6c6673dcd6\n(cherry picked from commit 15b68c15c1dd0c2df1cc49b550f2fa03a4c65e8e)\n(cherry picked from commit ffaf585d0f05ccd9123459bb0cdf345af5b09573)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/16a743ad52285dfefc8e1a45478f3b8e791688a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/16a743ad52285dfefc8e1a45478f3b8e791688a6"}]},"branch":"refs/heads/stable/stein"},"757fe57563ed54e53c14c8cfae780b71fb49b7ba":{"kind":"REWORK","_number":2,"created":"2020-02-05 17:41:47.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/61/706061/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/61/706061/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/706061/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/706061/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/706061/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/61/706061/2"}}},"commit":{"parents":[{"commit":"983bee4f95bf524fee9c0b9f87ec105d005eca96","subject":"Change fetch_fernet_tokens.py behaviour to check for minimum number of tokens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/983bee4f95bf524fee9c0b9f87ec105d005eca96"}]}],"author":{"name":"chenxing","email":"chason.chan@foxmail.com","date":"2019-09-16 06:46:41.000000000","tz":480},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-02-05 17:41:39.000000000","tz":0},"subject":"nova-libvirt: add UEFI packages to support UEFI instances","message":"nova-libvirt: add UEFI packages to support UEFI instances\n\nFix inability to run UEFI-based images/instances by installing UEFI\npackages also in nova-libvirt image which is not based on nova-base.\n\nAdapted for Stein and below (e.g. Rocky) by removing\nany C8 support.\n\nCloses-Bug: #1814552\nCo-authored-by: Marcin Juszkiewicz \u003cmarcin.juszkiewicz@linaro.org\u003e\nCo-authored-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\nChange-Id: I1d5cd3d9af98444acac5bedd7daeaa6c6673dcd6\n(cherry picked from commit 15b68c15c1dd0c2df1cc49b550f2fa03a4c65e8e)\n(cherry picked from commit ffaf585d0f05ccd9123459bb0cdf345af5b09573)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/757fe57563ed54e53c14c8cfae780b71fb49b7ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/757fe57563ed54e53c14c8cfae780b71fb49b7ba"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
