)]}'
{"id":"openstack%2Fkolla~706149","triplet_id":"openstack%2Fkolla~master~Ib995c6ebe5ee76ee1a39e4b5918a79bd61b87a01","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"Ib995c6ebe5ee76ee1a39e4b5918a79bd61b87a01","subject":"Add ssh clients to Ironic Conductor container","status":"MERGED","created":"2020-02-05 22:54:49.000000000","updated":"2020-03-11 13:31:39.000000000","submitted":"2020-02-07 14:02:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"706149-1581084160842-e685ee51","meta_rev_id":"a17bb3d64aecaece3c8f04614723d4613fdab1bd","_number":706149,"virtual_id_number":706149,"owner":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-07 14:02:40.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-02-06 23:08:43.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-02-07 10:48:43.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-02-07 08:47:30.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2020-02-07 10:48:43.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-06 23:08:43.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-07 08:47:30.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-02-07 10:48:43.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-02-07 14:02:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"688966f7dec68def1800b019294c3212556459d5","author":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"date":"2020-02-05 22:54:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"06a83929266c4ae622880a4d29f67099805e2ef6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-05 23:34:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/81341113240a479fa692b82e96adf7fd : SUCCESS in 28m 13s","accounts_in_message":[],"_revision_number":1},{"id":"cc8f40cf7ffc404b7aea01bc386be7cc86a8dba0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 00:32:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/ebee849ad5344e1089c0b514652a21dc : SUCCESS in 34m 59s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/f0703a3469d3416585b80229d1222f68 : SUCCESS in 48m 50s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/f8a2d043d8474992b2fa584ff77c8af1 : SUCCESS in 23m 21s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/2e3d627a4a204d5da99145e8dc9a572c : SUCCESS in 33m 27s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/4c48e0ec8cb84a3a8ce9e905052ec66f : SUCCESS in 1h 34m 31s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/3357e46f72c54f9a92e53e62db2cf6ce : SUCCESS in 1h 24m 39s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/3708437d7db34d2291109bab230646c9 : SUCCESS in 1h 25m 30s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/389df359163c4aa1b7b4592c7fd85972 : SUCCESS in 1h 05m 15s\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/6c01ab3e16b54298b685fe5d6da09334 : SUCCESS in 1h 32m 27s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/fe7530ed7a864399a426fee0bbcd314a : SUCCESS in 25m 11s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/76bea685dd7244eb8c30248a0503eb97 : SUCCESS in 21m 58s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/472ce49392cd49e8a3c54083faa1b2fe : SUCCESS in 56m 03s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cd3d98bcc61f4c2e84aff783bb3820c8 : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91503d6e5116400ba20f154bd9e16f40 : SUCCESS in 6m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c2016933d6cb457695e167f0bb8f3998 : SUCCESS in 5m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ae6c87d157284c59a28e92325ee8c8d0 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/66a3ffca277c438bafe07b26328d4817 : SUCCESS in 4m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08f00061d53048098ee306f6f98278d1 : SUCCESS in 6m 31s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9f4d3084fc1c4aa3b34843fa9e1f3a19 : SUCCESS in 35m 57s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/fb8a656329d14760a342174bd49816b5 : SUCCESS in 34m 18s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ec39561fcbbe4474af42877497a08492 : SUCCESS in 58m 58s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8fd0ee6fee3d4c908f340c78097a081e : SUCCESS in 56m 05s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7088f98a487a42cd8950848256771c53 : SUCCESS in 1h 17m 11s\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/43171c10cb094d3f96c792f6298d9efb : SUCCESS in 1h 00m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"08097ec924ac7f72c6bd3721c00dc04c20363611","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 01:21:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5560e92e5d41481e912fa75f3856e2f1 : SUCCESS in 2h 25m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5e723fdb6af5bbd0db8a432f85ffa2ee28829050","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-06 08:48:27.000000000","message":"Patch Set 1: Code-Review-1\n\nthe patch is fine but we require renos for fixes","accounts_in_message":[],"_revision_number":1},{"id":"0079417f521a11d0dab990b725b41c7bf7473817","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-06 10:10:36.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"02319a4eb8b64aef338d091d8b79be6feedb0600","author":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"date":"2020-02-06 21:02:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f8531089c5d8bf2992f03fe1ebdf0405d4b47eac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 22:48:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/9fb22e6c650a49d3998d94ce7f357dd2 : SUCCESS in 1h 14m 22s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/b21851ab005a467abe0fcf3c6827b075 : SUCCESS in 45m 50s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/86cfcd0892ea4678b177d4cd22dcaddc : SUCCESS in 37m 57s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/efc9ec6cc2234c24912d7a517c27d572 : SUCCESS in 41m 40s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/b2681dcf61ce465d844b81908323fd85 : SUCCESS in 1h 19m 35s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/de56e565691f4566aa1e8acf301a1c67 : SUCCESS in 1h 18m 36s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/c4726dec8b224d498de01bfa7adc5b3c : SUCCESS in 1h 29m 45s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/deec733010854c73b8b824098117b588 : SUCCESS in 1h 00m 40s\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/3bfa37e1bcf241dca9332103f0f5cb69 : SUCCESS in 1h 39m 33s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b10ee347ce0f4ac4ba878bde7ce6f5e3 : SUCCESS in 22m 41s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/7153125d2df24b66815713c18652a107 : SUCCESS in 21m 17s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d8ef97ca30fa43f79faa07cd6945bc7d : SUCCESS in 55m 25s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f915baa174cd412d82533521354f2876 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d7ee1d6cdfa4d12b73ceafc5d8fa16f : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e4afc4e25e904202869658b27b780103 : SUCCESS in 4m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2797e2ee3f5e495e9e14134c6b09cead : SUCCESS in 6m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d7ddcb70df3c4dc5a705b0438f34611c : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca402a16a43f4b269e870a9f0c554cc1 : SUCCESS in 6m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1e73ee6a88ae4cb59b510185c4e8b650 : SUCCESS in 6m 41s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/97b5aa995de74196a9de5e94bf2ca9ae : SUCCESS in 22m 30s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2cb00a684876456181316c7663024834 : SUCCESS in 25m 32s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e88bd6effa064bf6a3822a9cf598decb : SUCCESS in 55m 20s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4a803616bf534f0792a5ee89eb3c662a : SUCCESS in 1h 01m 11s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7dfe6f0925474e3a9ecc3cb20cd5c6af : SUCCESS in 1h 01m 54s\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/bc14c044c79245df97bc1fd51359651f : SUCCESS in 58m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"aa69fc2e8a93dce96b730a8b2dc8315180b19dcd","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-06 23:08:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/2f8366784ad44284abd8041aa5b94b57 : SUCCESS in 28m 01s","accounts_in_message":[],"_revision_number":2},{"id":"0568d3e54c0825c2581fac15cd6326a9859692fc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 01:06:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/af1a4d329fb34b0bb3dbf3d155676121 : SUCCESS in 2h 19m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8e907b8fab8f36484f055ae763e740f263a18709","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-07 08:47:30.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"6a74417bcdaea3448bb8e9a1e321fa96069d30f0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-07 10:48:43.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ef95f0de9778c88176e3dc55ef5a3116852e9206","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 10:49:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ae863667ce223c202aa59c598b713012bc3ca323","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 14:02:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"82009bba7719e2a8bdadd05fb7b76fac32acdb32","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 14:02:40.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/15501cf19fbd418ca6adb824fdc19c6a : SUCCESS in 38m 38s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/8c26bcd4ae174c83b7420412d59cbaf5 : SUCCESS in 28m 23s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/5ee07b3411514fdbaf348daa5a6a066b : SUCCESS in 1h 26m 55s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/308c5b8dd7c54d258dab95013eea77bc : SUCCESS in 1h 35m 04s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/813e77c550b64ec392f051a22b03eeff : SUCCESS in 50m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/66d93fe61ea2442ebe2d986a926e973c : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52221d32f8d8433088fb9d0d9e0fc2d9 : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/427ddd15201645b4925bd22c11130eb2 : SUCCESS in 4m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3b32dec08ed94487a796308285315cca : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd618445ba554e47b4f543c5b5b26625 : SUCCESS in 5m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cbadee2478ac4dcc90aa723e3ef83f76 : SUCCESS in 5m 05s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c15e96a15ee347babed8a2d38393999d : SUCCESS in 24m 41s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/27aea90ab57e4a12b70e626168a36d85 : SUCCESS in 53m 32s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5bf27eebb5c949cbbfd1dcdc31f50b38 : SUCCESS in 1h 14m 26s","accounts_in_message":[],"_revision_number":2},{"id":"25415b9fa7ba1557c803a5890a4527cd37a6209f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 14:04:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/376cce0881c24ea1973d207f167d9f62 : SUCCESS in 1m 20s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/219f08b8fe44469aa83d0ce82ee45799 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":2},{"id":"bbb4ec0d28003510fb0bcb367019656a2bb2ab4d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-11 13:31:19.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 15d9ce5b70ff3a78868a7d13995f75bf08f68263","accounts_in_message":[],"_revision_number":2},{"id":"090e6c6c58533f3faf910ebf23be283554e5d06a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-11 13:31:28.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit c139ac59163dec120fb57b78ae19b5bce7237d30","accounts_in_message":[],"_revision_number":2},{"id":"a17bb3d64aecaece3c8f04614723d4613fdab1bd","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-11 13:31:39.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit 1e146739da9254527e4fbb467cd7a503688a0cad","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7234e8b2f72275c85caa146a6d9e49d1ab8281b7","revisions":{"febed44233009b51179299554429b423f3c1e29a":{"kind":"REWORK","_number":1,"created":"2020-02-05 22:54:49.000000000","uploader":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"ref":"refs/changes/49/706149/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/49/706149/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/706149/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/706149/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/706149/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/49/706149/1"}}},"commit":{"parents":[{"commit":"ccd734b319ff3250444fb25f64d73e72dc70b94b","subject":"Merge \"CI: Replace ceph jobs with ceph-ansible\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ccd734b319ff3250444fb25f64d73e72dc70b94b"}]}],"author":{"name":"Antony Messerli","email":"amesserl@rackspace.com","date":"2020-02-05 22:31:06.000000000","tz":-360},"committer":{"name":"Antony Messerli","email":"amesserl@rackspace.com","date":"2020-02-05 22:31:06.000000000","tz":-360},"subject":"Add ssh clients to Ironic Conductor container","message":"Add ssh clients to Ironic Conductor container\n\nAdds ssh clients to Ironic Conductor container for\nansible deploy interface.\n\nChange-Id: Ib995c6ebe5ee76ee1a39e4b5918a79bd61b87a01\nCloses-Bug: #1862079\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/febed44233009b51179299554429b423f3c1e29a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/febed44233009b51179299554429b423f3c1e29a"}]},"branch":"refs/heads/master"},"7234e8b2f72275c85caa146a6d9e49d1ab8281b7":{"kind":"REWORK","_number":2,"created":"2020-02-06 21:02:50.000000000","uploader":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"ref":"refs/changes/49/706149/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/49/706149/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/706149/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/706149/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/706149/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/49/706149/2"}}},"commit":{"parents":[{"commit":"ccd734b319ff3250444fb25f64d73e72dc70b94b","subject":"Merge \"CI: Replace ceph jobs with ceph-ansible\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ccd734b319ff3250444fb25f64d73e72dc70b94b"}]}],"author":{"name":"Antony Messerli","email":"amesserl@rackspace.com","date":"2020-02-05 22:31:06.000000000","tz":-360},"committer":{"name":"Antony Messerli","email":"amesserl@rackspace.com","date":"2020-02-06 21:02:26.000000000","tz":-360},"subject":"Add ssh clients to Ironic Conductor container","message":"Add ssh clients to Ironic Conductor container\n\nAdds ssh clients to Ironic Conductor container for\nansible deploy interface.\n\nChange-Id: Ib995c6ebe5ee76ee1a39e4b5918a79bd61b87a01\nCloses-Bug: #1862079\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7234e8b2f72275c85caa146a6d9e49d1ab8281b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7234e8b2f72275c85caa146a6d9e49d1ab8281b7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
