)]}'
{"id":"openstack%2Fkolla~707338","triplet_id":"openstack%2Fkolla~master~Id37dca84c4eb918aaf2d3c036ef5387fe75988dd","project":"openstack/kolla","branch":"master","topic":"ceph-centos8","hashtags":[],"change_id":"Id37dca84c4eb918aaf2d3c036ef5387fe75988dd","subject":"Use StorageSIG repos for Ceph in CentOS8","status":"MERGED","created":"2020-02-12 08:52:33.000000000","updated":"2020-05-24 10:16:56.000000000","submitted":"2020-03-03 17:33:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"707338-1583256811377-30632db2","meta_rev_id":"4a0efb94cdb59d196cb29318cbc8daa2abfa8c4e","_number":707338,"virtual_id_number":707338,"owner":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-03 17:33:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-03-03 11:38:32.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"recommended":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":1,"date":"2020-03-03 13:27:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2020-03-03 13:29:39.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":2,"date":"2020-03-03 14:01:05.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":1,"date":"2020-03-03 14:01:05.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-20 15:13:50.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2020-02-20 15:13:55.000000000","updated_by":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"reviewer":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"state":"REVIEWER"},{"updated":"2020-02-20 15:13:59.000000000","updated_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"reviewer":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"state":"REVIEWER"},{"updated":"2020-02-20 15:14:00.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"},{"updated":"2020-02-25 19:25:21.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-02-26 10:00:07.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-03-03 11:38:32.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-03 13:27:35.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2020-03-03 13:29:39.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-03-03 14:01:05.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-03-03 17:33:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2ab0cc985d502b641311099fbf113769a726581b","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-02-12 08:52:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"40b1536faac65f67f034be1ec899ccd2a8fc8338","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-02-12 09:01:14.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fe0e60df671209aedb2c0e307a56e813f058c16f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-12 09:19:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/75433af664514e5486339ed1f331131c : SUCCESS in 26m 20s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/438544d2fa5340a1b306bb5a0eac617b : RETRY_LIMIT in 6m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"96a9ae62c71d91fdd848ff4f97d191b647547dbc","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-02-12 09:43:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ffae99299096bb373ee5d516f6537915034c7a31","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-12 10:11:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/fc82c607ea7d4faeba5ff4608447cbee : SUCCESS in 27m 00s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/139191d7cf6940a883c4b333922de52c : RETRY_LIMIT in 6m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0e6d235e9353fbb2ef8a5972464247ed0e7da2c2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 11:40:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9dc6fb4e7ea742b7a33bde9b509eeb69 : SUCCESS in 1h 55m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"952a7a41aa100a470ea6cba46b7d329335af79ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 12:16:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/ecc310bbd0e447a8b06c81d33aaf6fbd : FAILURE in 6m 17s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/e71ad1d6227b429fb60b6e888d03205a : FAILURE in 5m 32s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/f48860b2d90d49b09299092c6c27bda7 : FAILURE in 5m 08s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/b5e4101c6081462bb92c827e695e88fa : FAILURE in 5m 11s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/6fff8b73d3264ca0b0b88f28f02e3145 : SUCCESS in 1h 25m 16s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/3f32d4ea4065408aa1bba4564fcea231 : FAILURE in 11m 48s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/9a0d9e9f92454adeb01bbacaf5eb8f5e : SUCCESS in 1h 22m 59s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/cc3ae48d21614c8caa504f4e389af19e : FAILURE in 11m 40s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/c11e7f60afd048f282a88a29debe77d9 : SUCCESS in 1h 35m 44s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/1da50232c7e9474392584836e2d54124 : FAILURE in 15m 14s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/47211fb943e942aeb72f09d56e006706 : SUCCESS in 48m 53s (non-voting)\n- kolla-ansible-centos-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/557e4353e8814b769fde075b35769ec2 : FAILURE in 1h 29m 58s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/ba973c5891834cdb91b1a749b03c9fa1 : FAILURE in 2h 26m 54s (non-voting)\n- kolla-ansible-centos-source-zun https://zuul.opendev.org/t/openstack/build/2a3615014a0442879c84c0a4a4418a57 : TIMED_OUT in 2h 03m 25s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/92d317b659ab45378f9b5b594eddb3ba : FAILURE in 12m 59s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/ecd6704e1d0b46668ac341ce2cc3d141 : TIMED_OUT in 2h 02m 52s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/767734807d3a44af9db6aa482a159aef : FAILURE in 13m 30s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/e725f4342755418584fc10a4c5faaf6f : SUCCESS in 1h 28m 12s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/de64d29199c841dc86ea1cdd4df3ad51 : FAILURE in 12m 11s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/7eef6afd735542c5b77d5d8745318d5f : SUCCESS in 21m 40s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/170ff76f486f46578a3ece82548a47a7 : SUCCESS in 22m 09s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/cd3eb05e410447c9b22b995b4226d82a : SUCCESS in 59m 42s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/41145d01a5034bbb831f1c0c66ff12b4 : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7403a6355c7e48e199d0a69f871cc0a3 : SUCCESS in 6m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dae8cf3c426c444eab39c81b80f7c530 : SUCCESS in 4m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/115847d8aa6e42cfae57a961252c292b : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8b92a0a5a26b4595bb55da8de58be896 : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e53e8fd5cbf4ac0b815a2fce7c0e3ef : SUCCESS in 8m 03s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/579120b829b24e7887d04a5d956e628d : SUCCESS in 31m 13s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e2e4da93896a401c96b901e4adca5171 : SUCCESS in 34m 10s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d8dac5c9f4f64e29ad99087487a1c69b : SUCCESS in 54m 55s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/58664e1fc6f6429ba792194073d94509 : SUCCESS in 58m 25s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0f8b9bd0a8784a868edd8dd75559942d : SUCCESS in 1h 41m 24s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/c6cc13cbf4ab4641a56b59210e85aa12 : FAILURE in 30m 24s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/d6fc58d37e414ccc9098312a78a1b44d : FAILURE in 29m 10s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/8e9e062482084a4bbde3b95af6575f7b : SUCCESS in 1h 41m 47s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/f49167a926de41c1bdc9ca3cdcd39d67 : SUCCESS in 57m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0fc4c59d4ea10b3027e155af383de89a699aa9c0","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-02-12 13:07:57.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a3e0b3e16f71cb97e17f71826536e3e2c4300baf","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-12 13:36:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/3536e458f0604b95bb6ac9b914b1638c : SUCCESS in 27m 51s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/e8d85ded9dae408da17bb8088ece73c7 : RETRY_LIMIT in 6m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"325284a7eabe21c5b6b8da6896616b3b6626cc84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 15:18:57.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/6e0b6f9af8614e789e38d563272be25e : FAILURE in 5m 43s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/806e7cd18b5c48c0bb4477c7215cb793 : FAILURE in 5m 28s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/e78284b153624a19820e2f3a29efbadc : FAILURE in 7m 51s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/6c3bf686ace7495a810c53f3bc99e633 : FAILURE in 8m 16s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/3f0696718e6943debe12fef5a36cd085 : SUCCESS in 1h 17m 06s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/857377d5857642caa698cb290f37c15a : FAILURE in 15m 26s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/dafd71f1d37c4e0e87eea59656d08fd4 : SUCCESS in 1h 38m 27s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/7a3f7535350848ee9052854796d9e7f6 : FAILURE in 12m 03s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/76949d138cfb4e4ab30928cbf13df7e7 : SUCCESS in 1h 27m 11s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/4ba3afe01f664748b5da8aa0b93cad5c : FAILURE in 12m 11s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/b08ab46fd19a4667ac21ba30be660ffb : SUCCESS in 1h 03m 36s (non-voting)\n- kolla-ansible-centos-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/98aa529b10e04df19e7718a385e96754 : FAILURE in 1h 35m 07s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/ed084bbb9fba4872b05eebdcea54cf96 : FAILURE in 2h 05m 01s (non-voting)\n- kolla-ansible-centos-source-zun https://zuul.opendev.org/t/openstack/build/358bcb2861334c919231e64109511838 : SUCCESS in 1h 59m 49s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/69fb4b96823a4238b94e7966d2e6f547 : FAILURE in 15m 06s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/7c9c473c009f4692851216762182f5a5 : TIMED_OUT in 2h 02m 57s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/4ec6e205d47342d0b47c7c8be92dbf11 : FAILURE in 15m 57s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/a7367683cec84ca8b41161dc42fa7570 : SUCCESS in 1h 25m 43s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/6fee995df6dd4e8da36aa829cdb62ae1 : FAILURE in 15m 16s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/879ef5efab16422c801fe5d27a28e5b2 : SUCCESS in 44m 28s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/7c19b7e35ba34393b9e4923f0876b96e : SUCCESS in 27m 11s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/fef2845f005d4c0d841cd9bdd62f391d : SUCCESS in 58m 50s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3a1baf85545f438f8dcbf06e49ac61d5 : SUCCESS in 7m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/709f814be4ca4da6bd2c30ae8afaa4ec : SUCCESS in 11m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/05261b71330a4ed0ab642a0dad05f030 : SUCCESS in 6m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5f6a448f49f540dd8f99ea46b3ded21f : SUCCESS in 9m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/def811926fa2440f85cdbbd9c6c8b9e4 : SUCCESS in 9m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5578ed84c1604efeb16e0bf72fc1951c : SUCCESS in 14m 58s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f5e3c6f7817a4e16aa2d84949e158f3c : SUCCESS in 22m 55s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/baf322653ccb44c3872cc26445f9da9d : SUCCESS in 49m 11s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2e0afe81cca54c749166568cdfcdc24d : SUCCESS in 1h 15m 59s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d941ba42c2b146a6bff7f93626673908 : SUCCESS in 1h 06m 17s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/02cbfac6caa94fd99130e1f016cdad1d : SUCCESS in 1h 06m 23s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/7446ac7e86464034aeae6f4043e9d293 : FAILURE in 26m 06s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/004a364386ef485e8125eeb4d6e5f331 : FAILURE in 25m 17s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/7ef4845c9bcf45cfad13bf5fa0f59000 : SUCCESS in 1h 51m 51s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/2a5eb1ac45f046d0ba18f00583ba27a1 : SUCCESS in 56m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"665d82ae802e06a0fef4c3f52be32450bfcdcb46","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 15:49:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a854668b95be499a849fc069325d6b13 : RETRY_LIMIT in 19m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2ffdace676217e6e3a90d676917b1a9efac29d51","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-02-12 18:25:28.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"9e401d1a599ba11840ff63b94548e03bf2a17fec","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-12 19:02:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/5caeb7063004428dad8b1755a079e2ad : SUCCESS in 31m 28s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/9091ca7532a24c6c8bfd6f7d53df2929 : FAILURE in 28m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"611c361e8239c4406e431e007d2a7fdd1d8d81d8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 20:21:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4ac51f3cd90447578e0f89720597b820 : RETRY_LIMIT in 19m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"42698c3ff21ac133761b2de237d46e9998d626b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 20:33:03.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/f58b436b361242429fa38717272b3b80 : SUCCESS in 55m 09s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/708d499f781648688acd610721532b25 : SUCCESS in 39m 31s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/d4af3d93e5e7459b8a0e666a96fcbb13 : FAILURE in 39m 52s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/6b45055f8bee4819957797ffb3d6e0d6 : FAILURE in 38m 05s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/b766b530d77544bc86000def6f6971e4 : SUCCESS in 1h 21m 19s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/597c59762dc04ac3976595fc8bef1b33 : FAILURE in 30m 28s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/2fcd3c0c96794be49577a385a046efd4 : SUCCESS in 1h 17m 47s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/032efee2fd204cbaa05fc44191b35d06 : FAILURE in 23m 29s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/1d6c6a428f5f4e32b9f6b2d2f130390c : SUCCESS in 1h 59m 56s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/71680a64fdb44f74b060ad28e16ecb8e : FAILURE in 13m 40s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/c79018400a284eb49538fb25d8f0b2a7 : SUCCESS in 52m 33s (non-voting)\n- kolla-ansible-centos-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/fcebf4fc8d4e46508240c4b05baeb612 : FAILURE in 39m 36s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/9821118fa36244e49617e30ae4fa180b : FAILURE in 40m 02s (non-voting)\n- kolla-ansible-centos-source-zun https://zuul.opendev.org/t/openstack/build/2daa21fe8f5e48cb93fccc6dd4dc4b5e : SUCCESS in 1h 55m 18s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/4193a7a77f854bbcbf0a71e8439c1720 : FAILURE in 40m 51s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/cff9d15259c84ad197450099092e0ad2 : TIMED_OUT in 2h 03m 13s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/5d5999fb8d76434088e06a7b037188ef : FAILURE in 28m 38s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/24db9cefbc5f4b9bae232fb83503c26f : SUCCESS in 1h 30m 07s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/89119affcc264e1c835659bc34969e83 : FAILURE in 26m 54s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6e3a705927af4704b0890d3221c49754 : SUCCESS in 27m 53s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/63d8fe5a6d094ed2a13466ca9236f581 : SUCCESS in 27m 14s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/fc1305f789a24bc4810254b879c0091b : SUCCESS in 1h 08m 02s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bce401ed4a0b4170b811a451d3e5c923 : SUCCESS in 6m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b8f995726d947419548baf85a633bfe : SUCCESS in 6m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f3535d7f73541658dfc57831c09fc58 : SUCCESS in 5m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bf4454c3250b4923ba1c6f928cf02999 : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/426ec2edcae94dd49f2db93237a38a63 : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e990797ab03f4814aa5952cf735d0b90 : SUCCESS in 7m 25s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8ac51a949b7a4634bde120cc12595c46 : SUCCESS in 25m 36s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f961354f57d94427b9df04f7577b8a24 : SUCCESS in 26m 28s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/70aa01b67b10409e88e7caef532c467e : SUCCESS in 1h 02m 20s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ab8fd6e4500147ba94d2847919453a55 : SUCCESS in 53m 56s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7a10a4bd3ad449e78225776f790c6953 : SUCCESS in 1h 18m 56s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/44586a25e4c54256bc8c904d4a206141 : FAILURE in 30m 49s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/3071ea38fd6f42988ce11200387f2c5a : FAILURE in 26m 55s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/8182d4b333fa4b68b28bb4c951130e91 : SUCCESS in 1h 27m 57s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/8bbd29f8458442708991d97c8311561c : SUCCESS in 1h 14m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9e3d9bfa05ae0a0f8b45663165c7209d12699f74","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-02-13 11:33:33.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0dc8f74783f6593fbb741730ef7d3b40abb1c161","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-13 12:05:30.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/ffb5314ceeea4183b7ae239e589bf9e7 : SUCCESS in 31m 12s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/298c684da31c4c07846b23e7fff1d753 : SUCCESS in 29m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5914a3cf9b384056a58dfa8cae8779bdf0f2a26a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 13:41:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/a93389d81d914f2cb01f9c4eccd35dcc : SUCCESS in 1h 18m 42s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/12419eef749e445d8dc9f7b61328e4f5 : SUCCESS in 1h 07m 05s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/eb6c0f7cf4c7402e8042281e5ec6b2d1 : FAILURE in 8m 56s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/de8ef2fa95ed4f05909755a9aa3dfd86 : SUCCESS in 27m 03s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/e6281a1144d24dcc8f8b9c562be265c8 : SUCCESS in 1h 32m 02s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/cdb6fdf1c4054149a33c115a2005b796 : SUCCESS in 1h 06m 28s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/4a44d9a2e85541abb743585a2e8592dc : SUCCESS in 1h 14m 05s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/48102e87a0a9437d96ea3abc54e972aa : SUCCESS in 1h 02m 34s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/6f1bc7fa70434ae58396be3e75f099fb : SUCCESS in 1h 26m 39s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/d0ea11c0646540249035a16f8ca2f937 : SUCCESS in 58m 43s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/d24568d05c0a49a3b8b1efd39470ce5c : SUCCESS in 1h 01m 20s (non-voting)\n- kolla-ansible-centos-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/ae523950700d4a5988049e2581838484 : FAILURE in 50m 07s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/e92a35a5cf6e4f88a323a8b142d78296 : FAILURE in 40m 31s (non-voting)\n- kolla-ansible-centos-source-zun https://zuul.opendev.org/t/openstack/build/7a9794d6cc8e47a4a952aecd78c62dde : TIMED_OUT in 2h 02m 26s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/fa501aa40d704bc9a8901e868f082e29 : SUCCESS in 1h 40m 51s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/35554bb4aca84d1f8a74d52d2b8c8bb5 : SUCCESS in 1h 53m 22s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/7b0469e6aa1c4dddbd5821af3b39f43d : SUCCESS in 1h 44m 18s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/9c254f644e78426caedc0a1fd14a1d99 : SUCCESS in 1h 36m 53s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/588700608d2f4d939d1c20ec167dd766 : SUCCESS in 1h 15m 14s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2f642a6175954ec58a875fa30c47ebd3 : SUCCESS in 27m 05s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/98f10c13e7784f5284344aff541cf60e : SUCCESS in 22m 11s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7890ccc2f6724559b19c3b4ec035298c : SUCCESS in 1h 01m 07s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/29eafa5557544a8a9915b79f68c109dc : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a979e54c6d945f2915da29e392b9e8c : SUCCESS in 7m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d6b4de0cfdbb46299a05e7d835c8196a : SUCCESS in 4m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/69d7df2feb934f9996cf786eddd95866 : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/27e5167a002f453d89139134f9fb2959 : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2bb3b5c5c8ea4095a2f7e2186f0b1ea2 : SUCCESS in 8m 27s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7fe8d81cb64548a1ac3a5a3bc2d17dc1 : SUCCESS in 24m 45s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/070c8e217c3a4024b3ac42108564b6b9 : SUCCESS in 27m 52s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7653cb77c7bf4f01a3bd8402eaba8a3e : SUCCESS in 1h 00m 23s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9f434d92d1294d0d8e12359290e99a59 : SUCCESS in 1h 06m 24s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/145dcfb7c9144dab87a8f581110b2ac8 : SUCCESS in 1h 03m 41s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/be04aeed0c1f40938cf49c57c7c9fd37 : FAILURE in 28m 00s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/4a1a51c77d0c46698066c8a1fa286062 : FAILURE in 35m 37s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/4cdf1aed61fa46708a71d75a8aae10ae : SUCCESS in 1h 19m 58s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/c68279d4b1834c7aa65e3af93c589958 : SUCCESS in 1h 00m 27s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b195d4903cc8ae583a6cdd4979f6876baceb11d4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 14:51:03.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/17ada3441a84442cbc2705ff2e27b0e3 : RETRY_LIMIT in 19m 27s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"735c675eba3af506f840d8db5f30c2187c25a6a5","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-02-20 15:10:49.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"48e0ea6120228203294f3e1b33e8f6bee1831641","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-20 15:46:59.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/a1759760d01b4a6d9d01db3092c84e2e : SUCCESS in 29m 34s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/0882223d32f446aa9312701b7a152683 : SUCCESS in 33m 59s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6b8bf5971c3464227becd521871589e7de219732","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 17:39:36.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/e4494737b5674f8990666379fb50df5e : SUCCESS in 38m 40s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/b438ba0db2f3463eb67937140225b055 : SUCCESS in 42m 52s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/f25bcc4bb1d14fd6a05281c4de387ecf : SUCCESS in 31m 54s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/c9e48101c91245a9844a56f4862b5868 : SUCCESS in 30m 31s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/82010b015401495187e8722b641928e6 : SUCCESS in 1h 11m 23s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/7bcdf5203a8240a093d3058ff1eaaff0 : SUCCESS in 1h 29m 22s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/0d4fe5340608400a8ab641ca330ff469 : SUCCESS in 1h 33m 16s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/276255fc3abb46a5832389c02444808d : SUCCESS in 1h 28m 35s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/b2627e59e5144d04a926b976afb4676c : SUCCESS in 1h 40m 23s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/bc234e018972421fa284d0dae0f5cab5 : FAILURE in 44m 06s (non-voting)\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/a6ed9be3076d4f7a8c9dee0a9f99f2dd : SUCCESS in 52m 57s (non-voting)\n- kolla-ansible-centos-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/6e4a6c111e7148ebaa17c24b048fbbb7 : SUCCESS in 1h 55m 19s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/7323affa837b43aeb4704f0ec8031d06 : SUCCESS in 2h 21m 39s (non-voting)\n- kolla-ansible-centos-source-zun https://zuul.opendev.org/t/openstack/build/6dbc765fb2d644eda48647d6c60f22b2 : SUCCESS in 1h 54m 33s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/d039b0718aba44be9711266c4744f418 : SUCCESS in 1h 44m 46s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/832ac47732304457af0e99bd66a7a43e : FAILURE in 1h 47m 38s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/23387142b8454dbcbe0109b3d86ce6f5 : SUCCESS in 1h 44m 44s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/37b83dff2b804254a1b6dc18e42afac3 : SUCCESS in 1h 22m 47s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/866c0ac08cf24f6296dafc6797540046 : SUCCESS in 1h 17m 39s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/3d8cf15e4f2541ce9d009f43b5be81ba : SUCCESS in 25m 04s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a13e05c731384fc89482a3f399f75fb5 : SUCCESS in 20m 08s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e09417c4df694746b721cca588134eb8 : FAILURE in 50m 43s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cb3f5de519b447c28540949cddfc2aaf : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39f6cc5a877641fc818a8f8c0c2486a9 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dbfbc1a9b0cd4bf7ab0a203fb06a0428 : SUCCESS in 6m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0a2439a564844b1fbd35bb85f495bd17 : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b527737963d34ab8b1d6216d990a9639 : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6a765b6ce3d40899878a51a952fb058 : SUCCESS in 6m 01s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/db2568b27b984cfcb7e7bfa98b8fc1ab : SUCCESS in 23m 35s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/771965b80661435e937c51cc0f75d37b : SUCCESS in 27m 10s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1d88849ede624a55bca0b10d2a2bd363 : SUCCESS in 59m 42s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/42fb3f5f4ac144e59cf4b97460ecfd57 : SUCCESS in 57m 42s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8c9ce39a56c942938dfb5202bc29cde5 : SUCCESS in 1h 00m 31s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/6b8a881fcc794301aca043970368b1dc : FAILURE in 29m 43s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/25189a9542c046499636b258d2200eda : FAILURE in 26m 20s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/c47ce2a5a484443c97b01f1f12a5d2c7 : SUCCESS in 1h 25m 31s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/7a0cd47795bb4a00ab87aaaebc9d3f5b : SUCCESS in 1h 03m 30s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d6b99609603291b2cbd82d04d111329a89ca1486","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 19:26:07.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9d13b821e62c4423aeddc024673f57be : TIMED_OUT in 3h 01m 49s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3df19c4b996cd36fd1dee5c9ee3f51b0dce488eb","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-02-21 09:41:57.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"4f62207c2361767c414fadf6780547f0bd1a6f6a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-21 10:28:54.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/c75fc4b4d1b5458b805bb578cfb15325 : SUCCESS in 30m 42s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/ccf92286f602444d9a4fb43ade6bf505 : SUCCESS in 38m 35s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f08291d88e07900976086a9978647134df31e878","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 12:07:23.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/3fa31bbe6eb64d40a010b08ba642c8f3 : SUCCESS in 43m 30s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/a43ed2ffd32d42aba3a4c9e2b093c0f5 : SUCCESS in 43m 41s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/8922fd04036f4577b137ad0d999c910f : SUCCESS in 43m 28s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/1180186b71d34c6c9cac3e17359bd766 : SUCCESS in 48m 35s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/d52b0b43fb064116836b978126200064 : SUCCESS in 1h 32m 02s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/bd35ad3eb8734ddba5bab66afe0fb848 : SUCCESS in 1h 15m 39s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/523f9fc6fd7f4f23be7a7ec209413b16 : SUCCESS in 1h 27m 40s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/e0f18011937240eebe2ecca5df401384 : SUCCESS in 1h 24m 13s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/d153ba3d65bd454bb502abb9f546ab5b : SUCCESS in 1h 24m 51s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/b7704d97c1d64490b92b67192d500307 : FAILURE in 41m 35s (non-voting)\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/d448e1a89610476fa189e78e96ef77c8 : SUCCESS in 46m 50s (non-voting)\n- kolla-ansible-centos-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/d614eed54cf7486fbdfe7e02ab420034 : SUCCESS in 2h 09m 54s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/860f7297391f4d86ac0458800633abbd : FAILURE in 1h 39m 08s (non-voting)\n- kolla-ansible-centos-source-zun https://zuul.opendev.org/t/openstack/build/aac1f8cad8924b25b624da74b29a7bf9 : TIMED_OUT in 2h 14m 02s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/26501eed091943e1987c3d02999d9e1d : TIMED_OUT in 2h 10m 48s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/ce086f92e8464080ad59a3489813c938 : FAILURE in 2h 05m 05s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/278f61308944449a8a831f6ce8de3d26 : TIMED_OUT in 2h 10m 56s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/4debdb65757049aca3fd8f89c4a1673c : SUCCESS in 1h 29m 25s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/17b3b351cfb14f56b13c3aafc601a41e : SUCCESS in 1h 44m 24s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/74ef3aac83a540999fbe57c521db49e5 : SUCCESS in 23m 35s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/9f7599f0adff49bbb3b69e01f08006bc : SUCCESS in 40m 25s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/437c217f859248a78a14434f038372ef : FAILURE in 54m 11s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/214264f5bb424d9cb90310cb3dd72e8a : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c97c1a12b384b658631deb57c261e74 : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b410b946c8c4c2697ee74e85bd8009c : SUCCESS in 5m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9cf00348261946e2886f6662f24f8f58 : SUCCESS in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/beb09b1748124ad69322cc5f7fc2215f : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4048aa6a24342e798ec02ffa5991ee0 : SUCCESS in 7m 32s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0815b2c8b0314a5d951e50763865bfa3 : SUCCESS in 22m 59s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/25a16a17da4347f9999db25c036ac174 : SUCCESS in 35m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c98ab13b593849bd837eeace17894243 : SUCCESS in 1h 05m 58s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5c5fdeff55f64a43bb008a4c2cb3aeb7 : SUCCESS in 56m 54s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f41072fdb64f40c4afadde69800c92dd : SUCCESS in 1h 25m 21s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/c6b9844f795f4397b9935a0cf391ce37 : FAILURE in 34m 32s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/188700879e5446d1832894df43eec18b : FAILURE in 34m 03s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/0c698b003ecb452b83661d0e70b9fa9d : SUCCESS in 1h 41m 14s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/2c1f16cadd964b808da03344a4a5e024 : SUCCESS in 1h 00m 50s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5de04904a4053b711fd8d846dda5061dc77d8170","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 13:11:51.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0aa202f833a54a8b80b4fed9e3f2453c : TIMED_OUT in 3h 00m 56s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8f773111d3e7fef156469471161f10369fd95bd1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-02-24 10:59:48.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"cf88d3ff3c422acee3e9098f10acceb4f746ab54","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 11:05:51.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f6e636201a0e45c2ab9b7ec901e68869 : FAILURE in 5m 13s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4e0aeef25904db3050c411055740faf4f9416112","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-24 11:22:50.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/7fcefe82166042259a4c2523a2279698 : FAILURE in 6m 55s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/e9d28c4a5e3047078acde1f7e325caac : FAILURE in 9m 37s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"0bded828fd27332e0d6e83de0c52e5e2e283fdb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 11:28:27.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/da129b8ab0944ae5a7989dcb0efadafe : FAILURE in 8m 06s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/e120595189c94bf498860a0c5f029896 : FAILURE in 8m 09s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/bc19adc5a69b4977b8a7214ee7530a76 : FAILURE in 7m 32s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/60c59a79fa604a24a1e4a18ae093c130 : FAILURE in 8m 27s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/c362329b6a84408eae6bb6cba3cb5190 : FAILURE in 12m 57s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/8eab9cf1fe4343debdec4c9fb1caa5e2 : FAILURE in 11m 42s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/03db670ffa5a42e5be9b828cd40727e8 : FAILURE in 10m 05s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/9e400ab0e1204686b5553b58f9bdded2 : FAILURE in 13m 56s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/e55ee0604b5f490d8c32ae72cfd1c43c : FAILURE in 12m 14s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/511526ebae584ceab78c4a598eb53458 : FAILURE in 12m 39s (non-voting)\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/d1c7dd0ec80248ca8af05ec03b8deac6 : FAILURE in 11m 08s (non-voting)\n- kolla-ansible-centos-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/d5f74be072fe409fb97bf78a20a82fa8 : FAILURE in 14m 46s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/88c507e0cca343ccb34c60f950bc0c0c : FAILURE in 18m 05s (non-voting)\n- kolla-ansible-centos-source-zun https://zuul.opendev.org/t/openstack/build/cea0e4afac1047f98797f939985942a7 : FAILURE in 14m 54s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/31f1f3431ee444b495f506f95f25eb34 : FAILURE in 14m 21s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/7a4f2f07d5a4457093e5cf18faa2211b : FAILURE in 14m 50s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/5ce20a3b55d0483a8d24a778a380d606 : FAILURE in 15m 03s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/9325f95e82ca463bbaafc23bcad18704 : FAILURE in 10m 53s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/06953900fafb4f7e92e40b6903720af5 : FAILURE in 11m 29s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f3f11895deda4d5081c1a73ce4d60356 : FAILURE in 5m 16s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/4bbed742aaa54d5f9f90a5ed1367d3a6 : FAILURE in 6m 03s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8525f136461c43dbb752ce220ad44333 : FAILURE in 9m 27s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e556dfe81e1b4434ad08e5bf051cd5d4 : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f37c622735064536a21c7cee6ca66d19 : FAILURE in 6m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d4c26d50d944912999e2a4ba8c25945 : FAILURE in 6m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dc100ae30b3f44d1a4ed2111a6b14f30 : FAILURE in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/33d7807b72ed4d2db9390e39601596e4 : FAILURE in 8m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7a5e332db024c1898782b0e90c67a06 : FAILURE in 7m 23s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5464161a13cd42beb3a5df9c285d1856 : FAILURE in 6m 46s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ef5211ec311c43d0a18b7e9c2bf71444 : FAILURE in 4m 46s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3314eaabe530459eb3b0318416eae119 : FAILURE in 10m 57s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ececb36252374d33b4030108eca6bc29 : FAILURE in 11m 11s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/10bd6ecd59144a9ab79ee8bab7b1795d : FAILURE in 10m 54s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/59b10fec2fe2478494422daba83a815d : FAILURE in 17m 52s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/2b448519379d47c9b6a0c0c8c6c2c1c3 : FAILURE in 14m 08s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/7bd7b0c4ceed4a10a84fe3f92c750546 : FAILURE in 14m 31s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/b259aa5978da4ece8efc69effd9c4366 : FAILURE in 11m 45s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"884217bd21c250ac3f97f5cb88f5fd683e123b2e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-02-24 13:40:10.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"3906974915cac45a7f16aa3550f231f1644e5031","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-24 14:47:17.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/5dc8cb4761f447e79625be42fb6e7b30 : SUCCESS in 42m 05s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/e2183edbcdde4b70a37ed0a615026061 : FAILURE in 18m 17s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"af3f4db3aaade270cf54d0faf67b06125330608b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-24 15:28:47.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"005ec14c90e9d3ae317c18741c9b1aa21d76536c","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-02-24 15:31:17.000000000","message":"Patch Set 5: Code-Review+1\n\nwaiting for zuul","accounts_in_message":[],"_revision_number":5},{"id":"8b9cf7102b8c20bb9349556e9ff06fcc965edb56","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-24 16:35:19.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/7f14c7dc6612483ebb396a0d87352023 : SUCCESS in 29m 52s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/94932de1b7814241b425da163978c6f8 : SUCCESS in 32m 44s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"fdf64bc33de6920f73625946f0f2c5956de2bf1e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 19:02:48.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1334f6cd669f4e3eaf280d5a302d0e83 : TIMED_OUT in 3h 02m 34s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"692af8399e757263557e4d209fe55f10caed9b4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 19:52:13.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/4980b179d5cf4d1d8f8df24cd98567ec : SUCCESS in 34m 19s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/5e44239e0a7a4aad8624a8888d64b3f1 : SUCCESS in 43m 54s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/fe01d2ffc8664c7c93c3aeabd51f99ba : SUCCESS in 39m 01s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/9d661a53f60a4d39ae644b6d559949eb : POST_FAILURE in 31m 49s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/7967f6b136e54dbe88a7451a3e3c8c73 : SUCCESS in 1h 28m 48s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/4d6ccb344de541849d2c344d16f85cee : SUCCESS in 1h 14m 25s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/754f4469101f4b899236fb211e7bce79 : SUCCESS in 1h 30m 07s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/032772d8b17844bbae6956623ed3eaa0 : SUCCESS in 1h 24m 47s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/4768d25f59a24d688a494b4725c209f5 : SUCCESS in 1h 32m 18s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/4c21e11e1e064b1f8be22fc4d34175cd : SUCCESS in 46m 21s (non-voting)\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/f6c3c7e340a44185b11142b1f06e4ff0 : SUCCESS in 55m 45s (non-voting)\n- kolla-ansible-centos-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/593309d99ab34ccc983a354532d51ce6 : SUCCESS in 2h 06m 34s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/18cecc20fdf54a06a1b65cd52ae458c1 : POST_FAILURE in 2h 16m 39s (non-voting)\n- kolla-ansible-centos-source-zun https://zuul.opendev.org/t/openstack/build/769a4dd6bc464be6aa0eef915d8bd89f : TIMED_OUT in 2h 09m 35s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/a723abe3e7d14a7595aed97140a83635 : SUCCESS in 2h 08m 46s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/b4077a5a8e3047e29df8c6fe778c6288 : FAILURE in 1h 16m 32s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/0f9e02f292174971862783aba06874a6 : SUCCESS in 1h 54m 42s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/dd1ac7780fe74f958a62332a9ae279de : SUCCESS in 1h 28m 37s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/cc0b5d91cb954924ab7dfcec25a006e2 : SUCCESS in 1h 16m 23s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/3b1f6110fdc94052a0752f516df44a32 : SUCCESS in 30m 47s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/19540f4257a343ffa59ec99caff2d1f2 : SUCCESS in 20m 04s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4509cd3822634ef9b3371803688fbcab : SUCCESS in 1h 03m 21s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d2610c364dec4ea1942261ccd9a0c345 : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78fcef8f25cd4bdba78f6ed8b25ea435 : SUCCESS in 7m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b539dda0e6004da3a9521584f3c0d985 : SUCCESS in 6m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d8cfa04fff39486b87785e3aab0f14bb : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d69eab2bebad4fdeb23f37c6ff216511 : SUCCESS in 8m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbb20ae436f246a38e0bb03c9023fd60 : SUCCESS in 7m 09s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e78830ff1dd34dec8d23b95ee97d8729 : SUCCESS in 29m 36s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e0a64504c30a4c828e1c9906e78eb74e : SUCCESS in 24m 59s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8cc0ebf84ffb43378336a1d1acc75c6b : SUCCESS in 1h 15m 30s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4fdbea6da3d841fc9e081aab26001c8e : SUCCESS in 1h 09m 22s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/734d0b2483c642aeb5a3f7e3e745f39e : SUCCESS in 1h 07m 24s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/a9bc42f7f03f491c9c8dd13163ff2f8e : FAILURE in 29m 23s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/eebcd24ca9a3432fa5643b518c1b20bd : FAILURE in 32m 02s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/f9f3df608da44d018486e613a55ca70a : SUCCESS in 1h 43m 42s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/45090c9489484a4b9fa1637ce9af5824 : SUCCESS in 1h 05m 10s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"3163bcc8cb9352ae8cda579d89b45b8962c285f1","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-25 09:15:53.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"56e44e8eba93eb6646f080a080d6af8dcf8b5197","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-25 09:15:57.000000000","message":"Patch Set 5: Code-Review-1","accounts_in_message":[],"_revision_number":5},{"id":"15d6cc6f8b4479977f3f0895a0864153d4467705","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-02-25 10:50:41.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"100292dc5ac4b98c349470de5918783caedda03f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-25 11:31:27.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"596ccba2b63679898781f82c93c41d6a1c915aa4","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-02-25 13:21:15.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"be7ba86a0ec70b370070de0e620f3d62a442ec2c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-25 19:25:21.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"44511ac72e79a66f33a14436b906050b4657c91b","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-02-26 09:36:41.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"a874b44c916de31cb0d489c444f4900ce8fa6906","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-02-26 10:00:07.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"8dc3636aa1adb02e4740d6784a13734772668ebf","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-26 10:05:28.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/242f9b9195eb495fab3b40bc0dab9d57 : SUCCESS in 27m 05s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/d0f5c4f068e64a128dd00850a183f3bc : SUCCESS in 27m 58s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ecb24b65229c08e28773018fa59cd4f128f023b7","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2020-02-26 10:06:45.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"e1f33d876921f367f3c45b56eb32c481bd3c1e9e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 10:17:12.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1823b8187f00484b80e552c87a108c2e : FAILURE in 7m 06s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"c9638c43a8ba06200f482eaa8bd11d0c2284e085","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-26 10:42:10.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/5c45d5771f3c486989c920d0562c91ea : SUCCESS in 27m 13s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/3ac3f6dcc2db496bb6327b2d8c777291 : SUCCESS in 33m 14s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"5eae5bb8dfbfc8111ecc43d7f9aaaab2c5422b12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 12:49:00.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/6afd9b9123ed49a896e3663ab08e2cee : SUCCESS in 57m 00s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/48920219a21741c5970d027026d2a1e2 : SUCCESS in 1h 02m 22s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/d00196a49f04446bbd755c7aab7f6294 : SUCCESS in 1h 09m 09s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/39ae3b9e60fd4ca28ad7e3808e581e5e : SUCCESS in 1h 08m 33s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/cc71a9c3036846e0a6eb890f65d33e48 : SUCCESS in 1h 22m 44s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/a466cf4dee8144bb833c2fab5c7ffdb3 : SUCCESS in 1h 08m 19s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/6aae3674ba294d61be2e9a91612e67ee : SUCCESS in 1h 20m 35s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/72db8fc6377c4e53b9af414084722839 : SUCCESS in 1h 21m 55s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/ab49de09b281423ca32126c4a28fd628 : SUCCESS in 2h 05m 26s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/430e9727a3ff47b89dd2883ea5800575 : SUCCESS in 43m 49s (non-voting)\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/c21d448d3ddf4fcaa8e0a74415f0bd00 : SUCCESS in 49m 27s (non-voting)\n- kolla-ansible-centos-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/67ede2cbd419483999b07c2f3d03392c : SUCCESS in 2h 21m 22s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/8770bb50f6954dad9fce610cf6813970 : SUCCESS in 2h 20m 27s (non-voting)\n- kolla-ansible-centos-source-zun https://zuul.opendev.org/t/openstack/build/a154c8ef13784a23bf4bee9d68ec70c1 : SUCCESS in 2h 09m 27s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/dab25a7b973f401889119263c42a0873 : SUCCESS in 1h 58m 57s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/59395bd85b404ca3a637427633365f3b : FAILURE in 1h 37m 16s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/124191e9967f4b4cbebfd3baa5ecad72 : TIMED_OUT in 2h 09m 02s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/86e0a642aafc436e9fe9bf024e7f4e75 : SUCCESS in 1h 33m 00s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/45bfd42de7674054a8eee86b67038bd8 : SUCCESS in 1h 23m 15s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/fe602c2cab604dd4b68aa5ca75ef541f : SUCCESS in 31m 46s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e3e68ff41b634f85aadb32e6e5f91bf6 : SUCCESS in 28m 49s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/1cdb4a293e6844bcaff5d4dd7e3867d7 : SUCCESS in 1h 09m 40s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/625833e7fb5c44078e98d36ca069a8cf : SUCCESS in 7m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c32c1b9f92c46d38d89c2fa10eb4dec : SUCCESS in 7m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/51f33ce7aee648c5897c73d264a7510f : SUCCESS in 6m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c59b3e7534da47baaf1012de7ae8023b : SUCCESS in 6m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0bb64d78b3434f87a8abc6ed9686ddd6 : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/310598ebfa1447d1a48a952f5d766509 : SUCCESS in 8m 25s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8fb42a659e6c4c89bdf6e9eaf3eeacd0 : SUCCESS in 29m 24s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/47a5ea1786154b96b5fd47f1ed3eb71c : SUCCESS in 30m 00s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2495bf865d5e4a6099f71775baba0bfd : SUCCESS in 1h 11m 38s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ce139942d52a4f7e8fdc27f03ccbdedc : SUCCESS in 1h 09m 02s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5ba389b2e76a44d591d1bc572b28acb9 : SUCCESS in 1h 20m 22s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/c19742e5f05d4cf7b5976b4d5586f88a : FAILURE in 37m 34s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/1ce778f1a10b45ba831cbce0dff90062 : FAILURE in 27m 43s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/2463c74b3a2948bf9b6263701e8a032d : SUCCESS in 1h 36m 13s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/b9caad4d51b64f14aff901387a219ce3 : SUCCESS in 1h 03m 39s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"6af67d57d9deda3ffe0b3931409f696c8cc10de2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-26 14:01:50.000000000","message":"Patch Set 6: Code-Review+2\n\nThanks","accounts_in_message":[],"_revision_number":6},{"id":"711a21c458f890ec6e6a9650252415798b96be37","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-03-03 11:02:12.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"c208dcdc580df568d26ec2796fefdb24063e0e3c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-03-03 11:38:32.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/bdfbd24881ac4b2fab8c041995817f51 : SUCCESS in 25m 43s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/27ed08be762f4ef8ad2aa14f34e75e07 : SUCCESS in 34m 24s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"2035cdda5b97f2aa8f6a9442bd7931d2c2fdf209","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-03-03 13:27:35.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"a981ee571023536147a89b91491242ed876fe89f","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-03-03 13:29:39.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"c2561d12c71aa3240348aaa80b99a148443d5706","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 13:41:50.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/7e8fd39fd6094fdb94baf5caac02a82a : SUCCESS in 39m 27s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/3f71408b667c48a1ae16d362d5183c2a : SUCCESS in 56m 47s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/0c5bbd8a283c494593429de3a04fc379 : SUCCESS in 45m 25s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/2ba8f76bfdc448f4b7dd45d55534ea0f : SUCCESS in 38m 09s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/0b39a284c2be4d1abd600ec78998d85a : SUCCESS in 1h 28m 08s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/1084d923ea56493d9ce27e2a4b7de504 : SUCCESS in 1h 15m 59s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/8aa769af690649d9b4fcc1489b28f08c : SUCCESS in 1h 30m 05s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/49768c875f2f4da4b4cfa397a0c1d448 : SUCCESS in 1h 14m 08s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/a29276bceebe41e0bbb3ae3caf26dbba : SUCCESS in 1h 34m 57s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/3bfac9129cd8408f81382dbfac02be37 : SUCCESS in 38m 45s (non-voting)\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/ec8a24c9f4124dfbbcad537daf4b0f98 : SUCCESS in 1h 16m 19s (non-voting)\n- kolla-ansible-centos-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/f63927fc00ba4aa9acac6b22a4e6cc24 : SUCCESS in 2h 33m 39s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/4a6b2b9cf2384336b2c050a3b0ed93b6 : FAILURE in 1h 26m 02s (non-voting)\n- kolla-ansible-centos-source-zun https://zuul.opendev.org/t/openstack/build/f6c4dbbe6cb14e3fb2183c11ae1e41e6 : SUCCESS in 2h 07m 15s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/765c1b643ee04d97985f672c86dd0bf0 : TIMED_OUT in 2h 12m 30s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/6eecb5c61db349af92be1250f76482a3 : FAILURE in 1h 49m 24s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/fa7dd927f11246e28e70bf7fc2d538e1 : SUCCESS in 1h 59m 24s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/8406da6fce5d478a8100fe5bd0996ca5 : SUCCESS in 1h 29m 24s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/fa7ea1d3a2b1424ba61a84d13b376be8 : SUCCESS in 1h 45m 39s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/7e994de3dcb2494c89f7dfad095d61e5 : SUCCESS in 26m 54s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b1d7ef6addff4e018993aa8722cfbce0 : SUCCESS in 32m 03s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c84ed57383f24b25bd1add93f6a3fb7d : SUCCESS in 1h 25m 19s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/23c10fbbc72c4f04a644a2c13e8f29e8 : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a7d3afa9960481483d2dd5340ce0d94 : SUCCESS in 6m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd75392265984027b8650dfb4ce79072 : SUCCESS in 4m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/716c7bcbea794c69b1e6fe2ac8cd150e : SUCCESS in 6m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/beb66bce06fe4bb5bafecf66a22e72d0 : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b228ca0569a484b9dce3c21e9b418b9 : SUCCESS in 6m 54s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/24017c47d05b451785d6156eb63c0aaf : SUCCESS in 31m 45s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5f5953905b004e078b752a05622fe870 : SUCCESS in 43m 32s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8c71c2f2af2f422396f2ae7343560c9f : SUCCESS in 1h 27m 09s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7f5cf315e2774181820a40e6ca14bbfd : SUCCESS in 1h 21m 48s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/17c99fc96e414aeba2bea5fda02b0e8a : SUCCESS in 1h 24m 22s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/b2583a6e3fe34629aadb9ca3a336454a : SUCCESS in 1h 54m 13s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/f66c70320df94207b39264058de62b2e : FAILURE in 49m 57s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/00848b57112c4d0d8f8b0525ff84c895 : SUCCESS in 1h 50m 52s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/a6e1b765f1ff4e2189cc731fd81795c2 : SUCCESS in 1h 05m 11s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"3d3be8488a5dcdb1062347a41b794d9836b03949","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-03 14:01:05.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\nThanks","accounts_in_message":[],"_revision_number":7},{"id":"2bb495ab4a55e3efcd1a0e9d4a7fc1e2882c1559","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 14:01:19.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"33535184d61bde8c86acd9c3b1b3c0c0ad1e58f9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 14:09:01.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0d8460bce68c43c6ad3758abb7b01820 : TIMED_OUT in 3h 05m 52s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"457072d67319b84f50ffef459d3e60abd6f2595c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 15:36:14.000000000","message":"Patch Set 7:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"ceda495c807969604df5c6979aaacf9eee2ab9ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 17:33:31.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/53ed84a159aa4473869bc10c3b37c8ab : SUCCESS in 44m 07s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/386cb4042527400680dc83da78858575 : SUCCESS in 33m 13s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/225747a73a5c49df9bf98b1093d86308 : SUCCESS in 1h 30m 50s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/66637332490c40d0884347c10a3a3c8f : SUCCESS in 1h 30m 42s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/b0e6b091372d4ab084966681972d7124 : SUCCESS in 1h 36m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cfa2e1f25d384103aa879321bab37047 : SUCCESS in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acb0c985f3194240b5f483d3ecbcfd80 : SUCCESS in 7m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/984a8160e5f94b7cb6e37429a93c1325 : SUCCESS in 7m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ef2ab15a7ddf498aaf15d7c53248a32e : SUCCESS in 7m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d5f7fb131bd4436badf47ed2d3342da : SUCCESS in 10m 26s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/dc43ab5dd64241309775b4350ec5c9ae : SUCCESS in 27m 20s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d39ac74aa7b64fffb0529682ff08af34 : SUCCESS in 1h 12m 51s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3714a93d90324b51873c931898ff39d7 : SUCCESS in 1h 21m 19s","accounts_in_message":[],"_revision_number":7},{"id":"5cbaec3224e7ac3d39f4e0c76ae9d4304b6130c9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 17:33:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"3ef7c09e2b474e99e92a2ca31dd1cb025c7f70b7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 17:35:32.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07275d0c5b0f49e9ae9046dcdbd7ed47 : SUCCESS in 1m 25s","accounts_in_message":[],"_revision_number":7},{"id":"552f6fbdf5084522ec5bd4f176c34c39235e81f8","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-24 10:16:56.000000000","message":"Patch Set 7:\n\nre: https://bugs.launchpad.net/kolla/+bug/1880375 we seem to have switched to rdo trunk for master, not release...","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"0c6312968214e403b5a408d69df17bd0a9267095","revisions":{"9d1ee1236119e5ecae3e6f43fe4c991ce940ebbe":{"kind":"REWORK","_number":1,"created":"2020-02-12 08:52:33.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/38/707338/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/707338/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/707338/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/707338/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/707338/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/707338/1"}}},"commit":{"parents":[{"commit":"006eb8e5efb75567840a2925e074595ca4548953","subject":"Docs: show other versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/006eb8e5efb75567840a2925e074595ca4548953"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2020-02-12 08:51:02.000000000","tz":60},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2020-02-12 08:52:31.000000000","tz":60},"subject":"WIP Use StorageSIG repos for Ceph in CentOS8","message":"WIP Use StorageSIG repos for Ceph in CentOS8\n\nStorage SIG has built Ceph Nautilus and Ganesha for CentOS8 in CentOS\nBuild System.\n\nLet\u0027s switch to use them in kolla.\n\nChange-Id: Id37dca84c4eb918aaf2d3c036ef5387fe75988dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9d1ee1236119e5ecae3e6f43fe4c991ce940ebbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9d1ee1236119e5ecae3e6f43fe4c991ce940ebbe"}]},"branch":"refs/heads/master"},"7c4ee0f4d7d62125554e0f488550b42f58d8c0cd":{"kind":"REWORK","_number":2,"created":"2020-02-12 09:43:52.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/38/707338/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/707338/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/707338/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/707338/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/707338/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/707338/2"}}},"commit":{"parents":[{"commit":"006eb8e5efb75567840a2925e074595ca4548953","subject":"Docs: show other versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/006eb8e5efb75567840a2925e074595ca4548953"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2020-02-12 08:51:02.000000000","tz":60},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2020-02-12 09:10:09.000000000","tz":60},"subject":"WIP Use StorageSIG repos for Ceph in CentOS8","message":"WIP Use StorageSIG repos for Ceph in CentOS8\n\nStorage SIG has built Ceph Nautilus and Ganesha for CentOS8 in CentOS\nBuild System.\n\nLet\u0027s switch to use them in kolla.\n\nChange-Id: Id37dca84c4eb918aaf2d3c036ef5387fe75988dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7c4ee0f4d7d62125554e0f488550b42f58d8c0cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7c4ee0f4d7d62125554e0f488550b42f58d8c0cd"}]},"branch":"refs/heads/master"},"9908f074be6e0725654f7afbdbb860539a4058bd":{"kind":"REWORK","_number":3,"created":"2020-02-13 11:33:33.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/38/707338/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/707338/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/707338/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/707338/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/707338/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/707338/3"}}},"commit":{"parents":[{"commit":"006eb8e5efb75567840a2925e074595ca4548953","subject":"Docs: show other versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/006eb8e5efb75567840a2925e074595ca4548953"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2020-02-12 08:51:02.000000000","tz":60},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2020-02-13 11:33:31.000000000","tz":60},"subject":"WIP Use StorageSIG repos for Ceph in CentOS8","message":"WIP Use StorageSIG repos for Ceph in CentOS8\n\nStorage SIG has built Ceph Nautilus and Ganesha for CentOS8 in CentOS\nBuild System.\n\nLet\u0027s switch to use them in kolla.\n\nChange-Id: Id37dca84c4eb918aaf2d3c036ef5387fe75988dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9908f074be6e0725654f7afbdbb860539a4058bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9908f074be6e0725654f7afbdbb860539a4058bd"}]},"branch":"refs/heads/master"},"006118221a254730a4214bb12c25a1ff809d96e7":{"kind":"REWORK","_number":4,"created":"2020-02-20 15:10:49.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/38/707338/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/707338/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/707338/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/707338/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/707338/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/707338/4"}}},"commit":{"parents":[{"commit":"006eb8e5efb75567840a2925e074595ca4548953","subject":"Docs: show other versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/006eb8e5efb75567840a2925e074595ca4548953"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2020-02-12 08:51:02.000000000","tz":60},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2020-02-20 15:10:47.000000000","tz":60},"subject":"Use StorageSIG repos for Ceph in CentOS8","message":"Use StorageSIG repos for Ceph in CentOS8\n\nStorage SIG has built Ceph Nautilus and Ganesha for CentOS8 in CentOS\nBuild System.\n\nLet\u0027s switch to use them in kolla.\n\nChange-Id: Id37dca84c4eb918aaf2d3c036ef5387fe75988dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/006118221a254730a4214bb12c25a1ff809d96e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/006118221a254730a4214bb12c25a1ff809d96e7"}]},"branch":"refs/heads/master"},"d8f071b24bffe91f1c5de7744d30ab37b984b9d3":{"kind":"REWORK","_number":5,"created":"2020-02-24 15:28:47.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/38/707338/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/707338/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/707338/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/707338/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/707338/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/707338/5"}}},"commit":{"parents":[{"commit":"006eb8e5efb75567840a2925e074595ca4548953","subject":"Docs: show other versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/006eb8e5efb75567840a2925e074595ca4548953"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2020-02-12 08:51:02.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-24 15:26:35.000000000","tz":0},"subject":"Use StorageSIG repos for Ceph in CentOS8","message":"Use StorageSIG repos for Ceph in CentOS8\n\nStorage SIG has built Ceph Nautilus and Ganesha for CentOS8 in CentOS\nBuild System.\n\nLet\u0027s switch to use them in kolla.\n\nChange-Id: Id37dca84c4eb918aaf2d3c036ef5387fe75988dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d8f071b24bffe91f1c5de7744d30ab37b984b9d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d8f071b24bffe91f1c5de7744d30ab37b984b9d3"}]},"branch":"refs/heads/master"},"76f6fe79705cb745afcff5be1b152ac2a8d88527":{"kind":"REWORK","_number":6,"created":"2020-02-26 10:06:45.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/38/707338/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/707338/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/707338/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/707338/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/707338/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/707338/6"}}},"commit":{"parents":[{"commit":"006eb8e5efb75567840a2925e074595ca4548953","subject":"Docs: show other versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/006eb8e5efb75567840a2925e074595ca4548953"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2020-02-12 08:51:02.000000000","tz":60},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2020-02-26 10:05:57.000000000","tz":60},"subject":"Use StorageSIG repos for Ceph in CentOS8","message":"Use StorageSIG repos for Ceph in CentOS8\n\nStorage SIG has built Ceph Nautilus and Ganesha for CentOS8 in CentOS\nBuild System.\n\nLet\u0027s switch to use them in kolla.\n\nChange-Id: Id37dca84c4eb918aaf2d3c036ef5387fe75988dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/76f6fe79705cb745afcff5be1b152ac2a8d88527"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/76f6fe79705cb745afcff5be1b152ac2a8d88527"}]},"branch":"refs/heads/master"},"0c6312968214e403b5a408d69df17bd0a9267095":{"kind":"REWORK","_number":7,"created":"2020-03-03 11:02:12.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/38/707338/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/707338/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/707338/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/707338/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/707338/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/707338/7"}}},"commit":{"parents":[{"commit":"96d4e1ad3afec83de6e019a051686e05c6611c25","subject":"Merge \"Add Apache httpd setup to heat-api-cfn\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/96d4e1ad3afec83de6e019a051686e05c6611c25"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2020-02-12 08:51:02.000000000","tz":60},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-03-03 11:01:53.000000000","tz":330},"subject":"Use StorageSIG repos for Ceph in CentOS8","message":"Use StorageSIG repos for Ceph in CentOS8\n\nStorage SIG has built Ceph Nautilus and Ganesha for CentOS8 in CentOS\nBuild System.\n\nLet\u0027s switch to use them in kolla.\n\nChange-Id: Id37dca84c4eb918aaf2d3c036ef5387fe75988dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0c6312968214e403b5a408d69df17bd0a9267095"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0c6312968214e403b5a408d69df17bd0a9267095"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
