)]}'
{"id":"openstack%2Fkolla~707363","triplet_id":"openstack%2Fkolla~master~I75249055989309d0afdfc29c894beb6b1fcd1994","project":"openstack/kolla","branch":"master","topic":"virtualenv-temporary-fix","hashtags":[],"change_id":"I75249055989309d0afdfc29c894beb6b1fcd1994","subject":"Upgrade virtualenv in pre","status":"MERGED","created":"2020-02-12 11:10:15.000000000","updated":"2020-02-13 11:32:26.000000000","submitted":"2020-02-12 17:23:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"707363-1581528227108-4f5f8f67","meta_rev_id":"8897c9922faedb8b8f93c3cf9d4ae78cf27c10ed","_number":707363,"virtual_id_number":707363,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-12 17:23:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2020-02-12 14:18:49.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-02-12 15:09:08.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2020-02-12 15:09:08.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-12 14:18:49.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-02-12 15:09:08.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-02-12 17:23:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8fa89b222895a144a70ff74a13ab710380789d4f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-02-12 11:10:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"447b16b2576c4aadc0dd66b92d98f29f907a99b3","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-02-12 11:20:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d8403d83dc04163746d642b2df58dc0a5ff20ef6","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-02-12 11:37:26.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"98609716444c8929a6f424b622a514c73c1341e3","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-02-12 11:50:35.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"00e74ac9ffe1037a327d205e3da05690f94faf92","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-02-12 11:54:38.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"732819285edf7f31f2ec265411344221211c658f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-02-12 11:59:28.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"94a92fff638c25fc448b7cfed310e65780efd273","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-02-12 12:49:03.000000000","message":"Uploaded patch set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"769e4cd975cb016f9434d4c7e2216414c7d812b1","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-02-12 14:18:49.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"cbdb5c23ee72ffa72d63b364ca0244df07504730","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 14:38:14.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/7a9955c366ae4a83ad11a765c6a9ea22 : SUCCESS in 44m 42s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/8fe0a9aa9869421f8e11b3e81dd38679 : SUCCESS in 52m 12s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/f6e3403d7d7d4f10a33587e076760c82 : SUCCESS in 46m 21s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/fa11ce039fc54a56b08cdf28c04e7c6a : SUCCESS in 42m 32s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/5906e81e580f434fbdf569cf403d44b2 : SUCCESS in 1h 29m 35s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/835dcfd867454953b14d3f2e3e1cb05d : SUCCESS in 1h 21m 18s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/c0604897dc6a425f8a63c34531f75f0f : SUCCESS in 1h 26m 43s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/153ff485c7084bdfad539a946d06c11a : SUCCESS in 1h 24m 16s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/c0dc8f3e06f24f858114ae17f299ac36 : SUCCESS in 1h 44m 07s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/259167a8c16c4e10ac3589f0712f4350 : SUCCESS in 21m 54s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/3b0a60b9ce81401d967ab66c7025f386 : SUCCESS in 20m 38s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ccea21f934db4afa81a1ac39fd67ab50 : SUCCESS in 58m 04s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/95f27522b0fa4f5d8a38a4e9342f6161 : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cdb49f9c07884f03af313c87ddeda31b : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb209f3547b64272b7cf9defc7c70134 : SUCCESS in 4m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fe0ed156c3cc46b9b4cd7dd596a35cd5 : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/521e2101c57e42b7ae0761569f53979d : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05820e30ce524548bf804529c442371b : SUCCESS in 7m 30s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5024ae2885984cc991ff413b893313ac : SUCCESS in 24m 04s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bf2b658da7254e04b50021b8a141514d : SUCCESS in 1h 04m 47s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e87ae83bcc7944d792df7f4f5e61ba34 : SUCCESS in 59m 28s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7b18d4ad46214104b1b005b51a1f542f : SUCCESS in 1h 07m 52s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e8661a3b2f29454f89e44973c9db4624 : SUCCESS in 1h 38m 59s","accounts_in_message":[],"_revision_number":6},{"id":"c05d630f532f8ae7d0fdfd0fd940cc68187c979e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-12 15:09:08.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"55926615dd01cab58ba96ddaa7ad517d9b427290","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 15:30:27.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"96bc5b90a1e15c470b43ffcaa2e46733fe7da4cc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 15:43:38.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d857ac31641e4cbd84e34169fef74c3d : RETRY_LIMIT in 20m 40s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"1a344857e10a6f4b561d3a3a9460723053a33013","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-12 15:47:19.000000000","message":"Topic set to virtualenv-temporary-fix","accounts_in_message":[],"_revision_number":6},{"id":"9fd58c145721baa3970fee928dc322c0ea489706","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 17:23:46.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/07e9d1eb64ec4109a255fddbf4f36fb7 : SUCCESS in 48m 14s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/6f2200d90a1b4731988d2e1ae015b64c : SUCCESS in 30m 42s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/4287e347f9d840bfb187b6a061c71c05 : SUCCESS in 1h 35m 05s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/72b9cdd9bb834466a312f3e2d95f3120 : SUCCESS in 1h 12m 25s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/25dd48bb3f6a4f5fa72715a43c81d04d : SUCCESS in 1h 44m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fa900e2d6aa245a49f6046394c008d0f : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0db89f093e8146ebaf0656f70f3d973b : SUCCESS in 6m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ed108b50fb65463b80fe94e64d9a3db1 : SUCCESS in 8m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3915aaf564794ea88220dcec77922e59 : SUCCESS in 6m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29130bb12b774b9a97f85056c432ed3a : SUCCESS in 9m 23s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d8400b65181b42a89f16f85f92277c61 : SUCCESS in 38m 27s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b71c70ccaa924ff6a13e9f7aa1446978 : SUCCESS in 1h 02m 28s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/462e3009fc074a389268ac8b4b4f7efd : SUCCESS in 1h 17m 30s","accounts_in_message":[],"_revision_number":6},{"id":"6b3aed69d880838ae881d96998702f7a7ef01c4c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 17:23:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"cdccdcb29f788a8254a6c6127d58910ab8c72898","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 17:26:18.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef36c0fd65c34155aa802351956683a1 : SUCCESS in 1m 35s","accounts_in_message":[],"_revision_number":6},{"id":"8897c9922faedb8b8f93c3cf9d4ae78cf27c10ed","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-13 11:32:26.000000000","message":"Created a revert of this change as Ibf2848db954862f917a23eb0aa1594ccc4e7ef67","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"063a46a7de4471504535499ca2a972a79e4ce5e5","revisions":{"9d443e858cae499bac52714817b7c5c23b30cdc2":{"kind":"REWORK","_number":1,"created":"2020-02-12 11:10:15.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/63/707363/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/63/707363/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/707363/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/707363/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/707363/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/63/707363/1"}}},"commit":{"parents":[{"commit":"006eb8e5efb75567840a2925e074595ca4548953","subject":"Docs: show other versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/006eb8e5efb75567840a2925e074595ca4548953"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-12 11:09:38.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-12 11:10:13.000000000","tz":60},"subject":"DNM: Downgrade virtualenv in pre","message":"DNM: Downgrade virtualenv in pre\n\nChange-Id: I75249055989309d0afdfc29c894beb6b1fcd1994\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9d443e858cae499bac52714817b7c5c23b30cdc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9d443e858cae499bac52714817b7c5c23b30cdc2"}]},"branch":"refs/heads/master"},"384925e54778c77e96b85a0a87c8033c1d40fb07":{"kind":"REWORK","_number":2,"created":"2020-02-12 11:20:18.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/63/707363/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/63/707363/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/707363/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/707363/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/707363/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/63/707363/2"}}},"commit":{"parents":[{"commit":"006eb8e5efb75567840a2925e074595ca4548953","subject":"Docs: show other versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/006eb8e5efb75567840a2925e074595ca4548953"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-12 11:09:38.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-12 11:20:07.000000000","tz":60},"subject":"DNM: Downgrade virtualenv in pre","message":"DNM: Downgrade virtualenv in pre\n\nChange-Id: I75249055989309d0afdfc29c894beb6b1fcd1994\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/384925e54778c77e96b85a0a87c8033c1d40fb07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/384925e54778c77e96b85a0a87c8033c1d40fb07"}]},"branch":"refs/heads/master"},"e94d9cdb984518067ec83200a10ce690b66a477c":{"kind":"REWORK","_number":3,"created":"2020-02-12 11:37:26.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/63/707363/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/63/707363/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/707363/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/707363/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/707363/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/63/707363/3"}}},"commit":{"parents":[{"commit":"006eb8e5efb75567840a2925e074595ca4548953","subject":"Docs: show other versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/006eb8e5efb75567840a2925e074595ca4548953"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-12 11:09:38.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-12 11:37:06.000000000","tz":60},"subject":"DNM: Upgrade virtualenv in pre","message":"DNM: Upgrade virtualenv in pre\n\nChange-Id: I75249055989309d0afdfc29c894beb6b1fcd1994\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e94d9cdb984518067ec83200a10ce690b66a477c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e94d9cdb984518067ec83200a10ce690b66a477c"}]},"branch":"refs/heads/master"},"530ca0608ff60a42cfd0c01dd22030cdf85d2dca":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-02-12 11:50:35.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/63/707363/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/63/707363/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/707363/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/707363/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/707363/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/63/707363/4"}}},"commit":{"parents":[{"commit":"006eb8e5efb75567840a2925e074595ca4548953","subject":"Docs: show other versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/006eb8e5efb75567840a2925e074595ca4548953"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-12 11:09:38.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-12 11:48:40.000000000","tz":60},"subject":"Upgrade virtualenv in pre","message":"Upgrade virtualenv in pre\n\nSince virtualenv 20.0 (amongst other changes) six version \u003e1.12.0 is required.\nThis change adds upgrade of virtualenv and six in pre - to be reverted once\ninfra CentOS images are sorted out.\n\nChange-Id: I75249055989309d0afdfc29c894beb6b1fcd1994\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/530ca0608ff60a42cfd0c01dd22030cdf85d2dca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/530ca0608ff60a42cfd0c01dd22030cdf85d2dca"}]},"branch":"refs/heads/master"},"adcfde0780cb11ff38b7d157852a3bf745ea918b":{"kind":"REWORK","_number":5,"created":"2020-02-12 11:59:28.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/63/707363/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/63/707363/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/707363/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/707363/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/707363/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/63/707363/5"}}},"commit":{"parents":[{"commit":"006eb8e5efb75567840a2925e074595ca4548953","subject":"Docs: show other versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/006eb8e5efb75567840a2925e074595ca4548953"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-12 11:09:38.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-12 11:59:17.000000000","tz":60},"subject":"Upgrade virtualenv in pre","message":"Upgrade virtualenv in pre\n\nSince virtualenv 20.0 (amongst other changes) six version \u003e1.12.0 is required.\nThis change adds upgrade of virtualenv and six in pre - to be reverted once\ninfra CentOS images are sorted out.\n\nChange-Id: I75249055989309d0afdfc29c894beb6b1fcd1994\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/adcfde0780cb11ff38b7d157852a3bf745ea918b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/adcfde0780cb11ff38b7d157852a3bf745ea918b"}]},"branch":"refs/heads/master"},"063a46a7de4471504535499ca2a972a79e4ce5e5":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2020-02-12 12:49:03.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/63/707363/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/63/707363/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/707363/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/707363/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/707363/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/63/707363/6"}}},"commit":{"parents":[{"commit":"006eb8e5efb75567840a2925e074595ca4548953","subject":"Docs: show other versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/006eb8e5efb75567840a2925e074595ca4548953"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-12 11:09:38.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-12 12:48:46.000000000","tz":60},"subject":"Upgrade virtualenv in pre","message":"Upgrade virtualenv in pre\n\nSince virtualenv 20.0 (amongst other changes) six version \u003e1.12.0 is required.\nThis change adds upgrade of virtualenv and six in pre - to be reverted once\ninfra CentOS images are sorted out.\n\nDepends-On: https://review.opendev.org/707377\nChange-Id: I75249055989309d0afdfc29c894beb6b1fcd1994\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/063a46a7de4471504535499ca2a972a79e4ce5e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/063a46a7de4471504535499ca2a972a79e4ce5e5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
