)]}'
{"id":"openstack%2Fkolla~707430","triplet_id":"openstack%2Fkolla~stable%2Fstein~I0ac41463bd6ca1eb363e1c5cdc29f7222ec382b1","project":"openstack/kolla","branch":"stable/stein","topic":"bump-gnocchi-stable/stein","hashtags":[],"change_id":"I0ac41463bd6ca1eb363e1c5cdc29f7222ec382b1","subject":"Bump Gnocchi to 4.3.4","status":"MERGED","created":"2020-02-12 16:55:59.000000000","updated":"2020-02-12 19:57:21.000000000","submitted":"2020-02-12 19:55:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"707430-1581537317602-b4850ae3","meta_rev_id":"7d88a910704f0cc1c5f8e212716eb0d200275c6f","_number":707430,"virtual_id_number":707430,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-12 19:55:17.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-02-12 17:40:52.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-02-12 17:23:47.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-02-12 17:50:47.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-12 17:40:52.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-02-12 17:50:47.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-02-12 19:55:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f31ced4a6cc2a49ed7cfbdbd93c46311c395df21","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-02-12 16:55:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a5cd74682eb51528249756b81a18628f24ec2a6f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-12 17:06:26.000000000","message":"Patch Set 1: Code-Review-1\n\nplease add reno, this is important news for operators","accounts_in_message":[],"_revision_number":1},{"id":"d4fa111b9ce553fc14d9aeb79e944219262a50da","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-12 17:06:37.000000000","message":"Patch Set 1:\n\n(change itself +2)","accounts_in_message":[],"_revision_number":1},{"id":"a566351baf6f3c2bf4e1cc9c7d9fd7d4cc9786ef","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-02-12 17:20:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a4214e7bdf884865eab18d084266eb97d478aac7","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-12 17:23:38.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"455535bc4be34e56eb2bb39ed7e0648d16acf2ea","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-12 17:23:47.000000000","message":"Patch Set 3: Code-Review+2\n\nchanged that to fixes, many thanks!","accounts_in_message":[],"_revision_number":3},{"id":"909b3e9982d6e29e36f1bc6a74a35a4b302d1988","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-12 17:40:52.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"3aa06629767e313f8ed9147878f1862a1afd9367","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-12 17:50:47.000000000","message":"Patch Set 3: Workflow+1\n\nlooks sane to w+1 in advance","accounts_in_message":[],"_revision_number":3},{"id":"1b7c9de060ea6990a53352a415a95a134646c0db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 18:52:16.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/0525e28da9104858aa279cdb0969e628 : SUCCESS in 41m 22s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/6dd04696b30047ddbc01fd29283cf62e : SUCCESS in 46m 13s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/79b97a9596a24b1bb1b4c1e76f34b572 : SUCCESS in 1h 11m 27s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/44b20d83427144ad8a117208f77de626 : SUCCESS in 1h 06m 04s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/1f0b117282d943e4862bb9722e569c8f : SUCCESS in 1h 27m 05s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/f0a6c2f903db42898b050714e79e0832 : FAILURE in 12m 17s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/9632295a243540f2a35f49a5a68dc0ff : FAILURE in 57m 07s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/9f01e094839d4a62b25b03bb466b6b7c : SUCCESS in 47m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7e59b462ccd04aa7b59a455ffa251b07 : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33e52752993a4ea7a5b7b7c38dc5f162 : SUCCESS in 5m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0a6aaa9d7bea4513b8a1f98b4dcf45ee : SUCCESS in 8m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dffefed09ad9409fa6c010887f97ea29 : SUCCESS in 7m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/df237adb926d4728b65b83721170e943 : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b92007aba3a47dba463aac3bef85515 : SUCCESS in 7m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/81c7ad5d67bc4b89aed210ceb3fceb80 : SUCCESS in 5m 59s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/233ad99bf8884eba9ca8cbca9427f757 : SUCCESS in 26m 14s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/657efc729a4c4f8ba9a187ee805c647a : SUCCESS in 28m 15s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7a8e7220a2fa4a8fb170d2c5da20494d : SUCCESS in 54m 35s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0a693ffa70c24c68aa7a4f8ce0c8a353 : FAILURE in 37m 18s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/dc7e9c6599724817b1fd93c0b135b5aa : SUCCESS in 1h 02m 08s","accounts_in_message":[],"_revision_number":3},{"id":"5572d8f188df144485df1ee94f065960fee675ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 18:52:28.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e6fbcf566b7b0acd8be541a8b5e214913af3702e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 19:55:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"e7168689ee1830d225aeb9edf8a7a102dd2416b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 19:55:17.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/f268c2e4beac48b4a1f022e93663dbac : SUCCESS in 54m 29s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/0735d98804544f1f923e91e07e6b160d : SUCCESS in 57m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d573deeee6f64f888b663fc81a986db3 : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/002ee623ceb6481cbe9d1860c92c7dc6 : SUCCESS in 5m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c164c6a2116d4b6f94898e7981eef995 : SUCCESS in 7m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/962673b7ae7641efbbda09a4f5e17cef : SUCCESS in 7m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/73b7306c92b14bb984cd0d17eca2226f : SUCCESS in 6m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f015c1ba79ba438a87eef22dc9f071d4 : SUCCESS in 5m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8e14774c94b7457bb7cf77a00af3f2f8 : SUCCESS in 7m 41s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/016c85f7784e4d62bcd1506772d733fc : SUCCESS in 31m 09s","accounts_in_message":[],"_revision_number":3},{"id":"4bcbba2083c46faa28c18dbaf1691329b9e2a411","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 19:57:21.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46fbab691bbe4c71a27e488e533a6264 : SUCCESS in 1m 33s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d3b768cf7bb94bc482aa75ba6ff4d9db : SUCCESS in 1m 39s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"cbfded882cb951c95d3832aae0b1c38711e4a654","revisions":{"535877b5a718c4f5da7b74c7168b5d663077602a":{"kind":"REWORK","_number":1,"created":"2020-02-12 16:55:59.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/30/707430/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/30/707430/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/707430/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/707430/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/707430/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/30/707430/1"}}},"commit":{"parents":[{"commit":"1b3d1254c463fbf0700562b20a4e832ac7ab7380","subject":"[stein] Bump versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1b3d1254c463fbf0700562b20a4e832ac7ab7380"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-02-12 16:55:22.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-02-12 16:55:22.000000000","tz":60},"subject":"Bump Gnocchi to 4.3.4","message":"Bump Gnocchi to 4.3.4\n\nIt includes several bug fixes, in particular one for an issue when\ncoordination is disabled [1].\n\n[1] https://github.com/gnocchixyz/gnocchi/issues/996\n\nChange-Id: I0ac41463bd6ca1eb363e1c5cdc29f7222ec382b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/535877b5a718c4f5da7b74c7168b5d663077602a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/535877b5a718c4f5da7b74c7168b5d663077602a"}]},"branch":"refs/heads/stable/stein"},"56df064966a9b7b5c164cb04eb19f4e9dc3d8aa7":{"kind":"REWORK","_number":2,"created":"2020-02-12 17:20:38.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/30/707430/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/30/707430/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/707430/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/707430/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/707430/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/30/707430/2"}}},"commit":{"parents":[{"commit":"1b3d1254c463fbf0700562b20a4e832ac7ab7380","subject":"[stein] Bump versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1b3d1254c463fbf0700562b20a4e832ac7ab7380"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-02-12 17:20:15.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-02-12 17:20:15.000000000","tz":60},"subject":"Bump Gnocchi to 4.3.4","message":"Bump Gnocchi to 4.3.4\n\nIt includes several bug fixes, in particular one for an issue when\ncoordination is disabled [1].\n\n[1] https://github.com/gnocchixyz/gnocchi/issues/996\n\nChange-Id: I0ac41463bd6ca1eb363e1c5cdc29f7222ec382b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/56df064966a9b7b5c164cb04eb19f4e9dc3d8aa7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/56df064966a9b7b5c164cb04eb19f4e9dc3d8aa7"}]},"branch":"refs/heads/stable/stein"},"cbfded882cb951c95d3832aae0b1c38711e4a654":{"kind":"REWORK","_number":3,"created":"2020-02-12 17:23:38.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/30/707430/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/30/707430/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/707430/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/707430/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/707430/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/30/707430/3"}}},"commit":{"parents":[{"commit":"1b3d1254c463fbf0700562b20a4e832ac7ab7380","subject":"[stein] Bump versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1b3d1254c463fbf0700562b20a4e832ac7ab7380"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-02-12 17:20:15.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-02-12 17:23:34.000000000","tz":0},"subject":"Bump Gnocchi to 4.3.4","message":"Bump Gnocchi to 4.3.4\n\nIt includes several bug fixes, in particular one for an issue when\ncoordination is disabled [1].\n\n[1] https://github.com/gnocchixyz/gnocchi/issues/996\n\nChange-Id: I0ac41463bd6ca1eb363e1c5cdc29f7222ec382b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cbfded882cb951c95d3832aae0b1c38711e4a654"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cbfded882cb951c95d3832aae0b1c38711e4a654"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
