)]}'
{"id":"openstack%2Fkolla~707599","triplet_id":"openstack%2Fkolla~master~I2f66cdf18f2a845d97ad3512a35c7de040ace444","project":"openstack/kolla","branch":"master","topic":"wip/infra-images","hashtags":[],"change_id":"I2f66cdf18f2a845d97ad3512a35c7de040ace444","subject":"introduce non-infra-base image","status":"ABANDONED","created":"2020-02-13 10:03:22.000000000","updated":"2020-07-02 14:24:51.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"36662f5b753144129d443a035c322b1c870f4c96","_number":707599,"virtual_id_number":707599,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-06-30 17:09:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-06-30 16:05:44.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-24 15:23:31.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-02-24 15:23:33.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-02-24 15:23:36.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"},{"updated":"2020-02-24 15:23:39.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-06-30 16:05:44.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-30 17:09:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"aec7101282dec964e892dc4f1ace726e8bc50a60","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-02-13 10:03:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9d310d79e9334ddcff14ef690359501a49961243","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-02-13 10:06:19.000000000","message":"Patch Set 1:\n\nEasiest, minimal form of doing \u0027infra\u0027 images.\n\nMore a test rather than something to merge.\n\nAll infrastructure images built for binary/source will be the same. Building them on same machine \u003d\u003d using cache for second build.","accounts_in_message":[],"_revision_number":1},{"id":"a89ee30103837b11b3cac6c547c0cdd38eb155d7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 11:11:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bf9196344c2e6138cf2db68d9fece5287b4d8e88","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-13 11:12:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/824db78d5b5046248a679728c567e1db : SUCCESS in 25m 52s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/8f53130d53bf42e8bd26167b4045028c : SUCCESS in 39m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"941d08f8cc5a07454376363f7c38e71ac9331e0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 12:24:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/be0fe1e6eef64033901a5ec86c177c7a : SUCCESS in 32m 24s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/8856ea4533ec4616bf04c9b3e52a682e : SUCCESS in 45m 09s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/e84c45fa0180446380f5ae1ad982a209 : SUCCESS in 30m 22s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/ee0a826444a04233b676a9305011ce14 : SUCCESS in 32m 55s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/0f781b59f6964ca4b40312c15ffd1175 : SUCCESS in 1h 20m 31s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/4a27cde7e818465a9cb4bc2d3ae5f938 : SUCCESS in 1h 14m 08s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/37e74ef3571e46b1ab4b920e30be3792 : SUCCESS in 1h 24m 30s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/85b282380f454abab4e3a31a34e89550 : SUCCESS in 1h 08m 34s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/fc6b98e09e91477a9e4c45942551bab5 : SUCCESS in 1h 31m 11s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/fe2f718d787d409ea8f13c42671ff837 : SUCCESS in 59m 44s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/b83638a90dd8488b8db20dbf841eb01d : SUCCESS in 50m 20s (non-voting)\n- kolla-ansible-centos-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/661e8726723741ac84e87fbecd87ca3e : FAILURE in 39m 14s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/f815b95299154010bf77acea6f82a924 : FAILURE in 49m 07s (non-voting)\n- kolla-ansible-centos-source-zun https://zuul.opendev.org/t/openstack/build/dde44477c5fe4e2fb00f93d7b185cb5d : TIMED_OUT in 2h 03m 17s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/c0fd9dfb12ae4d8ea80e5015278e28cf : SUCCESS in 1h 50m 34s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/1ee45933730e43059e729ed6b6c4be83 : TIMED_OUT in 2h 03m 55s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/f60ad66470694ba984dce09cd683a7c9 : SUCCESS in 1h 42m 22s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/121cee6ed6fa4f518cd1cd64b395d554 : SUCCESS in 1h 24m 38s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/423c0f1879e34e968ff53daf1b30ea01 : SUCCESS in 1h 22m 09s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/84d02bdc850949fdb19f78c608085f1a : SUCCESS in 28m 29s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/191f52b85f3a4f32aca313ebacd5f2e9 : SUCCESS in 24m 34s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ce942c8495c04506bb9e390153954a74 : SUCCESS in 1h 11m 56s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aa24fe9fff5248309cf7d86f789e4b74 : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c651b52c8a814c4db787a3c45544073a : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0338e9b6400c48fab70045ccc9a51539 : SUCCESS in 5m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/facbbdba723c462cb430b49bc974edac : SUCCESS in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7589470d9bca47f88efe33d476808d86 : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f0b6300ba39494896af9d7366214ad1 : SUCCESS in 11m 12s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/38a879e174a940569cd5b69efe487459 : SUCCESS in 39m 34s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3a4c8e9423ac4584b3fa6e1b2245ebb8 : SUCCESS in 29m 04s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/eed1e481adf04c32b0c9ff86a63975f8 : SUCCESS in 57m 29s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/426616f2c10142ed92a6e005330ef9cb : SUCCESS in 56m 43s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2135685b9ad94c0b887bf2e690ea5a36 : SUCCESS in 1h 18m 42s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/0f2a95d9d31545b7b9e4e019939e1c4c : FAILURE in 28m 20s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/9e4d08a28cde471b8abc76ade7fe5aff : FAILURE in 29m 48s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/f60bb15169a34d45afb0ea5d24f5a684 : SUCCESS in 1h 25m 38s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/eb7e3fe3a858467987fd80851b24ae79 : SUCCESS in 1h 06m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b55ac3092a40e8d28d46f2f23ceffafdb5a3f1e6","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-02-13 15:44:07.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"175a99bbbce4202a7c7e3fa0dd59ab68dbe3dd05","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-02-24 14:44:03.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"e948a7ad1aa57b0bab991e24285acfc7a07fca07","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-02-24 14:44:07.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"684dbfde6b63aeb21c7d1046af2877d202ea5943","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-24 16:06:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/71d89079fb564cd3a10362c4d48d0bf7 : SUCCESS in 29m 14s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/d8a9e44ff68f4569860adea86370a988 : FAILURE in 19m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0739fa87dd9af3b9ab4b6162136e4827d834b268","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 16:55:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/cea0bec7c6974eb78831a2ab7ac4fc37 : RETRY_LIMIT in 20m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3a57e12a0d1187d02f46be3e10529bb232ca4e68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 19:23:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/9b352a345e914264b6825867753f6ce0 : SUCCESS in 33m 25s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/c5780ac08c15481ab2e2d70e0ed93c83 : SUCCESS in 1h 03m 24s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/5f19e294c94a4a00a7341d10cc53644a : SUCCESS in 48m 46s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/de5a218d7e264fbf803abc6c840c3de5 : SUCCESS in 51m 48s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/2222c960954f45698a2b55cb2b0dc02a : SUCCESS in 1h 35m 21s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/e9af7d267a6c495d8c3c706db803d9c0 : SUCCESS in 1h 28m 55s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/c1037e97a4ff4363bb580967a51cc405 : SUCCESS in 1h 39m 22s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/6cd1e5c6ce7947ec9147a6cfe3e640aa : SUCCESS in 1h 14m 25s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/e8dc1967b56243028f9de5816949944f : SUCCESS in 1h 54m 28s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/5c9c38873c954b40b8134b3cb70d8f8e : SUCCESS in 45m 52s (non-voting)\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/2ce7062744d041828874c74e81f6c34f : SUCCESS in 1h 08m 00s (non-voting)\n- kolla-ansible-centos-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/057ec7472fa04339ba3827ceacb7abd4 : SUCCESS in 2h 14m 47s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/9d68952c3f2745a5bc1ec5a8be84177f : SUCCESS in 2h 27m 48s (non-voting)\n- kolla-ansible-centos-source-zun https://zuul.opendev.org/t/openstack/build/d6b7e2edebb4448da150b74c43901f92 : TIMED_OUT in 2h 11m 39s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/6d226a9f29854b5c9b9ea864896781b0 : SUCCESS in 2h 02m 35s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/a71e30e2a8e74839b6fc69c0b28d6e86 : FAILURE in 1h 48m 18s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/9ccfca60aa8c4cacb7cc3fadb5ea7ae4 : SUCCESS in 2h 11m 46s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/149931c4bc33427ba82165d41c282ce9 : SUCCESS in 1h 41m 18s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/85f8ae1416494adb9cefbb280f548aa4 : SUCCESS in 1h 21m 45s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/237ea0b42cc74783b83893f59cbe1ff7 : SUCCESS in 30m 40s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/48635753ef6b4865841e4dffc669129e : SUCCESS in 22m 08s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2427e0d7b8444cfe8acb3ec10a12d97d : SUCCESS in 1h 16m 29s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0cad2f9a899a4dc0985c62cd9e16f48a : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad685f55d48340378cf178f71445e52b : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b895cb18cff443179591e97e54fdd5c4 : SUCCESS in 5m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0545f509033e4e10b1b29e83deedc25b : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42c45f3975574e9b9a4f15fff8e586b4 : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/147532a1f9f1469987df8f5854276713 : SUCCESS in 8m 56s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2395a79a007b49f6b36828b5ba929795 : SUCCESS in 32m 46s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8de4fdec69b84594a972a3eaa6bb7788 : SUCCESS in 27m 53s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d2ad676992074b33b911ac339f1379e1 : SUCCESS in 1h 16m 01s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7a4779d8fda7406f80dd93254199a2dc : SUCCESS in 57m 09s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6ced64f497b7444a866625cac620b844 : SUCCESS in 1h 05m 50s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/6db76550f50e4d0bbcd2b95f63350786 : FAILURE in 34m 44s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/d27a8d14c732470ba4d8705a96193759 : FAILURE in 25m 30s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/4b932c2e74e8483b993cd807183bf0e6 : SUCCESS in 1h 32m 51s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/359e6d2b9b8943389f72ee00b5de4ef5 : SUCCESS in 1h 05m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"58188917fed322abda2b94ae972493156039a09c","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-03-09 15:44:51.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"74d94fac49880b0ee8c29e0150e053bab2830f5d","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-03-09 15:56:16.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"8935611404afdb00a18b2a2e526552b0814ff3f6","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-03-09 16:14:24.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"059623948ca6ec5ec3a1bac5eb200b6e5842f82a","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-03-09 16:15:15.000000000","message":"Topic changed from wip/non-infra-base to wip/infra-images","accounts_in_message":[],"_revision_number":5},{"id":"a4b1714c49c8da135ecc750b36f7715200364050","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 16:19:05.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ce29155e876b4af68c5c635cfd9ae810 : FAILURE in 3m 37s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"400d6014047b0cd254e3e6913210371176e125a3","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-03-09 16:44:59.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/445a0e452dee442cb47fc0c3652b4dca : SUCCESS in 28m 04s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/8c9dce7dc8e744aeb1cd8df02d01075d : FAILURE in 16m 11s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"63a850a5f47484517c2c4f3b74177eff641fccc4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 18:33:32.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/0ecb0bd205634f3c96bae4eadcce9109 : SUCCESS in 32m 02s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/68acdd44b5e6423a99c35ff0c2512d17 : SUCCESS in 32m 31s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/469bf2657bc54f1a93e1537917afe8bb : SUCCESS in 1h 22m 56s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/d0bafd9792044780be599e5d20e3ab7e : SUCCESS in 1h 13m 36s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/9057bb06280244b3b78dda47d96ca458 : SUCCESS in 1h 25m 33s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/fed452216238401e894be19a5982951b : FAILURE in 22m 22s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/b4a60dac18164ae994dbc7eb42270af9 : SUCCESS in 1h 57m 09s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/86462b97d33e44fe82b5779edac5566b : FAILURE in 1h 52m 14s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/cf4d91f6095941aaac870bb39f731d60 : SUCCESS in 1h 14m 36s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2c1baa61527744bfaafeb1d55ec0e864 : SUCCESS in 31m 44s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/4805b2b6f1c94c8cbc144b99bd5a1c7a : SUCCESS in 38m 02s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/6c0653bf78ee40cdb2661bfad6735577 : SUCCESS in 1h 25m 36s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f0b4535312e24607ad00c4d7f1cdeeaf : SUCCESS in 7m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30c164b16fbd492899783c762ac447f2 : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d0801b854df43da99a661ce4070c2fd : SUCCESS in 7m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9312c7a29a484158b8dee35f7c6f4f19 : SUCCESS in 7m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a2e6ddbea3614acaac27bdde290c32dd : SUCCESS in 8m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65e7b083eb564cac96e4e9b72e17e20b : SUCCESS in 6m 33s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0afc78d7cebb411b884b88093385cfa0 : SUCCESS in 27m 54s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2916f92b861e467aac8e0272980b6b65 : SUCCESS in 32m 22s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c85b16acbf864a77a3b2b1c31744869b : SUCCESS in 1h 10m 22s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/799055a9ca37464cb594013b0265bdc7 : SUCCESS in 57m 29s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d5f1bba7989740eb88a45f5578c682d3 : SUCCESS in 1h 09m 45s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/0ef5b278b5bd4b30a6157f67468e0f50 : FAILURE in 22m 51s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/d5e44cc0c7d5471cb550cad4d3069d7d : FAILURE in 31m 30s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/b996ed54c26e4e00bdf77925c717d922 : SUCCESS in 1h 49m 02s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/fe1118fa1b2f4ea982a7c16c30e08566 : SUCCESS in 1h 06m 34s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"db0a756b014703e8139cdf7d002810a2e9d58b6c","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-05-11 15:29:11.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"f0b3f961a82d19ba7bc9d822e073816060e6ab8e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-11 17:18:15.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/0b0c0b2bd9d840949fc0d2f84a0591fb : SUCCESS in 27m 16s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f1bb544c22e253110e889f2b710094fd6b0ab1a8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 17:29:27.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/e687be133eb3432bbef10dc913403690 : SUCCESS in 1h 01m 41s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6408a26139464ae2863533ec2a6f5fde : SUCCESS in 1h 02m 37s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7a1a1cf2341f4fec9d7467b25234fc24 : SUCCESS in 1h 54m 43s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/42ad6d4fbed7403cab0aace1ab47fdf0 : SUCCESS in 1h 04m 22s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"76707cd3b1d12504ca3d69775bd1fb8d3fc688e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 18:03:48.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/eb30c25bc9c840f6acaab3d1536ad394 : SUCCESS in 23m 44s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/79afc55e615f478b82d3a1007190a94a : SUCCESS in 31m 54s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/0c7e0385a29144fd8923dea5d6a1ffa2 : SUCCESS in 1h 21m 29s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/14a0d84bce3446cbbe0817c889ca8a37 : SUCCESS in 1h 23m 11s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/ae09fc8f4c254e88a7eaa938597ce9af : SUCCESS in 2h 02m 56s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/4a4fe26ccdac47e3b5fa9e22a4207586 : SUCCESS in 54m 24s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/dda49cf411b448c7825e9c3159e05a93 : SUCCESS in 35m 56s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/62de3fff1b1a4774887f836982c0fe3f : SUCCESS in 1h 55m 40s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/2cc4563fc37d47458cd0f9c279f15943 : SUCCESS in 2h 19m 49s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/560735472be74e0494b90b360216dc95 : TIMED_OUT in 2h 15m 57s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/6c5e878efb904661a51785dfa0d55f31 : SUCCESS in 1h 58m 36s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/1e408b2e63de4fa591e213b057952c32 : SUCCESS in 1h 32m 43s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/2fc6ea930d644271ac01cfac57c2a25b : SUCCESS in 1h 47m 27s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2dd77502778144d3a9cfcdc5e2ad184d : SUCCESS in 27m 42s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/d6c2d128ecef4c3e9a52eeedc0216ca9 : SUCCESS in 31m 35s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ad7bbbb2ec3d44a99fa21249fb5ee32a : SUCCESS in 1h 06m 40s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/595ca34607f1474987956c8258a77203 : SUCCESS in 6m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7154e48ec464b9c86c6862ff5c41d2a : SUCCESS in 7m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/662287366ac24407ad50159fb1747a56 : SUCCESS in 6m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/35267e9cae644c30bbf74a1fe3721a93 : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c69948e697849a79eb0620bd1232f95 : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43b06a5c51f642a18ad2de8d19470b11 : SUCCESS in 5m 43s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/43f955f1d9694d9eae39bb7bb600073c : SUCCESS in 26m 03s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/dc599c931d4041c9bd02ee2dc979cb53 : SUCCESS in 28m 41s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7e722635c9634a5ca31ba1f8cffb9fbc : SUCCESS in 1h 05m 28s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f36a8d09a5204d769c492ff76da90610 : SUCCESS in 1h 03m 53s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/76b8ba5c94f54f3f81fa8064bf2e184e : SUCCESS in 2h 04m 13s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/852bfcbb37834590a90399a1804123bd : SUCCESS in 1h 49m 58s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/b34e577037094df0bd51727d0a1089fa : SUCCESS in 1h 48m 06s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/4842861973824605a0da9cb6b7f7cd38 : SUCCESS in 1h 39m 12s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/535920995085487e873b7f5cd0f26780 : SUCCESS in 1h 29m 44s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/24879877a8fa4ec5b91b2e99e5929df7 : SUCCESS in 2h 14m 20s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"e3c2f95ad2ce20f42f67c99c4fd31243a82fec20","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-05-17 13:46:33.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"b6650fd401ca93c33ac5a761efec04fd2bea3b30","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-17 14:12:08.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/681dcad6d22a4bb98f7ded55f8516226 : FAILURE in 22m 18s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/eb49a9bac3024757b8ea82cda4366a6c : FAILURE in 19m 46s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a6a6e1d0aada4b6db053a33b6448ad05 : FAILURE in 23m 09s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/310ca4e8125a4b62a6ce929b5ebcfa79 : FAILURE in 21m 10s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"ed1439efe19a52c4d0c9321dda0c77f4be731261","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-17 14:58:34.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/4bfb950036664410be38f8b81d33e890 : SUCCESS in 1h 10m 59s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"33c4a384d1733d6582624563dc6369ae3bab4303","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-17 16:52:00.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/a31bd1671d844530aa26145e5780bf68 : SUCCESS in 1h 11m 05s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/f863c745a25240bc82b947572d4c89f0 : SUCCESS in 37m 03s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/969080185f1a4f83b5b7cc69742bb7d7 : SUCCESS in 1h 50m 48s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/dcf1812e20c64ed194caedb1ecaef853 : TIMED_OUT in 2h 04m 25s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/3eed7784cbbd459f94a5b66fb74709bc : SUCCESS in 1h 50m 25s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/65613417700e4727aa1c28941a953ccd : TIMED_OUT in 2h 06m 19s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/3c5c551d1a2240e1bb46bf8d6ced08fc : TIMED_OUT in 2h 06m 02s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/3623452c9b63449689647aa007ad4f33 : TIMED_OUT in 2h 33m 43s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/e710e2864afc4653a716aad2e88e6b5d : TIMED_OUT in 2h 34m 24s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/ce07d1995b744c0395b1cadb9a0fc78c : TIMED_OUT in 2h 06m 43s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/f5eb2e32005249e3831bf5792b31cbfd : SUCCESS in 1h 49m 27s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/3cd17d542ffa4067a31f23ec4d78ec74 : SUCCESS in 1h 17m 55s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/4d30ba35ef3a4c15995fc2b5e5347935 : TIMED_OUT in 2h 06m 32s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/718567c7c21c43f793e782f6af4414c0 : SUCCESS in 21m 11s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/c746fa9fb1a749dcbc0fbc9985663d8f : SUCCESS in 2h 58m 46s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/053beba1b136417bb13a00a490a9b738 : TIMED_OUT in 2h 05m 28s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/674eb30347304b3e8349e7547fca97e1 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84a8cd96748d4907a5a533d96c977f4f : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/efd938d9b9be4db692d4f8412f532d94 : SUCCESS in 5m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a0c8f7537aff4d68bd671c58b74b7207 : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bdf9873757424a3fa9322d7fa18a82b3 : SUCCESS in 8m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/deab19df9d89486f88625b5f5d6a84be : SUCCESS in 30m 02s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f4437d17b32845a1bf9b176c3a2b40e4 : TIMED_OUT in 3h 01m 40s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c8bd54f68d1e4e02b94442e7508df065 : TIMED_OUT in 3h 01m 32s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/07dccc7add6e4088874ab2eb0449642b : TIMED_OUT in 2h 06m 36s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/774dd8521e0c446db1f4aca5454dba67 : SUCCESS in 1h 10m 01s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ca48b4ed32ab4e5dad54dca4947bcb81 : SUCCESS in 1h 22m 05s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/02bb1d5c26a04323a67e913c9c9b1edd : SUCCESS in 1h 47m 15s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/65de05d4ac144ab4a1df7c0882ef9b02 : TIMED_OUT in 2h 34m 42s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/72e5564c4bdb49cf8fc3e24115590c10 : SUCCESS in 1h 35m 29s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/01d52f174b47438a9e4d0bce625c466d : SUCCESS in 1h 14m 35s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/5e29a9af647843c5860379c0008be9c7 : TIMED_OUT in 2h 07m 24s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"5fbff45ab45c144d9a12d2848550e39c45fe31ef","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-05-18 19:09:44.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"87d1010626e81132266b1907144ed0ff712cc0f8","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-18 20:28:28.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/1c4a034589cf4b48b89a5d2aace93533 : SUCCESS in 29m 47s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"90154c7602dbf562a7a3b381216d7e8d0362331b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 20:55:46.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/61d793844e6b4c04a1710a609c65e5f5 : FAILURE in 54m 46s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/47dec87555b747b1b3653a1020bc22be : SUCCESS in 57m 07s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a49f4986af6541d497862218418d6e20 : SUCCESS in 1h 45m 25s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/e8adb1f39c1b45ef8aab5d1b4110a0b2 : FAILURE in 57m 36s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"d8908c1b139ec504210377004df31fb9532f68d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 22:37:45.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/e16ad28bdeba44df86c96a327a518e76 : SUCCESS in 26m 45s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/096222913d4b4490ac6b43d76cb1e34a : SUCCESS in 37m 03s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/f51d74df482e43058895ba079dad46dd : SUCCESS in 1h 14m 30s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/7fdddc25530e425cb9bd2e1b29d0c82b : SUCCESS in 1h 25m 50s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/d339c640cb64416dbf010f9da438c5b6 : SUCCESS in 1h 20m 00s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/df333b1205c14fdbbc1ba3499e381184 : FAILURE in 43m 34s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/9bbaa9c61f6a4308a9ffb4b01e97907a : SUCCESS in 1h 06m 23s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/7c4fb05ee29048358be5b50e15eac352 : SUCCESS in 1h 56m 07s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/3f3005af6e934fafbfc4851e3b5a827f : SUCCESS in 2h 20m 59s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/a0e897c914e744d581f5b977080a58c4 : SUCCESS in 1h 51m 17s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/6f9320234e414abea77fc8b4cc7f0420 : SUCCESS in 1h 50m 41s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/63dee9f621ab4afb9069c92f3f1e2449 : SUCCESS in 1h 26m 18s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/94346bce43d5468ab16fa7a22a411006 : SUCCESS in 1h 44m 43s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e80a424db9df4731b13f8c6c5b081492 : SUCCESS in 25m 46s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/2130f246dc4f46c4a501498d44fb3e42 : SUCCESS in 27m 25s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b640744da5c44aad9dbd86c4065dcc29 : SUCCESS in 1h 03m 28s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/82e776043eaa4f2ba330f0cda3394f29 : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c59e077e3e84c5a9e5fffa59327117b : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a29ac898d235437e8dd35fbbe22185a0 : SUCCESS in 4m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/186e037f7d5a440da1b08309fc088480 : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cde07fa8dd284d19b1d049e97a7c841e : SUCCESS in 5m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18e026232cab4b1abd9514310eed1e57 : SUCCESS in 6m 17s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9cc8504ede2b45c9bd3efcc3fc105b63 : SUCCESS in 30m 38s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a62d328f747242ae885f306e76bcac70 : SUCCESS in 36m 37s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4f4441dc4ec34142a7d69c90ad3d5730 : SUCCESS in 1h 15m 30s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5d367cb3760e4189ac747e6dd281b4af : SUCCESS in 1h 00m 31s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/631ed1dd7f3648f9bbbd528f08d2c050 : SUCCESS in 1h 29m 48s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/ccbbe16a9d374705b0480809fd209183 : SUCCESS in 1h 44m 19s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/e3603631db094592b36b76e9368a31b4 : SUCCESS in 1h 51m 30s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/89d1018fc265417da07a45607de72ad9 : SUCCESS in 1h 37m 56s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/3a569d6c10e64be49dde9bc3ebed5b6b : SUCCESS in 1h 11m 02s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/9b1edf68a8094e6db3b4caa860a5392c : SUCCESS in 1h 57m 08s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"c175936e9643f1e6fb5548fb926280e59275cd31","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-05-25 11:15:39.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"5f9f2a9ee159f97f72301196672273e0157352c3","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-25 11:42:17.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/db5befb66fc2436a9dcb5affa22be20d : SUCCESS in 25m 45s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"ea8c20d3fb557ba1acc0575c5502e21c14a31b2f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-25 13:16:15.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/3d8dd3c253dc4a34a460ef0d259d2236 : SUCCESS in 57m 51s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f33466953202487485fa92f6fe76aa8b : SUCCESS in 57m 21s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2e84503e366a4c84a7d659bd351303d6 : SUCCESS in 1h 50m 35s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/3e40bc62ac4e43ed968d885ded0211d1 : FAILURE in 7m 55s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"64ca7c0a0a0aabc579001a1b9fa0aa4ae9dedff9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-25 13:26:11.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/7bb6f98dbf944637bffa1cddac023f80 : SUCCESS in 24m 44s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/5f68ddc3e6d6427385b5afa078ceb2cb : SUCCESS in 30m 23s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/bed7f61fccb24af49397edc0b0f3ca88 : SUCCESS in 1h 17m 57s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/45c23efa380147ca815ed73bced47703 : SUCCESS in 1h 32m 03s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/09599fb4583d46568232eb87d73857c4 : SUCCESS in 1h 29m 59s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/d4343d98ef8a4f8aa2f315715271981d : SUCCESS in 38m 41s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/b7f8e0e1aa17460ba50028e86aca416a : SUCCESS in 32m 07s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/80bd6ce7ad4a4430a47359adc51ce2e3 : SUCCESS in 2h 01m 10s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/b96c882b7de245dcba9b32d509ed92da : SUCCESS in 2h 08m 38s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/2c5978e11cba44ffb77c71af5e1afb59 : SUCCESS in 2h 07m 12s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/fd46d4a81de5453c9e2cf0908a67cc08 : SUCCESS in 1h 57m 47s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/36a65320a0a8490185b269cda938f26c : SUCCESS in 1h 29m 04s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/6ebc0053488c49b097c324dfe9e61c64 : SUCCESS in 1h 49m 21s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f6349dd2762f429db174dbc889500882 : SUCCESS in 21m 59s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/7434b23243de4a448a94c038530afc17 : SUCCESS in 17m 44s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/10549d69e0264729905e245034afe98d : SUCCESS in 1h 01m 34s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5182777f532a4b249db05c5166194091 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d22c3a51cd643bc9fa07ee179be97de : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3c92b20525974f848a313cce6d603e4c : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a9381e8b19b94a1d801251981df0c06d : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eee8ebb80f3146fababa615ba32c7954 : SUCCESS in 5m 35s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ac31e2dde48c483aa9650f174da127e8 : SUCCESS in 21m 21s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0d7e60c56f18492b8dea77cf7c1dac76 : SUCCESS in 24m 27s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a66d358e5f8441769131033305bc9193 : SUCCESS in 1h 04m 32s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/796d14152ad5432d88cfb21bc0d79bfc : SUCCESS in 57m 49s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8bc7a71992e74aeda7c7efd35ea399c5 : SUCCESS in 1h 11m 25s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/b22340e49e4542a98bd4582822bddf77 : SUCCESS in 1h 36m 43s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/0b721fee92d149c3a1492865d03ceb92 : SUCCESS in 1h 52m 30s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/3b21e9d4ab394404b3c6b04088184758 : SUCCESS in 1h 29m 29s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/c9df4aa653174b429b7386b6c219dc63 : SUCCESS in 1h 04m 23s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/676bfcbf8bab4fb2b4acea37c0a0aee9 : SUCCESS in 1h 23m 15s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"cc09d2b81f384645bbf1d7c82ab63cdc8ac7e0ab","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-06-25 10:37:20.000000000","message":"Patch Set 8:\n\nhttps://etherpad.opendev.org/p/Kolla-infra-images has more information about patch","accounts_in_message":[],"_revision_number":8},{"id":"20211af9aa62f321bee8d73f8d88b922c3d0c332","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-06-25 10:38:17.000000000","message":"Patch Set 9: Commit message was updated.","accounts_in_message":[],"_revision_number":9},{"id":"9da506ea24e3865cc2d07ba622b35e69bf03fb32","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-06-25 10:38:21.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"53bef5b9f98bceaca71e8a8aed00a1fe2ef41f5d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-25 11:08:23.000000000","message":"Patch Set 10:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/a10a30094912460b94dfa066719c8b19 : SUCCESS in 27m 23s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"1f722aa477a7dbc0cc4369e39aae36d83265f8ea","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 12:11:19.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/502f2ee8afed410d84446f726370cc15 : SUCCESS in 58m 05s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/71fdc4aac43949c2844f0200fd3545d8 : SUCCESS in 35m 40s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/de2fbfb4813844e183a8ddbc1ee5cb88 : SUCCESS in 1h 31m 07s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/0262cb20ba3f40beb3dc41ca60f4254c : SUCCESS in 36m 49s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"2c54694f8b11446fd356acd7bf303ba3536c7096","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 13:05:46.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/ada8cd6c432447628bbdda88b2cf60f3 : SUCCESS in 24m 41s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/42c9bdc99a5b49ea9fb33671582a3428 : SUCCESS in 29m 06s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/dbd708f176e249c99d7a3476b858efa9 : SUCCESS in 1h 13m 58s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/352fa471e1e6479e93343de4f94d5b4a : SUCCESS in 1h 40m 05s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/378ed6f6207b4821825c2c0baf642c4a : SUCCESS in 1h 17m 24s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/7d100d0513b840f9a876b92a8f92fc79 : SUCCESS in 53m 07s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/d966519f17964158a8aa870a49e0deeb : SUCCESS in 36m 36s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/13d1da5098ce41b1be881849ec4a655d : SUCCESS in 1h 56m 50s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/123cac9d10ff4c2da0a2bb59be516872 : POST_FAILURE in 1h 04m 17s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/fc9523c835c249da947c0e1b5f9a23f1 : SUCCESS in 2h 01m 49s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/88699eb62bab4e0bb2ac82866d9e0a09 : SUCCESS in 1h 55m 56s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/4119c3b517ec45f3a49e2902e5cc2943 : SUCCESS in 1h 14m 46s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/db44ea2b781a464c8c08610e4f2d024d : TIMED_OUT in 2h 09m 00s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/1ac8d860b8da4bcda84d8d9c3266292b : SUCCESS in 2h 10m 41s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b8ad0a97f50d4bc4b03e6dbd1cf3adca : SUCCESS in 20m 27s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/6d65090a992346398c8104a42e6e52d0 : SUCCESS in 51m 36s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/41460546fc6443e99af880b1802dbfe3 : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89437a759d614ad7bb6b169168ac5740 : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac4d42e1d8fd48c799843112db4ef4f2 : SUCCESS in 7m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4629a60cfb844035ae4fdbdae267faf5 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f722aa39a544fb1ac35a0dcdff0b108 : SUCCESS in 9m 21s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c03c2c628c764647a42f62d0a09715b7 : SUCCESS in 22m 23s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/25bd8cb1540c4abba8ea647a6d76f03b : SUCCESS in 25m 35s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8f82cc92936c427392f26f846cd6e6f1 : SUCCESS in 1h 06m 57s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/76483359868247aea05254aea0255fe6 : SUCCESS in 1h 03m 31s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1fc4119d2fc54ebfaf20cfd4fb6f3186 : SUCCESS in 1h 12m 40s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/66fdfcf94e6940d0ba0a97640f3aa9a4 : SUCCESS in 1h 32m 14s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/6eb905e796174591a5a99abfb3b79d4a : POST_FAILURE in 1h 39m 36s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/0675be3231974852becf4f9a967a28d6 : SUCCESS in 1h 40m 23s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/d3913b40b0bb413695c2abb0c2804f53 : SUCCESS in 45m 19s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/b6fe478c4e0f457f815c9a0860d86cd1 : SUCCESS in 1h 33m 25s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"4522393cb2d9463f5731975e0b679cf3de48f3f5","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-06-30 13:38:56.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.","accounts_in_message":[],"_revision_number":11},{"id":"b814a8adb3556f67adfadac6d4289daeec1596e8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 15:30:44.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/30d6724ec99d43f7b96021fe45382c13 : SUCCESS in 1h 14m 31s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f418810a1567482fa982410fce408ef4 : SUCCESS in 1h 02m 36s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d4fc0a188cfe400aabd0b1f5be192bc2 : SUCCESS in 1h 48m 55s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/fff2875ced90422fb3f12624171243b5 : SUCCESS in 1h 05m 34s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"24432ce65c7f257767c7a30fdfa6f719b293a5d4","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-30 16:05:44.000000000","message":"Patch Set 11:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/00c9c62e787f4ae09b136f16b6a28759 : SUCCESS in 26m 36s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"f4c4e71c2e5b4e08525eb53ec4846c3118394f2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 17:09:44.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/b371b8f518a24f3cb232b89b8bad14b9 : SUCCESS in 21m 23s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/eaaac5a1256c4200a4bd696d32b2cc35 : SUCCESS in 31m 21s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/55bfe924f86c412fb5194aa3335ca118 : SUCCESS in 1h 12m 40s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/8b7b7912ee9c4a17bee9a6125c09815b : SUCCESS in 1h 25m 25s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/f124b293bf424cc6ac52b0b7750e608d : SUCCESS in 1h 19m 36s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/45e0581c45704e0b96d5afeb5c8246a9 : SUCCESS in 48m 56s (non-voting)\n- kolla-ansible-centos8-source-bifrost https://zuul.opendev.org/t/openstack/build/dc53e804bd344b7f923b504c8f449332 : SUCCESS in 31m 10s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/145be695f6194308812a258589e0f93e : SUCCESS in 1h 57m 42s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/4d4e587ea8f14793af8d15aaaf5bdbf1 : POST_FAILURE in 1h 03m 36s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/79a033e9e02a4fcc93d6980c0f7c7560 : SUCCESS in 2h 07m 20s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/2d40e6500f8e4bf2a516541997ca04d8 : SUCCESS in 1h 22m 49s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/6224e63145c84f949bcb0b99b36f85c6 : SUCCESS in 58m 45s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/07b84630bc464cf095b4d77bed279578 : SUCCESS in 1h 35m 16s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/df82895edca44b44b7d19a83ee0c9903 : SUCCESS in 22m 48s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/9e2d061b0eac4302a2883664ed18b200 : SUCCESS in 52m 06s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b277c79eece94d22a36feda4faad3c18 : SUCCESS in 1h 03m 13s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/127c017cf3c643dcae77ad9aa3958c80 : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2637b53e12a44d00abfa2b445c7b901a : SUCCESS in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5c0580c10fd74521b19e92652a283ad3 : SUCCESS in 6m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d0119b7c783a4b18be35783a4cad905b : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad53d21ff89d4fb99396799d2d209d47 : SUCCESS in 7m 00s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/56ba30082a514e9fa42705ba83fc286b : SUCCESS in 57m 26s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/05222705ecab46a7a2ed7f4cbe82f32d : SUCCESS in 23m 31s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c55df10c0cb0427b915ddbaec2c32903 : SUCCESS in 1h 09m 49s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d5db6bba620c4e0087ec6dd1a1b1bab6 : SUCCESS in 1h 04m 30s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e2111357b6f74e60bb9df6684fd86855 : SUCCESS in 1h 07m 44s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/1b1e0f5f6eac44dfaeb09a86b0b61811 : SUCCESS in 1h 30m 42s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/e2ee922d27914ce7805474d1193b492a : POST_FAILURE in 1h 02m 34s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/bd521eb2aa5643cab2ca5cf9462d5241 : SUCCESS in 1h 40m 50s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/41f3e3287efd441cb8eb9c8b1a34c783 : SUCCESS in 1h 21m 09s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/c543d07e37494b22824f13c11c2ecbf3 : SUCCESS in 1h 27m 47s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"36662f5b753144129d443a035c322b1c870f4c96","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-07-02 14:24:51.000000000","message":"Abandoned\n\nhandled other way","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"cdc3ffa98276abc4e83daebb2413a80aa9c11bd6","revisions":{"da34fb80e3c7ed5d5ef0305ce03a7099d35835e7":{"kind":"REWORK","_number":1,"created":"2020-02-13 10:03:22.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/99/707599/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/99/707599/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/707599/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/707599/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/707599/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/99/707599/1"}}},"commit":{"parents":[{"commit":"49445b68f8e195f4db7aebe972415d1d620665dc","subject":"Merge \"nova: use proper uefi package names under CentOS/RHEL 8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/49445b68f8e195f4db7aebe972415d1d620665dc"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-02-13 09:04:51.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-02-13 09:04:51.000000000","tz":60},"subject":"WIP: introduce non-infra-base image","message":"WIP: introduce non-infra-base image\n\nTo make sure that all our infrastructure images are the same for\nbinary/source builds we need to have exactly same \u0027base\u0027 image.\n\nThe only difference are KOLLA_INSTALL* variables. So let\u0027s move them to\nnew \u0027non-infra-base\u0027 image and rebase \u0027openstack-base\u0027 and\n\u0027kolla-toolbox\u0027 ones on it.\n\nChange-Id: I2f66cdf18f2a845d97ad3512a35c7de040ace444\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/da34fb80e3c7ed5d5ef0305ce03a7099d35835e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/da34fb80e3c7ed5d5ef0305ce03a7099d35835e7"}]},"branch":"refs/heads/master"},"ade8b5c9c17b2c811abb1ea19aa60a88c5b5af96":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-02-24 14:44:07.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/99/707599/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/99/707599/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/707599/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/707599/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/707599/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/99/707599/2"}}},"commit":{"parents":[{"commit":"89a66a45f958a8ef6a9a8730c5d19025a681fdfe","subject":"Merge \"Revert \"Upgrade virtualenv in pre\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/89a66a45f958a8ef6a9a8730c5d19025a681fdfe"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-02-13 09:04:51.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-02-24 14:44:07.000000000","tz":0},"subject":"WIP: introduce non-infra-base image","message":"WIP: introduce non-infra-base image\n\nTo make sure that all our infrastructure images are the same for\nbinary/source builds we need to have exactly same \u0027base\u0027 image.\n\nThe only difference are KOLLA_INSTALL* variables. So let\u0027s move them to\nnew \u0027non-infra-base\u0027 image and rebase \u0027openstack-base\u0027 and\n\u0027kolla-toolbox\u0027 ones on it.\n\nChange-Id: I2f66cdf18f2a845d97ad3512a35c7de040ace444\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ade8b5c9c17b2c811abb1ea19aa60a88c5b5af96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ade8b5c9c17b2c811abb1ea19aa60a88c5b5af96"}]},"branch":"refs/heads/master"},"327d460bcb3c85a399b62d9f3713d552b5f9eb9f":{"kind":"REWORK","_number":3,"created":"2020-03-09 15:44:51.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/99/707599/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/99/707599/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/707599/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/707599/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/707599/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/99/707599/3"}}},"commit":{"parents":[{"commit":"90e38f23d4e25c30c9807ab511ca4e101adcf688","subject":"Merge \"Deprecate Sensu images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/90e38f23d4e25c30c9807ab511ca4e101adcf688"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-02-13 09:04:51.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-03-09 15:44:36.000000000","tz":60},"subject":"WIP: introduce non-infra-base image","message":"WIP: introduce non-infra-base image\n\nTo make sure that all our infrastructure images are the same for\nbinary/source builds we need to have exactly same \u0027base\u0027 image.\n\nThe only difference are KOLLA_INSTALL* variables. So let\u0027s move them to\nnew \u0027non-infra-base\u0027 image and rebase \u0027openstack-base\u0027 and\n\u0027kolla-toolbox\u0027 ones on it.\n\nChange-Id: I2f66cdf18f2a845d97ad3512a35c7de040ace444\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/327d460bcb3c85a399b62d9f3713d552b5f9eb9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/327d460bcb3c85a399b62d9f3713d552b5f9eb9f"}]},"branch":"refs/heads/master"},"9eceaf46a179f32b066c0dd00c017f5c08c2a677":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-03-09 15:56:16.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/99/707599/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/99/707599/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/707599/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/707599/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/707599/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/99/707599/4"}}},"commit":{"parents":[{"commit":"90e38f23d4e25c30c9807ab511ca4e101adcf688","subject":"Merge \"Deprecate Sensu images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/90e38f23d4e25c30c9807ab511ca4e101adcf688"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-02-13 09:04:51.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-03-09 15:56:13.000000000","tz":0},"subject":"WIP: introduce non-infra-base image","message":"WIP: introduce non-infra-base image\n\nTo make sure that all our infrastructure images are the same for\nbinary/source builds we need to have exactly same \u0027base\u0027 image.\n\nThe only difference are KOLLA_INSTALL* variables. So let\u0027s move them to\nnew \u0027non-infra-base\u0027 image and rebase \u0027openstack-base\u0027 and\n\u0027kolla-toolbox\u0027 ones on it.\n\nImplements: blueprint infra-images\n\nChange-Id: I2f66cdf18f2a845d97ad3512a35c7de040ace444\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9eceaf46a179f32b066c0dd00c017f5c08c2a677"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9eceaf46a179f32b066c0dd00c017f5c08c2a677"}]},"branch":"refs/heads/master"},"e85d444ead23a672085d6d43cf1136430f2e2f67":{"kind":"REWORK","_number":5,"created":"2020-03-09 16:14:24.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/99/707599/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/99/707599/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/707599/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/707599/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/707599/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/99/707599/5"}}},"commit":{"parents":[{"commit":"90e38f23d4e25c30c9807ab511ca4e101adcf688","subject":"Merge \"Deprecate Sensu images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/90e38f23d4e25c30c9807ab511ca4e101adcf688"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-02-13 09:04:51.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-03-09 16:13:03.000000000","tz":60},"subject":"WIP: introduce non-infra-base image","message":"WIP: introduce non-infra-base image\n\nTo make sure that all our infrastructure images are the same for\nbinary/source builds we need to have exactly same \u0027base\u0027 image.\n\nThe only difference are KOLLA_INSTALL* variables. So let\u0027s move them to\nnew \u0027non-infra-base\u0027 image and rebase \u0027openstack-base\u0027 and\n\u0027kolla-toolbox\u0027 ones on it.\n\nImplements: blueprint infra-images\n\nChange-Id: I2f66cdf18f2a845d97ad3512a35c7de040ace444\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e85d444ead23a672085d6d43cf1136430f2e2f67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e85d444ead23a672085d6d43cf1136430f2e2f67"}]},"branch":"refs/heads/master"},"6d2317d0240412f49620e622bf6165765a723e2b":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-05-11 15:29:11.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/99/707599/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/99/707599/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/707599/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/707599/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/707599/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/99/707599/6"}}},"commit":{"parents":[{"commit":"cf06869684455c3ac36567d61c38ce77a6d869e6","subject":"Merge \"Enable mod_wsgi in Nova API container\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cf06869684455c3ac36567d61c38ce77a6d869e6"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-02-13 09:04:51.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-05-11 14:07:31.000000000","tz":120},"subject":"WIP: introduce non-infra-base image","message":"WIP: introduce non-infra-base image\n\nTo make sure that all our infrastructure images are the same for\nbinary/source builds we need to have exactly same \u0027base\u0027 image.\n\nThe only difference are KOLLA_INSTALL* variables. So let\u0027s move them to\nnew \u0027non-infra-base\u0027 image and rebase \u0027openstack-base\u0027 and\n\u0027kolla-toolbox\u0027 ones on it.\n\nImplements: blueprint infra-images\n\nChange-Id: I2f66cdf18f2a845d97ad3512a35c7de040ace444\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6d2317d0240412f49620e622bf6165765a723e2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6d2317d0240412f49620e622bf6165765a723e2b"}]},"branch":"refs/heads/master"},"488bd7674a3f1c43ddb776f6e2bf322763b3f897":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2020-05-17 13:46:33.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/99/707599/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/99/707599/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/707599/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/707599/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/707599/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/99/707599/7"}}},"commit":{"parents":[{"commit":"aabd532da9fe3cac64626d481c4b3113b5925e69","subject":"Remove Monasca Log API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aabd532da9fe3cac64626d481c4b3113b5925e69"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-02-13 09:04:51.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-05-17 13:37:38.000000000","tz":120},"subject":"WIP: introduce non-infra-base image","message":"WIP: introduce non-infra-base image\n\nTo make sure that all our infrastructure images are the same for\nbinary/source builds we need to have exactly same \u0027base\u0027 image.\n\nThe only difference are KOLLA_INSTALL* variables. So let\u0027s move them to\nnew \u0027non-infra-base\u0027 image and rebase \u0027openstack-base\u0027 and\n\u0027kolla-toolbox\u0027 ones on it.\n\nImplements: blueprint infra-images\n\nChange-Id: I2f66cdf18f2a845d97ad3512a35c7de040ace444\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/488bd7674a3f1c43ddb776f6e2bf322763b3f897"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/488bd7674a3f1c43ddb776f6e2bf322763b3f897"}]},"branch":"refs/heads/master"},"670546f012a429a8e4f8e4065ba7774b1084fa61":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2020-05-25 11:15:39.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/99/707599/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/99/707599/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/707599/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/707599/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/707599/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/99/707599/8"}}},"commit":{"parents":[{"commit":"9836fb9267afac298141e4a3f5a748c457029af7","subject":"Revert \"Switch to Ussuri release of networking-l2gw tarball\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9836fb9267afac298141e4a3f5a748c457029af7"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-02-13 09:04:51.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-05-25 11:12:48.000000000","tz":120},"subject":"WIP: introduce non-infra-base image","message":"WIP: introduce non-infra-base image\n\nTo make sure that all our infrastructure images are the same for\nbinary/source builds we need to have exactly same \u0027base\u0027 image.\n\nThe only difference are KOLLA_INSTALL* variables. So let\u0027s move them to\nnew \u0027non-infra-base\u0027 image and rebase \u0027openstack-base\u0027 and\n\u0027kolla-toolbox\u0027 ones on it.\n\nImplements: blueprint infra-images\n\nChange-Id: I2f66cdf18f2a845d97ad3512a35c7de040ace444\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/670546f012a429a8e4f8e4065ba7774b1084fa61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/670546f012a429a8e4f8e4065ba7774b1084fa61"}]},"branch":"refs/heads/master"},"3c3f0ac410177846a308e204c659de7075c705f7":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2020-06-25 10:38:17.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/99/707599/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/99/707599/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/707599/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/707599/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/707599/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/99/707599/9"}}},"commit":{"parents":[{"commit":"9836fb9267afac298141e4a3f5a748c457029af7","subject":"Revert \"Switch to Ussuri release of networking-l2gw tarball\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9836fb9267afac298141e4a3f5a748c457029af7"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-02-13 09:04:51.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-06-25 10:38:13.000000000","tz":0},"subject":"introduce non-infra-base image","message":"introduce non-infra-base image\n\nTo make sure that all our infrastructure images are the same for\nbinary/source builds we need to have exactly same \u0027base\u0027 image.\n\nThe only difference are KOLLA_INSTALL* variables. So let\u0027s move them to\nnew \u0027non-infra-base\u0027 image and rebase \u0027openstack-base\u0027 and\n\u0027kolla-toolbox\u0027 ones on it.\n\nImplements: blueprint infra-images\n\nChange-Id: I2f66cdf18f2a845d97ad3512a35c7de040ace444\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c3f0ac410177846a308e204c659de7075c705f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c3f0ac410177846a308e204c659de7075c705f7"}]},"branch":"refs/heads/master"},"522ae03e2620ab9089320f5a29bc0480ad988424":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2020-06-25 10:38:21.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/99/707599/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/99/707599/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/707599/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/707599/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/707599/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/99/707599/10"}}},"commit":{"parents":[{"commit":"c778749bc79bc235d58730367e8c8b8448f96512","subject":"Merge \"Fix horizon debian binary docker image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c778749bc79bc235d58730367e8c8b8448f96512"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-02-13 09:04:51.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-06-25 10:38:21.000000000","tz":0},"subject":"introduce non-infra-base image","message":"introduce non-infra-base image\n\nTo make sure that all our infrastructure images are the same for\nbinary/source builds we need to have exactly same \u0027base\u0027 image.\n\nThe only difference are KOLLA_INSTALL* variables. So let\u0027s move them to\nnew \u0027non-infra-base\u0027 image and rebase \u0027openstack-base\u0027 and\n\u0027kolla-toolbox\u0027 ones on it.\n\nImplements: blueprint infra-images\n\nChange-Id: I2f66cdf18f2a845d97ad3512a35c7de040ace444\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/522ae03e2620ab9089320f5a29bc0480ad988424"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/522ae03e2620ab9089320f5a29bc0480ad988424"}]},"branch":"refs/heads/master"},"cdc3ffa98276abc4e83daebb2413a80aa9c11bd6":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2020-06-30 13:38:56.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/99/707599/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/99/707599/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/707599/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/707599/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/707599/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/99/707599/11"}}},"commit":{"parents":[{"commit":"a4ee0d4ada9b82530f4cc6c21bbc134447abd2b9","subject":"Merge \"Add release note for FWaaS removal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a4ee0d4ada9b82530f4cc6c21bbc134447abd2b9"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-02-13 09:04:51.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-06-30 12:08:01.000000000","tz":120},"subject":"introduce non-infra-base image","message":"introduce non-infra-base image\n\nTo make sure that all our infrastructure images are the same for\nbinary/source builds we need to have exactly same \u0027base\u0027 image.\n\nThe only difference are KOLLA_INSTALL* variables. So let\u0027s move them to\nnew \u0027non-infra-base\u0027 image and rebase \u0027openstack-base\u0027 and\n\u0027kolla-toolbox\u0027 ones on it.\n\nImplements: blueprint infra-images\n\nChange-Id: I2f66cdf18f2a845d97ad3512a35c7de040ace444\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cdc3ffa98276abc4e83daebb2413a80aa9c11bd6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cdc3ffa98276abc4e83daebb2413a80aa9c11bd6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
