)]}'
{"id":"openstack%2Fkolla~707603","triplet_id":"openstack%2Fkolla~master~I9568261fdbbe4b9156d0e07c414ec911ca2e8557","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I9568261fdbbe4b9156d0e07c414ec911ca2e8557","subject":"Remove kolla Ceph container images","status":"MERGED","created":"2020-02-13 10:20:34.000000000","updated":"2020-02-14 14:31:41.000000000","submitted":"2020-02-14 14:28:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"707603-1581690532169-3390009c","meta_rev_id":"9ba6dffb84b27ca2702f80bfcc3e82544f9522b7","_number":707603,"virtual_id_number":707603,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-14 14:28:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-02-14 11:46:13.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2020-02-14 08:54:05.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-02-14 08:54:12.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2020-02-14 08:54:15.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2020-02-14 08:52:25.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-14 08:54:05.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-02-14 08:54:15.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-02-14 11:46:13.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-14 14:28:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"169689eb4a81e71c387cc0fd96db816b291a1231","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-02-13 10:20:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f032869ecae51b3eaff9a888def30f16c92aa4af","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-13 10:24:07.000000000","message":"Patch Set 1: Code-Review-1\n\nalso readme, zuul, reno","accounts_in_message":[],"_revision_number":1},{"id":"23ddd38adf9498a9ac2e2fd628ff3094441ea425","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-02-13 10:40:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cc3c10c48aaaa2062b5ffab06c94311e14e40a90","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-02-13 10:41:18.000000000","message":"Patch Set 2:\n\n\u003e also readme, zuul, reno\n\nreadme and reno fixed, where in Zuul?","accounts_in_message":[],"_revision_number":2},{"id":"e4be5a67f1982725ed7d4ef2f2f49d359c8b5ae7","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-02-13 10:41:18.000000000","message":"Patch Set 2:\n\nYou forgot to do \u0027git rm -r docker\u0027 as there are still some container images left...","accounts_in_message":[],"_revision_number":2},{"id":"9389e7de5eda8023f6f6653c5fc4e68b92bb77c0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-13 10:45:35.000000000","message":"Patch Set 2:\n\nfile matchers in zuul","accounts_in_message":[],"_revision_number":2},{"id":"f41285c16237fafe2e6245349217133f24495080","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-13 10:46:47.000000000","message":"Patch Set 2:\n\nah no, sorry, cached result :-)","accounts_in_message":[],"_revision_number":2},{"id":"51bddb2f371e50e0c20732e21651d7d0352b00b7","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-13 10:47:47.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"be079730580ab663852ec554873bf9abd50b42aa","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-02-13 10:59:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"01c9eac35ed388d1072ebb87a7ec79c73f021cf2","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-13 11:00:40.000000000","message":"Patch Set 2: Code-Review+2\n\nall ack","accounts_in_message":[],"_revision_number":2},{"id":"d29f56291e67d185d52943cae6c7548699f03714","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-02-13 11:04:04.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"890615201fb615efb87dd41627215c993118c71d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-13 11:06:32.000000000","message":"Patch Set 2: Code-Review-1\n\nhahahaha, good one","accounts_in_message":[],"_revision_number":2},{"id":"52ea18554fb5cd1099d08b148655bef775d428cd","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-02-13 11:07:56.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"46a9cfe9762d0db27bc61e5a073f283b9df01486","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-02-13 11:08:25.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"1a711334b636086e346b9cc4ccadd0581f2422c6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-13 11:08:56.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\njust go","accounts_in_message":[],"_revision_number":3},{"id":"8746e7dc6e8b56672a4459a814fa8cb706527765","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-13 11:37:03.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/cab15da827aa42569168928dbccf6a73 : SUCCESS in 28m 08s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/61b3c58385bf4e2eb2a671f5a40102a0 : SUCCESS in 27m 08s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"40599d49a6c51d3501ce1ec226e8db7f8fbae849","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 11:53:25.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5bce1a47c27fce1fc6c39d4cde6ae2d013320784","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 12:15:50.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/032dd726f5f844ec85733fb3f8265d58 : SUCCESS in 3m 24s\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/4eca2f50a58f4fec917b6b4bd4753f38 : FAILURE in 29m 13s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/4466e9830148474b9b22d0d97e96829f : FAILURE in 48m 13s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/97d297e8ca074ddfa2f2597919faffd4 : FAILURE in 39m 41s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/49f0ce0800954955b55056cdcd53a482 : FAILURE in 44m 09s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/ca2f2a2dd29d4bfb9c70818d1bf06be4 : FAILURE in 36m 15s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/06d44a86ac4a42c38eff2bcdb0637233 : FAILURE in 36m 57s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/c1b67dcdc1804d7b92a0e6582d378e16 : FAILURE in 44m 39s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/df17407fa2ac4689b0bb1075fd5cab29 : FAILURE in 33m 36s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/de0f17729b3c42519e6893eb8764d729 : FAILURE in 37m 57s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/67856d7b84e240a298d14208fbdcde3b : SUCCESS in 1h 06m 10s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/5ea4258ff16f452b86cdd77694c53b35 : FAILURE in 35m 37s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/dff07a33badd4a9bbb6d82f171af4264 : FAILURE in 22m 17s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/9bc347faf212461d822c21cb2dfa7daf : FAILURE in 21m 13s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/82aea7aa0cf445f78bbe816faef47de2 : SUCCESS in 8m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f98188886e04a78be9e6173884d7738 : SUCCESS in 7m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ff047431f42049a290171abf67ecc675 : SUCCESS in 4m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/29dee7bab70b4347a12117e9266efef0 : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/49ad413b7a044c52a9a0ac57cc4404a9 : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a015202cc5849c59998ae5242ef9370 : SUCCESS in 7m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/792ad1ea9da14b3f85f1a64a1a3946af : SUCCESS in 7m 21s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4418380b883e4cae91d6dc860959d685 : FAILURE in 29m 02s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6d8218ffdcfe49908932b855597b8ca4 : FAILURE in 27m 56s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6bf9d83b042145cd81e81f59238f7c29 : FAILURE in 22m 49s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8d3ce4aa5e6647c0beb541b44cc8d422 : FAILURE in 21m 58s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d55a0ccc39f1402a80d3333a0e4c56a1 : FAILURE in 22m 58s","accounts_in_message":[],"_revision_number":3},{"id":"b365e7c185d28ae79eb3543ed3a9783db2c5b19e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-13 12:26:02.000000000","message":"Patch Set 3:\n\nfind disk must not be copied if deleted","accounts_in_message":[],"_revision_number":3},{"id":"95abd0e905b19bcc18b3e87d1dc58273b5483a64","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-02-13 12:26:51.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"bd99686bd85664cc80e3a4619082496d0d5a0d69","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-13 12:27:01.000000000","message":"Patch Set 3: Code-Review-1 Workflow-1\n\nand better not delete it entirely since it is used for swift too","accounts_in_message":[],"_revision_number":3},{"id":"18a5b9a76e7d838333d4c76b775e5d5d4e4ff941","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-13 12:27:27.000000000","message":"Patch Set 4: Code-Review-1\n\nrestore that","accounts_in_message":[],"_revision_number":4},{"id":"b925fea8b544f8c9b07e9b07a2cd6885db3d0df7","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-13 12:52:43.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/12be23075a984aa0ba80f74b62d93b01 : SUCCESS in 24m 59s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/43a8739274ae462f8f9e00b2a1bf7b10 : SUCCESS in 24m 24s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2244d6657e965deb1447ffffbf4f7bc5a173233c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 13:29:52.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/bfb9ab3bc9c64c87a49aa9ef8e700cbe : SUCCESS in 3m 39s\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/fd6f0132f08b4fa39265961608afef1b : FAILURE in 46m 26s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/3824d3b25f4a4b9b9f5450885a317fe4 : FAILURE in 54m 17s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/6a14fc1246ee47cf886a9be6e0d3e711 : FAILURE in 48m 58s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/66d9851bde35460bbe30170781b5dbd3 : FAILURE in 33m 07s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/4e8a36caf09149d182e3e43bb86f8e76 : FAILURE in 32m 48s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/3c2cc5c0c45d4f2f956cd497c4c53f10 : FAILURE in 30m 34s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/e738a4f0941c4c86a896d186189029e2 : FAILURE in 35m 01s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/ae058471ce064bd9805b6f44382ce61e : FAILURE in 26m 26s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/30700fbed22a478e94ba2a150e0d308c : FAILURE in 42m 04s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/71218d14869c461b9fd7be4cffd42f4c : SUCCESS in 58m 19s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6456188ff88340a69e77013cf4f6681e : FAILURE in 41m 39s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/4101c0834a1c42efbdb525178dbab0ed : FAILURE in 45m 14s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/75e64c0dfb4240b8928c325492023121 : FAILURE in 20m 38s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dce444703773403ab641224f270af00f : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6db889c20ac549a59baa1cb4bc39153b : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f14529ef47fc4298ba69823b72e8155d : SUCCESS in 5m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0df58eac2f5c4205b6b96022eb0b37a4 : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b9467391fd144497a5a943d74c147409 : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8fbf84f4a8a4b4da6cf056ac574b3ce : SUCCESS in 7m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/190d41407fb74b19b3b9e2a667e81aa8 : SUCCESS in 6m 10s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/41a197bd85814d909a36e84f486b6642 : FAILURE in 24m 25s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6357fa413bfd4746963b4d2a78403c4e : FAILURE in 27m 10s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/110d1b73e1e949ea9902dde282a2edad : FAILURE in 24m 04s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8f3897104a8746909d78a6d07cf41e62 : FAILURE in 21m 29s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/93e6a30525ed49089db52904bf4a72a6 : FAILURE in 20m 11s","accounts_in_message":[],"_revision_number":4},{"id":"7b9d5396a8bb2eca79bd16bfcc58f0225f988a17","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 14:13:46.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7a3a9b7a6b53458d9f658455f434a95e : RETRY_LIMIT in 1m 33s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"31f83c395075ea8224061d01f5c8121f36821977","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-02-13 14:20:18.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"4266b0daf2e329ec5cdda5ffe94b271449ff2d7a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-13 14:53:38.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/1fb999e480074d1f8d7d4883cc29eee2 : SUCCESS in 26m 42s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/514788f86e3e423bbfebf01408afe342 : SUCCESS in 31m 59s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"6bbfe00a6ecf2dba501ef42dc1f5c7c978bd27c1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 15:38:19.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"0e68441bdf874d3d9f09964f3002218b0ab84633","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 16:02:22.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/49d2a7e39082452985c3bd9a6325e115 : SUCCESS in 4m 16s\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/987838c1ed4541d0a6deac0c6d156c91 : SUCCESS in 30m 56s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/ff6b1dc10ae449cf96b377ebfccce8a0 : SUCCESS in 39m 15s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/9848e173d0e44c99a3088d5c8bab321a : SUCCESS in 23m 51s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/8ece0a20f82c472883704da606add6a9 : SUCCESS in 28m 59s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/b96b43d55b5742f3ad19db8fb49f8b7f : SUCCESS in 1h 27m 19s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/680e3ef2a40f424bac5bcf6c072fa70f : SUCCESS in 1h 07m 18s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/456888077e004597ba5469c59137ff80 : SUCCESS in 1h 17m 36s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/827b97fd8aa542beab0933ea08d41579 : SUCCESS in 1h 10m 19s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/b67e7a2ce40f4e0c8b55786d846e66ca : SUCCESS in 1h 41m 31s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/c531f51852f2496d8afa04e5e4587375 : SUCCESS in 58m 45s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/cd60a7e2dc364a5cbe8055af869a7672 : SUCCESS in 24m 56s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/f89fc3e6ea464fea85cc3a941273a35f : SUCCESS in 21m 54s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d86f54830ad54461832168003a80bccf : SUCCESS in 1h 01m 56s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/30a11e18da3545cc8d190986c64eae02 : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b2f12e50dc643638e6d303d4f15c968 : SUCCESS in 7m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/276ea973861942b6b890bce8f6ac31da : SUCCESS in 4m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b9e9a610ca654e61abc96dd4cd5ab16d : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0f67a8d9612143a488ee4f6fdf21a184 : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db7b0880d9d843819f4a3ba5a7641b54 : SUCCESS in 6m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c12a479c9b0c422bb42954393e89a3a2 : SUCCESS in 5m 36s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8b39e7a5969a45a3b92ff0f9e0bd6271 : SUCCESS in 23m 35s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6c861d9dd998480e98bd3929a9d6194a : SUCCESS in 30m 44s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f669111ea7634fca91f0964c789ae0fc : SUCCESS in 1h 00m 06s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2e070683da8143eca5e9dd739bbdcb29 : SUCCESS in 58m 41s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4ca8e26a31a3423a842e2822a2541f04 : SUCCESS in 1h 03m 34s","accounts_in_message":[],"_revision_number":5},{"id":"d023c6544bd92530c39b905acb6d140dadd63fee","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-02-13 16:03:10.000000000","message":"Patch Set 5: Review-Priority+1","accounts_in_message":[],"_revision_number":5},{"id":"9956ab1525d97cba403a19fba5eff728c6fc6ea8","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-13 19:43:30.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"210c44408e22dc0fcb1dfc08b7a5dfe273f6745a","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-02-14 08:50:03.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e7a41d369c643f87da9c25cabc78373af1e916cf","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-02-14 08:51:09.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"da69f820ad05ca98757b2ab1220f0e2231261443","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-02-14 08:52:25.000000000","message":"Patch Set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"4b4809234b508bac60a64f2fb87bedc3d1678dad","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-02-14 08:54:05.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"2c11f79db0e2f282ff1caaf85fe0f8d318e96b49","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-14 08:54:12.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"0de99c17f8613785d49c7273dbb2ece78f9591f6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-14 08:54:15.000000000","message":"Patch Set 7: Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"5ff563d07a282a0608658f7c231c9d358b925d4b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-14 09:18:53.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/c1d9520cc9cf457fb01bfd9e0a8d980c : SUCCESS in 25m 41s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/7da7f67dcd1a4c54886e7505b56f5c8d : FAILURE in 12m 44s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"4ad788244f4d0df0d2a3dcb99f4bd149ac6b4d18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 10:30:49.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4d25316c01364c6388dabd57d9e98c81 : SUCCESS in 4m 17s\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/a7fa7327088e44d28e525283035115d3 : SUCCESS in 33m 34s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/7b1fbdab43484641a24918b500aee37e : SUCCESS in 45m 06s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/8a52c7862c1043c5844313f748a983f3 : FAILURE in 7m 31s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/cd7da86431d14e138c99480b9bbb36db : SUCCESS in 36m 36s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/5147657b53ab4222a8f421d440287dd6 : SUCCESS in 1h 24m 45s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/da9f212fbd2047bba6ce8325fd3dcd58 : SUCCESS in 1h 21m 30s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/05cfa435f430495fa04dcaa40f850e3d : SUCCESS in 1h 23m 47s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/0af1bca77b3649c0afd2e5b317cbfa33 : SUCCESS in 1h 19m 50s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/0508003a6cc24fd48b3b86cac8761a0e : SUCCESS in 1h 36m 19s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/6024547e30cb40f388be6fbfc3936c88 : SUCCESS in 53m 07s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/8f42cc3a72b4420cbbb5104fa470d65c : SUCCESS in 29m 19s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/cdde7fc1aada4dcd8ee0147de14239a5 : SUCCESS in 29m 58s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e517fd0da6894f41bdfbaf019dfc8984 : SUCCESS in 59m 01s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/04a4ea4c82ee4254992c61c42ff283e7 : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8659d62adc204331adf45255c88ea7e8 : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/94dde1247e5c4ad4b92afff44bf470ba : SUCCESS in 6m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ac8492c1ae3d4337a3984f1eabd33508 : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7157e20996bf4a20a1f2792da3eec82c : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d309276c22814da7ac3bb46f83d1bf39 : SUCCESS in 7m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/188d678904ec487fabb4548e59d7d3e5 : SUCCESS in 6m 21s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2f0ea0b0656c4b6aaa135e7ad99112c4 : SUCCESS in 30m 58s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0044b491e6764955bbb805c7f5271489 : SUCCESS in 29m 10s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1c72b0a2c67445caa9659df62700f445 : SUCCESS in 1h 08m 21s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ce33511292a741e48a0701e12b2219c7 : SUCCESS in 1h 06m 55s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/75730fee939f440c8722446c9ccfe5bd : FAILURE in 9m 59s","accounts_in_message":[],"_revision_number":7},{"id":"a9f0c79c539807fbe47da2c34c51292fb0856979","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-02-14 10:46:27.000000000","message":"Patch Set 7:\n\nrecheck (openev.org connectivity failures?)","accounts_in_message":[],"_revision_number":7},{"id":"63122e0130922efc6bc5f216fb89e0265b61ae5d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 11:12:29.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/853e75a3b7354816811d9ddaef219ccf : RETRY_LIMIT in 18m 52s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"cea368d890aa6b59a231443f0c26b13e9efc3c79","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-14 11:46:13.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/f06114cc51754045972443fa985a9f4a : SUCCESS in 28m 29s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/4ed3350110ba4c5fb957acd9581d282d : SUCCESS in 24m 53s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"3c5d10c79a61174fd68ae6079c89ea3e501acdd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 12:37:48.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/74281e271c8947e590ec43827c9577f9 : SUCCESS in 3m 28s\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/ce548736ecc243bab0f2e947971efd1d : SUCCESS in 32m 29s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/3d590ad917324c0e84bc9f3baa66ed12 : SUCCESS in 43m 01s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/d34dcb1c83e14a39bb8ff7dd528690c4 : SUCCESS in 28m 28s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/9d432ed5c8ef4de3a4ab500a97a010dc : SUCCESS in 33m 03s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/1bde120aaca34e35b8cd4aea61eb005f : SUCCESS in 1h 24m 11s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/ea6ec69c06a54764be57832039691d2d : SUCCESS in 1h 15m 25s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/222df048661640c394d5ee9411c8987e : SUCCESS in 1h 14m 06s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/b172121b2bfa4b0a97cc2329d2751a9b : SUCCESS in 1h 08m 56s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/0d81e300e90348f99b49b4512cd284e8 : SUCCESS in 1h 44m 17s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/8e64928a4fe147b1a3ca60a52b27a37a : SUCCESS in 51m 37s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/14cd7a09ecf34ac5922b0124c0a3b040 : FAILURE in 4m 08s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/35dab0f84bb440faa4d7edc4762ceec5 : SUCCESS in 23m 47s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/bf7dfe1d37b3402ebda1c742e66dd460 : SUCCESS in 1h 12m 57s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c52a8a5adc15443ead5fa771d4eb0784 : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31a3c0c2ad4a427c851cdb26032c1b76 : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/309fe2c32f0a4a0a962c8f15d46fd609 : SUCCESS in 4m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/32cbd08ddca44c2d97d9fd2af00fdde2 : SUCCESS in 5m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/129aa45508e24151b3866a1c1cdd7b40 : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3deb884337da469d8b0d17cf1d360db6 : SUCCESS in 8m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f7280ee9e78e4c12a5ff95bb8937b4cb : SUCCESS in 6m 26s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ccb8721c2ec74da6a430a18e9ba4c292 : SUCCESS in 26m 10s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d9c3957e63234661ab0f5919f3aa33b9 : SUCCESS in 30m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/65483d5c6ea844499d301628f1995aaa : SUCCESS in 1h 06m 20s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fdc5670c22d9456399e69b6c931cb5af : SUCCESS in 1h 05m 23s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9a4f9ee638ab44a69a8fad3ff1923e72 : SUCCESS in 1h 05m 13s","accounts_in_message":[],"_revision_number":7},{"id":"bd0961fbd39f3827e872d271e736c74a76c15883","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 12:38:02.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"e08ad6fa4fd978a773d5ce2e18a37407b0f966b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 14:28:51.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/aa31c6c04be143cbb7b293923bb8f225 : SUCCESS in 3m 51s\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/22f04ead738e42229d9372448de67699 : SUCCESS in 42m 55s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/983c1b4984784ad6a4d97995c9c7c26c : SUCCESS in 30m 24s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/2bda72cc58ff4a31a8bd7db61de97b5c : SUCCESS in 1h 19m 53s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/2cd4b8a1802e4c40b54aa2a0e6d567de : SUCCESS in 1h 12m 25s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/51b2ffdf7fae4d2b80b7d93ba5df6682 : SUCCESS in 1h 45m 49s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/406e11051b1345d69ae107830be2ec19 : SUCCESS in 53m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c052f0682ffe4b8f88bf795ab50694f8 : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/316adfeac1374115b92098e6e3aa368b : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d05e8a5297e459093dce67f8e136cf5 : SUCCESS in 4m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c95c846db307401f831364136a24b588 : SUCCESS in 6m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ec1145261c24a8eb4bab6633d62d927 : SUCCESS in 6m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2dd81d172f7046c9bb9baddeeecdc282 : SUCCESS in 6m 36s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/041a371ce6ce481eb92520734e2c0fb3 : SUCCESS in 26m 01s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/114e490b6f744337aa7251518cd9b105 : SUCCESS in 56m 39s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/326ec054fd0848d2abe8c148af57a0c6 : SUCCESS in 1h 06m 12s","accounts_in_message":[],"_revision_number":7},{"id":"5628fc961617b36ba5ad9e7890aa83fabf57435a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 14:28:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"9ba6dffb84b27ca2702f80bfcc3e82544f9522b7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 14:31:41.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21bac3b2dbdb49a28697b1adf9c58f0e : SUCCESS in 2m 15s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6f38bb8cad52427a827586fc27ba4c4b : SUCCESS in 1m 56s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"abce4d5398d44b46161de2bf21914e776d16d7e7","revisions":{"0a5b80fd943480de022200010596a7689245460b":{"kind":"REWORK","_number":1,"created":"2020-02-13 10:20:34.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/03/707603/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/03/707603/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/707603/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/707603/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/707603/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/03/707603/1"}}},"commit":{"parents":[{"commit":"063a46a7de4471504535499ca2a972a79e4ce5e5","subject":"Upgrade virtualenv in pre","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/063a46a7de4471504535499ca2a972a79e4ce5e5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-13 10:10:24.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-13 10:20:32.000000000","tz":60},"subject":"Remove kolla container images","message":"Remove kolla container images\n\nThis change removes Dockerfiles, find_disks.py script in kolla-toolbox\nand relevant kolla cli python code bits.\n\nKolla-Ansible part has been removed, and K-A was the only consumer.\n\nChange-Id: I9568261fdbbe4b9156d0e07c414ec911ca2e8557\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0a5b80fd943480de022200010596a7689245460b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0a5b80fd943480de022200010596a7689245460b"}]},"branch":"refs/heads/master"},"5e6ca802076adb3fa0f91afa7be2ae50005e9e59":{"kind":"REWORK","_number":2,"created":"2020-02-13 10:40:38.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/03/707603/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/03/707603/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/707603/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/707603/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/707603/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/03/707603/2"}}},"commit":{"parents":[{"commit":"063a46a7de4471504535499ca2a972a79e4ce5e5","subject":"Upgrade virtualenv in pre","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/063a46a7de4471504535499ca2a972a79e4ce5e5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-13 10:10:24.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-13 10:39:13.000000000","tz":60},"subject":"Remove kolla container images","message":"Remove kolla container images\n\nThis change removes Dockerfiles, find_disks.py script in kolla-toolbox\nand relevant kolla cli python code bits.\n\nKolla-Ansible part has been removed, and K-A was the only consumer.\n\nChange-Id: I9568261fdbbe4b9156d0e07c414ec911ca2e8557\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5e6ca802076adb3fa0f91afa7be2ae50005e9e59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5e6ca802076adb3fa0f91afa7be2ae50005e9e59"}]},"branch":"refs/heads/master"},"70db1eed8f0c9d4d5ad95a3ae834637635fbcd3e":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-02-13 11:07:56.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/03/707603/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/03/707603/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/707603/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/707603/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/707603/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/03/707603/3"}}},"commit":{"parents":[{"commit":"063a46a7de4471504535499ca2a972a79e4ce5e5","subject":"Upgrade virtualenv in pre","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/063a46a7de4471504535499ca2a972a79e4ce5e5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-13 10:10:24.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-13 11:07:27.000000000","tz":60},"subject":"Remove kolla Ceph container images","message":"Remove kolla Ceph container images\n\nThis change removes ceph* Dockerfiles, find_disks.py script in kolla-toolbox\nand relevant kolla cli python code bits.\n\nKolla-Ansible part has been removed, and K-A was the only consumer.\n\nChange-Id: I9568261fdbbe4b9156d0e07c414ec911ca2e8557\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/70db1eed8f0c9d4d5ad95a3ae834637635fbcd3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/70db1eed8f0c9d4d5ad95a3ae834637635fbcd3e"}]},"branch":"refs/heads/master"},"5290f94afe566c53848ac181198ea928cd29fddd":{"kind":"REWORK","_number":4,"created":"2020-02-13 12:26:51.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/03/707603/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/03/707603/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/707603/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/707603/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/707603/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/03/707603/4"}}},"commit":{"parents":[{"commit":"063a46a7de4471504535499ca2a972a79e4ce5e5","subject":"Upgrade virtualenv in pre","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/063a46a7de4471504535499ca2a972a79e4ce5e5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-13 10:10:24.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-13 12:26:41.000000000","tz":60},"subject":"Remove kolla Ceph container images","message":"Remove kolla Ceph container images\n\nThis change removes ceph* Dockerfiles, find_disks.py script in kolla-toolbox\nand relevant kolla cli python code bits.\n\nKolla-Ansible part has been removed, and K-A was the only consumer.\n\nChange-Id: I9568261fdbbe4b9156d0e07c414ec911ca2e8557\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5290f94afe566c53848ac181198ea928cd29fddd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5290f94afe566c53848ac181198ea928cd29fddd"}]},"branch":"refs/heads/master"},"c2bb8e0843c72997cdc77ea160c0e3a3cfb346f1":{"kind":"REWORK","_number":5,"created":"2020-02-13 14:20:18.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/03/707603/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/03/707603/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/707603/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/707603/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/707603/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/03/707603/5"}}},"commit":{"parents":[{"commit":"063a46a7de4471504535499ca2a972a79e4ce5e5","subject":"Upgrade virtualenv in pre","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/063a46a7de4471504535499ca2a972a79e4ce5e5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-13 10:10:24.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-13 14:20:14.000000000","tz":0},"subject":"Remove kolla Ceph container images","message":"Remove kolla Ceph container images\n\nThis change removes ceph* Dockerfiles, find_disks.py script in kolla-toolbox\nand relevant kolla cli python code bits.\n\nKolla-Ansible part has been removed, and K-A was the only consumer.\n\nChange-Id: I9568261fdbbe4b9156d0e07c414ec911ca2e8557\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c2bb8e0843c72997cdc77ea160c0e3a3cfb346f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c2bb8e0843c72997cdc77ea160c0e3a3cfb346f1"}]},"branch":"refs/heads/master"},"5bf79b9d6d69ac32f7b1294f0b78876cddcf6e41":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2020-02-14 08:51:09.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/03/707603/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/03/707603/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/707603/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/707603/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/707603/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/03/707603/6"}}},"commit":{"parents":[{"commit":"063a46a7de4471504535499ca2a972a79e4ce5e5","subject":"Upgrade virtualenv in pre","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/063a46a7de4471504535499ca2a972a79e4ce5e5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-13 10:10:24.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-14 08:51:06.000000000","tz":0},"subject":"Remove kolla Ceph container images","message":"Remove kolla Ceph container images\n\nThis change removes ceph* Dockerfiles and relevant kolla cli python\ncode bits.\n\nKolla-Ansible part has been removed, and K-A was the only consumer.\n\nChange-Id: I9568261fdbbe4b9156d0e07c414ec911ca2e8557\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5bf79b9d6d69ac32f7b1294f0b78876cddcf6e41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5bf79b9d6d69ac32f7b1294f0b78876cddcf6e41"}]},"branch":"refs/heads/master"},"abce4d5398d44b46161de2bf21914e776d16d7e7":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2020-02-14 08:52:25.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/03/707603/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/03/707603/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/707603/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/707603/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/707603/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/03/707603/7"}}},"commit":{"parents":[{"commit":"063a46a7de4471504535499ca2a972a79e4ce5e5","subject":"Upgrade virtualenv in pre","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/063a46a7de4471504535499ca2a972a79e4ce5e5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-13 10:10:24.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-14 08:52:23.000000000","tz":0},"subject":"Remove kolla Ceph container images","message":"Remove kolla Ceph container images\n\nThis change removes ceph* Dockerfiles and relevant kolla python\ncode bits.\n\nKolla-Ansible part has been removed, and K-A was the only consumer.\n\nChange-Id: I9568261fdbbe4b9156d0e07c414ec911ca2e8557\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/abce4d5398d44b46161de2bf21914e776d16d7e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/abce4d5398d44b46161de2bf21914e776d16d7e7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
