)]}'
{"id":"openstack%2Fkolla~707632","triplet_id":"openstack%2Fkolla~stable%2Fstein~I56525cdc357e187ee6f3adae068a37cd013a9fce","project":"openstack/kolla","branch":"stable/stein","hashtags":[],"change_id":"I56525cdc357e187ee6f3adae068a37cd013a9fce","subject":"add Linaro repo with fixed libvirt","status":"ABANDONED","created":"2020-02-13 12:19:55.000000000","updated":"2020-02-13 15:43:57.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bbcfa920cb556e2ce4dd3bc31a1de016572c1570","_number":707632,"virtual_id_number":707632,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-02-13 13:55:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-13 13:55:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b44579b81571f57d4c862406b99fea9877db7af3","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-02-13 12:19:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f236f9eee2be1ed31cddb492acff583f237cc2a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 13:55:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/5289eae25b7449f18b08e6117b547eae : SUCCESS in 51m 51s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/5fe44db40b9f4b1b85dac99fab4743c6 : SUCCESS in 1h 02m 03s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/ee3ce072030f4546b344c6856f937ef2 : SUCCESS in 1h 21m 45s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/fceb7440b6ae4922b7e007d1adf009fd : SUCCESS in 1h 29m 16s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/a8d155b8a57f4df7badbc70ac40bfbe6 : SUCCESS in 1h 31m 18s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/8a5762ec4db34d5498f02d150ce29e78 : FAILURE in 13m 50s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/04973bc9b45d41ca9b15c2fd1d847bea : FAILURE in 44m 58s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/f6ea3be1d6d749248f44aeb0ceeb9f4d : SUCCESS in 1h 00m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b1e0559349dc4d3695b478ba614b36fb : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/433fbd519d824c2e9172019451d2bfe9 : SUCCESS in 5m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4f04aca8b4f747308fa7954c1bf2b63e : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/575c2924eead41a89198aebd52ab161b : SUCCESS in 5m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3ada3ec3bd7e4a9f95c493ed87ff256e : SUCCESS in 6m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14235e6c33f24a1b9af56488b64abf07 : SUCCESS in 4m 39s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/254e78dbb5e24d6b91458bf8999a5fd1 : SUCCESS in 23m 34s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/568fe894ea484870ab451f2f54bf6c42 : SUCCESS in 29m 55s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/228ba8e9115148b88aa282f7ec54bb58 : SUCCESS in 1h 05m 23s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8b3b1a5b156248a6b1f5c1dddea5baf5 : FAILURE in 35m 21s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0b5c00b194c54dc4a9bdbe68f934c5a3 : SUCCESS in 1h 00m 14s","accounts_in_message":[],"_revision_number":1},{"id":"bbcfa920cb556e2ce4dd3bc31a1de016572c1570","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-02-13 15:43:57.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"efb2c982ff1bda269a5ccc6e86b29a62547b6149","revisions":{"efb2c982ff1bda269a5ccc6e86b29a62547b6149":{"kind":"REWORK","_number":1,"created":"2020-02-13 12:19:55.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/32/707632/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/32/707632/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/32/707632/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/32/707632/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/32/707632/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/32/707632/1"}}},"commit":{"parents":[{"commit":"cbfded882cb951c95d3832aae0b1c38711e4a654","subject":"Bump Gnocchi to 4.3.4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cbfded882cb951c95d3832aae0b1c38711e4a654"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-02-09 13:26:10.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-02-13 12:19:38.000000000","tz":60},"subject":"add Linaro repo with fixed libvirt","message":"add Linaro repo with fixed libvirt\n\nWe want to migrate Linaro Developer Cloud to train release. But this\nrequires fixing libvirt to work on ThunderX1 servers [1].\n\n1. https://bugs.linaro.org/show_bug.cgi?id\u003d3778\n\nThis change adds repository to default ones and pin it in a way that\neverything other than libvirt is ignored.\n\nmanual cherry-pick of https://review.opendev.org/#/c/706705/\n\nChange-Id: I56525cdc357e187ee6f3adae068a37cd013a9fce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/efb2c982ff1bda269a5ccc6e86b29a62547b6149"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/efb2c982ff1bda269a5ccc6e86b29a62547b6149"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
