)]}'
{"id":"openstack%2Fkolla~707787","triplet_id":"openstack%2Fkolla~master~Ie84ed771ffa9ecd70b945673727895c8edad30eb","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"Ie84ed771ffa9ecd70b945673727895c8edad30eb","subject":"kibana: enable for non-x86 on Debian/Ubuntu","status":"MERGED","created":"2020-02-14 09:02:04.000000000","updated":"2020-02-28 09:48:09.000000000","submitted":"2020-02-27 22:31:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"707787-1582842683057-d82724f5","meta_rev_id":"fdc6e37b61b15088c7cdd84abab9d7b49aceb6eb","_number":707787,"virtual_id_number":707787,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"date":"2020-02-20 10:07:43.000000000","_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"date":"2020-02-25 11:16:01.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-27 22:31:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-02-14 14:53:09.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2020-02-14 09:06:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2020-02-27 20:25:12.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":1,"date":"2020-02-27 20:25:12.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"date":"2020-02-17 06:04:48.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"date":"2020-02-17 06:04:40.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-14 09:06:03.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-02-14 14:53:09.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-17 06:04:40.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-02-17 06:04:48.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-02-20 10:07:43.000000000","updated_by":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"reviewer":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"state":"REVIEWER"},{"updated":"2020-02-27 20:25:12.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"},{"updated":"2020-02-27 22:31:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b19806a2ad0934038b7835d8fb81ca4261478c47","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-02-14 09:02:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9ff49aa2ee3104061653e084cd054838002a20b6","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-02-14 09:06:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"876f1afb7082d5b09f05c39d6753faa56b6bf2fb","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-14 09:31:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/6809e589abf24d7db959e6f6b50826ae : SUCCESS in 27m 17s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/4e3e32dc68e74995aaa877f6b298baa2 : FAILURE in 24m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"24db70d073691c7a0b7f5416c7927ee6a35364e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 10:33:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/5156fa05d8744fbc83ad3deee333a7f3 : SUCCESS in 29m 05s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/4eb19edadfed4bfba39d190c080c2c4a : SUCCESS in 53m 25s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/31fc089daf5d4a2a8caad129a1b74f5c : FAILURE in 7m 34s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/1f372a82df9a45baa81cadc5c6c0a33a : SUCCESS in 54m 02s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/0edcff20a01f4a26805ec93fbd39d19e : SUCCESS in 1h 23m 01s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/f5b1c6b86a704d4b9c9fb4af857bde03 : SUCCESS in 1h 29m 41s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/5be13fecc83e47aa8a3d1494c4bf0a2f : SUCCESS in 1h 28m 36s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/4cfba7651ada40d4b981eb9ff47fa1ad : FAILURE in 10m 26s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/e8e5662ad6e848d9a9f9d84e925999c6 : FAILURE in 11m 03s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/67f9b5921c4a4ca994980916b95c0979 : SUCCESS in 55m 36s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/992290b73f9e46629533237e1cb872f3 : SUCCESS in 49m 36s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/7c340f35de2f448285f47a8292d293ed : SUCCESS in 22m 25s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ddec5ff47f3d40e5a6411d1313b3c393 : SUCCESS in 1h 01m 52s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/17106c9799444d6f97005971a0e2946c : SUCCESS in 7m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be9e28db5c8947c1905c9ab83fa1fa9c : SUCCESS in 6m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c932d81711264ec5a54259b84b056f59 : SUCCESS in 7m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8fd08a2c62d840cd8724c47ddea0eb16 : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4983d04eda3940759e794dd544c00af2 : SUCCESS in 5m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2ee4d8b8dac4ced9ef520a37329e1ab : SUCCESS in 7m 28s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/803cd48f848147709f6d54adcfbc1a3f : SUCCESS in 25m 20s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/beb94126aa494bb7922193cecb5516ae : SUCCESS in 40m 08s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2fb3ba3daca3461382a3ab01945d2a9d : FAILURE in 10m 08s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5d8518e8463a4b159cd46cf85db0e6bf : SUCCESS in 1h 02m 03s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b97e8fa1ce694e6f8f167e0996bd2b1a : SUCCESS in 1h 03m 59s","accounts_in_message":[],"_revision_number":1},{"id":"41ffa63686296f6aae7dcc2a3e3ea77aca356c46","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 10:47:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/010cf618f3c248888eef82da36e2faa9 : RETRY_LIMIT in 19m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"96c6b47e91add00d05f03cbe5f8c68e44ca27de2","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-02-14 14:22:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a458fea642ea0d50c9f2196ecbd04c8dccca5a77","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-14 14:53:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/747fbc147d674491a49e2e859fc6742e : SUCCESS in 27m 21s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/3e083f38c3264f59be42e863c97fbba9 : SUCCESS in 30m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"170b78b503af0dde705e65f89952486b72caf8fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 16:03:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/5554486fb1354c208598dfb4a5334ff2 : SUCCESS in 47m 04s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/4e5db84bb8a048a7912815caf4168d01 : SUCCESS in 54m 30s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/51987218ea1b4d658baf3b853f4f4015 : SUCCESS in 41m 40s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/17f8d33b1c354c8cb427a27966743e8c : SUCCESS in 29m 24s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/ce712eddec4c465f92539f93b900d849 : SUCCESS in 1h 17m 08s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/322a63456c554f7792bca1fd76bd4fb1 : SUCCESS in 1h 25m 24s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/c01114120feb4634ba38e39f99b1a17c : SUCCESS in 1h 27m 25s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/0ad4ede218e14362a221dbdee1c9e6c4 : SUCCESS in 1h 13m 34s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/4c858290dc444b408cc640a0be319a5d : SUCCESS in 1h 40m 14s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/21222c765235435f88cad8529c957cb4 : SUCCESS in 56m 39s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/963d62b2af6242f78d31736a4438373c : SUCCESS in 22m 14s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/eca3e97ed75f419a86cd818739a35654 : SUCCESS in 37m 53s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a187fd0b13bf43718a71f8a296b06be4 : SUCCESS in 59m 58s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ad5558ef58614d57bcf5cecda60ac92e : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b9b797d34e24d4cb5a85f21d6c72907 : SUCCESS in 6m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/04817111c1be4eef87be68aff37a9db1 : SUCCESS in 5m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fe8fdd5e48f84774a226e0377a96657e : SUCCESS in 7m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/35c753dc2cf247098549d5c7baf071bf : SUCCESS in 6m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8b7d294d7b74d49921de3d60254828f : SUCCESS in 9m 07s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/17aac191d7d344139758dbd28d4ae782 : SUCCESS in 41m 16s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/61231f247fea47bbaf118cb239e74bbc : SUCCESS in 50m 41s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/249e23803ce0498da4fc9f16d41f5ada : SUCCESS in 1h 05m 04s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7023cadf0b7a4d14965d5c25a469c753 : SUCCESS in 59m 36s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2e6557baff7544a7a5ce84c558c2f6c3 : SUCCESS in 1h 08m 58s","accounts_in_message":[],"_revision_number":1},{"id":"f3b8dd6899aca3996d2d50217f7d270d812d2a51","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 16:06:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e12fde805e0e486f8351b981afbbe1b6 : RETRY_LIMIT in 20m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d2c5613cdf8cb9b3ab84bd3d332564d8b6253cb7","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2020-02-20 10:07:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ba9e7361f2e5c406472f6c33ef31bc5d0ee553fd","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-02-25 11:16:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f66c5ad2e9ef5a0b2f11ee550a14b773072533fa","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-02-27 20:25:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bafa773aeaf42a7d7b51f487235a56f8845bd2b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 20:25:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"86f6732429295562b39cf778bf3004c2e0f1c650","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 22:31:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/d8a2bb648f85444f957e9dca0f7cc54f : SUCCESS in 48m 35s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/827b6e8c15dc4a118feac0b45bdb16d3 : SUCCESS in 32m 36s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/2152194dec654e119b6902dbc3c5dd84 : SUCCESS in 1h 21m 10s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/2766e4a7abd042c29e98187c82950393 : SUCCESS in 1h 16m 23s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/efe021b14fb64b66b2acb126e341cf4b : SUCCESS in 1h 59m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bc8c87a740d443f6bf3b27e805fd0257 : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57548cf61cfe4cdc9431a2d36775fe2e : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/93877ab1a46a4c4ead97f6bc16bc547a : SUCCESS in 5m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fc82dfce512d4a2d9784dc499900daff : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71f4ccef96ca4ba5880d2d67afe444d0 : SUCCESS in 7m 18s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e654e4bf058648c3b0e6ec5db2ff8d4f : SUCCESS in 26m 32s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/90b385d90a0e4930a02d311a5febc168 : SUCCESS in 59m 39s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/25fae618756342aa80131722b3326cff : SUCCESS in 59m 54s","accounts_in_message":[],"_revision_number":1},{"id":"9e59ef1ae443f85cb99a9ba8394aba18afb0646e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 22:31:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"baedb89ad2802a463041ef9c7296d4539f93c3e3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 22:33:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e300404a64214e0eb6baf4ffa18bebc2 : SUCCESS in 1m 48s","accounts_in_message":[],"_revision_number":1},{"id":"74675297f172777066a9e1998cdaf7d652c6437e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-28 09:45:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fdc6e37b61b15088c7cdd84abab9d7b49aceb6eb","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-02-28 09:48:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b050ef63e81c5fdc210b0e99298b2738bad6b5d2","revisions":{"b050ef63e81c5fdc210b0e99298b2738bad6b5d2":{"kind":"REWORK","_number":1,"created":"2020-02-14 09:02:04.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/87/707787/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/87/707787/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/707787/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/707787/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/707787/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/87/707787/1"}}},"commit":{"parents":[{"commit":"063a46a7de4471504535499ca2a972a79e4ce5e5","subject":"Upgrade virtualenv in pre","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/063a46a7de4471504535499ca2a972a79e4ce5e5"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-02-14 09:00:57.000000000","tz":0},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-02-14 09:01:01.000000000","tz":0},"subject":"kibana: enable for non-x86 on Debian/Ubuntu","message":"kibana: enable for non-x86 on Debian/Ubuntu\n\nAs upstream generated \u0027all\u0027 package as \u0027amd64\u0027 one I took some time and\nrepackaged it.\n\nChange-Id: Ie84ed771ffa9ecd70b945673727895c8edad30eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b050ef63e81c5fdc210b0e99298b2738bad6b5d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b050ef63e81c5fdc210b0e99298b2738bad6b5d2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
