)]}'
{"id":"openstack%2Fkolla~707789","triplet_id":"openstack%2Fkolla~stable%2Ftrain~Ie84ed771ffa9ecd70b945673727895c8edad30eb","project":"openstack/kolla","branch":"stable/train","hashtags":[],"change_id":"Ie84ed771ffa9ecd70b945673727895c8edad30eb","subject":"kibana: enable for non-x86 on Debian/Ubuntu","status":"MERGED","created":"2020-02-14 09:13:37.000000000","updated":"2020-03-13 09:52:10.000000000","submitted":"2020-03-12 19:25:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"707789-1584041128029-eb552905","meta_rev_id":"eac2717c7ccfbb7412c9fd2000660c5531a32e26","_number":707789,"virtual_id_number":707789,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-12 19:25:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2020-03-12 15:34:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-03-12 15:34:57.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-03-12 15:35:00.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-19 10:05:17.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-03-04 09:48:11.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-03-12 15:35:00.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-03-12 19:25:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"76b8831c8b3e429402d0e77148718f6bb01710ca","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-02-14 09:13:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e58f099e075eb5e46fa211843a843f26ad1cf67d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 10:54:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/b1214ccd33e2418aacb59a8bae7ffe28 : SUCCESS in 50m 01s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/9fe800c9a6604022a7f243fc666a7d00 : SUCCESS in 47m 35s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/9d4ae55f2d6b4fe19882e93372ad75f2 : SUCCESS in 1h 24m 02s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/f3a79741856749018835512328e763bb : SUCCESS in 1h 30m 54s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/326e2114521d49b69210ebf3909aab5c : SUCCESS in 1h 39m 26s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/d4cca798df0b4954ad68e07aa1549447 : SUCCESS in 58m 23s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/07fbf24336e64893a21bb32337326a1e : SUCCESS in 29m 55s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/782a8a39a404475f8f617f412e35de5a : SUCCESS in 24m 22s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/11bf7b2ca7704749b2844ec9a713413a : SUCCESS in 1h 08m 02s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/63de9432e5494bca9afba16f539b26ad : SUCCESS in 6m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d375e1564a3c4ec48af4b33b3e7e20f2 : SUCCESS in 6m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/85a5a33fc30d4fa7af2070a091e1295e : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e27a778731604bdb87a5eb5c90571a03 : SUCCESS in 7m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4ee5f68d308c4f7b8172238c38b51e22 : SUCCESS in 7m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0c9390ea5ef4b4880ed6c94d9c038f5 : SUCCESS in 7m 09s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/40f88d891f6f4e8cad69b8ad36f6218a : SUCCESS in 27m 26s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4fe375b9fb2347d8b2e8d06c54abac07 : FAILURE in 34m 40s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f4e12edea2ca4d4c9f53ad473874bafe : SUCCESS in 1h 10m 23s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3577a04d2348416c99da44dc816b499b : SUCCESS in 1h 13m 42s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1379b8791062443f834d04a0b54cca75 : SUCCESS in 1h 17m 02s","accounts_in_message":[],"_revision_number":1},{"id":"a1ea7ef192fa086fc08fc2dcdc070bb5a5f9c212","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-02-14 16:01:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9a708ca0b6701d737f66f22d8b1abaea2a45d67f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 17:45:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/eb95f191a19b49668b4e749eedf038fe : SUCCESS in 1h 06m 28s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/7bccb795195749488fd3f0a4181e5384 : SUCCESS in 1h 17m 14s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/a9c644cf544c4de98477b0a437efc55f : SUCCESS in 1h 26m 05s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/85f4d0e364994c559526c20de5221cef : SUCCESS in 1h 22m 26s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/29009e134a3243c0a491907129089fbf : SUCCESS in 1h 39m 22s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/b033ff42345144c9a6923adf51f66cf0 : SUCCESS in 53m 25s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/4ce97a60a5d6417fb385062d768af1e0 : SUCCESS in 30m 11s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/7c0027d2ad404834853988e5730a3489 : SUCCESS in 29m 15s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/39dd1cc3a607428784ea9514223bc4e8 : SUCCESS in 1h 04m 51s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/181f4aa04dca4dc182c34139db5a92eb : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f5b22701c4242f0a45f6f549658c043 : SUCCESS in 5m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d575b880d10b4453bb7d3d00c720f524 : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e4a731aa58d4ef7b92fd4cbc1a130ae : SUCCESS in 5m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4b617fe4ab684900b042ceb352a85298 : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13a16fc4f9444c7ba30989060e047f11 : SUCCESS in 7m 59s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e33417ff661d4ca29a39c5c750080ccc : SUCCESS in 25m 16s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3574f83b3d634241b6acc9346ed31538 : SUCCESS in 31m 15s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a57c35b2b4e14aa2836e5de96020e0f1 : SUCCESS in 1h 08m 48s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/369b30238a3a4f1b902ce232544be5a7 : SUCCESS in 54m 28s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/cbbd175697224df8abaa6cca04bc88b1 : SUCCESS in 1h 08m 37s","accounts_in_message":[],"_revision_number":1},{"id":"a7e9c46b94b8ec2e964e345208d9c0ac40b23525","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-02-19 10:05:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f8e39a5ba4fef6edeafba179da6a31ed3f11bf00","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-04 09:48:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7fff9fd570494b228b65012b1bb365b23e24bd88","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 09:48:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5f9463ba6fcf46c3022faf00a1621808ce4d57c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 11:26:30.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/14d377023d3646cfb2477516284f3853 : SUCCESS in 56m 07s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/a998366373534d488be762e7c2adaf84 : SUCCESS in 28m 03s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/8de93fa75d6b4a00bb0be5c607a10bd0 : FAILURE in 7m 40s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/5b20c2776aed412e83d399a54a44ea5b : FAILURE in 7m 55s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/06e51488b5ad4da9b53143171c198ed9 : SUCCESS in 55m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3fb1f3c9a17349038f5d1e4b20779339 : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87fd5d3aa0db43ef904ecfae0c6cde6d : SUCCESS in 5m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6afd0782a51a43158a2f1eeed81335f8 : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/60cdae7a913c4fd68f27d897898c76c3 : SUCCESS in 6m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6d6dde83bd9340df92457cbed8530f4c : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58c5001685914cdf8c2c0270cae22d6c : SUCCESS in 6m 26s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a359bafe289d4f97b19002214dcac3a5 : SUCCESS in 29m 45s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/be078381f1e0477ea3c41866c5f0d485 : SUCCESS in 56m 49s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/06533efc7d6b4ef08ec92b2faf6f072a : SUCCESS in 58m 08s","accounts_in_message":[],"_revision_number":1},{"id":"a62aade57a0987cd2a86bba307312687a515c767","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-03-05 09:39:57.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"460f8ca8f799a53b840d3367447bb2aae811e32a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 11:09:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/7dd529f104ec4213b91541a6687eead2 : SUCCESS in 47m 55s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/8410aef448f64896a9f6523dc3336446 : SUCCESS in 55m 31s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/19633654f37540e0816e99ed0bcb3456 : SUCCESS in 22m 38s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/18f4f0640a424888bc6c236dbee79577 : SUCCESS in 31m 33s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/876e656acc6342d2b62f7a8a2279644c : FAILURE in 44m 01s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/89fb16283fc34e679dd342830fab6819 : FAILURE in 44m 41s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/094e42b324df434cbbb57fff3d74520e : FAILURE in 1h 25m 34s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/2ba2252343674804a3dd5027f4ba3dab : SUCCESS in 51m 19s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/5e82c1e8fde54cbe8a292798da9effde : SUCCESS in 28m 24s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/5adbb1dd409c443f807c7d42688d5e73 : SUCCESS in 20m 06s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5a8de3b1e1a44e71bd45f8c696eb1602 : FAILURE in 31m 13s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/79fdb28cc05f428a9cd2e74b29f2458d : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ecfd0297e574df38b07a755bf0aca10 : SUCCESS in 5m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ba031c1bd7424a1393f1ab6c0d91850d : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4fa082b254de407e86f43f1892277373 : SUCCESS in 4m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a81f147608c8487fa5852ea0ae94e58d : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/243a6ee550e54929807c8e869dc13304 : SUCCESS in 6m 02s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/37c7b2fc7e5a45efb833e054cb0b01bb : SUCCESS in 23m 41s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9bce1a85a8944b928514e3392908153b : SUCCESS in 33m 26s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5326c1b8219d43839f79fe3927169704 : FAILURE in 29m 12s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3d9c153407934574b0e5de28b0d8afa6 : FAILURE in 28m 11s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4927ae294729410ea465f9d9bab69b66 : FAILURE in 54m 11s","accounts_in_message":[],"_revision_number":1},{"id":"3e4b53a126313da98e2db5da72e26d33918687f7","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-12 15:34:57.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"e610407a4fe0ae2976eaca208e3dd0c9684cec9a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-12 15:35:00.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c05d38aa46de64e7ca6e5807f7d6e5e980851fe0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 17:31:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/913dae7018214660b1e50a484947c5b4 : SUCCESS in 43m 49s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/31a975e076884dc3a1878160e843aa18 : SUCCESS in 53m 35s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/bf49e171141741fdb89d496a59a927d3 : SUCCESS in 39m 34s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/880195003935449a86a17fe4be05cb2f : SUCCESS in 33m 32s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/853c0c37da814f1daedc4f0351084d94 : SUCCESS in 1h 31m 17s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/50a469e288f742a19c4a0cb78c72411d : SUCCESS in 1h 10m 12s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/ff43917de3c54aeba3c9879a09a00720 : SUCCESS in 1h 28m 12s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/b7d973b0d1184fd88799eee5768bcccd : FAILURE in 50m 45s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/c81386fe397942b39d81d35083d26f13 : SUCCESS in 1h 24m 40s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/be47c186b60a445a8e0ca9a64197ebd2 : SUCCESS in 1h 03m 11s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/febde578f0694494a4d585e6cb1cc36b : SUCCESS in 27m 51s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/69af69d37d524e579fb7a82354ec99e1 : SUCCESS in 20m 46s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/edf30a2b9e0b4a0492b290d46b8cc3f8 : SUCCESS in 1h 05m 41s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d5f61d04338b40dcb1f0b12faad5c17b : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c9beac90ce84623bff5f40ffdfdd615 : SUCCESS in 6m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ed709acd9d1740f5bc7b30ffdf6705c9 : SUCCESS in 6m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b2ec234d86284369843fc49316ca23da : SUCCESS in 6m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ed0d714795a84dd0853e152f28c24266 : SUCCESS in 7m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77928b86060d486abcdc4b2b63f4d417 : SUCCESS in 9m 20s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9caf9a32e8814b65bba8a5c230f04939 : SUCCESS in 31m 24s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6f01e7e0a26f4062a33ac2eef6a1998c : SUCCESS in 37m 30s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d57f5e062ca64001ac4d1b0680f365f7 : SUCCESS in 1h 01m 07s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/21398303e4d6427198e8475b6e55df87 : SUCCESS in 54m 01s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b79b7e6964ea4e43a58c84918a14d3b2 : SUCCESS in 59m 54s","accounts_in_message":[],"_revision_number":2},{"id":"83a2c9046b8cc3747e75493d438ef966aacdc17c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 17:41:44.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e5d6ba2fef38df6f8ebc860268e564bfa62b0fc3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 19:25:27.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/3079b1e6f2304fd2b0ea369be7eb8cee : SUCCESS in 1h 08m 11s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/434efe95b72a4c89b0d267789e054569 : SUCCESS in 49m 20s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/09e0fc52f216417ab13dc7d291485094 : SUCCESS in 1h 27m 59s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/ee94111c296145bb85a3bf4f342f40a2 : SUCCESS in 1h 08m 13s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/de978f2527c64b8e8ac20f7fb1fa0340 : SUCCESS in 1h 37m 37s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/a5c3697992bb449ea32e1780ab11eb19 : SUCCESS in 50m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3dbbb6e2fe9642a7bff2a9ff01bea153 : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d260d9efb5b404bad2ae04979dc3524 : SUCCESS in 4m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e6532a9ec4d942f09e3f93513ef3c2a7 : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d52895abaef84c6ab1ac725bab698975 : SUCCESS in 3m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/46f4956263564d5f842c404318e31d34 : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5269c077a564904b8fb3f5d5b5f9e07 : SUCCESS in 6m 13s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/eb239b43541e4b86a865241aeea45a61 : SUCCESS in 47m 56s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/129a0a6b602447d285219b6cd9ff9e20 : SUCCESS in 58m 39s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/579b4bd44a134a0890cc05720b5f8faf : SUCCESS in 1h 03m 58s","accounts_in_message":[],"_revision_number":2},{"id":"7d43634192d5b450f990b76f671d89cea4f4e1c8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 19:25:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"156ef6be85f9d14ed1c990ee93e38d85db9b7630","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 19:27:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43e09fbb69964a948cd6f04dd39c88a8 : SUCCESS in 1m 49s","accounts_in_message":[],"_revision_number":2},{"id":"eac2717c7ccfbb7412c9fd2000660c5531a32e26","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-03-13 09:52:10.000000000","message":"Patch Set 2:\n\nHave to improve that patch as we just got kibana package updated...","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"331b3903b9b1177fdadd0ae6ad6f14282a85dd71","revisions":{"543b9c7ddfd93098562cf55169f245fb859a5bbd":{"kind":"REWORK","_number":1,"created":"2020-02-14 09:13:37.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/89/707789/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/89/707789/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/707789/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/707789/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/707789/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/89/707789/1"}}},"commit":{"parents":[{"commit":"9f7c7b56845a5920d7335949ca0d7562e8391a9f","subject":"Change fetch_fernet_tokens.py behaviour to check for minimum number of tokens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9f7c7b56845a5920d7335949ca0d7562e8391a9f"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-02-14 09:00:57.000000000","tz":0},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-02-14 09:12:48.000000000","tz":60},"subject":"kibana: enable for non-x86 on Debian/Ubuntu","message":"kibana: enable for non-x86 on Debian/Ubuntu\n\nAs upstream generated \u0027all\u0027 package as \u0027amd64\u0027 one I took some time and\nrepackaged it.\n\nmanual cherry-pick from https://review.opendev.org/#/c/707787/\n\nChange-Id: Ie84ed771ffa9ecd70b945673727895c8edad30eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/543b9c7ddfd93098562cf55169f245fb859a5bbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/543b9c7ddfd93098562cf55169f245fb859a5bbd"}]},"branch":"refs/heads/stable/train"},"331b3903b9b1177fdadd0ae6ad6f14282a85dd71":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-03-12 15:34:57.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/89/707789/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/89/707789/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/707789/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/707789/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/707789/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/89/707789/2"}}},"commit":{"parents":[{"commit":"ca8b61516401d61d90e2940fa130c8f55a8b6e45","subject":"openstack-base: pin setuptools \u003c 46 to get horizon working","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ca8b61516401d61d90e2940fa130c8f55a8b6e45"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-02-14 09:00:57.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-03-12 15:34:57.000000000","tz":0},"subject":"kibana: enable for non-x86 on Debian/Ubuntu","message":"kibana: enable for non-x86 on Debian/Ubuntu\n\nAs upstream generated \u0027all\u0027 package as \u0027amd64\u0027 one I took some time and\nrepackaged it.\n\nmanual cherry-pick from https://review.opendev.org/#/c/707787/\n\nChange-Id: Ie84ed771ffa9ecd70b945673727895c8edad30eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/331b3903b9b1177fdadd0ae6ad6f14282a85dd71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/331b3903b9b1177fdadd0ae6ad6f14282a85dd71"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
