)]}'
{"id":"openstack%2Fkolla~708273","triplet_id":"openstack%2Fkolla~master~Ic82c59a5e78078b4fea10df9d30b35da14cad922","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"Ic82c59a5e78078b4fea10df9d30b35da14cad922","subject":"Zun: add zun-cni-daemon image","status":"MERGED","created":"2020-02-17 23:07:04.000000000","updated":"2020-03-23 10:18:28.000000000","submitted":"2020-03-23 10:16:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"708273-1584958600796-577aa814","meta_rev_id":"50ce57da1b35160d7d9f9abe2171675025a0b55c","_number":708273,"virtual_id_number":708273,"owner":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-23 10:16:40.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-03-21 18:33:48.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"date":"2020-03-21 16:56:11.000000000","_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"all":[{"value":0,"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":2,"date":"2020-03-23 01:36:59.000000000","_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":2,"date":"2020-03-22 17:24:21.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":1,"date":"2020-03-23 08:35:22.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-02 12:26:16.000000000","updated_by":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"reviewer":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"state":"REVIEWER"},{"updated":"2020-03-20 22:29:59.000000000","updated_by":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"reviewer":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"state":"REVIEWER"},{"updated":"2020-03-20 22:29:59.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-03-20 22:29:59.000000000","updated_by":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"reviewer":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"state":"REVIEWER"},{"updated":"2020-03-20 22:29:59.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-03-20 22:29:59.000000000","updated_by":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"reviewer":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"state":"REVIEWER"},{"updated":"2020-03-20 22:29:59.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-03-20 22:29:59.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"},{"updated":"2020-03-21 18:33:48.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-23 01:36:59.000000000","updated_by":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"reviewer":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"state":"REVIEWER"},{"updated":"2020-03-23 08:35:22.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-03-23 10:16:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"12648e160bb5de73b7aeb3fe34323dc3faf14930","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-02-17 23:07:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"59f888541d637a536bcf2a1fd1e06c5a05cc7ab6","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-02-17 23:07:33.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"160ebb7c19c56ba19f1c6d693344c70e43e3672d","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-02-17 23:07:56.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"7b360ed078ae94f1b7bc9f9f4b3a96bf420f894a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-17 23:57:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/514a8ddf4bb040a9a8cf2702f403259a : SUCCESS in 26m 54s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/9b8dae055faf4c9889ffc3b994ad731e : SUCCESS in 28m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0bbd1b78de6b38b2553396d4c12d60b1ae7ee48b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 01:25:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/58ed1460fe534444a6b729aff0c62c33 : SUCCESS in 34m 13s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/32a61a1c4c5a4e6dbcca076df168cb2b : SUCCESS in 41m 26s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/30b5b755b7944d4f9bf97911c785fc2b : SUCCESS in 40m 18s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/297b7dcd56d6492abb4619b2475c0de2 : SUCCESS in 31m 56s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/cd17cdd10e534d0a949c8e0791773a63 : SUCCESS in 1h 30m 16s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/c89d56da74634c87bac71d98f81b9955 : SUCCESS in 1h 15m 27s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/35f7bb28e9684ff4bf5d72169d7812d5 : SUCCESS in 1h 26m 09s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/d3c32af3f71d43d3ae37e0cea71d2f41 : SUCCESS in 1h 20m 07s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/44f06b0ca51b495ab8e95735af67ffe1 : SUCCESS in 1h 44m 55s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/30a47d3e8a1d4563876a0b73c52ee0ee : SUCCESS in 48m 24s\n- kolla-ansible-centos-source-zun https://zuul.opendev.org/t/openstack/build/6c0ca20465734941add244b9e4e497b4 : TIMED_OUT in 2h 04m 18s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/df865e400c54413c96126c0361b04d43 : TIMED_OUT in 2h 03m 09s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/62353ac4e71147baad131dd5ddab998d : SUCCESS in 41m 37s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/88d4a8daa2174e5cb62befb075f05f10 : SUCCESS in 36m 18s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/29d30527418241009699fe0ed4fab133 : SUCCESS in 1h 00m 59s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8ad0145018d6417da43cf9900601f162 : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a822ec84f744847abbbaf758021e8e0 : SUCCESS in 6m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/54f49e194e3d477d94b3a807c8b33338 : SUCCESS in 5m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cd622e741e6245e88b6b45a4a10af35d : SUCCESS in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f64b4241e084c55a912bd3fac52be37 : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9b3037616824b059027baa93b5983d7 : SUCCESS in 7m 03s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/77a5640376fe4ea493655572ae9f380a : SUCCESS in 26m 39s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f05640a88aa24e4eb4753193a7ee76ca : SUCCESS in 25m 08s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2d3f29a8fbc64f498e462a5b703c4133 : SUCCESS in 55m 56s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a11b3a0df529494e8b039ef084f500a7 : SUCCESS in 54m 38s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/584f1d2a8d7d43d6a331cf6dd578836b : SUCCESS in 1h 13m 40s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/6d6e5eb53d0544f9b0019f09f5bd27ca : SUCCESS in 1h 31m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"db1a83a5d4b6554e4dbc097bbfb7ac11e87c8a83","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 01:36:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9751ce1f9319467fae72821f6b4da4e7 : RETRY_LIMIT in 18m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"386d966a9e61705aeae47cbc1d2c106fecd21912","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-02-18 03:12:16.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"eeaddabb2fc09ff12defc4e7427de67a85b65aeb","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-18 03:44:07.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/91744e299d0541458b738f7ee87e6e37 : SUCCESS in 25m 52s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/ac132bc64f0345ee9e9a8c53ae447001 : SUCCESS in 30m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f4be224ef400ec069b4aa15b14c54271baa2d6cb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 04:56:51.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"eec4632b6e927945e4c7ce69fa8fa88f89985d6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 05:04:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/76bf2e1d2f9c4427944950ca78fea113 : SUCCESS in 32m 03s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/257bf7bdb57247fca7307706b5a5e0f8 : SUCCESS in 41m 33s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/6468b4293ae043d2a4c4960e0c774175 : FAILURE in 10m 19s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/337022156c7649059b94d6f18dd107ba : SUCCESS in 33m 33s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/1343d18a674e4eb78ec85bfdf1b0a43d : SUCCESS in 1h 28m 05s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/46dc3cedf29e4f7fa58c43ea8da0c857 : SUCCESS in 1h 11m 41s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/4fb6f72228b543ddbfb1bad4600b2781 : SUCCESS in 1h 13m 06s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/5d18af66a76f4ca18acdd6279fd235a2 : SUCCESS in 1h 08m 35s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/cf06a98025b5471eb0857df71d16eb34 : SUCCESS in 1h 31m 11s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/b7766e8f8d7f40b69cd469f32c855ed9 : SUCCESS in 51m 20s\n- kolla-ansible-centos-source-zun https://zuul.opendev.org/t/openstack/build/71a1d3cd0dff43f6aebbdff9975d45f8 : SUCCESS in 1h 49m 29s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/a067980936e34c1ba77fb6854f22c1ef : SUCCESS in 1h 39m 34s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b088a9bebca34cb187c67913d4909760 : SUCCESS in 21m 58s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/08180447d27640f4b9becfd588b831e6 : SUCCESS in 19m 40s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/855f384cf94c436cab8eeac9b1241e57 : SUCCESS in 1h 04m 59s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fd261965b9b945668748bf8cd8f2da6b : SUCCESS in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f49963da177b48c7a81e05f79acd7e0d : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/834518277de846b78e35ac4439679830 : SUCCESS in 5m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/68c343b2474f479f8fe05117ff42a3f7 : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/59514c1e2def463e8e5598f324968f73 : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bf21f07a4ed475bb1a5b9bc33fb99ed : SUCCESS in 7m 12s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ddc8af2af11a40a8a4f6f949efb06103 : SUCCESS in 40m 27s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/fee660c7754745248a6182e89deda877 : SUCCESS in 37m 32s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a294031c9b544505ac21f7400585153e : SUCCESS in 1h 05m 40s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/21bd916c450a4a35bc64bce621a03f9c : SUCCESS in 55m 01s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9693c19b659b4750b443f6f74dff57d3 : SUCCESS in 1h 04m 14s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/96362cb375c4406e879e0004ff41fadf : SUCCESS in 1h 23m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7108fcde4c21f5b73454b11f38bf5f1a3b9099b5","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-02-18 05:11:13.000000000","message":"Patch Set 3: -Workflow","accounts_in_message":[],"_revision_number":3},{"id":"af9b1fbdb24624c003f6cf98cc3a943bbc73106f","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-02-20 02:13:21.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"44e011d168e8e4b510609e97e0036959ce79e7bb","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-20 02:45:52.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/1108ddce86254a82bff8c4e647f8d41e : SUCCESS in 24m 39s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/8f68a709187d41da846c30b2737a2651 : SUCCESS in 31m 07s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"0bbf023d729c83bd11c79423a0c25800150dbab8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 04:17:14.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/0cccfaa4f8574d08b81e71157cf1bcfe : SUCCESS in 30m 57s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/20cbc198463f499a94fda84d111a56ec : SUCCESS in 43m 23s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/e4bdf34a786449b79a259ca08ceefc66 : SUCCESS in 24m 41s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/b56146ae36884b91b4482b9df1f9f9f4 : SUCCESS in 27m 47s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/d8b289c4321f43849c051c7e01075246 : SUCCESS in 1h 25m 14s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/01738e44df074ceb8105de93f9c45694 : SUCCESS in 1h 14m 24s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/e47bc6406a0d40e49175100685aa945b : SUCCESS in 1h 41m 59s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/174b9b6cc7ed4b388e4ad89a9b1dd7a6 : SUCCESS in 1h 09m 46s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/c764c634cf3c4b71bf69557841589771 : SUCCESS in 1h 41m 06s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/acfff5bbe5574c42b9505aa5078b1dd4 : SUCCESS in 35m 09s (non-voting)\n- kolla-ansible-centos-source-zun https://zuul.opendev.org/t/openstack/build/195719e9c17a4f2683a9253d033d48fe : SUCCESS in 1h 59m 04s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/c943ca0211e7427e892ba4dd941c0ec1 : SUCCESS in 1h 52m 51s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/8f22c7ccacf44796bdb42628a6cfb2b1 : SUCCESS in 22m 18s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/3394f3d7af6f497bad32f1cb208eb149 : SUCCESS in 19m 04s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a4ce3fd57d734c6fadd856612ffbca8d : FAILURE in 53m 19s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1ed3ab53829148e7aefb95e00f998e76 : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f813f3efa2894050a3c8faac7572e87a : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3459f0132734d16a8dfbeb81c4e85d3 : SUCCESS in 4m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4afed2c0c7ac4eb683666936c6dd915d : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1a36f10580b145958551dcd2e18fbebb : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3d4fc9749954a3f9e82e1de6bbeb0f6 : SUCCESS in 5m 51s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ce87b7e538f3491fa7cc1a4fc4f42e13 : SUCCESS in 23m 39s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f7a7ed8b486d407097604c8979e04433 : SUCCESS in 32m 29s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2558efdd75c64aabbee1a43b105a58b9 : SUCCESS in 1h 02m 37s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fce9d1ff69b546f1938f9ea5f0b52d16 : SUCCESS in 1h 03m 29s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0cd385ecbdb1490c98387e76eff0f582 : SUCCESS in 1h 00m 27s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/51fbaa2dc2bd442c9b2440f36e7595fd : SUCCESS in 1h 23m 50s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f8e7442f0db6f13dd0a4dfad7c61a33f3c185566","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 05:15:03.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1e17f81c8d124e98920c483073db4833 : TIMED_OUT in 3h 01m 11s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d472ac237e6751cb4409b42efa81cd2767497252","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-02-23 17:15:07.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"65867d962ce824613ab67b1363b2d454c0fc8712","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-23 17:47:14.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/23d82ed0a2704c79a8a20b4c3b3bec93 : SUCCESS in 31m 12s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/61a69147a0524aa7ac8e57cd0264a560 : FAILURE in 15m 55s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a9f7f07409eb9f1b4e9e6b646d3d058f4e036dfe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-23 19:21:18.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/2c94aed9c0a746dcb6a8cbcb23cabf4b : SUCCESS in 47m 13s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/7ad210c4fea247fb804949137a459de5 : SUCCESS in 51m 22s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/a819dde63a8b46138bb7b1225ea0ec35 : SUCCESS in 30m 51s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/7e2e6845fc95411fbcd3a1f4434cfd63 : SUCCESS in 41m 27s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/d1c0506ab7234cce8f97337aeeedd035 : SUCCESS in 1h 22m 25s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/c885a222d2e6461f8ff40381096e56e4 : SUCCESS in 1h 25m 51s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/a4b912085a384f1fb57cf0eee1680c59 : SUCCESS in 1h 21m 52s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/571613ea4acb4933bdeb154ca1a5e103 : SUCCESS in 1h 27m 09s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/aa8a5ee99de04ae887d569dae8209018 : SUCCESS in 1h 31m 06s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/7db0dca1e25f493ebbb5ea4d75c4420c : FAILURE in 21m 34s (non-voting)\n- kolla-ansible-centos-source-zun https://zuul.opendev.org/t/openstack/build/5a84aea8250c400983b57f1156472be8 : SUCCESS in 2h 03m 54s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/7f1330be5bfe42fdb9c464c082918312 : SUCCESS in 1h 46m 28s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/5aabb3ab27bb4d23a4eeed01e24a7c72 : SUCCESS in 22m 19s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/4d24ad6a205144749f77656d8d1206fd : SUCCESS in 19m 20s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/da11a5d957554af58ee2b4cf22ad5b4b : SUCCESS in 1h 19m 57s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4f6b16f932284f3a9ee091daecdc176e : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d9fc848896c45cfadb4e14a1f9411ec : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/11ff0802b934429bb24747a351467fa2 : SUCCESS in 3m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dbf87a5f65824a21a8db5bd6b2a6cf13 : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c2e37d0b01e24150ae9468ee708262a1 : SUCCESS in 4m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93e3301ab39143358a6523d031cb064c : SUCCESS in 6m 36s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/19e780d892b04efa9677b5bf9bb3a10e : SUCCESS in 23m 29s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5e0f2610a5f94f5c96fff5087b09c65c : SUCCESS in 24m 40s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1c0fab8f54854499a7002d36acd0ef4a : SUCCESS in 1h 00m 06s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b045e1858fb8419b8da5a054fc5ff949 : SUCCESS in 58m 32s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5aa8380bcc20452a90f8907f7085c9e4 : SUCCESS in 1h 07m 51s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/d7f86e7d770349cb963bf103aa1525b2 : SUCCESS in 1h 33m 57s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"44e595c581fdd230ebf3bed551bd32aee9d302ce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-23 20:44:20.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a93d79711af249ec95c7194296f60821 : TIMED_OUT in 3h 01m 00s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d91bb16dc5537104ad7a0074c5e60ddc3ce9a677","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-02-24 03:43:34.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"54839c59121f92d6b106de03ef3d1c87c66cacab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 03:50:12.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e6bc4cd7b5b3487795ed41abfd1628e9 : FAILURE in 3m 41s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"367a7fcd1376bd3607ab2d85809b293a2c3ca46f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-24 03:52:38.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/9c6ef67a04b5428b8900c5fbce3561a2 : FAILURE in 6m 35s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/d54bd1efddc945eabe79607007c8d60e : FAILURE in 8m 01s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"6f6dbf62e8193f0440a9c58425230199a07cf97c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 04:03:12.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/c5c07ffb7bde4e0f86eef9487da05945 : FAILURE in 5m 53s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/2f5d5290c7da4085837db253e620d64a : FAILURE in 6m 00s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/291095c230044f2f970aa15fd4ab1554 : FAILURE in 5m 34s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/a48e349fe3f94e259b216afc78e50050 : FAILURE in 5m 40s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/b38d059154b14a0c957b0d7e11e95000 : FAILURE in 10m 08s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/1ca698ade6e642f1879a1ecd683c169b : FAILURE in 10m 14s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/9b6830dc1ae34c4ca95ff3fe6584e4a0 : FAILURE in 9m 25s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/2bd7d093149a4a1ea8d95971d9b20c43 : FAILURE in 10m 09s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/8f6d9c652db64466a8d9ee4c75a18f93 : FAILURE in 9m 41s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/d5b3fd62011e4f5fbb085469f55f6e0c : FAILURE in 14m 17s (non-voting)\n- kolla-ansible-centos-source-zun https://zuul.opendev.org/t/openstack/build/8bb812d5ea3249dd9f45414490d441ec : FAILURE in 13m 10s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/e98a2eba401347a89fe82707f34dadde : FAILURE in 13m 37s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b575afd3794841cd8d1b1a2335eeea6a : FAILURE in 5m 33s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/4d683b9b89a246008cced3c516d141a5 : FAILURE in 3m 46s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/bc0ee1060727425592297f9eee4f4319 : FAILURE in 8m 08s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9883d4304cc142849693e09b55b254ab : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89bb94ebf59f4a7882b5d1e7118232e3 : FAILURE in 3m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e98add89483d46ff9ce4e58504468c8e : FAILURE in 3m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/633bc7c6dee04136b93f2d4b05ea9021 : FAILURE in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ab21febcb9294f44b307ed79c0f51bee : FAILURE in 3m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af12480d00c3477996df2ed3a3bd27de : FAILURE in 4m 47s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f74b381c305f491ba55e1d1267fc8dab : FAILURE in 3m 39s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/05e38009298f410282d210010467d345 : FAILURE in 3m 53s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/942892137f654de297e26984aa5dd471 : FAILURE in 8m 13s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/605d8ebb298f4d95abfdd43fc9bf3cdb : FAILURE in 8m 34s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ef5d89ed424640b59db029447de0b0cf : FAILURE in 9m 06s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/d6c7f2aa3d89400b9c5a4ee121e2319c : FAILURE in 11m 48s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"db9cc574250278910d4ff78e38659bb63a985c26","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-02-24 04:30:28.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"8d2ea855b152b80e27eae08b52cbf7b9cbd839a8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 04:37:13.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/532c5cb7f1dc49eea8b603d0cae43355 : FAILURE in 3m 58s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"b4bfeee89ff937adde9f70347474d8be2e278159","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-24 04:38:25.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/1bfc4f7f3fba4bef9764ab564fdcf822 : FAILURE in 6m 20s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/6291cab1a9084487965a9dc3e61a3a57 : FAILURE in 7m 10s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"80a1bc8a7ee3f1eb449afc7d43b25d54ee2f2c0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 04:49:38.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/a6ca23fcbd1548c2890c707f552aef53 : FAILURE in 5m 18s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/54716e86ff97489896b6d4743ed91780 : FAILURE in 5m 17s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/0f409a2c447c4782b395c1f974edfcb8 : FAILURE in 5m 20s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/c3018d95e9134747afa6121f0f6380c3 : FAILURE in 5m 26s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/31ab6431697b438bac7e804ae1ac62d8 : FAILURE in 9m 49s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/200586c849d74170aef93b52d257ac32 : FAILURE in 9m 57s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/1740581322b74b8ca77882f6627090f7 : FAILURE in 9m 25s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/b4fca19566604b52ba508d6835126ca7 : FAILURE in 10m 25s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/18012f32b9d24119bd28cb4e197d69f4 : FAILURE in 9m 16s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/c87c6a8a1e1f46359477f7e08dbf8841 : FAILURE in 12m 00s (non-voting)\n- kolla-ansible-centos-source-zun https://zuul.opendev.org/t/openstack/build/68f60274d36c4454a2182e842aec3fc1 : FAILURE in 14m 18s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/2fc5dbcbebec413aab172c523528903e : FAILURE in 14m 04s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/09c4cea45fcb4153a23e78f6440ca214 : FAILURE in 3m 49s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0d05b95cf49c4905904f56dd8e2ba6d1 : FAILURE in 3m 58s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/fac9a5519cb34e69acea66f8332027e9 : FAILURE in 8m 35s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b321dc4387b948ceac7f8c57d8c7b30e : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c35a868947f43009ff006636cbe06c0 : FAILURE in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8cafa14fa23d410898be42673ae71329 : FAILURE in 3m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/262d68f2c549480c929d3e97b34e3f1b : FAILURE in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/228db387d62643d2a9b13c069a133613 : FAILURE in 3m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ff2de3c9ae246cf934692484d979ad0 : FAILURE in 5m 15s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4719fc366e94491b88c84e8a9d8cad29 : FAILURE in 3m 55s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3cc29547911c4321bb16da6c814df6a1 : FAILURE in 3m 59s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1f8acd73d25442aba0147cadce81db43 : FAILURE in 8m 36s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3783280e312545fda72598264e15398b : FAILURE in 8m 07s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2fbcdafd8bfc48cfb974ad6fedcb4800 : FAILURE in 10m 30s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/bcd46c89e7194876a3b7192c3a8dd41c : FAILURE in 11m 13s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"92f5b2dcc7bec3c167fe131860b1a50fa2574cf1","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-02-24 05:02:03.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"6cae973fe547d2092b6a5fe18eb4674a43069b19","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 05:08:48.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ef089aa9697842d8b4880052258c07f8 : FAILURE in 3m 38s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"c348b36aeab605a8306b1858dcd82d644ffcb301","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-24 05:14:21.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/171d73a82b844376b013457c0b88ac65 : FAILURE in 6m 13s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/6e7462bb5cef4ae3a7f6b068eda733ed : FAILURE in 9m 35s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"140f4045d55db98088424d0ea73b9ac19795338a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 05:20:41.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/27c9713b27824a34b7e49faf1073ebc1 : FAILURE in 5m 23s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/fe01016aa9124a04bd681c5a91c8a01f : FAILURE in 5m 39s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/83ad29261c9c475a9f0475f3146d61c6 : FAILURE in 5m 36s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/41a8bdbccf024bb09b3421c03124f008 : FAILURE in 5m 34s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/7abbfb8fb45a40b8bb714e1530840eb1 : FAILURE in 10m 05s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/a68c0252a5c94c78a761a67d0ac0c525 : FAILURE in 10m 18s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/797f4b94c8f643b5b8a47ed8aaf3ac0b : FAILURE in 9m 26s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/4a452be2066445cb8aa6baf08866c885 : FAILURE in 10m 17s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/8f4a051a1374440a99717bb539608be6 : FAILURE in 9m 45s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/c5c0af23fa604e50951e2eb1b1945c71 : FAILURE in 11m 42s (non-voting)\n- kolla-ansible-centos-source-zun https://zuul.opendev.org/t/openstack/build/3209ad40c5694343bc0e79b56aad6393 : FAILURE in 12m 53s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/927f063039a840e8aa5b5bd3951e3ba1 : FAILURE in 14m 34s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6b5c9ce9a9f148a9a968d81f1728155e : FAILURE in 5m 32s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/53aa62b77d504a6886de493a2bea2cb8 : FAILURE in 4m 15s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e39005fa81ed429a8a4bec06b05a6c5d : FAILURE in 8m 48s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/91a53247934041d995d7b48452780bfe : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26b8e97b8cd74542bf3506d74515950e : FAILURE in 3m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b13a2c943bbb46d29a65f26b756e1c2a : FAILURE in 3m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e8a7673e20424ce780ec05af2e7b32c2 : FAILURE in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/877e916946974416a5a15c3da58bfa58 : FAILURE in 3m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/186b4d4c068e4c3ba0cd074092091b7c : FAILURE in 4m 23s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f03f5b15acf648c7aab33f47e0804130 : FAILURE in 3m 51s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f5a19c0cbd444ba78ec5c639b2d7a303 : FAILURE in 3m 55s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e7a71b222e6f4eceba493f11278247bd : FAILURE in 7m 49s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/50297e8faaf54b1c96c97432b7fc4f84 : FAILURE in 8m 44s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5084cf1710bd4111a98b6e5374f8e6b5 : FAILURE in 7m 38s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/6880eb34ca3c46a5ae72e3bef616a08b : FAILURE in 11m 51s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"be6e711e9a4d0da35da80d93b52255c76d374796","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-02-24 20:49:11.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"c673af8cfc54d63a608b3e4a2ab8c40656165c3d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-24 22:46:54.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/e1ac36300e7b4e55832745cad0507851 : SUCCESS in 40m 44s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/03bfc9325ff746e7bd3a67cb09769b62 : SUCCESS in 35m 06s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"b8ed29665bbf28786c8b6236ab482b3ac99a8cc2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 00:25:57.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/43ebaf64033a4af780a3eee08a5cf443 : SUCCESS in 41m 02s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/f1a27c8c69aa40fc8dacd57deff014e0 : SUCCESS in 43m 04s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/fe287c88c2014ffc9f322b5ab7eda482 : SUCCESS in 32m 11s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/abcb6b99a2184914be6a588caf1dbcbe : SUCCESS in 36m 58s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/b58ccd4c62b54fd7933b07ddf1bd63de : SUCCESS in 1h 33m 10s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/198f99b75c234bca866fd86fdbb141bf : SUCCESS in 1h 46m 49s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/1897b9276c6d45c0b9347d6d6d2c038c : FAILURE in 1h 01m 01s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/4265bb5fed204465b75831974514bf37 : SUCCESS in 1h 21m 29s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/9e7b67305bca485ab4fa4f63f67ecb19 : SUCCESS in 2h 09m 56s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/55bda4d5b6f1419f80ea3a3becaf7b99 : SUCCESS in 44m 29s (non-voting)\n- kolla-ansible-centos-source-zun https://zuul.opendev.org/t/openstack/build/f9946587f97348c5a971640bd7c73ef2 : TIMED_OUT in 2h 15m 15s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/397e301da5264690a412da0d9f219ae7 : TIMED_OUT in 2h 09m 50s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2eac2a064af340479155c5bb600f6234 : SUCCESS in 42m 58s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/74eca1eaa0634ffb8fb4f992c26000ce : SUCCESS in 41m 32s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/92a31700c9cf47e2b5d8212d203f13b8 : SUCCESS in 57m 12s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/697f8939da2f43ad8a8c42291a3b288f : SUCCESS in 9m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ae4b68ec93944509d67ed9c7c327539 : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ee1e933faed54a28be2745a4db33c7c8 : SUCCESS in 4m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/096fd48236f445ac938160e16f91163d : SUCCESS in 7m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f4d83c9c39e34ad88ff118b8762b7347 : SUCCESS in 12m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b73289f7c24d428bbab419c3d865a94d : SUCCESS in 6m 01s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/641ad911d60848238b11854e3d76823a : SUCCESS in 25m 22s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/be15d915ac284c4a81113e4da2229370 : SUCCESS in 44m 28s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d6385b64f4194c08941af686f82b0a74 : SUCCESS in 1h 37m 30s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4313999297db4764bc49d5707feda313 : SUCCESS in 1h 04m 32s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3c5cb991f5c04452894f0ced1d015b53 : SUCCESS in 1h 23m 07s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/ce03e541845043f3b6a52750fc345e08 : TIMED_OUT in 2h 12m 45s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"9e62286e626f3a22ae08419d2e68ac3fa4fe9ae9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 01:41:15.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/34e2112b85c140b08d913ef27f09ef28 : TIMED_OUT in 3h 01m 12s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"803aa7899a7564f1b868d99b2ec4647fdb8232bf","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2020-02-25 10:20:15.000000000","message":"Patch Set 8:\n\nPlease add a release note.","accounts_in_message":[],"_revision_number":8},{"id":"35aef77f876ac235be25a15e2759479b042906cf","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-02-26 03:48:19.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"57fd40723a875194e6e83163778e6bb503d4c827","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-02-26 03:53:11.000000000","message":"Patch Set 9:\n\n\u003e Please add a release note.\n\nDone","accounts_in_message":[],"_revision_number":9},{"id":"aca6b2e98176372dd55e031a59da4daa972af5dc","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-26 04:41:17.000000000","message":"Patch Set 9:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/e40fe06242294c5d9a868acdb5df0546 : SUCCESS in 30m 45s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/924c52be68af4a59a601683c22731a67 : SUCCESS in 36m 32s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"8348a57d96229e0f4f344c3c2738c4a54300b2c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 06:08:13.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/3e8dc49965804010b46cbaceca6a3020 : SUCCESS in 34m 14s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/d5272e86bf514b90bc1f61d2c388abff : SUCCESS in 44m 47s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/239eda297a4342a0ad95601545e0f04f : SUCCESS in 25m 29s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/a26eeb4dacf14073b0f44163e878de0a : SUCCESS in 30m 17s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/21f54b83d8ff429ea597e369f3bfbd60 : SUCCESS in 1h 32m 50s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/6834693b07e34ae6bc4ef5807a6bbdd7 : SUCCESS in 1h 13m 39s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/7c399c4ef1634694bc711aee387b2436 : SUCCESS in 1h 27m 21s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/62f48ac117cc4063be886c03e9aa9366 : SUCCESS in 1h 13m 41s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/72121aaeb06d4cdba05cb04ad0ca1ca4 : SUCCESS in 1h 43m 43s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/49f5135160d6475db06c13f91a80312d : SUCCESS in 51m 20s (non-voting)\n- kolla-ansible-centos-source-zun https://zuul.opendev.org/t/openstack/build/693d348ee86f42d09a4307fceef7efbe : TIMED_OUT in 2h 14m 37s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/53c0e6f26717458d94e23cc39c149f32 : SUCCESS in 2h 09m 45s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/c7219d61c92b4673b8e7cd8d18174e71 : SUCCESS in 21m 28s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/737c0b8ca0c64c2fb970332a79bee205 : SUCCESS in 21m 27s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e7fb821ea5af41dd8a68e5555fb1efdd : SUCCESS in 57m 02s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4e89c58ba3b04a94a4b1736cf2bb5a64 : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c67eb8d99a2d4a0da1f097cc3fd9f544 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc4b786db8be43e980ac65971835696e : SUCCESS in 4m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b4fc8357309240f4ae487198d967da8d : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ceca9837ed6f45a2ba451f3cb43dcc59 : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41f9bcefe7df4ca2bcd69ddda395bd96 : SUCCESS in 6m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d5b5d300f55e4565b797da56dbb95c3a : SUCCESS in 6m 13s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/670391fdad2c49d29990a7324af21ec3 : SUCCESS in 23m 10s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/fd8fc03756f74ba19c0496ac928f400e : SUCCESS in 36m 12s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/08327329b3af405ebfa1b3c642ec9b5b : SUCCESS in 52m 01s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0d1e638b08a94bb5843b087e8d0c09d4 : SUCCESS in 1h 01m 11s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/96bb3a7184fe42d7b02a93dfe7db7f76 : SUCCESS in 1h 03m 22s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/5f004b13bd274090a8f8634a0e555030 : SUCCESS in 1h 30m 21s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"d7a7281a11d9383c4ce1ed2969b4020722f8e1fc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 07:18:07.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/479560b1119f41908bc12bf17d27554a : TIMED_OUT in 3h 00m 57s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"2d17731b8e6884a4d81b851ad4e40e59a44084a7","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-02-29 23:25:20.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"7b18a0fc471cfc4928a6d59f2a769be832254e7a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-29 23:58:07.000000000","message":"Patch Set 10:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/dcbb6ad676dd409bbafb426f2c893caa : SUCCESS in 30m 44s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/592b3902b33f469cbcefa687f788f139 : SUCCESS in 31m 09s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"f4c5c42fe8b09dba4a967b47664979412937f31f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-01 01:39:49.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/25779e0e71c74316aca538d4389e1f2f : SUCCESS in 39m 42s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/879b1b8364b8455eba5e8679cb900466 : SUCCESS in 36m 41s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/b7d2ab5465be43659e8a4ddfbe6be708 : SUCCESS in 27m 06s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/12d14d2e2eb74fbf8f4f9047a5063340 : SUCCESS in 41m 51s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/9a70ec934591429bb3e5919d10c98b30 : SUCCESS in 1h 38m 33s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/304ecc317f3b46cfb47a4e301d6ace89 : SUCCESS in 1h 36m 44s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/69e625d66dda47ee816a42807e3ef475 : SUCCESS in 1h 23m 14s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/137e040c43da48c9bd8f9d4c86c599fb : SUCCESS in 1h 31m 25s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/92178d46e1bc4697b4b257838b7f87cf : SUCCESS in 1h 37m 46s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/13b89c2c9f3642c8a7978338e544491c : FAILURE in 36m 52s (non-voting)\n- kolla-ansible-centos-source-zun https://zuul.opendev.org/t/openstack/build/39e26d29609b41f88d87aa321c0109ae : TIMED_OUT in 2h 08m 06s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/f08c46b3ee1b492b98362cee00f2b7ce : SUCCESS in 1h 58m 08s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/44c87a22c7724668a2f8796d899c9170 : SUCCESS in 32m 36s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/00ba6c4ed0e84f769020b9ec1bf367b2 : SUCCESS in 24m 23s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3045ef3f820542eea2897b8e00d37b90 : SUCCESS in 1h 01m 32s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/36b34d2745564a459c4fefc60eb22f4b : SUCCESS in 6m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b17abb6486744ff39319c725d5e9406a : SUCCESS in 10m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ceac5a4843449c293ac1728fd4d5bb8 : SUCCESS in 4m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/00f6dcb40a4f482ea720df06726ff6e5 : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa6eeb54b5ba4d86a32d36bc24c25527 : SUCCESS in 4m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f873fc5f5b424dd28b2e7be934cb531c : SUCCESS in 11m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fd5a74c97e31426eb7344f408ff9fe6f : SUCCESS in 5m 27s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/dd738cf26220425d8a4fd4878c405f6a : SUCCESS in 37m 22s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3c3988ea31894dc489383672ba9cc51c : SUCCESS in 26m 41s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e72c6a154af844abb85695ec023eb348 : SUCCESS in 1h 06m 53s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4114b61291544b70bba74cebf39b26a6 : SUCCESS in 59m 34s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/388d592b2a6341208abed051e8502ea3 : SUCCESS in 1h 14m 02s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/e4499aed819b4327b6b3e181c8f7237e : SUCCESS in 1h 49m 57s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"6bb7a0c870751b58648d8ef8ad4b9ea9f0188f80","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-01 03:19:43.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a093536f23ff49d580e5519bc09b0e59 : TIMED_OUT in 3h 00m 54s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"23cfa72d3a831645a6cdfa8f224d75ac9b4ba38d","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-03-01 20:33:35.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"17d4681fe8790d0662725239d4ad8c40d30961c5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-01 20:42:21.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ca7cf8acd6ac4bb0877f0c4c42cc2f59 : FAILURE in 5m 28s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"7837959d3b95c2a7a9292f34a500f9c11999faa0","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-03-01 20:53:36.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"c276710dedbbd939d39f597bbed0514bc507450b","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-03-01 20:56:39.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"8312d1dd5b59a2d4f19022d7ca54c4b9e6a38886","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-03-01 21:27:45.000000000","message":"Patch Set 13:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/f298b94521ef4f159fecbffba6e8f94e : SUCCESS in 29m 44s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/b865d9dbfd944c208c396be5b4274a2c : SUCCESS in 27m 47s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"45936993956791f2f3d8d8f9b767cce6dc142120","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-03-01 23:05:06.000000000","message":"Uploaded patch set 14: Commit message was updated.","accounts_in_message":[],"_revision_number":14},{"id":"433a9689358836f32e89977670bf1aa4c086dc3b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-03-01 23:34:45.000000000","message":"Patch Set 14:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/815f36896f2e4933b79139d7df1c184a : SUCCESS in 28m 17s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/9b77422bde2f44fa81eef930ef611df7 : SUCCESS in 26m 53s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"758f4c624f91a80be4904b4a9210dda4289fc9a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 01:25:20.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/c666685e894142fca6356f4df1a0b525 : SUCCESS in 29m 44s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/8243aeb788e04c249d65b1d1ee9261ce : SUCCESS in 44m 47s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/8d2a4fd63df843f5932a0079b0594033 : SUCCESS in 30m 32s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/cd61c6da886942a3a87d803985604ba3 : SUCCESS in 32m 09s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/6b01ad5b5f9d4905af1eaf95e87c3b5e : SUCCESS in 1h 37m 26s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/76e3fd8fb40d42069eca2a3fd4a9ff49 : SUCCESS in 1h 21m 23s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/2b833a1c2b9945dcb43d227f5aaa411c : SUCCESS in 1h 36m 27s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/e5a1eefcbc374a018d1f3280e043b665 : SUCCESS in 1h 18m 39s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/cf539dc28340482692f511da73fc6fcb : SUCCESS in 1h 48m 32s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/d82da63b0f674c3392c00d5dc2a5c0ec : SUCCESS in 38m 52s (non-voting)\n- kolla-ansible-centos-source-zun https://zuul.opendev.org/t/openstack/build/6ac54abf09ef49ab9441cedc8b605d74 : SUCCESS in 2h 10m 40s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/22b20350170e4c048fbeead4d47fc0ac : SUCCESS in 1h 44m 28s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/c459ee9f6b804a6399b0b3b15927b507 : SUCCESS in 22m 15s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/d7340432b9ce4bae84fe90adb9df208d : SUCCESS in 42m 25s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/27b8376f1e14497381a9891d1fb60332 : SUCCESS in 1h 19m 32s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b76ada22dd06422ab7c3dde7fae8e17c : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efdfd67f591d4dcb861e7390499a53bf : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/901ff9c0dc6c4675941e7ffe4623e132 : SUCCESS in 5m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ac8868ca199140ff9509f5e23c22887c : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eb9f6eb9ef68463280497746fa5def53 : SUCCESS in 4m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/885f3a2f63bb4b888f96a451002d1080 : SUCCESS in 13m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a6bd4aeb18d24ff9bd2a0b90a8fecc09 : SUCCESS in 5m 13s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/19e63688844549ee89d9430758383bee : SUCCESS in 26m 26s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3afb8cfb236d40c6b1d3312309277061 : SUCCESS in 25m 09s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/69c68fc723a441d2bae4b389588d67f2 : SUCCESS in 54m 12s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/adfa3b14caa0441ea7f782337aa3b964 : SUCCESS in 58m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8d61fa1b67204dc3847fe91f8bd2b394 : SUCCESS in 1h 02m 29s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/a665dcd6a63a47198df33002287ee422 : SUCCESS in 1h 40m 13s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"065accd602ed0cc58b0868188a6aa47aa30a211b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 03:09:22.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/03a6874c08944cc8b532d1e9062a294f : TIMED_OUT in 3h 01m 00s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"692c02760e8cb98c2065283e0bc87cf9983fef86","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2020-03-02 12:26:16.000000000","message":"Patch Set 14: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"2ecdf208d23ac0b2379086afec46989cf8088cc2","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-03-05 02:41:38.000000000","message":"Patch Set 15: Published edit on patch set 14.","accounts_in_message":[],"_revision_number":15},{"id":"cee237daf9fbe98e84398b310b01e29249ba792e","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-03-05 02:42:04.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"a3507113ad0a31080c28ae1f7d53538d76b4cfb8","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-03-05 03:15:24.000000000","message":"Patch Set 15:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/09246bd2a0134d5d9f8cf95abd579737 : SUCCESS in 25m 08s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/84abb59d0a564fb686eabcb068573149 : SUCCESS in 32m 55s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"e5b83f8eea59f1039438384d143658fdd2cd1551","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 04:52:31.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/415b90b6af23422b9225cd9dc1412dc6 : FAILURE in 35m 18s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/41806ed6676945e49d98f22df1e1dcc8 : FAILURE in 45m 02s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/6309b0c306c146b183cac046197af298 : SUCCESS in 33m 10s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/ac496f41ff4140f2ac7ef8da037db28f : SUCCESS in 42m 20s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/f2864c6946d34659b90b3cc26ac8ccc8 : SUCCESS in 1h 30m 35s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/59235d5099234af7b4ebe339681a67f5 : FAILURE in 50m 34s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/2cee2934aa944c0e8e75c5e49c24f41b : SUCCESS in 1h 30m 59s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/1a35a1b916b54dc19a7c9ae2112e6fe8 : SUCCESS in 1h 19m 31s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/19af0c9dadb14475a9d12befba4a6f18 : SUCCESS in 1h 31m 13s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/2224022ffa8b464ab6218fd1a30bb894 : SUCCESS in 47m 03s (non-voting)\n- kolla-ansible-centos-source-zun https://zuul.opendev.org/t/openstack/build/f674466b13be47f6b78ffd9c526bae18 : TIMED_OUT in 2h 08m 57s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/6c4e5f2e4eff432ea14d800626e26ec0 : FAILURE in 48m 31s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e389fb15db614a02ae73c57bc98dc3b3 : SUCCESS in 33m 42s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0c05e33a14004c09b72d5decaa572c0a : SUCCESS in 23m 05s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/73942b3399884700a9664f6d2aa186ad : FAILURE in 47m 56s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8415b838d1fb4d2689969d24a9183b7a : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01d2daf1e1c8405f853b6cfa0bf23f42 : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d6b2a9adef1c4c03ab9225989715b9d1 : SUCCESS in 3m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/af581cb8f26843d1bd11c4ae64482692 : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cb89d14566a64530b9ce0b86b04c7acd : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc3d56313a964c848d00205517447b48 : SUCCESS in 5m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/811fe33455ae4625b774e0cb8eca5ca9 : SUCCESS in 5m 36s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ff4377bbc4334c53802130ddbad9d1db : SUCCESS in 29m 33s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/893797f6d9d74f2c8e006fccec1838b7 : SUCCESS in 38m 15s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4e79bd4a2df747f8b8525dd0214d5e81 : FAILURE in 26m 34s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d0f0c84057ce4c18873c4dcc5100447f : SUCCESS in 56m 52s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6482caeedcd642adbe042e4ab0c39b58 : FAILURE in 1h 16m 05s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/f4205e50d65743abaec3ae934e585fab : FAILURE in 36m 13s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"e7e8567f94177a0a57a075c229a4683bba7de253","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 05:43:58.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/51b80cead25648e5a9c3f2204745405a : TIMED_OUT in 3h 01m 03s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"f1e864f28f601be06d4ed718e555e5e9a33f53b3","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-03-06 01:30:54.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"c905daf7d214c2ecda91ba0298d12763535b173d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-03-06 02:12:16.000000000","message":"Patch Set 15:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/868aee97404442c79f0e7cf4a50817e3 : SUCCESS in 27m 03s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/ce5bd888feec4208a738da4d34f3095d : SUCCESS in 40m 01s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"46062197878ad611b390dde11f75b078f45309fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 03:34:23.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/8ea195d72e03452e9d0d58fc7f9b52af : SUCCESS in 29m 05s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/7151678378034195acb79be091925ee9 : SUCCESS in 34m 52s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/74a7610bd514464aa1a157ebcb4e4ec2 : SUCCESS in 1h 23m 49s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/89948ac95da14b30a1bbb35fa412f287 : SUCCESS in 1h 15m 08s (non-voting)\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/44062aa1ef8f4c83973363e1d9b10777 : SUCCESS in 40m 41s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/f677ebb226ac4d228a0bb6f2f32c1da2 : SUCCESS in 1h 56m 33s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/45ba0390bf9041eba170a5d0761325fb : SUCCESS in 24m 10s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0a82a20692e14fc1bab38b64c3254b52 : SUCCESS in 18m 52s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/677bdaca71eb44329d2e64258d099f6a : SUCCESS in 58m 37s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ca1def1ebc244c0abdcaed20b63c79a1 : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21507e62de084ff799cc0819bda1ee6c : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9ea71e606cb04bc5b942939d78b21575 : SUCCESS in 5m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cad9f7b0bbbf4e4d8eb45f106edf6895 : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01977bb1e4cb499992472a61723ff0f8 : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ccf88a36e5a4da29d10f2748e696075 : SUCCESS in 5m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4056ffb1ad2d444da288d36b375a0f83 : SUCCESS in 4m 24s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/db42abf91930451f9eeb4296b4fcaa6c : SUCCESS in 25m 18s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3f3d2e103e454436ab38ce63b61e3f35 : SUCCESS in 32m 35s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/728c3996f1d74fe0b84d37749ab7ccf4 : SUCCESS in 1h 00m 47s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9427920e73b940128dbbcf135d2919a5 : SUCCESS in 1h 01m 26s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f79570c7a2734d3d81af4b3d3375c20d : SUCCESS in 1h 04m 57s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/56a475a8c3824826b8bc6cd13356060f : SUCCESS in 1h 28m 03s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"2ce21e487cf9915a15d0f5135c5495f265cf0795","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 04:32:38.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/76cb0537d4204d8b8674c8c96532b31d : TIMED_OUT in 3h 01m 02s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"e541690aad0a3dc11dc772f21627e8df9ade48be","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-03-13 03:59:05.000000000","message":"Patch Set 16: Patch Set 15 was rebased","accounts_in_message":[],"_revision_number":16},{"id":"536ddaf0889bd50abbb722c6fbbf86b04406d08f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 04:02:13.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1ee1f3329d4a4ea4a9bfe1aa8f8a004d : FAILURE in 2m 38s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"f7e909696781ff19ac6c42b5381f3a36f86a4ca8","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-03-13 04:40:29.000000000","message":"Patch Set 16:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/1d7f0a17c0ec45c4b1e2a701299358f8 : SUCCESS in 30m 34s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/ea3f55750b854c85a4ec7082204aa580 : SUCCESS in 39m 55s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"6b07202f5a2a327b3143e56fd2d2b18165cc958a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 06:05:25.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/9bd943c815ca41a1a51b490ae93ecc6c : SUCCESS in 22m 13s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/be5a1321c0594ff59656b11085d84a70 : SUCCESS in 30m 57s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/79c4fa55fd854dd09a832ceaa1dbc8c3 : SUCCESS in 1h 18m 09s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/cc4b3f54e89b40f5b8ef6c5b6d2904b3 : FAILURE in 53m 09s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/dc70638d47894bd4a5901405bcdfca94 : SUCCESS in 1h 53m 38s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/684e174ce8d14bf484caefd1bc136e87 : FAILURE in 44m 21s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/1e3b6046d8834ab18fdc4eb0925f08af : SUCCESS in 2h 03m 47s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/13599674f703494dba535827d1f1465f : SUCCESS in 19m 32s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/177b3503e7e442edb139b5e916e8496c : SUCCESS in 22m 00s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8bf84e381d9b46adbe4756479f56c464 : SUCCESS in 1h 02m 23s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/786f8650f2dd4f6090e5549aefc51642 : SUCCESS in 3m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65b85996dafd4253aaf27fa5934257c9 : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/808519912c5e45bfb81274bf32ee56b6 : SUCCESS in 3m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8754e918403140f98443ca56969362a8 : SUCCESS in 3m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ec910f5414534a1fa8dd520d08c7e549 : SUCCESS in 3m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a9d74df66454366b1eea0bf40d32b7b : SUCCESS in 4m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c6dc1b8d8a64fa39308c319fe0ca055 : SUCCESS in 4m 54s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8c13d4b55403455ca58dc9b42be7ddf9 : FAILURE in 23m 10s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e7cedb9c00a14c918a4e7973ff7a3fc6 : FAILURE in 23m 23s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2ac3e2f60e05458b94b00bcc854c4699 : SUCCESS in 57m 15s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/83d7e46c72c54f6ca1d86b05d518b607 : FAILURE in 17m 56s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1196c630dec74c459802c81a48ae63bb : SUCCESS in 1h 12m 48s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/b2bbb6488d9d478e8c59960fe0481dd1 : SUCCESS in 1h 28m 47s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"ac3cf20ad2166ff829d421d2ae683c584001ce26","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-03-13 21:30:19.000000000","message":"Patch Set 16:\n\nrecheck","accounts_in_message":[],"_revision_number":16},{"id":"54be652ad953e67803aa2357bdf0752007df8d29","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 21:34:42.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/21b98031652b4225940d8660a8aef208 : FAILURE in 2m 55s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"beaee8f4fef1a39b30fa3d4fad70c6bb03a79d24","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-03-13 22:02:40.000000000","message":"Patch Set 16:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/8b435a19f6d040a1bff8b71bef543ad9 : SUCCESS in 26m 13s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/ec7b83236c05400886aa7d4984d32112 : SUCCESS in 30m 51s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"16a6f4cd7fc2362f894b6748b348d848432ea1a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 23:52:23.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/f51a277402544aa18613a52120ff35c9 : SUCCESS in 1h 13m 47s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/d5250e6a1f844c7aa887bfc6a18e02a6 : SUCCESS in 37m 34s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/a46bb91bcb924ab4b7e6953193a83479 : SUCCESS in 1h 19m 13s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/c872403411a54d93afccd493f93f5c37 : FAILURE in 1h 01m 44s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/6bc929fee0104cbe9aedebe552378c01 : SUCCESS in 1h 26m 34s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/861f4877d5264072b1de360e45a115f8 : SUCCESS in 52m 13s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/c7e9a2bb418443038491b441ec94aabe : SUCCESS in 2h 14m 34s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/0125fe7d9e384e45bb774279aa620482 : SUCCESS in 27m 22s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ddb4b86c0af649e39130bd939d3de57f : SUCCESS in 19m 36s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/1578684867d94e53955b8850bacf5fcf : SUCCESS in 1h 00m 54s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/da9a77761cad438ab2fe83623f334222 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dde5fca1f4cf4129889230877b6c0a08 : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c5d33f3dcae478a915b9f6a84caa23e : SUCCESS in 4m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ad7aeacabdd44444b934209083dfa37d : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/65375517995a4732aba61dcb281ba3b2 : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe2345e8b4514e2bac30e2121a7004b6 : SUCCESS in 4m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0eabe4dd7f4d4993bec1d041e3af43fa : SUCCESS in 5m 18s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bfa6dc8eddeb4963b7a17e46411c2d6a : FAILURE in 22m 18s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/cec2de0a04e14a58b43650ea8917d775 : SUCCESS in 24m 12s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/74a2d1bf68ef47c0b9da9c353f2c05ee : SUCCESS in 58m 01s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/edc37c0b8ecf402b863b7b61172bb5ae : FAILURE in 17m 55s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6e44b4874f4f4b1ebe0c8254b0874c29 : SUCCESS in 1h 05m 32s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/38b63ef6b71845879e078f3a2715d261 : SUCCESS in 1h 31m 56s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"e26a0833358bc6d41416e55e666cfc163ddbcc6f","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-03-18 02:36:37.000000000","message":"Patch Set 17: Patch Set 16 was rebased","accounts_in_message":[],"_revision_number":17},{"id":"745a41d4e78705a192e6369d6ecb76cd63e7e367","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-03-18 03:01:41.000000000","message":"Patch Set 17:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/e6486158d35a4bbeb84a4d0fc260d070 : FAILURE in 6m 04s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/036b9c95ced04012aff13c493acd45fc : SUCCESS in 24m 11s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"a8fa559c60c0b5a624ae243b8b771faaa69d529b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-18 03:22:08.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f7b324e75bbc437d8243934fef5407e0 : FAILURE in 44m 41s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"4297df970c818372563e22e972c909b0b08e2195","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-18 04:30:44.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/99bd1d54191b45b0a44383581b5379e3 : SUCCESS in 20m 09s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/a65ba54e7ddb4a0f8a55e7862d43a689 : SUCCESS in 25m 43s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/b213a481fcb24a24801f0eb0d9700cb1 : SUCCESS in 1h 15m 26s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/c200182ee57d48fd9f0f72036ceae0c8 : FAILURE in 50m 37s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/72cdcc4cf8b04fa4862afbc1a3fe93da : SUCCESS in 1h 18m 03s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/501b5d2c3dcd4983a759bd9f33095a2b : SUCCESS in 45m 55s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/453a042fd6054e5991cb88f0a26a8155 : SUCCESS in 1h 49m 30s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/cbeb65417efd4b78b7041eab2a59d98c : SUCCESS in 21m 24s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b3f0068d69404af9a8d67ae2d41388b8 : SUCCESS in 19m 49s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c2986467a3d14a7e999bb49fdebdf56a : SUCCESS in 58m 29s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a4260c9424944c4a81168807874fb428 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2a891ded21341279b3ece9e500b3cb2 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f275e2e7b07448eaa06aa9f5c0c08cc9 : SUCCESS in 3m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0f62d33abcf24ff0bafe852cd158a344 : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0f6647e6c3e943bebad8105942cdbc99 : SUCCESS in 4m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2440a167f86947ca8af4edbf25fc22b1 : SUCCESS in 4m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/68ae810f1f8944bfbe839ff7f7e1111b : SUCCESS in 4m 34s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f04926a0613448d29777a01e201210ad : SUCCESS in 26m 25s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3e571f3c1d8a4173a61c9641317e76b9 : SUCCESS in 25m 18s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8d10844495d04bcf93fe8bd3496cd547 : SUCCESS in 57m 28s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d46aff95cbf34b7e8abe4c3a5c551d3e : SUCCESS in 56m 29s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7d6911bd375148fb9e4f32ce7b62a4f1 : SUCCESS in 1h 08m 41s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/b52c9da75b6a4f91860a65e08113b757 : SUCCESS in 1h 28m 11s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"b1114f2af0bc4073d29d496ca4a7033f26516563","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-21 09:17:26.000000000","message":"Patch Set 17:\n\n(4 comments)","accounts_in_message":[],"_revision_number":17},{"id":"a96645c67dbffff39ba33ebc482a0c6d6523de0a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-21 09:24:34.000000000","message":"Patch Set 17:\n\nalso, this looks like kuryr-libnetwork clone? is it going to play similar role here?","accounts_in_message":[],"_revision_number":17},{"id":"bd248f317ac14cb76be607ec75ceea755398cbca","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-03-21 16:44:25.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"334e7bfaaa951b52c81edbd22eb838c87d3316de","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-03-21 16:44:57.000000000","message":"Patch Set 18:\n\n(4 comments)","accounts_in_message":[],"_revision_number":18},{"id":"fbd41c5ffde8963df7919f1f00d474706f920eef","author":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"date":"2020-03-21 16:56:11.000000000","message":"Patch Set 18:\n\n\u003e also, this looks like kuryr-libnetwork clone? is it going to play\n \u003e similar role here?\n\nProbably its role is similar to kuryr-libnetwork but not a direct clone.\n\nKuryr-libnetwork is a CNM (Container Network Model) plugin for Docker. Zun-cni-daemon is a CNI (Container Network Interface) plugin for CRI runtime (i.e. containerd).\n\nZun have been using kuryr-libnetwork for networking docker container. Since U release, Zun adds support for CRI runtime (i.e. containerd) in parallel with Docker. CRI runtime basically requires a CNI plugin as its networking model. Therefore, we introduces this daemon process, which basically implements the CNI model for Zun capsule (similar to pod in k8s).","accounts_in_message":[],"_revision_number":18},{"id":"82fa5a6d12d5c62aac82ac015ac0d18c12dfaa8a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-21 17:31:15.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/658e0887751b4f628dcaf22a356cfcb3 : SUCCESS in 46m 17s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"146265e1912199c2318eaacb9fea7af626024894","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-03-21 18:33:48.000000000","message":"Patch Set 18:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/f4ff2186dafc415f9a1c9021000f154b : FAILURE in 5m 38s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/fe039d9168ee47ee979fdd3d74d55c5b : SUCCESS in 28m 47s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"6bf26fe8d6364818e225a49201f17ef6e866c377","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-21 19:05:09.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/f9feba3f0f5f40428069e299954389cd : SUCCESS in 43m 58s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/4c77f0d0554e4e0a97375f6a5f8bec7b : SUCCESS in 28m 47s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/75972e35aa9d4636ac2eca3b2a12763d : SUCCESS in 1h 10m 08s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/8009ab91860141409d7886cac5d5a789 : FAILURE in 48m 14s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/fc497cbccd8f47b8ac0e6f2ab0cc95f6 : SUCCESS in 1h 18m 13s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/813ea51b7379461ab891310ead203425 : SUCCESS in 57m 32s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/5ed97e7f9dbf4e43ab60171091e57b01 : TIMED_OUT in 2h 15m 34s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/07b2217248104325b74087059f04692f : SUCCESS in 30m 57s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/3e4b60f861104520949da229581afa2c : SUCCESS in 28m 22s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f71ee5639f5445b4a47222e20ded1134 : SUCCESS in 1h 14m 28s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/412c1d4294ea403da668bc283ba940d1 : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1741d07c32a44edbaf7af46ee04b0eba : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/55e9cf650d794cc48b1424bf90f28eb0 : SUCCESS in 4m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9ad249894b78447bb81f6649182cbbf7 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c6eb5ad8cf144034aeee90444f717ad7 : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9536c9593e4d43ba8455ff7470e90a79 : SUCCESS in 7m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/452561538cb04c6e83e9388ac7d311be : SUCCESS in 5m 08s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f5c34dcd457e40b1aa6b24d1f5d4d188 : SUCCESS in 23m 50s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/70246cc803854fcbbee6c026fc168ba6 : SUCCESS in 22m 42s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e7495f4be0004f408f932d2b16528dfa : SUCCESS in 1h 14m 21s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5814272f64c94cc6b634cbee90205b06 : SUCCESS in 59m 36s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/84c780ab09a24a3f89b82b8e7fd6894b : SUCCESS in 1h 06m 28s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/5600ecea9c714e52bcf4f754da60ca74 : SUCCESS in 1h 50m 44s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"ec1c4cf12ed04ac92c581e37e66d9e420cac83ff","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-22 17:24:21.000000000","message":"Patch Set 18: Code-Review+2\n\nA\u0027ight. Then it\u0027s more like kuryr-kubernetes. I see you actually copied swaths of its code to zun, not sure if this is the best approach but let it be.","accounts_in_message":[],"_revision_number":18},{"id":"e3a4b10e48da54878ff87e322809c72f1e6000f4","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2020-03-23 01:36:59.000000000","message":"Patch Set 18: Code-Review+2","accounts_in_message":[],"_revision_number":18},{"id":"912faa892e30ff33a78bb82e4f9d596a70113130","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-23 08:35:22.000000000","message":"Patch Set 18: Workflow+1\n\ngoing in","accounts_in_message":[],"_revision_number":18},{"id":"911d040bffdc6a5d397b54b1690266b34ace8994","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 08:35:35.000000000","message":"Patch Set 18: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":18},{"id":"02a30ea0a614f5558c6753a382f0173204af46d7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 10:16:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":18},{"id":"d560bda3df7ce5436ca65e98d7956b0328e4a2b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 10:16:40.000000000","message":"Patch Set 18: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/d6a6e070fb9242bf91f4419dca54f83b : SUCCESS in 30m 04s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/835bbd7da84f414c8246b799e6eb7b3a : SUCCESS in 1h 22m 51s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/30eb8c4ef66244978a6085e8587943f6 : SUCCESS in 1h 18m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8ac95687088445809810274eb1b2cac2 : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b71ae67461a41028b6086008bf8ceb2 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/384f0db9f4f6476eb04825f4825d043f : SUCCESS in 3m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c007de452b404359a632132703ad80ff : SUCCESS in 3m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8eee9d2859243968b0b2a93be66a39b : SUCCESS in 5m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e464821ebd8d473faeb412f1b5453809 : SUCCESS in 4m 59s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2ac8f951ae754d2184ae3eb9de2bb20b : SUCCESS in 26m 18s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4ecc0576a48f44efac01c67804ad39cb : SUCCESS in 53m 14s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3a69f2242f6f4ff5b154757f98c68e14 : SUCCESS in 1h 07m 40s","accounts_in_message":[],"_revision_number":18},{"id":"50ce57da1b35160d7d9f9abe2171675025a0b55c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 10:18:28.000000000","message":"Patch Set 18:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89ac1a5fa17040b9b0de0417a24de7db : SUCCESS in 1m 21s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/157193750f1649cdada50f606d2623ef : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":18}],"current_revision_number":18,"current_revision":"39820c4abb83c55e66e7760c5844754fc07e4599","revisions":{"91d570a4143e064f21fd7f937b80f1e17615c526":{"kind":"REWORK","_number":1,"created":"2020-02-17 23:07:04.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/73/708273/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/73/708273/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/708273/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/708273/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/708273/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/73/708273/1"}}},"commit":{"parents":[{"commit":"abce4d5398d44b46161de2bf21914e776d16d7e7","subject":"Remove kolla Ceph container images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/abce4d5398d44b46161de2bf21914e776d16d7e7"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-02-17 23:04:22.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-02-17 23:07:04.000000000","tz":0},"subject":"Zun: add zun-cni-daemon image","message":"Zun: add zun-cni-daemon image\n\nZun-cni-daemon is a new process for implementing CNI plugin for Zun.\nIt will be used by CRI runtime to connect podsandbox to neutron.\n\nChange-Id: Ic82c59a5e78078b4fea10df9d30b35da14cad922\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/91d570a4143e064f21fd7f937b80f1e17615c526"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/91d570a4143e064f21fd7f937b80f1e17615c526"}]},"branch":"refs/heads/master"},"8f79957bc81d428fe0d341fb60d8b5b1f9c6a406":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-02-17 23:07:56.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/73/708273/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/73/708273/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/708273/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/708273/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/708273/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/73/708273/2"}}},"commit":{"parents":[{"commit":"abce4d5398d44b46161de2bf21914e776d16d7e7","subject":"Remove kolla Ceph container images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/abce4d5398d44b46161de2bf21914e776d16d7e7"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-02-17 23:04:22.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-02-17 23:07:44.000000000","tz":0},"subject":"[WIP] Zun: add zun-cni-daemon image","message":"[WIP] Zun: add zun-cni-daemon image\n\nZun-cni-daemon is a new process for implementing CNI plugin for Zun.\nIt will be used by CRI runtime to connect podsandbox to neutron.\n\nChange-Id: Ic82c59a5e78078b4fea10df9d30b35da14cad922\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f79957bc81d428fe0d341fb60d8b5b1f9c6a406"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f79957bc81d428fe0d341fb60d8b5b1f9c6a406"}]},"branch":"refs/heads/master"},"eef537a8766d6b2d4e50c4ec11fe486b8856b5c0":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-02-18 03:12:16.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/73/708273/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/73/708273/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/708273/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/708273/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/708273/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/73/708273/3"}}},"commit":{"parents":[{"commit":"abce4d5398d44b46161de2bf21914e776d16d7e7","subject":"Remove kolla Ceph container images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/abce4d5398d44b46161de2bf21914e776d16d7e7"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-02-17 23:04:22.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-02-18 03:12:00.000000000","tz":0},"subject":"Zun: add zun-cni-daemon image","message":"Zun: add zun-cni-daemon image\n\nZun-cni-daemon is a new process for implementing CNI plugin for Zun.\nIt will be used by CRI runtime to connect podsandbox to neutron.\n\nChange-Id: Ic82c59a5e78078b4fea10df9d30b35da14cad922\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eef537a8766d6b2d4e50c4ec11fe486b8856b5c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eef537a8766d6b2d4e50c4ec11fe486b8856b5c0"}]},"branch":"refs/heads/master"},"36dc5697b7d51e5ca824fa3ef2b30c1aab845708":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-02-20 02:13:21.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/73/708273/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/73/708273/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/708273/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/708273/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/708273/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/73/708273/4"}}},"commit":{"parents":[{"commit":"abce4d5398d44b46161de2bf21914e776d16d7e7","subject":"Remove kolla Ceph container images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/abce4d5398d44b46161de2bf21914e776d16d7e7"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-02-17 23:04:22.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-02-20 02:13:17.000000000","tz":0},"subject":"Zun: add zun-cni-daemon image","message":"Zun: add zun-cni-daemon image\n\nZun-cni-daemon is a new process for implementing CNI plugin for Zun.\nIt will be used by CRI runtime to connect podsandbox to neutron.\n\nNeeded-By: https://review.opendev.org/#/c/708213/\nChange-Id: Ic82c59a5e78078b4fea10df9d30b35da14cad922\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/36dc5697b7d51e5ca824fa3ef2b30c1aab845708"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/36dc5697b7d51e5ca824fa3ef2b30c1aab845708"}]},"branch":"refs/heads/master"},"1eebac32942b402a21eca5d69959fb095f6fc052":{"kind":"REWORK","_number":5,"created":"2020-02-23 17:15:07.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/73/708273/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/73/708273/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/708273/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/708273/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/708273/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/73/708273/5"}}},"commit":{"parents":[{"commit":"abce4d5398d44b46161de2bf21914e776d16d7e7","subject":"Remove kolla Ceph container images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/abce4d5398d44b46161de2bf21914e776d16d7e7"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-02-17 23:04:22.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-02-23 17:14:58.000000000","tz":0},"subject":"Zun: add zun-cni-daemon image","message":"Zun: add zun-cni-daemon image\n\nZun-cni-daemon is a new process for implementing CNI plugin for Zun.\nIt will be used by CRI runtime to connect podsandbox to neutron.\n\nNeeded-By: https://review.opendev.org/#/c/708213/\nChange-Id: Ic82c59a5e78078b4fea10df9d30b35da14cad922\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1eebac32942b402a21eca5d69959fb095f6fc052"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1eebac32942b402a21eca5d69959fb095f6fc052"}]},"branch":"refs/heads/master"},"7dbb2509164c9423feb03bffb2773f44e6ec6ac0":{"kind":"REWORK","_number":6,"created":"2020-02-24 03:43:34.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/73/708273/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/73/708273/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/708273/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/708273/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/708273/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/73/708273/6"}}},"commit":{"parents":[{"commit":"abce4d5398d44b46161de2bf21914e776d16d7e7","subject":"Remove kolla Ceph container images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/abce4d5398d44b46161de2bf21914e776d16d7e7"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-02-17 23:04:22.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-02-24 03:43:26.000000000","tz":0},"subject":"Zun: add zun-cni-daemon image","message":"Zun: add zun-cni-daemon image\n\nZun-cni-daemon is a new process for implementing CNI plugin for Zun.\nIt will be used by CRI runtime to connect podsandbox to neutron.\n\nNeeded-By: https://review.opendev.org/#/c/708213/\nChange-Id: Ic82c59a5e78078b4fea10df9d30b35da14cad922\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7dbb2509164c9423feb03bffb2773f44e6ec6ac0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7dbb2509164c9423feb03bffb2773f44e6ec6ac0"}]},"branch":"refs/heads/master"},"6000c582ce42e5b067c235acf94c6940d7c1647a":{"kind":"REWORK","_number":7,"created":"2020-02-24 04:30:28.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/73/708273/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/73/708273/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/708273/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/708273/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/708273/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/73/708273/7"}}},"commit":{"parents":[{"commit":"abce4d5398d44b46161de2bf21914e776d16d7e7","subject":"Remove kolla Ceph container images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/abce4d5398d44b46161de2bf21914e776d16d7e7"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-02-17 23:04:22.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-02-24 04:30:22.000000000","tz":0},"subject":"Zun: add zun-cni-daemon image","message":"Zun: add zun-cni-daemon image\n\nZun-cni-daemon is a new process for implementing CNI plugin for Zun.\nIt will be used by CRI runtime to connect podsandbox to neutron.\n\nNeeded-By: https://review.opendev.org/#/c/708213/\nChange-Id: Ic82c59a5e78078b4fea10df9d30b35da14cad922\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6000c582ce42e5b067c235acf94c6940d7c1647a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6000c582ce42e5b067c235acf94c6940d7c1647a"}]},"branch":"refs/heads/master"},"d8052f5aba26335db300d000a3587e026aca440d":{"kind":"REWORK","_number":8,"created":"2020-02-24 05:02:03.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/73/708273/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/73/708273/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/708273/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/708273/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/708273/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/73/708273/8"}}},"commit":{"parents":[{"commit":"abce4d5398d44b46161de2bf21914e776d16d7e7","subject":"Remove kolla Ceph container images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/abce4d5398d44b46161de2bf21914e776d16d7e7"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-02-17 23:04:22.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-02-24 05:01:56.000000000","tz":0},"subject":"Zun: add zun-cni-daemon image","message":"Zun: add zun-cni-daemon image\n\nZun-cni-daemon is a new process for implementing CNI plugin for Zun.\nIt will be used by CRI runtime to connect podsandbox to neutron.\n\nNeeded-By: https://review.opendev.org/#/c/708213/\nChange-Id: Ic82c59a5e78078b4fea10df9d30b35da14cad922\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d8052f5aba26335db300d000a3587e026aca440d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d8052f5aba26335db300d000a3587e026aca440d"}]},"branch":"refs/heads/master"},"43a8fe9bc1355652dba564625f1a67e778e89d30":{"kind":"REWORK","_number":9,"created":"2020-02-26 03:48:19.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/73/708273/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/73/708273/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/708273/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/708273/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/708273/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/73/708273/9"}}},"commit":{"parents":[{"commit":"abce4d5398d44b46161de2bf21914e776d16d7e7","subject":"Remove kolla Ceph container images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/abce4d5398d44b46161de2bf21914e776d16d7e7"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-02-17 23:04:22.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-02-26 03:48:12.000000000","tz":0},"subject":"Zun: add zun-cni-daemon image","message":"Zun: add zun-cni-daemon image\n\nZun-cni-daemon is a new process for implementing CNI plugin for Zun.\nIt will be used by CRI runtime to connect podsandbox to neutron.\n\nNeeded-By: https://review.opendev.org/#/c/708213/\nChange-Id: Ic82c59a5e78078b4fea10df9d30b35da14cad922\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/43a8fe9bc1355652dba564625f1a67e778e89d30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/43a8fe9bc1355652dba564625f1a67e778e89d30"}]},"branch":"refs/heads/master"},"19932f1468b928e9053707aa6dc25c977b0b7db3":{"kind":"REWORK","_number":10,"created":"2020-02-29 23:25:20.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/73/708273/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/73/708273/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/708273/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/708273/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/708273/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/73/708273/10"}}},"commit":{"parents":[{"commit":"abce4d5398d44b46161de2bf21914e776d16d7e7","subject":"Remove kolla Ceph container images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/abce4d5398d44b46161de2bf21914e776d16d7e7"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-02-17 23:04:22.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-02-29 23:25:03.000000000","tz":0},"subject":"[WIP] Zun: add zun-cni-daemon image","message":"[WIP] Zun: add zun-cni-daemon image\n\nZun-cni-daemon is a new process for implementing CNI plugin for Zun.\nIt will be used by CRI runtime to connect podsandbox to neutron.\n\nNeeded-By: https://review.opendev.org/#/c/708213/\nChange-Id: Ic82c59a5e78078b4fea10df9d30b35da14cad922\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/19932f1468b928e9053707aa6dc25c977b0b7db3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/19932f1468b928e9053707aa6dc25c977b0b7db3"}]},"branch":"refs/heads/master"},"872c7e4b94d9e8d52774a91c97b35cbddb1444ad":{"kind":"REWORK","_number":11,"created":"2020-03-01 20:33:35.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/73/708273/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/73/708273/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/708273/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/708273/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/708273/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/73/708273/11"}}},"commit":{"parents":[{"commit":"abce4d5398d44b46161de2bf21914e776d16d7e7","subject":"Remove kolla Ceph container images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/abce4d5398d44b46161de2bf21914e776d16d7e7"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-02-17 23:04:22.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-03-01 20:33:28.000000000","tz":0},"subject":"[WIP] Zun: add zun-cni-daemon image","message":"[WIP] Zun: add zun-cni-daemon image\n\nZun-cni-daemon is a new process for implementing CNI plugin for Zun.\nIt will be used by CRI runtime to connect podsandbox to neutron.\n\nNeeded-By: https://review.opendev.org/#/c/708213/\nChange-Id: Ic82c59a5e78078b4fea10df9d30b35da14cad922\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/872c7e4b94d9e8d52774a91c97b35cbddb1444ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/872c7e4b94d9e8d52774a91c97b35cbddb1444ad"}]},"branch":"refs/heads/master"},"8e7336d86eb9a059d426adb3c45c1799f1e82bd5":{"kind":"REWORK","_number":12,"created":"2020-03-01 20:53:36.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/73/708273/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/73/708273/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/708273/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/708273/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/708273/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/73/708273/12"}}},"commit":{"parents":[{"commit":"abce4d5398d44b46161de2bf21914e776d16d7e7","subject":"Remove kolla Ceph container images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/abce4d5398d44b46161de2bf21914e776d16d7e7"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-02-17 23:04:22.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-03-01 20:53:30.000000000","tz":0},"subject":"[WIP] Zun: add zun-cni-daemon image","message":"[WIP] Zun: add zun-cni-daemon image\n\nZun-cni-daemon is a new process for implementing CNI plugin for Zun.\nIt will be used by CRI runtime to connect podsandbox to neutron.\n\nNeeded-By: https://review.opendev.org/#/c/708213/\nChange-Id: Ic82c59a5e78078b4fea10df9d30b35da14cad922\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8e7336d86eb9a059d426adb3c45c1799f1e82bd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8e7336d86eb9a059d426adb3c45c1799f1e82bd5"}]},"branch":"refs/heads/master"},"5ba52749fdc0c02426a8f2f0523ba5f86bc0f43c":{"kind":"REWORK","_number":13,"created":"2020-03-01 20:56:39.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/73/708273/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/73/708273/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/708273/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/708273/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/708273/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/73/708273/13"}}},"commit":{"parents":[{"commit":"abce4d5398d44b46161de2bf21914e776d16d7e7","subject":"Remove kolla Ceph container images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/abce4d5398d44b46161de2bf21914e776d16d7e7"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-02-17 23:04:22.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-03-01 20:56:34.000000000","tz":0},"subject":"[WIP] Zun: add zun-cni-daemon image","message":"[WIP] Zun: add zun-cni-daemon image\n\nZun-cni-daemon is a new process for implementing CNI plugin for Zun.\nIt will be used by CRI runtime to connect podsandbox to neutron.\n\nNeeded-By: https://review.opendev.org/#/c/708213/\nChange-Id: Ic82c59a5e78078b4fea10df9d30b35da14cad922\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5ba52749fdc0c02426a8f2f0523ba5f86bc0f43c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5ba52749fdc0c02426a8f2f0523ba5f86bc0f43c"}]},"branch":"refs/heads/master"},"425396d1795648410e87753ce5b07d3c975d8ca0":{"kind":"NO_CODE_CHANGE","_number":14,"created":"2020-03-01 23:05:06.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/73/708273/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/73/708273/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/708273/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/708273/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/708273/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/73/708273/14"}}},"commit":{"parents":[{"commit":"abce4d5398d44b46161de2bf21914e776d16d7e7","subject":"Remove kolla Ceph container images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/abce4d5398d44b46161de2bf21914e776d16d7e7"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-02-17 23:04:22.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-03-01 23:02:11.000000000","tz":0},"subject":"Zun: add zun-cni-daemon image","message":"Zun: add zun-cni-daemon image\n\nZun-cni-daemon is a new process for implementing CNI plugin for Zun.\nIt will be used by CRI runtime to connect podsandbox to neutron.\nThis image is based on the zun base image and includes additional\npackages such as openvswitch.\n\nNeeded-By: https://review.opendev.org/#/c/708213/\nChange-Id: Ic82c59a5e78078b4fea10df9d30b35da14cad922\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/425396d1795648410e87753ce5b07d3c975d8ca0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/425396d1795648410e87753ce5b07d3c975d8ca0"}]},"branch":"refs/heads/master"},"9b059e1e8315bb64f5eb7718239db4c01849d4f7":{"kind":"REWORK","_number":15,"created":"2020-03-05 02:41:38.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/73/708273/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/73/708273/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/708273/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/708273/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/708273/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/73/708273/15"}}},"commit":{"parents":[{"commit":"abce4d5398d44b46161de2bf21914e776d16d7e7","subject":"Remove kolla Ceph container images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/abce4d5398d44b46161de2bf21914e776d16d7e7"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-02-17 23:04:22.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-03-05 02:41:30.000000000","tz":0},"subject":"Zun: add zun-cni-daemon image","message":"Zun: add zun-cni-daemon image\n\nZun-cni-daemon is a new process for implementing CNI plugin for Zun.\nIt will be used by CRI runtime to connect podsandbox to neutron.\nThis image is based on the zun base image and includes additional\npackages such as openvswitch.\n\nNeeded-By: https://review.opendev.org/#/c/708213/\nChange-Id: Ic82c59a5e78078b4fea10df9d30b35da14cad922\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9b059e1e8315bb64f5eb7718239db4c01849d4f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9b059e1e8315bb64f5eb7718239db4c01849d4f7"}]},"branch":"refs/heads/master"},"23de65933b0be6a6db201467b7655209037dbc92":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2020-03-13 03:59:05.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/73/708273/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/73/708273/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/708273/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/708273/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/708273/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/73/708273/16"}}},"commit":{"parents":[{"commit":"5498671b344e48d6db865887537ff15cf4028510","subject":"openstack-base: pin setuptools \u003c 46 to get horizon working","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5498671b344e48d6db865887537ff15cf4028510"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-02-17 23:04:22.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-03-13 03:59:05.000000000","tz":0},"subject":"Zun: add zun-cni-daemon image","message":"Zun: add zun-cni-daemon image\n\nZun-cni-daemon is a new process for implementing CNI plugin for Zun.\nIt will be used by CRI runtime to connect podsandbox to neutron.\nThis image is based on the zun base image and includes additional\npackages such as openvswitch.\n\nNeeded-By: https://review.opendev.org/#/c/708213/\nChange-Id: Ic82c59a5e78078b4fea10df9d30b35da14cad922\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/23de65933b0be6a6db201467b7655209037dbc92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/23de65933b0be6a6db201467b7655209037dbc92"}]},"branch":"refs/heads/master"},"5cb29b145ff953b51e3d861e92d1e53a103f0e8f":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2020-03-18 02:36:37.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/73/708273/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/73/708273/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/708273/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/708273/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/708273/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/73/708273/17"}}},"commit":{"parents":[{"commit":"2e465a16fe108e0edd7cc92893d089747ddf6807","subject":"Merge \"base: fix pinning of libvirt for aarch64\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2e465a16fe108e0edd7cc92893d089747ddf6807"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-02-17 23:04:22.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-03-18 02:36:37.000000000","tz":0},"subject":"Zun: add zun-cni-daemon image","message":"Zun: add zun-cni-daemon image\n\nZun-cni-daemon is a new process for implementing CNI plugin for Zun.\nIt will be used by CRI runtime to connect podsandbox to neutron.\nThis image is based on the zun base image and includes additional\npackages such as openvswitch.\n\nNeeded-By: https://review.opendev.org/#/c/708213/\nChange-Id: Ic82c59a5e78078b4fea10df9d30b35da14cad922\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5cb29b145ff953b51e3d861e92d1e53a103f0e8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5cb29b145ff953b51e3d861e92d1e53a103f0e8f"}]},"branch":"refs/heads/master"},"39820c4abb83c55e66e7760c5844754fc07e4599":{"kind":"REWORK","_number":18,"created":"2020-03-21 16:44:25.000000000","uploader":{"_account_id":11536,"name":"hongbin","email":"hongbin034@gmail.com","username":"hongbin"},"ref":"refs/changes/73/708273/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/73/708273/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/708273/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/708273/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/708273/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/73/708273/18"}}},"commit":{"parents":[{"commit":"fe8820be560c32832968d939ea7733ebd0848acd","subject":"Merge \"CentOS 8: enable monasca-grafana image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fe8820be560c32832968d939ea7733ebd0848acd"}]}],"author":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-02-17 23:04:22.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin034@gmail.com","date":"2020-03-21 16:44:16.000000000","tz":0},"subject":"Zun: add zun-cni-daemon image","message":"Zun: add zun-cni-daemon image\n\nZun-cni-daemon is a new process for implementing CNI plugin for Zun.\nIt will be used by CRI runtime to connect podsandbox to neutron.\nThis image is based on the zun base image and includes additional\npackages such as openvswitch.\n\nNeeded-By: https://review.opendev.org/#/c/708213/\nChange-Id: Ic82c59a5e78078b4fea10df9d30b35da14cad922\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/39820c4abb83c55e66e7760c5844754fc07e4599"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/39820c4abb83c55e66e7760c5844754fc07e4599"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
