)]}'
{"id":"openstack%2Fkolla~708298","triplet_id":"openstack%2Fkolla~master~Ia97d09f335d495b791c71c66843d085d00a4fab2","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"Ia97d09f335d495b791c71c66843d085d00a4fab2","subject":"[Do Not Merge] test arm64 CI","status":"ABANDONED","created":"2020-02-18 04:17:41.000000000","updated":"2020-03-19 07:13:05.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8d74f10dd310c6c13debc76915df3369a3d89049","_number":708298,"virtual_id_number":708298,"owner":{"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-02-18 06:15:56.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-02-18 04:54:04.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-18 04:54:04.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-18 06:15:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a8cabf542d94e343811cfe6727e99c48a639fb02","author":{"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},"date":"2020-02-18 04:17:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c64874ea955ecdeb0dc8442f5663435025f0f4b6","author":{"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},"date":"2020-02-18 04:18:41.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"2dc840882a6e33cf6a994d7a36862c6c4eb2f6b1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-18 04:54:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/d4667b62c84140c786a6600d467c729d : SUCCESS in 25m 24s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/b1d578fb438a4a37b7329836118fc73d : SUCCESS in 33m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7a584d7feb31528f3602146c21633612bfcc0f26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 06:15:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/407228b7f54747a7bef45c3b6046bd9d : SUCCESS in 31m 29s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/c595abb75f1f485cb3716be97462893f : SUCCESS in 44m 27s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/f8dfc4dac0264612a00aaec6fb7f4fc6 : SUCCESS in 44m 23s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/a65c1fba9e02421282a46c3abd9e28ce : SUCCESS in 41m 09s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/a48b0f73c2d043ac805246e973f815f5 : SUCCESS in 1h 22m 01s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/8a7ebc9e1951414eb5517136ef1b2ce8 : SUCCESS in 1h 21m 39s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/f82868ef903949c4beb4054d78ff86ea : SUCCESS in 1h 18m 00s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/91299f58ed094dec9245d90054e25126 : SUCCESS in 1h 18m 10s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/747526cb6a5e407f96df68382e02854c : SUCCESS in 1h 33m 49s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/f9c47d6f5e834dd8a4e2da6596a0cf52 : SUCCESS in 52m 16s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/8534a6cf944b40c399799037b4f50ad0 : SUCCESS in 27m 02s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a657e728c817448eaf22952c79542e6b : SUCCESS in 23m 07s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0a7fa00f8f094886a3fca69994084438 : SUCCESS in 59m 42s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c04028ec1a544e76976754a96781d45c : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61d481b9d73547e7a3e8da1685316d4d : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b68eb39d1a5844968eed1a3c5c24c2c3 : SUCCESS in 4m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d27dae9f928645eeb2775633b94e88f6 : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e8dec01a376840a59fd577e0f6db2ed3 : SUCCESS in 4m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9099acd559f04ae2a5c908ba985aa111 : SUCCESS in 6m 15s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a89a6edaf78c4eecb4d2beb27511bfc4 : SUCCESS in 25m 03s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/af33ac67e2584cbc804c228646e97491 : SUCCESS in 26m 22s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/bc4f5e4db3f04a1493dee8d0be1486ff : SUCCESS in 1h 32m 24s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/faea3386d2ca4ef799d15d9626ce240e : SUCCESS in 1h 00m 19s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f3818ff7d9a94ad397b7cff7fc157518 : SUCCESS in 1h 03m 31s","accounts_in_message":[],"_revision_number":2},{"id":"9af641c8d74f1ea4f28d8c71755dca1398d16cb3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 08:04:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/26605f015a5d48a393440cf01e3a45b6 : TIMED_OUT in 3h 01m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8d74f10dd310c6c13debc76915df3369a3d89049","author":{"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},"date":"2020-03-19 07:13:05.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3f1f468cecd3e69ec149a0272a71679850e60ede","revisions":{"e40ba27e92fb88cc1eea3aec28d10ae59c3b597c":{"kind":"REWORK","_number":1,"created":"2020-02-18 04:17:41.000000000","uploader":{"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},"ref":"refs/changes/98/708298/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/98/708298/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/708298/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/708298/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/708298/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/98/708298/1"}}},"commit":{"parents":[{"commit":"abce4d5398d44b46161de2bf21914e776d16d7e7","subject":"Remove kolla Ceph container images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/abce4d5398d44b46161de2bf21914e776d16d7e7"}]}],"author":{"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","date":"2020-02-18 04:07:38.000000000","tz":480},"committer":{"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","date":"2020-02-18 04:17:38.000000000","tz":480},"subject":"test arm64 CI","message":"test arm64 CI\n\nChange-Id: Ia97d09f335d495b791c71c66843d085d00a4fab2\nSigned-off-by: Kevin Zhao \u003ckevin.zhao@linaro.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e40ba27e92fb88cc1eea3aec28d10ae59c3b597c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e40ba27e92fb88cc1eea3aec28d10ae59c3b597c"}]},"branch":"refs/heads/master"},"3f1f468cecd3e69ec149a0272a71679850e60ede":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-02-18 04:18:41.000000000","uploader":{"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},"ref":"refs/changes/98/708298/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/98/708298/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/708298/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/708298/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/708298/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/98/708298/2"}}},"commit":{"parents":[{"commit":"abce4d5398d44b46161de2bf21914e776d16d7e7","subject":"Remove kolla Ceph container images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/abce4d5398d44b46161de2bf21914e776d16d7e7"}]}],"author":{"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","date":"2020-02-18 04:07:38.000000000","tz":480},"committer":{"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","date":"2020-02-18 04:18:11.000000000","tz":480},"subject":"[Do Not Merge] test arm64 CI","message":"[Do Not Merge] test arm64 CI\n\nChange-Id: Ia97d09f335d495b791c71c66843d085d00a4fab2\nSigned-off-by: Kevin Zhao \u003ckevin.zhao@linaro.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3f1f468cecd3e69ec149a0272a71679850e60ede"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3f1f468cecd3e69ec149a0272a71679850e60ede"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
