)]}'
{"id":"openstack%2Fkolla~708733","triplet_id":"openstack%2Fkolla~master~I5514c53002a9e772b96e767b10fddf6342c7eb1d","project":"openstack/kolla","branch":"master","topic":"drop-weird-swift-conf","hashtags":[],"change_id":"I5514c53002a9e772b96e767b10fddf6342c7eb1d","subject":"Drop weird swift rootwrap config (xenapi with placeholder values)","status":"MERGED","created":"2020-02-19 19:54:43.000000000","updated":"2020-02-25 20:14:56.000000000","submitted":"2020-02-25 20:13:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"708733-1582661609414-3cc46f25","meta_rev_id":"0fdf1ca8f60156dd3f5e29094e94e0f85a3df627","_number":708733,"virtual_id_number":708733,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"date":"2020-02-25 14:44:06.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"date":"2020-02-25 09:04:46.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-25 20:13:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-02-25 17:55:02.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2020-02-24 10:46:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2020-02-20 10:01:54.000000000","_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2020-02-24 10:46:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-20 10:01:54.000000000","updated_by":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"reviewer":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"state":"REVIEWER"},{"updated":"2020-02-24 10:46:54.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-02-25 14:44:06.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-02-25 17:55:02.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-25 20:13:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"da71301b47683030dc7943ed1771720f2811f316","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-19 19:54:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"61873cdc30a9ad455e4f51db6a03a85668ac7770","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-19 21:12:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/686e296197224dcca5c0dc3b99069e18 : SUCCESS in 25m 31s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/8eb5f1051c984975a1fc5ff2c7d61218 : SUCCESS in 1h 16m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a9f6a571e6df91f67bd64ba95f9d114d642a1e45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 21:42:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/a949b3723d9f4f0cbf783ebe2d0210a8 : SUCCESS in 31m 37s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/5a554bd17c9c4eca93cd5ea61518aaad : SUCCESS in 50m 28s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/05c6121b180447c7987595076648b7b7 : SUCCESS in 32m 16s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/dad44c9e5a5f4a50bd619c0867306fe3 : SUCCESS in 38m 17s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/d038116541a44615916d93e0766aa508 : SUCCESS in 1h 21m 47s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/a4c7b1a6260340769a0c4fea85efe6d6 : SUCCESS in 1h 10m 21s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/97b8d5012a1c4ddc9363579ca330939a : SUCCESS in 1h 19m 34s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/0cc4e448eea4421c8ffd67778cb9abc6 : SUCCESS in 1h 44m 09s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/86e6f538f6b84260a1eb4be62a4b3b78 : SUCCESS in 1h 40m 06s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/b19449867d334cfd9771808120c94bbe : SUCCESS in 35m 55s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/60321b8f861d4b5bbfa5dcabeed6f9b0 : SUCCESS in 38m 18s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/54c96178b4ec4b9f9973b45edced5ec1 : SUCCESS in 31m 02s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d4dc50590d5a4f35865fd389bfe239f2 : SUCCESS in 1h 01m 12s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/191cf75bdd0d478da2f8cbf67e5863cf : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea77afe80d3647aa8be7c39e42f3fd8e : SUCCESS in 6m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b5824f85b51f485396e4ea9bde49dc20 : SUCCESS in 5m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/34b49eddb86d4f9c93ebf9ba9bb62fd4 : SUCCESS in 6m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ecea85b84ba4659a0b5969ab516a583 : SUCCESS in 7m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf1afe9c66774d1cb38895621b5d2f98 : SUCCESS in 11m 59s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6e2d7cff8c484405ac5df84d8be01fc9 : SUCCESS in 38m 57s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f93838065ddc4eaa8258d34c0bbb89be : SUCCESS in 38m 16s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f018d638a370424480a69b17a047f841 : SUCCESS in 56m 05s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ed061d2e08a642e6b467ec70c5e81894 : SUCCESS in 1h 02m 14s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3c18abefe916409ebc861c2a043fb894 : SUCCESS in 1h 06m 26s","accounts_in_message":[],"_revision_number":1},{"id":"57a493d7efca3ff0424f3f57d870903f51ef182f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 23:18:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d79cf0fbadab4feaa58295b8b99cb3bb : TIMED_OUT in 3h 01m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a20ea0aaa29c25a573fd0f97ed63721b6cf3b033","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2020-02-20 10:01:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5a6bbd74279f9c3dec4378446cf8d3e68237379e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-02-24 10:46:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8c2faaa185ceeb72d49b4c7eb72ce37ee9828124","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 10:47:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"956f84a3587033b1de2cde32ad9aa67403ebf16c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 11:50:35.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/fe280e3fe3cb4ac4a3bb47c803165d04 : FAILURE in 6m 34s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/585631314d704a09b9a2660ebf2d2530 : FAILURE in 7m 36s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/459976852cc945a986d040f4345f7d4d : FAILURE in 13m 28s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/df04b53878ef43988d946a49e0dcaf9f : FAILURE in 15m 17s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/e0ef921bc96449798d09638adb883a74 : FAILURE in 13m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ce81baae7ae54d4daa6e44568a0c0487 : SUCCESS in 8m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b89f68b866eb4b03b7b839d153b1f1e9 : FAILURE in 7m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8da00e4eea26492e8a2ae8d4f9d72bbe : FAILURE in 7m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/171b83de979542569f99e6d0e688787b : FAILURE in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac9baede50314ed8a475a81ae1d77b33 : FAILURE in 6m 07s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/daeb9bb9b0604109900474538017a6b1 : FAILURE in 9m 37s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/46315fce413b4b0f850fe36cb9e37086 : FAILURE in 15m 28s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/404f9ddfcb834cb2915d7c4d456e97d2 : FAILURE in 11m 47s","accounts_in_message":[],"_revision_number":1},{"id":"2c3a7921d0f1b0a25e508031a9f8732450045752","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-25 09:04:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9ce4db5b5a9c13c6f797f4da0935bc75d35ef211","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-25 09:36:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/a18388c133db46238b78cea6a1eb3969 : SUCCESS in 29m 57s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/e42645196005427fa8785e07609eadc6 : SUCCESS in 29m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f5b215f09f321ee521a14bdf06a7b0ea34cdf2e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 11:10:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/55f621a806464724b37268bb2106ccff : SUCCESS in 33m 35s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/5a01f24f8dcf45dab9f88ddcda2dcf40 : SUCCESS in 1h 47m 04s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/b903fbc768054876b0c5954c01ba3fc9 : SUCCESS in 32m 10s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/5a75c6286dbb44c3a4a166df544a54cd : SUCCESS in 33m 05s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/bcbcf8a50b5b4af29b54fde7cfd6260e : SUCCESS in 1h 22m 12s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/28c47cc2a1074ef48a6ffb138d80c6e8 : SUCCESS in 1h 27m 04s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/2844ba5d26a34253aa3bc291b2ec0da4 : SUCCESS in 1h 29m 25s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/4fe8be7bb47a4f858cf188c4850e515b : SUCCESS in 1h 23m 58s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/b72fee657e8d4e36b18f4eaa26790537 : SUCCESS in 1h 44m 54s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/95e6a6c273864435a06e44117c5ddd1e : SUCCESS in 43m 26s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/8fe1d0a58c4a4cf388af34c96aa78286 : SUCCESS in 29m 58s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/75b30ebbf60143008464acb90a53d878 : SUCCESS in 35m 17s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a7e37d08517d4b29bb5a17e1245e6001 : SUCCESS in 1h 06m 49s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2a62c210ba954cf6a66cf484c71a6c4d : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7e87a68a9b140f5b495774c536dad2b : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f351eb7942e74ddb947a45abfe31c676 : SUCCESS in 5m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/824694123a7f4d9a9d328ee347b236a8 : SUCCESS in 5m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0e3fe906b2a249abbe0f85186e11a2da : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad1c5ca03d3647b9b968784783c30ea0 : SUCCESS in 6m 54s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/774d581481be4b30b9580bc8136b44ff : SUCCESS in 26m 53s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e481d1e4948f44a89145404f98f3b3a2 : SUCCESS in 36m 02s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b4bea036c0f14e57b89df7e14adc602f : SUCCESS in 1h 19m 00s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8601fb2288ab4f5897e52e8fc741d3cb : SUCCESS in 1h 04m 12s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/29588b1d5504479489423bb020f7a33c : SUCCESS in 1h 14m 13s","accounts_in_message":[],"_revision_number":1},{"id":"e0280f9620c9aac3779da2eea5ab54c9a57f0b58","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 11:10:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1b3129e61786768e81d28f5f133d7de90324cc54","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 12:07:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a79f2c169c1b4fc28bb8d9349ea18c6a : TIMED_OUT in 3h 01m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8bfbc481c1313b313eb27d9422b9323a1c88eba1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 13:10:45.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/87035dcf2b75473f9e4161cf7e14eafc : SUCCESS in 47m 34s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/75dca35b2e89434384a3ffcbdb4e5a97 : SUCCESS in 29m 13s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/02da41f30d8a499e86e3b65ced88a992 : SUCCESS in 1h 35m 37s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/a6a202114cd046ffb01233e95a83f497 : SUCCESS in 1h 15m 54s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/624a94d7de8540cd9c669015948f6239 : SUCCESS in 1h 33m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/15f76c079dee493e944aab22a195db3b : SUCCESS in 6m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/534b9db4d440484aab7d9426af7a3ebf : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb3cd681698e493898b9c337f3f47896 : SUCCESS in 6m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/97cbc803c5674965a8c1ff03cb077508 : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/233668d6119c4d729a97eeab5240cef9 : SUCCESS in 6m 37s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5bfa386e76fe4fc8a81416c289d21889 : SUCCESS in 31m 40s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d99c7b8f1d51454a9124d00c20227439 : FAILURE in 9m 31s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7f38ad7b68904010b405fe9d37d30b8d : SUCCESS in 1h 05m 33s","accounts_in_message":[],"_revision_number":1},{"id":"6895acc3b186fd5c4a15c54163dae0d463da0496","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-25 14:44:06.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"02cead74dc5be1fc9764ed7017ef238b592ad545","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 15:43:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a5ebe5d635944510aba830f658b9725e : FAILURE in 32m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1e0e5f67c0a934b773d901a31a452b1290fd7148","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 16:25:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/a9c2afb47756429fae539c722c5e647e : SUCCESS in 35m 50s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/213e10aa6bc741d6a42806c0fdc3cc9b : SUCCESS in 42m 50s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/baebbfdb08034d3695d071252d717f23 : SUCCESS in 38m 36s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/b8488c5099ad4c0989ca9a94087917a9 : SUCCESS in 30m 40s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/50428b122ab54c309b2ed299a6325f33 : SUCCESS in 1h 22m 51s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/94f5f611456d4245bcf2b7a8bef65e1d : SUCCESS in 1h 18m 09s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/0e77796075574d36ad9028c29206479f : SUCCESS in 1h 36m 38s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/a21456a3bf054fd08b8fdf01925c384e : SUCCESS in 1h 14m 51s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/dbb0e80dc1804f50ad726110280d8d12 : SUCCESS in 1h 26m 33s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/28b8e804bd7149a09758aa15703ed416 : SUCCESS in 40m 29s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e9a09473487b4573b6d8cfec4ed760d9 : SUCCESS in 27m 04s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/70bb46439cb644fc8cdb18bf7a0c3146 : SUCCESS in 31m 16s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0f8edc129d6140f691b3708f846fcb0d : SUCCESS in 1h 06m 30s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6155a616b843401998ebf8aa89b5ec35 : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fb64b5a74c74c6b9b164600c0127250 : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/375f3b92e21a41388277a59cfbdaa184 : SUCCESS in 5m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1bff6ae47550467ca406faa622f9b598 : SUCCESS in 6m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c442ede1aab47dfba1fb7f0e4c19363 : SUCCESS in 7m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c5c3c2726f542898906120c020a6c77 : SUCCESS in 9m 30s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6c0537637a4e4206b911282de7113b20 : SUCCESS in 48m 21s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4e15eba109a9452c95bceebae5eea810 : SUCCESS in 49m 53s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/95dba64e0b904110837e9342f0f45011 : SUCCESS in 1h 07m 14s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/43baa18dbf9947c48e4c05f374873da7 : SUCCESS in 1h 05m 39s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0a2d37d685b04d24ada65e4dd8619a4f : SUCCESS in 1h 10m 40s","accounts_in_message":[],"_revision_number":1},{"id":"1ec3b15e92c3e424f63d386e59fd414df32cfaca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 16:25:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e075b9f4a0ea6fcc682fb00d752a9942fe6808f8","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-25 17:55:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/02504fe664614857884bfb5310780ff5 : SUCCESS in 28m 37s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/1a9a64305b7b41f8ad16c2cab86671ac : SUCCESS in 46m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5748b7496171f93f1f0249a33b7ae290c6be2611","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 20:13:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"58101fd49052dff820800ad2b596ec8befdcbbd1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 20:13:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/c858bcce13e64ffba744ab7db04d7605 : SUCCESS in 40m 18s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/08998a4ec72a4150809ee2c822407c5e : SUCCESS in 32m 26s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/b0325b1ef4ad4f36a063ce406b1792b6 : SUCCESS in 1h 37m 31s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/1ee23d74ada540dc8b50591b070f79b3 : SUCCESS in 1h 24m 22s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/25e3de2fb4554f5a9b29fb3e74f3f548 : SUCCESS in 1h 44m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c4539d457559461b846f8ab90ae1983d : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0be358d8fb3e42f5be66957212ec0830 : SUCCESS in 7m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f5824291858f4c248f4f600e2311f2a7 : SUCCESS in 6m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/57b16a4f2b264f288ba067f354591ac6 : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30bc323dda4f4da3aa1a620c3acc9902 : SUCCESS in 8m 28s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c035e52b0dbc4ba9878dc02e0fb049aa : SUCCESS in 38m 22s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3e48591e8e60448fac2bcbe94fb4b0c3 : SUCCESS in 57m 57s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/71b3d11792764d3ca4aa53a4ec907737 : SUCCESS in 1h 24m 43s","accounts_in_message":[],"_revision_number":1},{"id":"043afb591fd6443761b9d5bfe2300fdc128f2c1f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 20:14:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a037311573554e0db0ac451792e800c3 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"faec906090ec5f59f6e4fc8b9c4f64e14a0025ce","revisions":{"faec906090ec5f59f6e4fc8b9c4f64e14a0025ce":{"kind":"REWORK","_number":1,"created":"2020-02-19 19:54:43.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/33/708733/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/33/708733/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/708733/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/708733/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/708733/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/33/708733/1"}}},"commit":{"parents":[{"commit":"ceae02f09d1c4e67013485c7323c11f8655c9c1e","subject":"Get rid of Python 2 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ceae02f09d1c4e67013485c7323c11f8655c9c1e"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-02-19 19:54:06.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-02-19 19:54:06.000000000","tz":60},"subject":"Drop weird swift rootwrap config (xenapi with placeholder values)","message":"Drop weird swift rootwrap config (xenapi with placeholder values)\n\nChange-Id: I5514c53002a9e772b96e767b10fddf6342c7eb1d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/faec906090ec5f59f6e4fc8b9c4f64e14a0025ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/faec906090ec5f59f6e4fc8b9c4f64e14a0025ce"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
