)]}'
{"id":"openstack%2Fkolla~708758","triplet_id":"openstack%2Fkolla~master~I9692dda817ef134d647247431565e1b58cf9da41","project":"openstack/kolla","branch":"master","topic":"block-label","hashtags":[],"change_id":"I9692dda817ef134d647247431565e1b58cf9da41","subject":"Add block labels to all Dockerfiles","status":"MERGED","created":"2020-02-19 21:59:46.000000000","updated":"2020-03-02 12:29:42.000000000","submitted":"2020-03-02 12:26:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"708758-1583152017545-cb471cbf","meta_rev_id":"5b3164fe0262613b5e049931571a042c7e9b6bb7","_number":708758,"virtual_id_number":708758,"owner":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-02 12:26:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-03-01 17:55:52.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-03-02 10:35:16.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-03-02 10:39:45.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-03-02 10:39:45.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-25 09:04:52.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-03-01 17:55:52.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-02 10:35:16.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-03-02 10:39:45.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-03-02 12:26:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fd8bd6e232c5ecf80860e9d8407dc67c900e3512","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2020-02-19 21:59:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3ec499637f12431be86d85459e7f9b82bfd426bc","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2020-02-19 22:01:06.000000000","message":"Patch Set 1:\n\nReason: we do not want to have the maintainer label and build date label in our images. At the moment we have to remove them manually. This way it is simply possible by using the template overrides.","accounts_in_message":[],"_revision_number":1},{"id":"117a89f103b14119444f8bcd9cd626b30b143816","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 22:49:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b83e8be47ba648c9b097c176f31a028f : FAILURE in 27m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"34412bd11e5aeea3eecca0aa86bba852de3f8010","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-19 23:00:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/e239b4bce309453aa47110e33db7ba21 : SUCCESS in 27m 28s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/2485fcbfe0b8473dbcba304a04e34a7c : SUCCESS in 36m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bb17c9f9b41eeab6f8af2907ae924f3e20402adb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 00:27:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/57eb0a39eb9a4b208c3cfc4563650cb1 : SUCCESS in 33m 02s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/516c89c7bc674fc69277c5cf5d37ea1f : SUCCESS in 41m 35s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/5a20752652fb46918485ebd8727b46d3 : SUCCESS in 28m 10s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/6128e193911c4021a4a641ca9aaaa695 : SUCCESS in 34m 50s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/35b61f6ab23740d18aa5875464d150f2 : SUCCESS in 1h 22m 17s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/d88de051c28f426e988a8fff073159a4 : SUCCESS in 1h 12m 53s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/418e4aa057564ed7a73f9cef341874dd : SUCCESS in 1h 10m 43s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/339b457d41e84c0e897a43f4fae1901b : SUCCESS in 1h 07m 57s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/7080054a99af48a09cfe5e7959a0c033 : SUCCESS in 1h 29m 32s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/acb668c519e74378b64db7deb041a6ef : SUCCESS in 42m 03s (non-voting)\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/9cdb583501dc41108876b3bad97c8082 : SUCCESS in 43m 23s (non-voting)\n- kolla-ansible-centos-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/4cec421debce4b8f94cf0750d9440681 : SUCCESS in 1h 51m 11s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/a0e726ce3eb04c0f8e1614abb07399ab : SUCCESS in 2h 22m 31s (non-voting)\n- kolla-ansible-centos-source-zun https://zuul.opendev.org/t/openstack/build/414b4f0b284846028bd3d3386f40a198 : TIMED_OUT in 2h 02m 43s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/ef72af2d58c94c52acf496f98c369d70 : SUCCESS in 1h 45m 26s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/1d706cf1fa3e4a19ab91551381618ffc : FAILURE in 1h 33m 21s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/8b620dfb77814226a318d8f27f50fc57 : SUCCESS in 1h 44m 56s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/510be92102654dfc8e21d8100c797173 : SUCCESS in 1h 36m 12s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/c748790c3e6847b1b5c711d5a30d4ba7 : SUCCESS in 1h 12m 23s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2365abe094614439bd456566498649c5 : SUCCESS in 38m 02s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/fd8332e5d3944213afd91347b9331be2 : SUCCESS in 30m 29s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f378e93fc4b046fd9a9a56f38b335111 : SUCCESS in 1h 03m 39s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/06798878eec54e8799fa2bd9dcb77099 : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15ee5504c6ea4e58a1247f10e4be6a5b : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c124f26d3e7b413daa68c6ed6320901f : SUCCESS in 5m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a1a6f4029dfa4cc49bab5fcd315e157f : SUCCESS in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0f60d5182bac43b18e8a678c0ba1a0e4 : SUCCESS in 6m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f71be557976d45c7a9b07a775c8be40c : SUCCESS in 7m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e8a6850fed5a4217b0d5b95093e0fd0a : SUCCESS in 6m 44s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5d834403425e4831803e27c723d466b7 : SUCCESS in 24m 37s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/89d2a343a00448ba948c14f792411550 : SUCCESS in 41m 48s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8c8c5ea28409497b9444b19abaae6c20 : SUCCESS in 1h 00m 19s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e93b356180544080b64b323d0a794e78 : SUCCESS in 59m 50s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/34406bfd474747cc9e12c5d6124ca1e0 : SUCCESS in 1h 13m 46s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/115a032ce82d445a921199803da90706 : FAILURE in 26m 19s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/302aa12e44844d7093e0946c3d03a3f3 : FAILURE in 27m 31s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/1fb66add8e6b42ffa33a41734173ac1a : SUCCESS in 1h 27m 37s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/1eeeb3edae384d9fbe1c7c80525cc6c3 : SUCCESS in 1h 06m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e8c9f37ee159d87c33b31f5e89b4a9570b868d8c","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2020-02-24 15:30:54.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"5cf719d05a3f30d794828ce6ffbcf22e92ddd69c","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-02-24 15:32:25.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"f48a424fc1bd5728996c15c80f5922d9e42fbe28","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-24 16:58:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/c24d25e36dcf4f8eabdbbe84a21ff362 : SUCCESS in 28m 19s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/90ae18b1135b43aab3ac56681b8f5180 : SUCCESS in 35m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"196ed426177beeac5e56824a1b9a18ad824e4218","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 18:33:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9aef675be2b141688ca478de01b899e0 : TIMED_OUT in 3h 01m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"21c419d13baf9c4022ddae47b6a9ee32d3cf61cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 20:01:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/c460e138bef749aeb27e85d5683ba542 : SUCCESS in 38m 59s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/e43875b05a3741949d44ec66ac72debb : SUCCESS in 50m 14s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/4140a20bd8ed481fb2553b98b9bf11f0 : SUCCESS in 42m 31s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/3c7eb2a0d44647ca8a9a8cdee6c5bd35 : SUCCESS in 30m 39s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/7a4c678b595d487abefe18a38d797156 : SUCCESS in 1h 32m 15s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/169c9643b0704c6fb7c77be71103481b : SUCCESS in 1h 18m 56s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/6214e5457f2c462aa6c4ca984b9b143c : SUCCESS in 1h 31m 25s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/94581483b7694c08bf3288791cd29111 : SUCCESS in 1h 18m 28s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/238f5afa7a154a869215d76b36baf0ff : SUCCESS in 1h 40m 18s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/c705dfa483744d47b2a421c0629d9610 : SUCCESS in 48m 29s (non-voting)\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/406db19331d8475ea3f0d97eaeb2e59e : SUCCESS in 1h 09m 43s (non-voting)\n- kolla-ansible-centos-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/ee09cc5ceabc4304ab63e59ab251372b : SUCCESS in 2h 26m 30s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/1dadcb3798f749eea9734ed745b46a6f : FAILURE in 1h 57m 46s (non-voting)\n- kolla-ansible-centos-source-zun https://zuul.opendev.org/t/openstack/build/9163b75264c4487a8fb4c1a8eb03ec7f : TIMED_OUT in 2h 09m 52s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/3244e17139d145638d386e5ad2e7dc3e : SUCCESS in 1h 55m 58s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/fbc4fb0a58b7491688c2bb08d62d3ab0 : FAILURE in 1h 33m 45s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/84ee019719eb4b5496a162be749fcaca : SUCCESS in 1h 55m 55s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/ef375e56c3b24a94aaff708947113bbf : SUCCESS in 1h 39m 22s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/a1af2517c826449694d7413e709f36bb : SUCCESS in 1h 17m 04s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f98633a6f3ca451f8acda57d5256eba5 : SUCCESS in 35m 15s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0f70ca65de504eaf8aef744b1dc803c7 : SUCCESS in 24m 31s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/31042cef5c824a76875f5a05282c6fc1 : SUCCESS in 1h 02m 19s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/77dafd36e21748178c4bde1b9c2276f4 : SUCCESS in 7m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e9a4384ae21403ca1c85405b1c099f3 : SUCCESS in 6m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d4738c469b2c4fab8ccb175d34b4ef2c : SUCCESS in 6m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/15bb620d25b64c22b7c8ae44c469275f : SUCCESS in 7m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf177f3507da46cb9a6a199a966365f4 : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2296fc93f957482a82e7205856e6761a : SUCCESS in 6m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/55583645f92546e4a30136a04867e1ef : SUCCESS in 8m 19s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/577923234c734e789d44421a709c9db5 : SUCCESS in 25m 03s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8ff38d9fe87446569e6951fccfe1c7e1 : SUCCESS in 31m 31s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/18916269fc0e455facd913a74e54e3d1 : SUCCESS in 1h 00m 20s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/09a2bb80016545dfaf90a3833c3b12ac : SUCCESS in 1h 02m 12s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7ee710dec86647f1943bc5914bbf6cb4 : SUCCESS in 1h 17m 20s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/fb97e0c415d6409b8cbad47ed5f57683 : FAILURE in 31m 59s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/628cfdf437c04139a8fb20e32827700a : FAILURE in 30m 44s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/c4ee91cf9b2548bf8ec555f96ae09c28 : SUCCESS in 1h 34m 44s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/21772a54f6e34426bcb301e6046561dd : SUCCESS in 1h 05m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0c0ef88290d8e0b034e87bcc3e60a674d9d09b6e","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-02-25 09:04:52.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"63b32ca4719b8e0c8ee31d41e900b4bfa1accad8","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2020-02-25 10:35:22.000000000","message":"Patch Set 2:\n\n@Marc @Radoslaw: Can we merge this? We would need it and I wouldn\u0027t like to wait long because of new incoming Dockerfiles.","accounts_in_message":[],"_revision_number":2},{"id":"eef0d9736b76b570f7aa60e596c75057c90bea65","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-25 14:08:06.000000000","message":"Patch Set 2: Code-Review-1\n\nI was sure we had this already, but it seems not.\n\nHere\u0027s how I understand the headers and footers to work. Assuming we have 3 images: base -\u003e openstack-base -\u003e my-image\n\nEach of these would have its own pair of headers and footers.\n\nbase_header/base_footer\nopenstack_base_header/openstack_base_footer\nmy_image_header/my_image_footer\n\nThen the final image in the chain has a footer right at the end. Therefore the \u0027header\u0027 block should go at the start of the base image, and probably be empty like the others.\n\nIf you want to make it possible to swap out the labels, I suggest you add a different block, e.g. block labels","accounts_in_message":[],"_revision_number":2},{"id":"c80dcf1a7e19917879fbe4f5b69c96579bc490c2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-25 14:09:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f541844d6f9649f6c90b59c18549b2b75970651d","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2020-02-25 15:54:50.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2f36a42d2d1338b364b0e1947febc0956819dc65","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2020-02-25 15:56:26.000000000","message":"Patch Set 3: Workflow-1\n\nDoh. Didn\u0027t read the comment correctly. I\u0027m renaming the block to Label.","accounts_in_message":[],"_revision_number":3},{"id":"08fcbdb81880a9ce7d9964829f23c76812f7b399","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2020-02-25 15:56:58.000000000","message":"Topic changed from global-header to block-label","accounts_in_message":[],"_revision_number":3},{"id":"5e8bc43d8689fb78a81450e885175dd4574d8ab4","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2020-02-25 16:04:06.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"536f647dc50b060c466e485dcf74d63ee265894c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 18:59:20.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/f084769a00304e12a2e89d4112cccead : SUCCESS in 39m 48s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/446605a3cee04b1b8974f72039d47d9a : SUCCESS in 41m 23s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/57acbb0837a8461d91a9ec71c870c455 : SUCCESS in 30m 00s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/647cda1df6054ecdbf856d2ebe66f45b : SUCCESS in 35m 54s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/3ff8ce523e7b4bb380e082b4bf09e586 : SUCCESS in 1h 30m 52s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/9d2d6cf7cb954285aab3285b38de694c : SUCCESS in 1h 14m 12s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/9ff313d72b974b53bbb90f97e1249fd7 : SUCCESS in 1h 18m 35s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/5b13ff8ef21f49baaa2a88c8b8b41c99 : SUCCESS in 1h 09m 37s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/2541a5e6aeed49218b33c37ae182a4fb : SUCCESS in 1h 40m 20s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/d641d0900ef74ed6a86ff4436fd78d98 : SUCCESS in 49m 06s (non-voting)\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/36f4c6b934074d35baa13193dc154cfd : SUCCESS in 1h 13m 19s (non-voting)\n- kolla-ansible-centos-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/de5ce91b13894deeabe648537fc7c4f1 : SUCCESS in 2h 13m 45s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/024cd85afce14d988618435d7f049242 : TIMED_OUT in 2h 46m 17s (non-voting)\n- kolla-ansible-centos-source-zun https://zuul.opendev.org/t/openstack/build/1826225dbc354450923afe386e9f8fbd : FAILURE in 1h 33m 23s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/954b4001a55d400abbd2a5a1d4187d56 : SUCCESS in 1h 58m 19s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/36bf51f413054bd69e6f140b267c8c49 : FAILURE in 1h 38m 30s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/0e8a8a6e880f458f9acd3322485c2395 : TIMED_OUT in 2h 14m 54s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/db8a7b7878a2462082dd865ccfbc1370 : SUCCESS in 1h 27m 56s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/5f170bde04314e799480a8b724f5c542 : SUCCESS in 1h 38m 12s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/7a2bdc5c91b848c086a5e071c0fa3d3d : SUCCESS in 28m 14s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/39d60b46e6aa41759d7a5776599e2353 : SUCCESS in 23m 51s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/9375421bd431430bbabc0732f741f124 : SUCCESS in 1h 00m 18s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f733f8634933470ea89315cca4641e5b : SUCCESS in 8m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acb99b15be3e4679896b1a9f4c462757 : SUCCESS in 8m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/faf2f0688fbf4180aaeae6bfb5913ea0 : SUCCESS in 6m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/104f5e7ad68d4401b7330ad91d5442ef : SUCCESS in 7m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b906b7d262345ec87937f02fb8d1519 : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a926af2d518340b69290523edbea957a : SUCCESS in 7m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/81cafcc3ce7d4fb1b646d0cf106e56f2 : SUCCESS in 6m 49s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/54da995b661846da9321a5d3ffdaa7a5 : SUCCESS in 26m 43s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/95e1276fcd184b0b960424f764dd435f : SUCCESS in 33m 29s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ba0c17c8c526449c8dd741bb7cb39c31 : SUCCESS in 1h 01m 20s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7dd83067cc0640a78ccf587fbc47ed31 : SUCCESS in 1h 08m 56s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/39fbe5d0dade42d69b829f3592c61f28 : SUCCESS in 1h 12m 27s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/dc585931de194ad2ab438c74c5a80896 : FAILURE in 28m 45s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/6aa6aeb2971644afb8f30f413c519b79 : FAILURE in 27m 29s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/051985ed4994479b94078b2043b44f72 : SUCCESS in 1h 40m 13s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/5d29cd1a5b954b6c9ad16c4998509d2b : SUCCESS in 1h 05m 18s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2f7be193c072e7808c2d67c49e95d9d70eb7a189","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 19:08:22.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1cfa45b80a9447e58d13d89fa0715ff0 : TIMED_OUT in 3h 01m 02s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2df3bc61d37284bb003b4c410fe4bdcdf83804b3","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-25 19:18:45.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/8fb25fcd60244ba39120bb59a62b4202 : SUCCESS in 32m 31s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/7c32cd90b54a4e7cb2495b9a34c7c0b6 : SUCCESS in 39m 17s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d67d702aa65d52ab46bc2fab18aefaf1b4382e55","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2020-03-01 17:25:58.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"da85ff281a582e256f1c88fb6b70f5d781c98a3f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-03-01 17:55:52.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/98b01f917a9340a7b3a8725b38784b94 : SUCCESS in 27m 47s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/0766a12270524b8da59b45a966fc8bd1 : SUCCESS in 29m 03s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"04b0560d31891e49491a4e8c49620665b46c09fb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-01 17:59:38.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7687fc2ef6794068a53523d9a9b1af71 : FAILURE in 30m 52s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d1ff577b384e27e7b564cfb6ae73af3128f476df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-01 19:58:43.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/f35ca345708a4b69bf1efad8a572ab40 : SUCCESS in 28m 08s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/572dc54b7890412488e21df7c181a5c1 : SUCCESS in 42m 58s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/b41254fb2d2e4a2ea5400d9ae246f137 : SUCCESS in 29m 22s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/8f31dbc768014ebf862f2e7787c2656b : SUCCESS in 37m 20s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/7c66ca7abf4241288bce933a60da372d : SUCCESS in 1h 22m 45s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/aa3f1bfdaa7f4a4e82d65189f109f8d0 : SUCCESS in 1h 20m 16s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/f67d97940f9047cb86028a41c51be603 : SUCCESS in 1h 16m 49s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/3382d4624e5f49599d6b638920d5360d : SUCCESS in 1h 09m 15s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/54123074e6ed4b918966049dc113ed05 : SUCCESS in 1h 31m 33s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/fda081ee29f84efbb810dd82278fcc7a : SUCCESS in 37m 45s (non-voting)\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/1f2c40c2f56c456f8d1b29643e7982d9 : SUCCESS in 54m 59s (non-voting)\n- kolla-ansible-centos-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/fea6f57f56024673ab56b91c4ced44fa : SUCCESS in 2h 02m 40s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/eb715e93704a4032b0e646ff032378cb : SUCCESS in 2h 16m 04s (non-voting)\n- kolla-ansible-centos-source-zun https://zuul.opendev.org/t/openstack/build/4be42cc988d242efaa7c4e7df8ab14af : SUCCESS in 2h 10m 03s (non-voting)\n- kolla-ansible-centos8-source-zun https://zuul.opendev.org/t/openstack/build/65ac73df2cf14ec382fc5166d24818b0 : SUCCESS in 1h 55m 40s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/d2cbd84ffa794bbd9db9b08d5fdf6421 : FAILURE in 1h 26m 44s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/39b75b5a15934b05ab39da24aeef0727 : SUCCESS in 1h 52m 30s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/aac5a6997eb84919b9cfe4a1cf201fc9 : SUCCESS in 1h 24m 43s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/47752725f66c4cc2809789c54b36671d : SUCCESS in 1h 18m 57s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/21c79798198042e4b3b12fe6f1770280 : SUCCESS in 24m 54s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/9f7161a9d7974c8d8e762e1fef52ed80 : SUCCESS in 21m 22s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/bec9d2337f5e45d9ba03f19bc835d8d6 : SUCCESS in 1h 03m 35s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/032d3eea445e4494857956c52824f674 : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0968ae70cfa34ed0b1d2ae99e67636ef : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d197c993eb5240ac87162beaeff53688 : SUCCESS in 5m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f65bd73c0c03446597ddd769c66838e1 : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/16a18823b9d145479737599e399001cb : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/977f40c4102e444885b3064c107a2d80 : SUCCESS in 6m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3d3f98a8eac1418e96541c65b87fc57a : SUCCESS in 5m 42s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0247bb058a574da5a2614cf91da75485 : SUCCESS in 24m 42s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/391f7ac88f45445190d6a635d69acf04 : SUCCESS in 25m 33s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/67f898174bf7462eb3bb243a676f8084 : SUCCESS in 1h 02m 10s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9653c97969864298b7d1b4aac358e6f5 : SUCCESS in 57m 38s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ea76da717f4546d8913bf3a457352b80 : SUCCESS in 1h 05m 43s\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/73ed926e7cff4dd0be1effbbbbc9ae8d : SUCCESS in 1h 30m 12s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/2ac7de97fe7f4aba97691a36318274b2 : SUCCESS in 1h 42m 39s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/66b2db7eee434977b0d356e0bf1e3265 : SUCCESS in 1h 32m 19s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/a4bb552179a14e0cbfd58536394eb34e : SUCCESS in 1h 02m 20s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7317c52787f3cf93b5321242bab06f95839d0340","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-02 10:35:16.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"c03b816af768dc02d77eb63b2b173f27d893733d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-02 10:39:45.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"9cd950d061acad939bd8a62a25d9763412dfff20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 10:40:02.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"0b799a640a5b9bf705a4c011aa5862c2adfc4a0b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 12:26:57.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/204019a265734c1eb9957630e63ed2f1 : SUCCESS in 45m 53s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/daaf9bbb43dd4df284c93dd6898fa7e4 : SUCCESS in 39m 52s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/78b5ed24415b4d8d84fbf555f705c365 : SUCCESS in 1h 40m 02s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/c1431ddfe7a04ed298ea11598ae1c7f9 : SUCCESS in 1h 31m 52s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/5182631df5924f749b4c0be7b9c52908 : SUCCESS in 1h 41m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e7cd6def630e4c8b92a3f64786777145 : SUCCESS in 7m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d83c771944746db9f46e2b6a65700f0 : SUCCESS in 8m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/962b47652c404d75bc67e8d20f059aaa : SUCCESS in 5m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d7a25308a6424f299aba0cd46b0c7cc5 : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f28e610ed86f48fbb6ef5efd5bf983ec : SUCCESS in 6m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2a18dbc9ef534c89b6735a68d1502db2 : SUCCESS in 7m 11s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/77400183a0a84307bf8bc8de139bfd44 : SUCCESS in 36m 14s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c8cb566f456143c8ab238477b92cf19a : SUCCESS in 1h 00m 15s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4dfce732482144a5b2808423f40c0948 : SUCCESS in 1h 29m 12s","accounts_in_message":[],"_revision_number":5},{"id":"76a7cfdca284907b42c7f0756f86e680114d4e65","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 12:26:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"5b3164fe0262613b5e049931571a042c7e9b6bb7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 12:29:42.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0c4fb24f9434453840b5ace18c02665 : SUCCESS in 2m 08s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a16901471cc4b56a29eab7ada4ad931 : SUCCESS in 1m 46s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"861f55fbfd7558d56008e6275b0c6f5993346b95","revisions":{"eb16e14b74f440df4cf50b20d4de5b884de69786":{"kind":"REWORK","_number":1,"created":"2020-02-19 21:59:46.000000000","uploader":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"ref":"refs/changes/58/708758/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/58/708758/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/708758/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/708758/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/708758/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/58/708758/1"}}},"commit":{"parents":[{"commit":"ceae02f09d1c4e67013485c7323c11f8655c9c1e","subject":"Get rid of Python 2 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ceae02f09d1c4e67013485c7323c11f8655c9c1e"}]}],"author":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2020-02-19 21:39:33.000000000","tz":60},"committer":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2020-02-19 21:58:52.000000000","tz":60},"subject":"Add a generic header","message":"Add a generic header\n\nThere is already a generic footer block. This change introduces\na generic header block.\n\nBy default, the label is set in this header block.\n\nChange-Id: I9692dda817ef134d647247431565e1b58cf9da41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eb16e14b74f440df4cf50b20d4de5b884de69786"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eb16e14b74f440df4cf50b20d4de5b884de69786"}]},"branch":"refs/heads/master"},"4e140f39fad0b0e3ab9ab6f8c63be49b962e04ac":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-02-24 15:30:54.000000000","uploader":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"ref":"refs/changes/58/708758/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/58/708758/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/708758/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/708758/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/708758/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/58/708758/2"}}},"commit":{"parents":[{"commit":"89a66a45f958a8ef6a9a8730c5d19025a681fdfe","subject":"Merge \"Revert \"Upgrade virtualenv in pre\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/89a66a45f958a8ef6a9a8730c5d19025a681fdfe"}]}],"author":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2020-02-19 21:39:33.000000000","tz":60},"committer":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2020-02-24 15:30:54.000000000","tz":0},"subject":"Add a generic header","message":"Add a generic header\n\nThere is already a generic footer block. This change introduces\na generic header block.\n\nBy default, the label is set in this header block.\n\nChange-Id: I9692dda817ef134d647247431565e1b58cf9da41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4e140f39fad0b0e3ab9ab6f8c63be49b962e04ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4e140f39fad0b0e3ab9ab6f8c63be49b962e04ac"}]},"branch":"refs/heads/master"},"a3a0e849cc78ff9f2928f3241087631f966ad6f1":{"kind":"REWORK","_number":3,"created":"2020-02-25 15:54:50.000000000","uploader":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"ref":"refs/changes/58/708758/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/58/708758/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/708758/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/708758/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/708758/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/58/708758/3"}}},"commit":{"parents":[{"commit":"89a66a45f958a8ef6a9a8730c5d19025a681fdfe","subject":"Merge \"Revert \"Upgrade virtualenv in pre\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/89a66a45f958a8ef6a9a8730c5d19025a681fdfe"}]}],"author":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2020-02-19 21:39:33.000000000","tz":60},"committer":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2020-02-25 15:54:28.000000000","tz":60},"subject":"Add a generic header","message":"Add a generic header\n\nThere is already a generic footer block. This change introduces\na generic header block.\n\nBy default, the label is set in this header block.\n\nChange-Id: I9692dda817ef134d647247431565e1b58cf9da41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a3a0e849cc78ff9f2928f3241087631f966ad6f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a3a0e849cc78ff9f2928f3241087631f966ad6f1"}]},"branch":"refs/heads/master"},"49ad631624a729b28dc19c1b6525e27d52d552c4":{"kind":"REWORK","_number":4,"created":"2020-02-25 16:04:06.000000000","uploader":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"ref":"refs/changes/58/708758/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/58/708758/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/708758/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/708758/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/708758/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/58/708758/4"}}},"commit":{"parents":[{"commit":"89a66a45f958a8ef6a9a8730c5d19025a681fdfe","subject":"Merge \"Revert \"Upgrade virtualenv in pre\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/89a66a45f958a8ef6a9a8730c5d19025a681fdfe"}]}],"author":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2020-02-19 21:39:33.000000000","tz":60},"committer":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2020-02-25 16:03:50.000000000","tz":60},"subject":"Add block labels to all Dockerfiles","message":"Add block labels to all Dockerfiles\n\nChange-Id: I9692dda817ef134d647247431565e1b58cf9da41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/49ad631624a729b28dc19c1b6525e27d52d552c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/49ad631624a729b28dc19c1b6525e27d52d552c4"}]},"branch":"refs/heads/master"},"861f55fbfd7558d56008e6275b0c6f5993346b95":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-03-01 17:25:58.000000000","uploader":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"ref":"refs/changes/58/708758/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/58/708758/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/708758/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/708758/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/708758/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/58/708758/5"}}},"commit":{"parents":[{"commit":"fecd325c67fb7700d86f682be57c1ef456ed7707","subject":"Merge \"Actually disable EPEL, and epel-modular\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fecd325c67fb7700d86f682be57c1ef456ed7707"}]}],"author":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2020-02-19 21:39:33.000000000","tz":60},"committer":{"name":"Christian Berendt","email":"berendt@betacloud-solutions.de","date":"2020-03-01 17:25:58.000000000","tz":0},"subject":"Add block labels to all Dockerfiles","message":"Add block labels to all Dockerfiles\n\nChange-Id: I9692dda817ef134d647247431565e1b58cf9da41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/861f55fbfd7558d56008e6275b0c6f5993346b95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/861f55fbfd7558d56008e6275b0c6f5993346b95"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
