)]}'
{"id":"openstack%2Fkolla~708810","triplet_id":"openstack%2Fkolla~master~I07c79c078cd1620291e78a491f63bfba505296b7","project":"openstack/kolla","branch":"master","topic":"add-apache-to-neutron","hashtags":[],"change_id":"I07c79c078cd1620291e78a491f63bfba505296b7","subject":"Add Apache packages to neutron server containers","status":"ABANDONED","created":"2020-02-20 07:06:50.000000000","updated":"2020-04-24 09:23:39.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fa3608e61d63400baf8cb462295a67b1a962264b","_number":708810,"virtual_id_number":708810,"owner":{"_account_id":29344,"name":"白永君","email":"bai.yongjun@99cloud.net","username":"yj.bai"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-03-04 10:13:51.000000000","_account_id":29344,"name":"白永君","email":"bai.yongjun@99cloud.net","username":"yj.bai"},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-02-20 14:05:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-02-20 13:04:16.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29344,"name":"白永君","email":"bai.yongjun@99cloud.net","username":"yj.bai"},{"value":-1,"date":"2020-02-24 10:05:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"date":"2020-03-04 09:50:25.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":29344,"name":"白永君","email":"bai.yongjun@99cloud.net","username":"yj.bai"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":29344,"name":"白永君","email":"bai.yongjun@99cloud.net","username":"yj.bai"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":29344,"name":"白永君","email":"bai.yongjun@99cloud.net","username":"yj.bai"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":29344,"name":"白永君","email":"bai.yongjun@99cloud.net","username":"yj.bai"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-20 13:04:16.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-20 14:05:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-24 10:05:09.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-03-04 09:50:25.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"fa878004d4709236697918a624d824d3bf164ee2","author":{"_account_id":29344,"name":"白永君","email":"bai.yongjun@99cloud.net","username":"yj.bai"},"date":"2020-02-20 07:06:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dfcee6413f2b4ced84c8972b219aa00b97d2516a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 07:37:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1173bacd8d184f0aad175e929a65dbdb : FAILURE in 29m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"23b73e127ac6db6698fb853930e7961341c7e6e5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-20 07:38:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/17989fc1eb284d169971e875fd9d74e4 : SUCCESS in 30m 53s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/bb4611116f014d788b0726b68f04a942 : SUCCESS in 26m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3edf973f258c2ab986dcae4048eef6d811fa4623","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 10:03:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/393fdf399a7f4cd49058c9b218c8cee4 : FAILURE in 36m 54s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/cee8c1dff60249f28883dd1eef6785cd : SUCCESS in 45m 01s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/30f121f5b5474c558b5e5bf0ca94eb88 : SUCCESS in 35m 11s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/68ff6142eb0b4d9e8d21b74b47fec98e : SUCCESS in 31m 13s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/adb40fc5d8b140d1b96a1d3473437483 : SUCCESS in 1h 27m 54s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/2f1138f715954160a966179d05e23b0f : SUCCESS in 1h 10m 43s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/b21362e9d4ef4d7c8f9038e0c6c3e821 : FAILURE in 37m 56s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/09f9a2b19ae448c6b38d08999b82d918 : SUCCESS in 1h 15m 06s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/51abbf496f0244e8ac82682588710ede : SUCCESS in 1h 40m 19s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/ee420dfc321246ce87f0190130ce35e9 : SUCCESS in 52m 13s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/37ae277364bf4517a95b433dc3b6ea0d : SUCCESS in 25m 24s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/86009c5962d34e01a7ce8f820e3d7e19 : SUCCESS in 24m 23s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/dbe11aaf217d405e8ff3fe84557a1a07 : FAILURE in 57m 10s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4d7c5cdee760499aa06cc3ab339fb391 : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a46a920908774264823b5c705c364882 : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/40b948c2f2a34672aa456ab0196aa57e : SUCCESS in 5m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2039c5d04eef4d5d913c8ce47a75acfa : SUCCESS in 6m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9e1f915b1d9d42d6b5456da036e9312d : SUCCESS in 9m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97cc3dc67b134030a5ee0540bc2c8bd0 : SUCCESS in 8m 36s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e2a4410cbafb43399c84f1a5beac0e54 : SUCCESS in 26m 58s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/066e70fbe5dd43a5b403ada66d374f27 : SUCCESS in 34m 59s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/74a43054f5e64ded8a3146ffd96f4174 : SUCCESS in 56m 59s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b5cc416fb80249b389f5da59c391329a : SUCCESS in 54m 56s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/61a9ec71fc4c411194020f091cffe938 : SUCCESS in 1h 05m 08s","accounts_in_message":[],"_revision_number":1},{"id":"3771b94021eb2c45ad040cc7e984e7d245e0165a","author":{"_account_id":29344,"name":"白永君","email":"bai.yongjun@99cloud.net","username":"yj.bai"},"date":"2020-02-20 12:32:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"940ffd8905cf2100924ee90e50204bd82775889c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-20 13:04:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/6614639f071b4e50b5bb01dd270ef412 : SUCCESS in 30m 18s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/774089b710484624b13776757c3eb49e : SUCCESS in 28m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e7ac89ff0e0b418aca5c0117d49d338e8b4bdf86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 14:05:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/3310ef1e4756412686ae616c5a95a572 : SUCCESS in 36m 16s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/978ed22036044ba8aa7137cd8947c083 : SUCCESS in 45m 44s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/8d6d9eede7b54a0d8784ed9e1aeb162d : SUCCESS in 24m 12s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/378e4ae2a3e64ee3bb52c544db0affcf : SUCCESS in 33m 25s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/095b192c8667406f91290ffdfe74ffa8 : SUCCESS in 1h 16m 33s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/2c5b3e3057d841798b74866ab7593768 : SUCCESS in 1h 07m 47s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/5f18607127c54919bab3e1da92bf75a3 : SUCCESS in 1h 22m 55s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/4a75a31372c0418a94866c61e527d273 : SUCCESS in 1h 16m 23s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/f28edd8204b74c28ba62b2bb3f9e594d : SUCCESS in 1h 30m 53s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/34d43e94e6aa47b1a5ae86a1873fa7dc : SUCCESS in 35m 15s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6462f3ae16724834aed31abd943d38a9 : SUCCESS in 25m 09s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/bc9a6bc79b304a40ae66902bee705e23 : SUCCESS in 26m 45s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/79359de98e9f42bf9346059fe39e57d7 : FAILURE in 59m 39s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e433a2c6d194836862ce98a48b12985 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d517e689eea242e8ba78b50c571a23cb : SUCCESS in 6m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0786bb0e877a4143bed190d0b51bc8d8 : SUCCESS in 5m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/958aef6082244a54b6853e00bc721dbf : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fcf1f2f8e7e043d98f136ecc8e69d0c6 : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2097c5f18dc6430288718dd4f9c36cda : SUCCESS in 5m 57s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f18935029b754a1da118a34290c6391b : SUCCESS in 24m 54s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1e2becfdcc61411eb21e0a39beec6751 : SUCCESS in 33m 35s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1363007cbed644769242319c24f99c5b : SUCCESS in 1h 03m 47s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e8e190e27b5a4e66a611cfad5647c310 : SUCCESS in 1h 00m 30s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/049bccbb65714db899d0d916bb57237d : SUCCESS in 1h 10m 50s","accounts_in_message":[],"_revision_number":2},{"id":"76c2b2562cfd495c031b579656154f4a1217c8e3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 15:56:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/756f8754699a48d4be089b8b2e2be4ef : TIMED_OUT in 3h 01m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"68493ec02bf2750c1ae386eeb6684f6cf14da2a4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-21 16:33:08.000000000","message":"Patch Set 2: Code-Review+2\n\nThanks","accounts_in_message":[],"_revision_number":2},{"id":"8c67eda7e4c958aab6b981ff398fe0215bf0d9c1","author":{"_account_id":29344,"name":"白永君","email":"bai.yongjun@99cloud.net","username":"yj.bai"},"date":"2020-02-24 06:25:03.000000000","message":"Patch Set 2:\n\n\u003e Thanks\n1.  I started the neutron api with apache in the docker container use kolla.There was no problem with the first three requests (use openstack networking list) after startup. but the fourth started to be problematic.\nhttps://bugs.launchpad.net/neutron/+bug/1864418\n\n\n2. If the neutron api use apache alone. an additional container would be needed to start the RPC service: \n# /usr/bin/neutron-rpc-server --config-file /etc/neutron/neutron.conf --config-file /etc/neutron/plugins/ml2/ml2_conf.ini\n\ni think the first problem is very troublesome.  We can\u0027t move on until the first problem is solved.\n\ncan we use  mod_proxy?","accounts_in_message":[],"_revision_number":2},{"id":"805827e77ded749ef37e49eede6d4bd9c28d0e90","author":{"_account_id":29344,"name":"白永君","email":"bai.yongjun@99cloud.net","username":"yj.bai"},"date":"2020-02-24 07:45:35.000000000","message":"Patch Set 2:\n\n\u003e \u003e Thanks\n \u003e 1.  I started the neutron api with apache in the docker container\n \u003e use kolla.There was no problem with the first three requests (use\n \u003e openstack networking list) after startup. but the fourth started to\n \u003e be problematic.\n \u003e https://bugs.launchpad.net/neutron/+bug/1864418\n \u003e \n \u003e \n \u003e 2. If the neutron api use apache alone. an additional container\n \u003e would be needed to start the RPC service:\n \u003e # /usr/bin/neutron-rpc-server --config-file /etc/neutron/neutron.conf\n \u003e --config-file /etc/neutron/plugins/ml2/ml2_conf.ini\n \u003e \n \u003e i think the first problem is very troublesome.  We can\u0027t move on\n \u003e until the first problem is solved.\n \u003e \n \u003e can we use  mod_proxy?\n\nTripleo seems to use mod_proxy method ?","accounts_in_message":[],"_revision_number":2},{"id":"b20d22c9388b768338f63324b1752ba3ce4f59dc","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-02-24 10:05:09.000000000","message":"Patch Set 2: Code-Review-1\n\nyes, we need neutron-rpc-server when we use WSGI in Neutron - I don\u0027t think we should use mod_proxy, but rather spawn an additional container image with neutron-rpc-server (it will fork processes, makes sense to isolate it) - but we could use the same container image, just run it second time with another name (like neutron_rpc_server) in kolla-ansible.\nI\u0027d like to see a kolla-ansible change related to this change, before we merge this - is there one? Can\u0027t seem to find it.","accounts_in_message":[],"_revision_number":2},{"id":"4aaf768ecccaaec3c5acb8c11ac25ac61dcf9bb2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-24 10:17:58.000000000","message":"Patch Set 2:\n\n\u003e yes, we need neutron-rpc-server when we use WSGI in Neutron - I\n \u003e don\u0027t think we should use mod_proxy, but rather spawn an additional\n \u003e container image with neutron-rpc-server (it will fork processes,\n \u003e makes sense to isolate it) - but we could use the same container\n \u003e image, just run it second time with another name (like\n \u003e neutron_rpc_server) in kolla-ansible.\n \u003e I\u0027d like to see a kolla-ansible change related to this change,\n \u003e before we merge this - is there one? Can\u0027t seem to find it.\n\nSounds like a reasonable plan.","accounts_in_message":[],"_revision_number":2},{"id":"3eca135fe1b67b59afd374c414dcbdafc525e3b2","author":{"_account_id":29344,"name":"白永君","email":"bai.yongjun@99cloud.net","username":"yj.bai"},"date":"2020-02-24 11:48:24.000000000","message":"Patch Set 2:\n\n\u003e yes, we need neutron-rpc-server when we use WSGI in Neutron - I\n \u003e don\u0027t think we should use mod_proxy, but rather spawn an additional\n \u003e container image with neutron-rpc-server (it will fork processes,\n \u003e makes sense to isolate it) - but we could use the same container\n \u003e image, just run it second time with another name (like\n \u003e neutron_rpc_server) in kolla-ansible.\n \u003e I\u0027d like to see a kolla-ansible change related to this change,\n \u003e before we merge this - is there one? Can\u0027t seem to find it.\n\nBernard Cafarelli (bcafarel)  said need some package?\n\nImages built with this patch`s dockerfile have this problem\n\nhttps://bugs.launchpad.net/neutron/+bug/","accounts_in_message":[],"_revision_number":2},{"id":"b2cfa2d8f3ade7c574851b47d578643f87efb116","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-04 09:50:06.000000000","message":"Patch Set 2: -Code-Review","accounts_in_message":[],"_revision_number":2},{"id":"74de413b75e88696bd4e5df86562a79c59a7a68f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-04 09:50:25.000000000","message":"Patch Set 2:\n\nLet\u0027s hold off on this for now until we decide on wsgi vs. uwsgi.","accounts_in_message":[],"_revision_number":2},{"id":"e95165068f53733d964fbff4dcd042845f5e1d1b","author":{"_account_id":29344,"name":"白永君","email":"bai.yongjun@99cloud.net","username":"yj.bai"},"date":"2020-03-04 10:13:51.000000000","message":"Patch Set 2:\n\n\u003e Let\u0027s hold off on this for now until we decide on wsgi vs. uwsgi.\n\nok","accounts_in_message":[],"_revision_number":2},{"id":"9d3e6e9a49882676a210e3e261880ed8293c4094","author":{"_account_id":29344,"name":"白永君","email":"bai.yongjun@99cloud.net","username":"yj.bai"},"date":"2020-04-24 09:23:39.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e76d553883a4b1d8ab4a8a9e83f24f05fb1c9f4d","revisions":{"b806d7d94366ec51a424000eee7d2a32171b07dd":{"kind":"REWORK","_number":1,"created":"2020-02-20 07:06:50.000000000","uploader":{"_account_id":29344,"name":"白永君","email":"bai.yongjun@99cloud.net","username":"yj.bai"},"ref":"refs/changes/10/708810/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/708810/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/708810/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/708810/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/708810/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/708810/1"}}},"commit":{"parents":[{"commit":"1b0884af8a4dc90b459c2557acdf987ad0a7df75","subject":"change tripleo centos-7 build containers to centos-8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1b0884af8a4dc90b459c2557acdf987ad0a7df75"}]}],"author":{"name":"yj.bai","email":"bai.yongjun@99cloud.net","date":"2020-02-20 07:03:12.000000000","tz":480},"committer":{"name":"yj.bai","email":"bai.yongjun@99cloud.net","date":"2020-02-20 07:03:16.000000000","tz":480},"subject":"Add Apache packages to neutron server containers","message":"Add Apache packages to neutron server containers\n\nWill add ssl for neutron api and change startup from python to apache for\nneutron api so add apache packages to neutron server container.\n\nRelated: blueprint add-ssl-internal-network\n\nChange-Id: I07c79c078cd1620291e78a491f63bfba505296b7\nSigned-off-by: yj.bai \u003cbai.yongjun@99cloud.net\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b806d7d94366ec51a424000eee7d2a32171b07dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b806d7d94366ec51a424000eee7d2a32171b07dd"}]},"branch":"refs/heads/master"},"e76d553883a4b1d8ab4a8a9e83f24f05fb1c9f4d":{"kind":"REWORK","_number":2,"created":"2020-02-20 12:32:28.000000000","uploader":{"_account_id":29344,"name":"白永君","email":"bai.yongjun@99cloud.net","username":"yj.bai"},"ref":"refs/changes/10/708810/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/708810/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/708810/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/708810/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/708810/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/708810/2"}}},"commit":{"parents":[{"commit":"1b0884af8a4dc90b459c2557acdf987ad0a7df75","subject":"change tripleo centos-7 build containers to centos-8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1b0884af8a4dc90b459c2557acdf987ad0a7df75"}]}],"author":{"name":"yj.bai","email":"bai.yongjun@99cloud.net","date":"2020-02-20 07:03:12.000000000","tz":480},"committer":{"name":"yj.bai","email":"bai.yongjun@99cloud.net","date":"2020-02-20 12:32:09.000000000","tz":480},"subject":"Add Apache packages to neutron server containers","message":"Add Apache packages to neutron server containers\n\nWill add ssl for neutron api and change startup from python to apache for\nneutron api so add apache packages to neutron server container.\n\nRelated: blueprint add-ssl-internal-network\n\nChange-Id: I07c79c078cd1620291e78a491f63bfba505296b7\nSigned-off-by: yj.bai \u003cbai.yongjun@99cloud.net\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e76d553883a4b1d8ab4a8a9e83f24f05fb1c9f4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e76d553883a4b1d8ab4a8a9e83f24f05fb1c9f4d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
