)]}'
{"id":"openstack%2Fkolla~708816","triplet_id":"openstack%2Fkolla~master~Ifa6157cf495ff735e1d45899ecaf83a84bf9f793","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"Ifa6157cf495ff735e1d45899ecaf83a84bf9f793","subject":"Add Apache httpd setup to heat-api-cfn","status":"MERGED","created":"2020-02-20 07:59:21.000000000","updated":"2020-03-02 20:08:20.000000000","submitted":"2020-03-02 20:05:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"708816-1583179541210-b95df229","meta_rev_id":"dfd7854879171624b59036781f78d13ca8084fce","_number":708816,"virtual_id_number":708816,"owner":{"_account_id":29344,"name":"白永君","email":"bai.yongjun@99cloud.net","username":"yj.bai"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-02 20:05:41.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-03-02 16:06:03.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2020-03-02 12:43:17.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-02-25 14:52:23.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2020-03-02 12:43:17.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2020-02-26 15:03:13.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-25 14:52:23.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-02-26 15:03:13.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-03-02 12:43:17.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-03-02 16:06:03.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-02 20:05:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3db8f0e48da0bab3b40da34ef5a5fae000a9b964","author":{"_account_id":29344,"name":"白永君","email":"bai.yongjun@99cloud.net","username":"yj.bai"},"date":"2020-02-20 07:59:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dd2b9fdb53d24d881526f25b297d901e66b12ee8","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-20 08:38:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/b7181fc5714c48fc89195da344fdcf05 : SUCCESS in 31m 25s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/0a8f19fee92e448f963d53a342b91ed4 : SUCCESS in 38m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0d70e4d504d4de5908dedfba776ad2eb8c21e74e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 09:51:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/3d15835cb9ba4f0994699c7d603126e6 : SUCCESS in 33m 23s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/f354759c283f47dc8ed6e2d1db49620d : SUCCESS in 44m 45s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/d145a24bdf57477eb7d22337058d6fdf : SUCCESS in 24m 31s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/c1c63ee45a6d4a8c97d9824ddde8499d : SUCCESS in 27m 10s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/ddd9e773856e4085ab20b53531dd559a : SUCCESS in 1h 27m 52s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/34b0b421b5a0489f9eed1955e9eb84c5 : SUCCESS in 1h 11m 08s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/f57c7d44ebd449c889802013b5e1a783 : SUCCESS in 1h 31m 08s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/f096049a0cbc4346ac7aa01386aa2f83 : SUCCESS in 1h 09m 36s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/c434c6e6c1634a84aba341786ae7eb53 : SUCCESS in 1h 31m 53s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/6ae8d4287e134c7bb96bfb0a5ad8f4e7 : SUCCESS in 47m 16s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/b0fb968cf0894b0caf8d9c0fb9e2981f : FAILURE in 1h 39m 47s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/974643ef468446f784f1c6b38aef463b : SUCCESS in 1h 47m 38s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6987a579b2f74dc98916f5e6c6265855 : SUCCESS in 25m 10s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/757df9271164469f9d9270425f449171 : SUCCESS in 21m 11s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/9caf99b7e2674dfd900efbbfb19b9af0 : FAILURE in 50m 48s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/132b8c56a09947f9a36c79e5ba08d30a : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1edab90b15144cc29826dd2cac281e6f : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e22112af1fb1419ba5295cfa32d11d1c : SUCCESS in 4m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7184705d2b7b4460b99b44741618c49d : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/318f6afa4890464d89b5b21af877b479 : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bff186f772d4d6f99a6ae10c5347d7c : SUCCESS in 6m 11s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6b7868a9d0f14baabdd567581ff06ff7 : SUCCESS in 25m 48s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a4c20437a3234bb9910d6bf0ff4a9a72 : SUCCESS in 49m 30s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2e115634cff1430fb260a0b59aa3532a : SUCCESS in 1h 00m 38s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/880650cda713468ca89730e5ea3b5f27 : SUCCESS in 57m 26s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a7cd5c425de3480a9313b6c0f689c06c : SUCCESS in 1h 25m 26s","accounts_in_message":[],"_revision_number":1},{"id":"a4d3d6ca88bd4a472656bed8138e81a8a591a697","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-20 11:04:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0bc6a292c99e4420a2948a4314a365c6 : TIMED_OUT in 3h 02m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e93ce156123c4d2cf1201aa4a3a5b3a023138806","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-21 16:30:23.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nNeeds to be added to the image:\n\nCOPY extend_start.sh /usr/local/bin/kolla_heat_extend_start\nRUN chmod 755 /usr/local/bin/kolla_heat_extend_start","accounts_in_message":[],"_revision_number":1},{"id":"1cca72f832715dcca24fc885ac71ac9e21b93c63","author":{"_account_id":29344,"name":"白永君","email":"bai.yongjun@99cloud.net","username":"yj.bai"},"date":"2020-02-23 07:18:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a4cfbe2082dc5939a4c54e8ff1b069f589fca377","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-23 07:45:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/3969242d75944c29a67ef1b2218ac530 : SUCCESS in 25m 58s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/e777d16cc80b4889a92ff3632d339fba : FAILURE in 15m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"129a3cfa1c7bd79d36f43e9e63d403454251122d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-23 09:38:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/ab77129ff1904a2aa1133aae99cc2c4a : SUCCESS in 36m 44s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/910f5bd3554949eab6b4de56e3543918 : SUCCESS in 43m 55s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/91f0001647d346838af25f0849fcb3f5 : SUCCESS in 25m 55s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/e65cafa3853a4472ade0c9c86b8f7f92 : SUCCESS in 28m 27s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/4c2a77b4236d4e31b1db0ad1a45f6948 : SUCCESS in 1h 42m 32s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/9594ad3fb1e7437f9c370169e04b177b : SUCCESS in 1h 40m 32s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/3bdc0f20488640f7816465a7aba74372 : SUCCESS in 1h 43m 34s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/a6f8f92ce35f477a82e6637b59440b4d : SUCCESS in 1h 36m 24s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/0d9c784713014d70b51ecbb69d7a9d83 : SUCCESS in 1h 37m 55s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/d3b1c11f5f944b188e646130308f20b6 : FAILURE in 24m 37s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/b6ed887161cc40ae8e725e52e66b61cf : FAILURE in 1h 40m 09s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/15eaa1eb2f894d1a8771b5c38160aaad : TIMED_OUT in 2h 10m 38s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/4ed896ec67034953849d43a7554b33df : SUCCESS in 26m 20s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0d6d1d80ee494f259bb2083984ee5010 : SUCCESS in 19m 27s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a277c3f32c624ac18052e8a49aded4ad : SUCCESS in 1h 34m 20s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8917edf800bb46cf847d043c8433faef : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cb48557b454429e8db50500e9c833d4 : SUCCESS in 7m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c96989c697294e42a5b051f0c6182c42 : SUCCESS in 4m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fe224c5d99924b82b8630f6f1a5c9d0d : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc5a1daee7534ca9a048cc0c906e0e53 : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6605f4053534f4d9b444fb2d77d8675 : SUCCESS in 9m 59s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a83407741c1949c89f57b5a0535a716e : SUCCESS in 28m 45s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4fb9acaec0cb4d7899537f4b14faeaf2 : SUCCESS in 27m 52s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/36e38b414830499384d026ddd1efc0c2 : SUCCESS in 56m 48s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9a7f8435fb9b4c798badfc6b121643ed : SUCCESS in 1h 01m 33s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7f5d72b4464b4dff8bda47b0d800b92d : SUCCESS in 1h 42m 40s","accounts_in_message":[],"_revision_number":2},{"id":"a7436356d5ceef4741cc9745c8e8a5aabf19a564","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-23 10:19:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9c08806488a8474aa51ac7db55142e03 : TIMED_OUT in 3h 01m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"af51931d15a44c0665727c7743b3c0f335b41b50","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-24 14:46:24.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"da130d99f71fb812ccfb700104f2f12199a993ad","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-24 16:06:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/9b9d1729c51a4935807cf3edaa42b6af : SUCCESS in 29m 02s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/d66f454d3eb14118b806bc251bcafe3d : FAILURE in 19m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6b6ac2136ced7719a4abb69564565f34aca355eb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 17:55:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5ad2b955cc4b468495133f0d55b4b0cc : TIMED_OUT in 3h 01m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5544a2ad91b08efc1a44878935211d3d21f59a77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 19:17:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/f24d943291aa4f9eb4694afa3f4dddc1 : SUCCESS in 29m 21s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/db3c89784ce44fdf9d742f8477c4f32c : SUCCESS in 50m 23s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/abb08b9c8c124e7bb9232731b1a7185d : SUCCESS in 24m 44s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/ce3e23de214d4fcaa3169c84cfde1bc8 : SUCCESS in 29m 41s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/842ce3636a07464389be3341ad79416c : SUCCESS in 1h 35m 39s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/5fea1596675f43af971c8643b126eb31 : SUCCESS in 1h 16m 23s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/1770a8d00965471cbb2f1c0d09b17048 : SUCCESS in 1h 23m 23s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/c93d5ac584794f66b5dc56dfbbe23b62 : SUCCESS in 1h 15m 18s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/a350e2faeed94d48b39b31b111dd2e56 : SUCCESS in 1h 49m 38s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/cfd17f855f874cf1adbd05d59eb0ff65 : SUCCESS in 58m 09s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/a449a06936544a589a033838a5f67c13 : FAILURE in 1h 54m 23s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/476b84b8e6c147c6a77bf8ca4544a05c : SUCCESS in 2h 10m 45s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6273e6cb50f8487c8019e72d4c8cf16f : SUCCESS in 29m 00s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/359047838fd649d1b7b7551c062f4b1a : SUCCESS in 25m 54s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/64bff8df95114fb9b299572dd783251d : SUCCESS in 1h 10m 03s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2ba8a930aa9e43dfb401643e9accc2aa : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9360308a0ab445ff9f0149e509aa3ac0 : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/736d96ccb7d145f29b5f0d69dc62635c : SUCCESS in 4m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a4d82a281e8f47ec9731a8b9461a4bbc : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/550a0a948fa34e9bae98052c38dc16f6 : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3de6f88c2676426e8bda97791223daea : SUCCESS in 7m 42s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/61faf23ac24a47eaafa730de647e62f4 : SUCCESS in 26m 42s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c74fb689b40c4e0b8e1cbd98b026934c : SUCCESS in 36m 41s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f194ad42ac3a4316a69c8a56d58ebf7c : SUCCESS in 1h 08m 29s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2f6bf4e91584497a8e4bf4640ee7b756 : SUCCESS in 57m 08s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1c5a2dd7f7f144f899ef23bb18d29801 : SUCCESS in 1h 07m 53s","accounts_in_message":[],"_revision_number":2},{"id":"ffaf0a480e9ef278def496b863f52c679b37d07e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-25 14:52:23.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b7b5510e450eccb345c64624ed9f6b88a0516474","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-03-02 12:43:17.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c212a9291d5385e56cca40bc54f830e35defdf04","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 12:43:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"90c5497953a151200952aeef249073dfe6194b3f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 14:41:07.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/5b13dc450f164580bd8c2207526f7a61 : SUCCESS in 46m 20s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/d6becd6cd61e409193aed27c126665c7 : SUCCESS in 28m 46s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/4d5681dd8aa04ba58b32b10def9e2a0d : SUCCESS in 1h 44m 14s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/8011e51426f047b09b0a94792885649e : SUCCESS in 1h 19m 44s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/6d8d94fa219741209f3237bea312f569 : SUCCESS in 1h 51m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ffcadf296b7f44e5b8d7924d6a140c56 : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4cb876ef2ee431a8c65696a1669cd35 : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/158a8f97857845f686f3c3899a9998fc : SUCCESS in 4m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/38123dbc36814c5e83f4d8e11edc1c37 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22d1500212014d59b3af1e16e5efdfbd : SUCCESS in 6m 32s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1b007d12ff8d4ac0a3f7c6c0ba04f462 : SUCCESS in 35m 12s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/55d68b1fd32b43cbb23a57c9662fa9fb : SUCCESS in 1h 03m 53s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/849aeac730044e3bbd0f8c62235e2390 : FAILURE in 1h 15m 47s","accounts_in_message":[],"_revision_number":2},{"id":"c7042858b79b29b80a06983827da2c08b56aeb35","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-02 15:17:06.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"2cd0ad7c0425d1cb340eceda109f8d52297860ff","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 15:44:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b4eca2de0b0f465f90e483567ccb8896 : FAILURE in 24m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c73c88d59b0065c3b32f5232b7728e57295c36d3","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-03-02 16:06:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/62149249756442e59d05a54b2f0593bd : SUCCESS in 31m 36s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/7f54785288344c1bbb037a64941d95ec : SUCCESS in 47m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3969b70b70c12ab755adab7eb1936f5e0548b176","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 17:38:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/662d270978894124b06eb8f7eb37e6fd : SUCCESS in 34m 25s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/89828c73b61e443f9ff55956b3749d84 : SUCCESS in 43m 27s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/e58bada6312b4df79c493742cabce200 : SUCCESS in 40m 03s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/af24ae1d38fb4e95a53e9d7207f05b65 : SUCCESS in 34m 32s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/40cca812101347cdb789b39283d3a91d : SUCCESS in 1h 34m 28s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/917b56a3509a4356b3a2e1f3c75d27ed : SUCCESS in 1h 24m 00s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/535a44dbdf014740a8f70212c6abd2a3 : SUCCESS in 1h 29m 55s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/33dafdecb8844f81925267204e65217b : SUCCESS in 1h 29m 33s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/f7ab97d6bf484ef9a2ac36adba46a768 : SUCCESS in 1h 35m 43s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/ad3e451ec4e94d40b3723bf05637d0d4 : FAILURE in 46m 43s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/eaf905aae73641e7aa95cf47f4f599b3 : FAILURE in 1h 35m 46s (non-voting)\n- kolla-ansible-centos8-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/9dff36dd6c074eb7ae23a3ce98789b3d : SUCCESS in 2h 08m 19s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6899642bed754e14aee4bb6925dd6c02 : SUCCESS in 22m 22s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/dd66c323042c4ac8b669084ff842bbd1 : SUCCESS in 23m 52s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c8c19bcdf1c64298a81ec0d5b2b194ab : SUCCESS in 1h 04m 01s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e67dbdac6b084ee890d4c61ce410cb09 : SUCCESS in 7m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/973ebb4146424d0b8e762b7c8a92d6d7 : SUCCESS in 7m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5b6a84c3d68487d8cd6e35df917c2ad : SUCCESS in 6m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/93fad35a90364595a2a2e406631d511a : SUCCESS in 6m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4b2a268453d94d689feb55f28d56ffc9 : SUCCESS in 6m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e706cbdbd5dd4b9fb3da76408d15ac92 : SUCCESS in 7m 22s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/059ff24e2284476cb45791e692f2068e : SUCCESS in 32m 46s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d50e347403c248229ff06ceb459bb4a6 : SUCCESS in 33m 57s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/29df473383e4447d8ea255f5cce20e88 : SUCCESS in 1h 01m 12s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2f75168e62d445eeaaac398e6a485425 : SUCCESS in 1h 03m 14s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a8d49cb0de6b4ba5adffce3231236a01 : SUCCESS in 1h 14m 22s","accounts_in_message":[],"_revision_number":2},{"id":"8dc89077722e2f45a444061f4308d2bb3b7380d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 17:39:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d1ed58e640a647ea2108cad178bca9ad40a7e54d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 20:05:41.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/e1fb39be8a3e40b48b3c47a8911b1228 : SUCCESS in 42m 35s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/e07f931a98ff460891c01fdd2d8f1abe : SUCCESS in 30m 11s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/051b7ffe9c2741f58937396378844834 : SUCCESS in 1h 25m 53s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/c5c9cdcc29d248ef9412c75bfd7bf3d4 : SUCCESS in 1h 20m 57s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/767c513457e24dc4acd2d1165e2a1574 : SUCCESS in 1h 43m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e79a81d159d7497aa9f0bdb49dc017be : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f65b642addaf478588a76854c3ca73c8 : SUCCESS in 6m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/44faa593895a44b6a450d2059210c00d : SUCCESS in 7m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a5fc5fa10a344b8fb4bc2583c42406c4 : SUCCESS in 6m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23f9e05a33554260b1b06f3028e2e373 : SUCCESS in 10m 03s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8ed2727a27414c958a1f17fd24c4850a : SUCCESS in 34m 50s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c7916802fe604b408fa0b53a09b83d7e : SUCCESS in 1h 13m 43s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9f67c76fe408450d8f7ef858a68b99c2 : SUCCESS in 1h 06m 29s","accounts_in_message":[],"_revision_number":2},{"id":"5cf398b6f4bb4bb02bdb6a53ed5e45d5473a1998","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 20:05:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"dfd7854879171624b59036781f78d13ca8084fce","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 20:08:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b17a7d189b94472b70d16a2999e8d39 : SUCCESS in 2m 06s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c402f20e3e4c42bf59094c8d298cc679adeb85ee","revisions":{"b4b3023e58982c40d1fe772c8baf1123a20e9d7b":{"kind":"REWORK","_number":1,"created":"2020-02-20 07:59:21.000000000","uploader":{"_account_id":29344,"name":"白永君","email":"bai.yongjun@99cloud.net","username":"yj.bai"},"ref":"refs/changes/16/708816/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/708816/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/708816/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/708816/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/708816/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/708816/1"}}},"commit":{"parents":[{"commit":"1b0884af8a4dc90b459c2557acdf987ad0a7df75","subject":"change tripleo centos-7 build containers to centos-8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1b0884af8a4dc90b459c2557acdf987ad0a7df75"}]}],"author":{"name":"yj.bai","email":"bai.yongjun@99cloud.net","date":"2020-02-20 07:54:42.000000000","tz":480},"committer":{"name":"yj.bai","email":"bai.yongjun@99cloud.net","date":"2020-02-20 07:54:45.000000000","tz":480},"subject":"Add Apache httpd setup to heat-api-cfn","message":"Add Apache httpd setup to heat-api-cfn\n\nwill change heat-api-cfn up way from python to apache.\nand add tls.\nhttps://review.opendev.org/#/c/707131/\n\nRelated: blueprint centos-rhel-8\n\nChange-Id: Ifa6157cf495ff735e1d45899ecaf83a84bf9f793\nSigned-off-by: yj.bai \u003cbai.yongjun@99cloud.net\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b4b3023e58982c40d1fe772c8baf1123a20e9d7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b4b3023e58982c40d1fe772c8baf1123a20e9d7b"}]},"branch":"refs/heads/master"},"c402f20e3e4c42bf59094c8d298cc679adeb85ee":{"kind":"REWORK","_number":2,"created":"2020-02-23 07:18:13.000000000","uploader":{"_account_id":29344,"name":"白永君","email":"bai.yongjun@99cloud.net","username":"yj.bai"},"ref":"refs/changes/16/708816/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/708816/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/708816/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/708816/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/708816/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/708816/2"}}},"commit":{"parents":[{"commit":"1b0884af8a4dc90b459c2557acdf987ad0a7df75","subject":"change tripleo centos-7 build containers to centos-8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1b0884af8a4dc90b459c2557acdf987ad0a7df75"}]}],"author":{"name":"yj.bai","email":"bai.yongjun@99cloud.net","date":"2020-02-20 07:54:42.000000000","tz":480},"committer":{"name":"yj.bai","email":"bai.yongjun@99cloud.net","date":"2020-02-23 07:18:01.000000000","tz":480},"subject":"Add Apache httpd setup to heat-api-cfn","message":"Add Apache httpd setup to heat-api-cfn\n\nwill change heat-api-cfn up way from python to apache.\nand add tls.\nhttps://review.opendev.org/#/c/707131/\n\nRelated: blueprint add-ssl-internal-network\n\nChange-Id: Ifa6157cf495ff735e1d45899ecaf83a84bf9f793\nSigned-off-by: yj.bai \u003cbai.yongjun@99cloud.net\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c402f20e3e4c42bf59094c8d298cc679adeb85ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c402f20e3e4c42bf59094c8d298cc679adeb85ee"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
