)]}'
{"id":"openstack%2Fkolla~710068","triplet_id":"openstack%2Fkolla~stable%2Fstein~I51ebb3e35c08ab0e2402508c9f833aa0fa71cef0","project":"openstack/kolla","branch":"stable/stein","hashtags":[],"change_id":"I51ebb3e35c08ab0e2402508c9f833aa0fa71cef0","subject":"Bump stein versions","status":"MERGED","created":"2020-02-26 16:01:35.000000000","updated":"2020-02-27 11:19:00.000000000","submitted":"2020-02-27 11:16:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"710068-1582802197989-b16e9fe7","meta_rev_id":"2d416613f3268f009e4f42d31cba350d26be0bcf","_number":710068,"virtual_id_number":710068,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-27 11:16:37.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-02-26 16:02:56.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-02-27 08:51:16.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-02-27 08:51:16.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-26 16:02:56.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-02-27 08:51:16.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"},{"updated":"2020-02-27 11:16:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f0a0b9341e8fa77be631baae7b515dbdbefcfc81","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-02-26 16:01:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cfdd4273c538cf67b26b088b106d35bf85b416de","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-26 16:02:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f0d270c7b9bc62380dd7800ad56e2363d8ebcedf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 17:47:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/99704a7f973e448d8518e64531c0bfca : SUCCESS in 52m 31s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/cb041625f6264e968af9752abe29e29b : SUCCESS in 57m 26s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/f3722ae9332c45ca888f4dfa71d5f974 : SUCCESS in 1h 36m 24s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/c69afa19a80e4b35bffb9a05062dee91 : SUCCESS in 1h 21m 53s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/12f0c981773d4182b842f88375120576 : SUCCESS in 1h 34m 01s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/52c92f42e91c4a96bbb14ef9d8e61691 : SUCCESS in 49m 03s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/3597f1feb20149ab849f2b702f10e642 : FAILURE in 48m 28s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/2ca01fe3a3594ca3a378bc18a847f107 : SUCCESS in 47m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/833d8f76459944ffa8ecfe91b2e11587 : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e918f8141bd4125b2ab8fa70448bd05 : SUCCESS in 8m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fe8ec5d4194c47dfb19538032ee8b69e : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/86d4a1ecac8b463faae8bf4b209ee11b : SUCCESS in 6m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/360ef53ce63f4834931a2a1f64dfdd34 : SUCCESS in 7m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4cbc80d462c477981571563df4cc44c : SUCCESS in 4m 59s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/776d6a09dbd346f0a39c4b782cbd5dc4 : SUCCESS in 28m 47s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/050aa1c2907349669dbfc46189a9dc1e : SUCCESS in 29m 45s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d210af3add824c62a181f2042f8ad8b3 : SUCCESS in 57m 59s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e55a5256512f4c63b985b0e38bb125e3 : FAILURE in 40m 23s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/bab05ad514ab43f583e2b735b833682e : SUCCESS in 1h 07m 50s","accounts_in_message":[],"_revision_number":1},{"id":"3923439c5538ec2d19be3f758a87a47a6d5743e9","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-02-27 08:51:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4bf6f802cfd3b44288c70bf33322afeeecbd1c08","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 08:51:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8a760f874d61134ea150a4d49985c95c2ea89a36","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 11:16:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b5cbdf9b1984780d0f9af43fb9061a242f7c874c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 11:16:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/fb34543fa4964c0193b53d0d92b89057 : SUCCESS in 1h 06m 15s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/3ec151c0b90f450887789ba85a977d7c : SUCCESS in 1h 04m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/763e0b4a1ee2456295c074d89c3755c2 : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0946fb3d1f4420a9b8fc25d8fa48367 : SUCCESS in 7m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ae753c66661848068073e10c92c2476c : SUCCESS in 7m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0f58c81cd0114eadbf0e086e1aa5ab5a : SUCCESS in 7m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/85aaf577f2ba4bfebcbb0ddc002b0bba : SUCCESS in 6m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e191c200b1aa401a80fe7a7e5c509eb8 : SUCCESS in 6m 54s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0cf5ca5767b54c9d9db14d738cefdb23 : SUCCESS in 27m 34s","accounts_in_message":[],"_revision_number":1},{"id":"2d416613f3268f009e4f42d31cba350d26be0bcf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 11:19:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8949758f17b84e75b1101832834a913e : SUCCESS in 1m 34s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"85e1ef6092e96ef96a695fd6574f8d20a4fd52e5","revisions":{"85e1ef6092e96ef96a695fd6574f8d20a4fd52e5":{"kind":"REWORK","_number":1,"created":"2020-02-26 16:01:35.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/68/710068/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/68/710068/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/710068/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/710068/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/710068/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/68/710068/1"}}},"commit":{"parents":[{"commit":"cbfded882cb951c95d3832aae0b1c38711e4a654","subject":"Bump Gnocchi to 4.3.4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cbfded882cb951c95d3832aae0b1c38711e4a654"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-02-26 16:01:25.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-02-26 16:01:25.000000000","tz":60},"subject":"Bump stein versions","message":"Bump stein versions\n\nChange-Id: I51ebb3e35c08ab0e2402508c9f833aa0fa71cef0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/85e1ef6092e96ef96a695fd6574f8d20a4fd52e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/85e1ef6092e96ef96a695fd6574f8d20a4fd52e5"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
