)]}'
{"id":"openstack%2Fkolla~710073","triplet_id":"openstack%2Fkolla~stable%2Fstein~If552f7ffc22d4b25769c14c083179eecf5a93158","project":"openstack/kolla","branch":"stable/stein","hashtags":[],"change_id":"If552f7ffc22d4b25769c14c083179eecf5a93158","subject":"Ignore EM releases in version-check.py","status":"MERGED","created":"2020-02-26 16:18:25.000000000","updated":"2020-02-26 21:11:21.000000000","submitted":"2020-02-26 21:09:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"710073-1582751388277-d1a6a7d1","meta_rev_id":"a6bd8a7b09128e77ea7fecd670809ce77ceb56b5","_number":710073,"virtual_id_number":710073,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-26 21:09:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-02-26 16:38:55.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-02-26 16:42:17.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-02-26 16:42:17.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-26 16:38:55.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-02-26 16:42:17.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-02-26 21:09:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"37615c1f4d97eacccfa68215ce403e66937ddb03","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-02-26 16:18:25.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"eedfd7150268f027540f0a127444269dccdfec22","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-02-26 16:18:36.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit bd10e34510f29c53adb8f141dab5cc323c207b09","accounts_in_message":[],"_revision_number":1},{"id":"08ef844a84531e96bf88241fa417e17b5fb3d26c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-26 16:38:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"41ddd6d0a5dc88174dfdf5d6923c69c120bfa7a6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-02-26 16:42:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"674ce45083a33cb02f123f3759f2024725b55892","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 17:58:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/509b9f24a9b54b99b0a970e878479374 : SUCCESS in 58m 57s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/9a5cdb9b3533478aaeb9b7c95aa7a075 : SUCCESS in 1h 03m 03s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/71fd7899dd4c41edaab2d0cc9ca02762 : SUCCESS in 1h 18m 03s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/4c70377c692446929dbf27791adc5972 : SUCCESS in 1h 12m 23s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/6e4f25c44a0c4df28348fe489a422dae : SUCCESS in 1h 35m 48s\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/3bb65f88e1e84582a9315deec5df8b92 : FAILURE in 1h 00m 28s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/a330b1a8cefa4cda8304be0f997928c8 : SUCCESS in 57m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/105767ea066d403c80f1b0625f4afd0e : SUCCESS in 7m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cb28cc3242040a98c24057ab2a7a882 : SUCCESS in 10m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/54babf990e9e41038cfd46bc4dad34d3 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3522be76798442fbbfa34000ba3f936 : SUCCESS in 6m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c3d6941c40c2483ab7fc9789c1c36a14 : SUCCESS in 8m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ac43b384dfd480dbfb73d7da4555a9c : SUCCESS in 7m 28s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e377e9e996e147669b99a1eb315a8b59 : SUCCESS in 43m 57s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a0be5fe821d640efb930d942206a0271 : SUCCESS in 41m 22s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c17a8220d59e40c4a2012c1413c77659 : SUCCESS in 49m 33s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d51d3cba5bac4bb7a61ab0b091512e4d : FAILURE in 35m 13s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6acbad5f48b94e71b37bfb84bf4cae80 : SUCCESS in 1h 18m 49s","accounts_in_message":[],"_revision_number":1},{"id":"904c40d9a2a3342a21d28beca107859c49f02757","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 17:59:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cdddfb0dd8b3fce7f543afc6952d5de13e39d033","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 21:09:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b15680e5d2cd93f368a83fcd662f48c188392856","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 21:09:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/239c07bd1fe54ad7b90472ac7bda3269 : SUCCESS in 43m 13s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/836b3b1ee6074bb2b282476016f0ebb0 : SUCCESS in 47m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ce57b728f23748c2a7bc85b734a7485f : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a843986bdaa46de87569782c6e3990f : SUCCESS in 5m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/005c455982774c919ba6a2dc1cf1d7ec : SUCCESS in 7m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/49ea7be7d9b6490dbd134da75c7e105e : SUCCESS in 5m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6e73a4fe54fa41089fd37ddc22b4d378 : SUCCESS in 6m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a56ee6226c74c249b1c2f75a28d63b2 : SUCCESS in 5m 30s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/21b7b8be60dc450b9da7e4eccae06445 : SUCCESS in 29m 12s","accounts_in_message":[],"_revision_number":1},{"id":"a6bd8a7b09128e77ea7fecd670809ce77ceb56b5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 21:11:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d289cba2e5e5496ba8328f4a585e8448 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"aabd2ac49013f8f489691a0132edf9fc919b1243","revisions":{"aabd2ac49013f8f489691a0132edf9fc919b1243":{"kind":"REWORK","_number":1,"created":"2020-02-26 16:18:25.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/73/710073/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/73/710073/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/710073/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/710073/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/710073/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/73/710073/1"}}},"commit":{"parents":[{"commit":"cbfded882cb951c95d3832aae0b1c38711e4a654","subject":"Bump Gnocchi to 4.3.4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cbfded882cb951c95d3832aae0b1c38711e4a654"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-05-20 18:00:33.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-02-26 16:18:25.000000000","tz":0},"subject":"Ignore EM releases in version-check.py","message":"Ignore EM releases in version-check.py\n\nWe don\u0027t want to use Extended Maintenance (EM) tags, since these are\njust markers of a point in time and not available as released artifacts.\n\nChange-Id: If552f7ffc22d4b25769c14c083179eecf5a93158\n(cherry picked from commit f528ad81026526af02610aa3e08200fd53b87ab9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aabd2ac49013f8f489691a0132edf9fc919b1243"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aabd2ac49013f8f489691a0132edf9fc919b1243"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
