)]}'
{"id":"openstack%2Fkolla~710318","triplet_id":"openstack%2Fkolla~stable%2Fstein~Ie7a5d74e4640e5ec5e9885b4733551cf050d0853","project":"openstack/kolla","branch":"stable/stein","hashtags":[],"change_id":"Ie7a5d74e4640e5ec5e9885b4733551cf050d0853","subject":"Bump stein versions","status":"MERGED","created":"2020-02-27 17:10:50.000000000","updated":"2020-03-13 12:16:04.000000000","submitted":"2020-03-13 12:14:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"710318-1584101661235-c6fdaf72","meta_rev_id":"78e4f34dbc87805c1975bcfc59cd765f2d5d6485","_number":710318,"virtual_id_number":710318,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-13 12:14:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-03-13 11:17:09.000000000","_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-03-04 09:47:17.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2020-03-13 11:20:08.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-13 11:20:08.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-03-13 12:14:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ed88e99fb5f910e4d8f37d92d0c69fb35df81c2b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-02-27 17:10:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7fa112d51eca61543acdeaa545c35ac55af88c80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 22:19:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/9aadd4c58a97427db530aff55273b6a0 : SUCCESS in 41m 14s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/8e2f3100b7d14e27b9150b134926f375 : SUCCESS in 1h 11m 48s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/47b852767daf4d69afcecd3b785365af : SUCCESS in 1h 23m 16s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/e2f60af945724ff5866c43ae1f4d93ae : SUCCESS in 1h 20m 21s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/a08ac1fe0fdc46da8e09f1912ab869a8 : SUCCESS in 1h 03m 16s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/92330583f9a143fc856b3ce3705cb25a : FAILURE in 44m 49s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/59e13da5d1374e918cb4df872b5f067f : SUCCESS in 59m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f2c2ae3e12704ffcbf15553b0e8b116d : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a12ad94ff494494924f9157cd555556 : SUCCESS in 5m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d41f0a1a739849ada65323bfca57bab0 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d09624a7a1ea4816903fb039e4f9d986 : SUCCESS in 5m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/eccee4c289f1449897b49051ce410cf0 : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0152f744e8b45fdbad5d81f231e3e66 : SUCCESS in 6m 11s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/40ca267ee31d4db6a4f054a248ffc1f1 : SUCCESS in 26m 40s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/32823d3dea7640988d1a22ae34ed59da : SUCCESS in 40m 28s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/805109f9b8434bdf83a2b03ccdb81bfa : SUCCESS in 57m 58s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a0c2a43bbc4c4c7f821c51572d4de7b6 : FAILURE in 35m 41s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0591c26ac3814b468abf8c605a0abc3d : SUCCESS in 1h 03m 19s","accounts_in_message":[],"_revision_number":1},{"id":"303b6c03020e1e9a55f53e209a7e67196b501c1c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-04 09:47:09.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ee6bf5ec13d30084f0c89720c57121ef14c2c2da","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-04 09:47:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"db3859ef5f9345c831514703d018fdbfeec3b9f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 10:57:11.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/f221cf0b72664a00a301cabe489e9440 : SUCCESS in 1h 02m 26s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/ecd42ffa528e487db293c5082140bc8d : SUCCESS in 1h 08m 51s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/8fcd958fbe9d420bacd3b0b1821671b1 : FAILURE in 6m 36s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/3284c5749f01444bbc4b9af85a58dd9e : FAILURE in 7m 19s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/885ad6d98eca42149537dbeb68c5756a : FAILURE in 6m 53s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/8872cf8883114d12947f7627b757be01 : SUCCESS in 56m 53s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/a93bb3c73b0540c8aad9a3816ce88f9b : FAILURE in 49m 13s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/d713a6c3f28f4fe89371561b032dd442 : SUCCESS in 49m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a5a6f383bd5d44e7bced467255959666 : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/585f8ac20d6841fea37c07d18a674fc8 : SUCCESS in 5m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5e87e26ae9924a31afade3cee0e6f53e : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e1dddbf02c242d9ae65921aa6b73fc8 : SUCCESS in 4m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/99cd744a73f4447baa87b838b67aedb9 : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f46310212da446779e982a0906fc587c : SUCCESS in 4m 23s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9556cc2c4a0c487689d8c30fc02ccc37 : SUCCESS in 24m 09s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/707da0d10c1b44939c32a4d52b534feb : SUCCESS in 25m 38s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/94203ad546db47a78cee1c6362868542 : SUCCESS in 56m 24s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c022d591b6eb475da5abbc7c8109a167 : FAILURE in 33m 43s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/96c64a567010462e9d90c53c8854096a : SUCCESS in 56m 26s","accounts_in_message":[],"_revision_number":1},{"id":"f766a07e5a48cae9afc4fef906a32df89aa19ce5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-10 13:02:36.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bd902179f7670e2c344738c622403007b8c81682","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 15:11:56.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/40d1d5344d964d7f80dfee7b6a1d23b4 : SUCCESS in 1h 13m 25s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/335de31f116844ddb982122627d973d0 : SUCCESS in 43m 04s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/441be747d09a4106b781ea35c5979e0e : SUCCESS in 1h 12m 09s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/bd7bcf0ff2364ffca40bfee34161f941 : TIMED_OUT in 2h 02m 46s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/424e5d1ac1304ca1baefd8d8e518ac48 : SUCCESS in 1h 30m 25s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/d4f03a7de24e4b1089146a91c7bea41a : SUCCESS in 59m 42s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/23fb9a6837b34813bbf4db63d0718860 : FAILURE in 50m 32s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/c5db015eeee8479280aef8a26e0a293f : SUCCESS in 55m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5932194f30164a31927a22de47138462 : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca176e03e53d49fca777307d1289c24a : SUCCESS in 5m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/57d38737a4f64438afb4db52b6629cc1 : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/add2ce223d1a445cbe6e4bcf9229f708 : SUCCESS in 5m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/32e1357febaa4c72aebaf03f0d2f1e68 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13cf3508c63b469caaaa83902ade6b6b : SUCCESS in 5m 27s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/dcecbaad64b24fa5bf025cc16bacd6aa : SUCCESS in 24m 24s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f3778f8b55eb4a31b879c5b7b9d4169c : SUCCESS in 34m 05s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/09cf717ad4d348969aa861cac848c465 : SUCCESS in 1h 10m 42s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d8cbded6d9fb4004b7fe93407a219a10 : FAILURE in 15m 56s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/cc78a83223e042fbb7769d020307a51d : FAILURE in 11m 57s","accounts_in_message":[],"_revision_number":1},{"id":"74bbae0adebe0fabe2d33a7f8c4df612f2d2512c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-13 09:22:01.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5ba28cc1dc6adf8f6f38af468cb663ef293f2fcf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 11:00:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/627144a1851f43f5bf3c732c2e6ba20a : SUCCESS in 1h 00m 59s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/0e4f3f34260349e8b133ae80278d70a4 : SUCCESS in 46m 23s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/038727730adc4544a2d244fe6658d2ef : SUCCESS in 1h 26m 42s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/53422f8a70b243cb8b550f7b46b8ab02 : SUCCESS in 1h 26m 02s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/99f85c7a199049ef982f4c5ca8db7a6a : SUCCESS in 1h 35m 37s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/2f8e47ae927b484c9ef8b6a3e846b929 : SUCCESS in 55m 47s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/73907df0452f4259a9d88127bdb4ff5a : FAILURE in 53m 47s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/c798232364c14a0eb840c45d0edc23ca : SUCCESS in 54m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8da1af2f1be54c299302e62aece8a505 : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ce28dd7760b481988f1d47a80ace3bf : SUCCESS in 4m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5086101443654c9385dd72870ef8f54e : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/27916a9b7d304d5e856846416bdb3aa2 : SUCCESS in 4m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8f0685c8ba1c4430a88bcf22390a5c85 : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cdac07d331244e5a06b783cf4542b28 : SUCCESS in 4m 21s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9b1c5ba645894a8994e52cfddfaf20d4 : SUCCESS in 21m 04s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/09af271816c44e17a5a81377436e75f2 : SUCCESS in 26m 26s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5e711080073a4ac1b600916cbbeedc96 : SUCCESS in 1h 13m 27s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4da10c0e21fc409ab3e6b9236cc9a6f4 : FAILURE in 30m 44s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/fcef9b6d49214ff9bb9b881a2d8cdb37 : SUCCESS in 1h 06m 39s","accounts_in_message":[],"_revision_number":1},{"id":"6056dacc4ed1cc414a9f4a6a3430d306edd7f693","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-03-13 11:17:09.000000000","message":"Patch Set 1:\n\nFinally :-)","accounts_in_message":[],"_revision_number":1},{"id":"b183130bd3fd5e039d6cf20beb620f5de5813456","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-13 11:20:08.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"328f95ca21e68853d1ae90b8bfc5e473727dea0b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 11:20:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"86de7baddd1fc3f640e76d3c05f4a8590c767bcb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 12:14:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/533a0d7850e54baa980aed8841f8a8e7 : SUCCESS in 42m 44s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/55a6a9de22a940adb07e0f28ba305904 : SUCCESS in 50m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c129fa5f271346dd9735702970bd4cc4 : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e51a26ac05dd4e1c89384e641f1a511f : SUCCESS in 3m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a62cc972ab9b448899d09bdcdd0766f5 : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1ebed094b6f8425abbe2c21e415795f3 : SUCCESS in 3m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3e3f00df7f2a4506919610e0f12ffe2b : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b541377aeb4447aa8785c78fe7c40c6e : SUCCESS in 3m 57s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1c57cb659eb5423eb656550ce34b969a : SUCCESS in 28m 19s","accounts_in_message":[],"_revision_number":1},{"id":"7bf0e80fa6afdeb61a1be3ff664d2189793126e6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 12:14:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"78e4f34dbc87805c1975bcfc59cd765f2d5d6485","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 12:16:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d64cd6caf5ad4bbb9e4228498922f3aa : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7c2199665a928c2ccb54ea4f7376629e67426240","revisions":{"7c2199665a928c2ccb54ea4f7376629e67426240":{"kind":"REWORK","_number":1,"created":"2020-02-27 17:10:50.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/18/710318/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/18/710318/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/710318/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/710318/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/710318/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/18/710318/1"}}},"commit":{"parents":[{"commit":"7195b7016ab85772556ae395fe8747b92af81af6","subject":"Merge \"Bump stein versions\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7195b7016ab85772556ae395fe8747b92af81af6"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-02-27 17:09:48.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-02-27 17:09:48.000000000","tz":60},"subject":"Bump stein versions","message":"Bump stein versions\n\nChange-Id: Ie7a5d74e4640e5ec5e9885b4733551cf050d0853\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7c2199665a928c2ccb54ea4f7376629e67426240"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7c2199665a928c2ccb54ea4f7376629e67426240"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
