)]}'
{"id":"openstack%2Fkolla~710326","triplet_id":"openstack%2Fkolla~stable%2Ftrain~I4417ad735933df786651f3190cc3250b9b306027","project":"openstack/kolla","branch":"stable/train","topic":"bp/centos-rhel-8","hashtags":[],"change_id":"I4417ad735933df786651f3190cc3250b9b306027","subject":"CentOS 8: Install python in the base image (Train only)","status":"MERGED","created":"2020-02-27 18:07:42.000000000","updated":"2020-02-28 14:33:03.000000000","submitted":"2020-02-28 14:30:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"710326-1582900240178-577aa814","meta_rev_id":"c04489e95e040de7cca821a51aea07fa8119d5b7","_number":710326,"virtual_id_number":710326,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-28 14:30:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2020-02-28 12:28:49.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-02-28 12:30:06.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-02-28 12:30:06.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2020-02-28 12:27:15.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-28 12:28:49.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-02-28 12:30:06.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"},{"updated":"2020-02-28 14:30:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4b8e635059635c2c7f403175e3d01677a7f3a6a0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-27 18:07:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3aab9327257c4c1569dceb8a0abf92c7277eb836","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 23:10:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/132a00f64d0a4b4d9f2266cdcbcdac95 : SUCCESS in 44m 16s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/3701a334035e49b5a68d1d988055c424 : FAILURE in 7m 13s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/215b57e490f04c5a9094bb7b450736fd : SUCCESS in 25m 21s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/63d63602d8cd4950853c5bc734d87922 : FAILURE in 7m 05s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/9ac5a73d4cda4c7eade85cae60149e15 : FAILURE in 11m 08s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/d17b010b16c3486888bb1114ecae2e03 : SUCCESS in 1h 24m 01s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/d4e10363a3154c0f8cf50969d882845e : FAILURE in 11m 07s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/9589e1a8c1e347329d22fb82408cb43f : SUCCESS in 49m 51s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/fd19ac30d92e4de6a707a57efb9e4a38 : FAILURE in 14m 07s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/0cac39cdb1b642f480bf8336397acf0b : FAILURE in 16m 02s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-ansible-centos-source-zun https://zuul.opendev.org/t/openstack/build/bd5b1754d0614f44bd38f84e5b56a824 : FAILURE in 19m 20s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/e11ce38933fd49b592cb8335707361e5 : FAILURE in 14m 43s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/4c57417b6da844d5b523cad33278c4c3 : FAILURE in 10m 50s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/c07a457bc3ec4b3693e375e7ded8d052 : SUCCESS in 32m 02s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/9e31c09bf812405792c60f2d232a2730 : SUCCESS in 21m 14s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/1011eae7e23540908725b387925a450c : SUCCESS in 1h 02m 01s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a2cdf534f08643448bceafbe21c3b349 : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd7942e086c64828a6d4bc0c3ec8d7a3 : SUCCESS in 5m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/44e430e13de440af95da8de4be284c18 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f1fce5cbfef14ebe8e6585b8b360fd68 : SUCCESS in 5m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/92c6df1af36d47dfb69c4ac816fed8ea : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c002153de03147d497ccd9190a17334f : SUCCESS in 8m 01s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c71a74a459184f738890927a901be1d7 : SUCCESS in 26m 16s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7d6687fa114746e9bb99b4b3add35888 : SUCCESS in 35m 05s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/301d667c004047e9aa7ad1ecadfe3a4a : SUCCESS in 57m 41s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/78c32971b92c4c0aad763913a3817451 : SUCCESS in 58m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d32cbee34c1c45f3aacb4f2a7b0464af : SUCCESS in 1h 02m 25s\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/6f3bec7da18e41faa5867be9e5284c1e : FAILURE in 1h 04m 52s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/91fd37ea339f484ea44bbda7a9adc30b : SUCCESS in 1h 55m 06s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/ccd20c44e4bb42628891db1cfc17bd2b : SUCCESS in 1h 39m 15s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/1d1fe822fb3a4ff4bd9e245eee06058f : SUCCESS in 57m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dc7f8984df54e11a22e0c4f9fe7f2ddd1734ccc1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-28 09:17:11.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"e67be7bbc2e321b4ce65cf579eaf1f97b624ac70","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-28 09:19:38.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"1286b8ec994e2d94019133bcb22f2596c401e5cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 11:55:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/6d392bacdc59489180899091acdf5a1d : SUCCESS in 40m 52s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/324cc137be774aef81596f74c8b3babe : SUCCESS in 49m 54s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/7553035ed0204a9b8c2b9efe79fc3f5f : SUCCESS in 39m 41s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/1151f7fdcfef4e21a5c7d2dd8601dca7 : SUCCESS in 29m 05s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/d4fa6dc219b448229c836139c3af8541 : SUCCESS in 1h 42m 01s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/867fab8459ff4117afa689c95125e2c3 : SUCCESS in 1h 28m 02s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/02410866f19640d6973fbbd02474f745 : SUCCESS in 1h 35m 35s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/78e9caf6b4d846168d7c8df06eee62dc : SUCCESS in 56m 36s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/086a4030f22a451b8bc80ec8b241c4b0 : SUCCESS in 58m 25s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/2d80ae3ad608451a9c2d0c14317bf2d9 : SUCCESS in 2h 27m 28s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/5feadd932a4c42c09dda9dfe35d520de : SUCCESS in 2h 04m 52s (non-voting)\n- kolla-ansible-centos-source-zun https://zuul.opendev.org/t/openstack/build/6ac46f74792f433887505f193cedb221 : SUCCESS in 1h 58m 14s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/393e51ae033f4543aa83bca92354adc0 : FAILURE in 1h 43m 06s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/5b9287a0a8cb41c98f4e4153f1723c69 : SUCCESS in 1h 45m 41s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/53cd59ebdb7c43a1be270eb97a60080f : SUCCESS in 28m 55s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a36263473e814dd18ab8147c5c3de813 : SUCCESS in 26m 06s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5f24cca79c5f45c79cbd8af35282f2b4 : SUCCESS in 1h 02m 45s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5e5b5217026948849e82931eb6641844 : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f473800da1a4cc88c985d6bf57e16f0 : SUCCESS in 5m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/240db8d23d66496d89241764c4fd4f4f : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/535f078b9f4645c0b52dc76b52696384 : SUCCESS in 5m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/db438c02b0a840f4a0f7a956e2dc5d22 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/830676ec437844909381e21c0d5346b2 : SUCCESS in 6m 54s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/13cbe95e6ef84a4484d3bf57f978c33c : SUCCESS in 26m 59s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/097641a38f3a46379208acbd3d4d5e3d : SUCCESS in 33m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a564a1898f9143cc810747da1efe5c9d : SUCCESS in 1h 12m 20s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/645224d9fda94820bd421697e9d54468 : SUCCESS in 1h 09m 02s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/47d59cd9304e4e8d8ed0d16d3c338c8a : SUCCESS in 1h 18m 19s\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/73fce2d4c8974cc08d083598d3158668 : FAILURE in 1h 23m 41s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/d566692b5fb6466f91937cb8f5a592de : FAILURE in 1h 51m 18s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/b6e3ce1c5ec0422e903e3ff3ebf3b329 : SUCCESS in 1h 25m 33s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/2179cc6fb6df48ed903b7bc24d2f1602 : SUCCESS in 58m 13s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b75d1cc33fc8dd27209f88c315e7fa5485e2c562","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-28 12:27:15.000000000","message":"Patch Set 3: Review-Priority+1","accounts_in_message":[],"_revision_number":3},{"id":"d20161fdc1e3eefd619f86cd7ed567a2a1222aaa","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-02-28 12:28:49.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"63d82dd7af013771859b0f58a42ad215fb1fa79b","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-02-28 12:30:06.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"6b7a2048b247d007020df05a229709859794ac66","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 12:30:22.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ec66e5f64573d38bbded814de9b98d078d855f09","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 14:30:39.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/9749d89f8d2e467789b5cddf7e6097b0 : SUCCESS in 55m 40s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/e1290550ac1e4e13bdc5d7c3d33da32f : SUCCESS in 30m 32s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/97863c6a2f424e429a9be1265d2ecd5f : SUCCESS in 1h 40m 01s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/9ae489c26c214e28b3db3b5fc9fcc54c : SUCCESS in 1h 55m 15s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/2970b70bbcc04d2cad1caefd4a2423af : SUCCESS in 1h 03m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/96cb4e760c3c4144ae060060bbf9ef54 : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/979c9d37c6bb411b9e51004d8ca4b215 : SUCCESS in 6m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/87d1205d28ca4a4da14f21da2d7ddb9b : SUCCESS in 6m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e97dfd413924be8ad80a867a49820a7 : SUCCESS in 7m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3a2afdcd325c4e1da87186b99470e0ac : SUCCESS in 7m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/573909cae425459cac565fb6c238f892 : SUCCESS in 8m 36s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/26fb02706e3d4ed0ac26d9f2844ba6d4 : SUCCESS in 28m 04s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0de02dd673e14c60a59efb17943ac43c : SUCCESS in 59m 34s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3bd485f53a8b4a3b842758b993a6acff : SUCCESS in 1h 17m 24s","accounts_in_message":[],"_revision_number":3},{"id":"acb171bdb4301b11bf738d0824b88452bbad561c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 14:30:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"824fa7cbb8eab58868a62c2f96b4d23208e08c1c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 14:33:03.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22bc7ce1ab5643218f9e8ac4aa12e0c8 : SUCCESS in 1m 46s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8583243e893cc559480d706285fbeaaffed05faa","revisions":{"49a72349604909e4b162c277000d573dfcd12f56":{"kind":"REWORK","_number":1,"created":"2020-02-27 18:07:42.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/26/710326/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/26/710326/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/710326/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/710326/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/710326/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/26/710326/1"}}},"commit":{"parents":[{"commit":"594f468bad94f50c2451d450e3c241aa08bf7f07","subject":"Merge \"CentOS 8: Update packages in images\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/594f468bad94f50c2451d450e3c241aa08bf7f07"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-27 18:07:12.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-27 18:07:16.000000000","tz":0},"subject":"CentOS 8: Install python in the base image (Train only)","message":"CentOS 8: Install python in the base image (Train only)\n\nThis was missed due to the difference in base package lists between\nmaster and stable/train, with train having separate lists for binary and\nsource, and master having a single combined list.\n\nPartially-Implements: blueprint centos-rhel-8\nChange-Id: I4417ad735933df786651f3190cc3250b9b306027\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/49a72349604909e4b162c277000d573dfcd12f56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/49a72349604909e4b162c277000d573dfcd12f56"}]},"branch":"refs/heads/stable/train"},"d6f94985b6e4f3a4bf5742ffb779b9212a71a95a":{"kind":"REWORK","_number":2,"created":"2020-02-28 09:17:11.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/26/710326/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/26/710326/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/710326/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/710326/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/710326/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/26/710326/2"}}},"commit":{"parents":[{"commit":"594f468bad94f50c2451d450e3c241aa08bf7f07","subject":"Merge \"CentOS 8: Update packages in images\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/594f468bad94f50c2451d450e3c241aa08bf7f07"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-27 18:07:12.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-28 09:17:06.000000000","tz":0},"subject":"CentOS 8: Install python in the base image (Train only)","message":"CentOS 8: Install python in the base image (Train only)\n\nThis was missed due to the difference in base package lists between\nmaster and stable/train, with train having separate lists for binary and\nsource, and master having a single combined list.\n\nPartially-Implements: blueprint centos-rhel-8\nChange-Id: I4417ad735933df786651f3190cc3250b9b306027\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d6f94985b6e4f3a4bf5742ffb779b9212a71a95a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d6f94985b6e4f3a4bf5742ffb779b9212a71a95a"}]},"branch":"refs/heads/stable/train"},"8583243e893cc559480d706285fbeaaffed05faa":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-02-28 09:19:38.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/26/710326/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/26/710326/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/710326/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/710326/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/710326/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/26/710326/3"}}},"commit":{"parents":[{"commit":"f367d2303f3fe22cdd9ce55b20dd77918676ab36","subject":"CentOS 8: Use upstream Ceph/master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f367d2303f3fe22cdd9ce55b20dd77918676ab36"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-27 18:07:12.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-28 09:19:26.000000000","tz":0},"subject":"CentOS 8: Install python in the base image (Train only)","message":"CentOS 8: Install python in the base image (Train only)\n\nThis was missed due to the difference in base package lists between\nmaster and stable/train, with train having separate lists for binary and\nsource, and master having a single combined list.\n\nPartially-Implements: blueprint centos-rhel-8\nChange-Id: I4417ad735933df786651f3190cc3250b9b306027\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8583243e893cc559480d706285fbeaaffed05faa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8583243e893cc559480d706285fbeaaffed05faa"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
