)]}'
{"id":"openstack%2Fkolla~710970","triplet_id":"openstack%2Fkolla~master~Iff80392357ef5e2dd3f2e24dbf4df299660805b9","project":"openstack/kolla","branch":"master","topic":"ci-docker-network-mode-host","hashtags":[],"change_id":"Iff80392357ef5e2dd3f2e24dbf4df299660805b9","subject":"CI: Use network mode \u003d host","status":"MERGED","created":"2020-03-03 10:54:00.000000000","updated":"2020-03-14 13:35:13.000000000","submitted":"2020-03-14 13:33:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"710970-1584192825998-eb552905","meta_rev_id":"b10c9c9bbb28a7ecb86c1e41af770bafd75d8589","_number":710970,"virtual_id_number":710970,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-14 13:33:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":2,"date":"2020-03-13 18:29:44.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-03-13 18:29:44.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-03-13 18:29:51.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2020-03-13 18:29:44.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"date":"2020-03-13 18:29:44.000000000","_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-03 12:12:46.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-03-12 19:02:28.000000000","updated_by":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"reviewer":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"state":"REVIEWER"},{"updated":"2020-03-12 19:02:31.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-03-13 11:21:06.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-03-14 13:33:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4e6c0a065b07fcc4f0891c7199976e86a52fdcca","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-03 10:54:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"85e4b22bf18f2a733946e95b60a9218802902867","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 11:54:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7c83eb18709141378649a544b3e580a7 : SUCCESS in 55m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"256f4572e4a91d520061fb456ca3f80340985887","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-03 12:10:41.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"f95e5b9ac6bacee82aab309cf908085d4ca6abbf","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-03-03 12:12:46.000000000","message":"Patch Set 2: Code-Review+2\n\naarch64 works again!","accounts_in_message":[],"_revision_number":2},{"id":"7205344f7e3a9f1624b470e00364616aefa139ac","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-03 12:15:09.000000000","message":"Topic changed from wip-network-mode-host to ci-network-mode-host","accounts_in_message":[],"_revision_number":2},{"id":"f3a4f69c9ebf5d710930990f74a768ef85f26efb","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-03 12:15:16.000000000","message":"Topic changed from ci-network-mode-host to ci-docker-network-mode-host","accounts_in_message":[],"_revision_number":2},{"id":"b255fefefa1c0c2c8248e37526ca0ce731873948","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 13:03:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4c210abb2d4045059ef19656f68685e4 : SUCCESS in 51m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"403b33d15d0a4635e15c7ce04e1216ddac1bb230","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-03 14:03:53.000000000","message":"Patch Set 2:\n\nCould this be a problem for users running without this configuration option?","accounts_in_message":[],"_revision_number":2},{"id":"41229b7f28134f57217d0f828c2577fe4ff7425f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-03 14:04:28.000000000","message":"Patch Set 2:\n\n\u003e Could this be a problem for users running without this\n \u003e configuration option?\n\nPresumably we\u0027d also need the same config in kolla-ansible.","accounts_in_message":[],"_revision_number":2},{"id":"da1c757127c883e28becf33b64ad7f5078234621","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 14:09:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/061746c8ae844c23b1015250de1b8816 : SUCCESS in 29m 47s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/a0f469e920964f4ca5525bff2693e7d9 : SUCCESS in 42m 00s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/d061053b6c7843e1bf527a59bc5ef891 : SUCCESS in 34m 29s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/56e63ca5d65b4992a31b35c5b3749429 : SUCCESS in 30m 19s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/0e610cff1cc14f8ba6c49fea57d3dbe7 : SUCCESS in 1h 50m 36s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/2420749f6af34b489e39d10f82f6c965 : SUCCESS in 1h 27m 15s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/c895645f6f144ba7a3e584b8ba2b1248 : SUCCESS in 1h 29m 18s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/aadda1f2be0e467cba784c38c1eac4c3 : SUCCESS in 1h 17m 23s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/bab8a109b468416da46f187dc9c7eefe : SUCCESS in 1h 34m 59s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/26a8557c271041b08a48930707fc7e4a : SUCCESS in 34m 49s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/3ebf1d0d126540bd98c79b248c45086f : SUCCESS in 19m 57s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7fa1b3c20d0b43ab80b3b619a5daf965 : SUCCESS in 1h 12m 34s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2cbe42f895ee4b098175ec0335f80f97 : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75f27ed2f7324a37ae18b132af7f901e : SUCCESS in 7m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/82f6ba7b9ca94fffa3655d17404207ed : SUCCESS in 7m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cf1c84714c494b63a0353421ffb10f51 : SUCCESS in 5m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7448294bc1674650a3a7a7e82e06ff91 : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/358c197f08764bf38bfba0db96411e8a : SUCCESS in 8m 10s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/84dcbedfe4d6431cb761c22bfe01f90f : SUCCESS in 26m 42s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/51dafe8706b44b47bc7d72d0d63417f6 : SUCCESS in 26m 54s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7173083144da4b5f86d7f20881f85c49 : SUCCESS in 58m 52s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/697c2fdb3f95487e9243a7ae8e8d190c : SUCCESS in 1h 05m 09s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/50b6c777bbb0438eb4e9ea655da3b92e : SUCCESS in 1h 10m 07s","accounts_in_message":[],"_revision_number":2},{"id":"bf09054d64d8e64222c1ed1618241233015e38ef","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-03 17:03:24.000000000","message":"Patch Set 2:\n\n\u003e \u003e Could this be a problem for users running without this\n \u003e \u003e configuration option?\n \u003e \n \u003e Presumably we\u0027d also need the same config in kolla-ansible.\n\nYou don\u0027t say: https://review.opendev.org/710987 :-)","accounts_in_message":[],"_revision_number":2},{"id":"827f278b623b6b68700dba1bfa1236864c384f34","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-03 17:09:23.000000000","message":"Patch Set 2: Workflow-1\n\ndiscussion pending","accounts_in_message":[],"_revision_number":2},{"id":"63d17940706cc428f6ed32c51c48c1f4a2beabb3","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-03 17:10:11.000000000","message":"Patch Set 2:\n\n\u003e Could this be a problem for users running without this\n \u003e configuration option?\n\nI guess: if they wanted to build on IPv6-only sites like we? Consumers should not care.","accounts_in_message":[],"_revision_number":2},{"id":"6b794db43af9cb4988ca2cc5c45f31bbdc74511a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-12 18:56:45.000000000","message":"Patch Set 2: -Workflow\n\nopening as https://review.opendev.org/712765 is PITA","accounts_in_message":[],"_revision_number":2},{"id":"cf592fa3174f6b4c739a136bccc155fad6c36f12","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-12 19:03:34.000000000","message":"Patch Set 2:\n\nI guess, for users, follow-up with docs that reasonable builds with IPv6 connections require usage of host networking. Or maybe later switch kolla to default to host networking (with notice - reno)?","accounts_in_message":[],"_revision_number":2},{"id":"0a81d4524bd4ba5ae86f6dcf99db2c4ec1a5f85b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-13 11:21:06.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"14e59a2af1acf1e2b91957ba8e9eb9d4b36ba0b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 11:21:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"79c37a800b9a669d95fdf7a03475827fa410049c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 12:57:47.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/038480bd431144b1b8e240da5d2f8c16 : SUCCESS in 45m 25s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/c67c28dd649844ff8e4d369c2dc37d07 : SUCCESS in 1h 29m 05s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/1a4e4198a274412099fb39d3ee2039b5 : SUCCESS in 1h 20m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c0707d5dc4bd443e803e65300427a19a : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9829b794ad04c60a3b9a366239a888a : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2a38bec602f9408d8adbdf3bb73e3f68 : SUCCESS in 4m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/eaaf59a206b749f8897cc7622ae7b043 : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/083554e3f0c346108ff042f2b1cf788b : SUCCESS in 5m 28s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/42bcf7e102364b8080ef5e02ff504efe : FAILURE in 32m 31s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/514c1f8978ce4214a7d86dd03c1035d2 : FAILURE in 15m 03s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5dbb6596b4cb4e4b97c64c3c78ed0ebd : SUCCESS in 1h 06m 21s","accounts_in_message":[],"_revision_number":2},{"id":"9c27cb4179dbdd50545a4b4f37b16337ad5a4522","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-13 18:29:44.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"42849c46f65885f80e848df93e1fde89d25f36a8","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-13 18:29:51.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"faa172e579e2d82cfaacb02b533243e0d5ce636f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 18:35:39.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/787960d85ddc443d93ae4f596e612c4f : FAILURE in 2m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"76f03a8dda7aee77135f852383f0b74931c5c20e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 20:36:29.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/03ec5f29204f41dba74fb25bf23a304d : SUCCESS in 29m 39s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/ef7bd5613e1f41d79acbc44a6e597304 : FAILURE in 2h 06m 02s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/2f4503dc60dd4904a7a80c5e1c752265 : SUCCESS in 1h 21m 54s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/59afdd8bba234b5d9c00f74c15545065 : FAILURE in 52m 15s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/65f974ffc2504586abf26dc56c4f1990 : SUCCESS in 1h 21m 52s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2042695d082246b0a4ac4a4db470ffef : SUCCESS in 22m 30s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b924d191d96147ca8e6d7c33d507bd4f : SUCCESS in 19m 37s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/9a1960fd3705498ab7a914ea1bf550d4 : SUCCESS in 54m 45s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6eba100d1e5f45ed94ecf9d4844e4bb1 : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26395dfc5534447ba9b0e97d5f635555 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e58f463b4d5c466980cf795fdfa89915 : SUCCESS in 4m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/db013a08f33f4d10a67161fecb2a8b9c : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/94a9a9edf7fc4ff1a408638e86da73f2 : SUCCESS in 4m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f267e2c1d15243d9a3343fd6444c679f : SUCCESS in 6m 12s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5cb3744390a941058b883dfc7e39c658 : FAILURE in 24m 16s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b4615d277d4d4561bc1a9baf37936bf9 : SUCCESS in 29m 07s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ada98d581c3345f1882da0bc7ee078cf : SUCCESS in 54m 30s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e74ffc8c13d74362ba1df6efeb7652b8 : FAILURE in 17m 57s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/691d6d8c4d944e06a29d4003a3e2b7e9 : FAILURE in 13m 09s","accounts_in_message":[],"_revision_number":3},{"id":"b59c0a02797bb4c90337c9de6b04d1f7abcbaa62","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-14 09:19:31.000000000","message":"Patch Set 3:\n\nrecheck\n\nnetwork issues","accounts_in_message":[],"_revision_number":3},{"id":"0e1852749e7911fde0211ff0489afb83506524bc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-14 09:22:54.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/107b7605d0be455780cde04feb52d05a : FAILURE in 2m 42s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"42f6167566780ed9b333b565330f7cf0850d4987","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-14 12:00:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/34abaefda8c1414fa17137357eb7c067 : SUCCESS in 2h 21m 18s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/2f7594a5e731485ea20216a4102c5449 : SUCCESS in 2h 17m 35s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/61aa8f5ee1a24d18bce5a59fd7226555 : SUCCESS in 1h 22m 26s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/21e73f129e764182b0f91a16ab197f22 : FAILURE in 59m 36s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/cd38ea6b29a24e77a2f8fed5dd916433 : SUCCESS in 1h 25m 28s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/18a4d34959cd46f5b12fa9be2acd7de2 : SUCCESS in 48m 21s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ce52054615294e47af3a894560317aed : SUCCESS in 22m 28s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5f0af4c93d0e425a8cb809351f448568 : SUCCESS in 57m 07s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/709e84d5919c40e5a07501cf0e82a96d : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5b993aa0ac74da9bc4491b96165bcd8 : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a6c4ca9d4d0f4d90912b6860474fc03a : SUCCESS in 3m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e9fde90c70a64c8cb996323e846a5cbb : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b367f029c074d6bae0da9fe7a3ac8f8 : SUCCESS in 4m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/080a8322d66241a09c822224cd8fcbb6 : SUCCESS in 5m 03s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/53474c2f1fbb40b094f6011c8fe0d542 : FAILURE in 24m 01s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4bca3f8671bd4f4593d074f6deb11c28 : SUCCESS in 2h 23m 04s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1c751346b85349989935eeba9dbb2708 : SUCCESS in 1h 06m 29s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/53e40f7d8f914623bd18ad967c5d6fd7 : FAILURE in 19m 22s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b070591b011a4369b7335df4a5a68c70 : SUCCESS in 1h 14m 22s","accounts_in_message":[],"_revision_number":3},{"id":"eb43844241693767a59ff6c97e5f971be3cde9a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-14 12:00:27.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"94b850a77c9340c71ca5cb0b9dd9a0b054c836b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-14 13:33:45.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/cfbe3267b8394f33b3efe051cafac09a : SUCCESS in 28m 44s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/c31567c2d6164454b48041d6ba8cb3f0 : SUCCESS in 1h 15m 41s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/2de7d0264e6448138f1a361051cd986f : SUCCESS in 1h 30m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cf93968469134991a6b7b2ce0dbe5924 : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4299061438446d2937fd685124339fc : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c697acbe87264b26a513def8aad2d02f : SUCCESS in 4m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/20411981eb1d44799841ddf1db7e0169 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f6e6896f37f4342b9627b5fc3191cde : SUCCESS in 6m 41s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ca2707d50c014689b97b94235614f352 : SUCCESS in 27m 23s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9b5b28d6e9f74557b219b04da8dfd7cc : SUCCESS in 58m 08s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/80ada7bbba814420a63de992e3ac4be0 : SUCCESS in 1h 11m 03s","accounts_in_message":[],"_revision_number":3},{"id":"d41b2d9f461e1065d9b939ee4d9737b0c8cc1a55","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-14 13:33:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"b10c9c9bbb28a7ecb86c1e41af770bafd75d8589","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-14 13:35:13.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b44b0a9ce0c44319e0e7827e16c202c : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5912f2c16e68b4b581cb3b2d7e974e438b9333f5","revisions":{"50636ab65a050dc9c5708648dc32c69e2d56a821":{"kind":"REWORK","_number":1,"created":"2020-03-03 10:54:00.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/70/710970/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/70/710970/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/710970/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/710970/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/710970/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/70/710970/1"}}},"commit":{"parents":[{"commit":"10570292f4e801b653124d5fc1ac1611f9d2f007","subject":"Merge \"Add collectd-pmu package for Intel PMU (C7 only)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/10570292f4e801b653124d5fc1ac1611f9d2f007"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-03-03 10:53:01.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-03-03 10:53:58.000000000","tz":60},"subject":"WIP: Network mode host","message":"WIP: Network mode host\n\nDoing testing for linaro ipv6-only nodes (flaky aarch64 CI).\n\nChange-Id: Iff80392357ef5e2dd3f2e24dbf4df299660805b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/50636ab65a050dc9c5708648dc32c69e2d56a821"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/50636ab65a050dc9c5708648dc32c69e2d56a821"}]},"branch":"refs/heads/master"},"2253e2a77841134a214d68d84ca9dfc4566a87e4":{"kind":"REWORK","_number":2,"created":"2020-03-03 12:10:41.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/70/710970/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/70/710970/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/710970/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/710970/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/710970/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/70/710970/2"}}},"commit":{"parents":[{"commit":"10570292f4e801b653124d5fc1ac1611f9d2f007","subject":"Merge \"Add collectd-pmu package for Intel PMU (C7 only)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/10570292f4e801b653124d5fc1ac1611f9d2f007"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-03-03 10:53:01.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-03-03 12:10:27.000000000","tz":0},"subject":"CI: Use network mode \u003d host","message":"CI: Use network mode \u003d host\n\nThis avoids issues with IPv6 not enabled in docker.\n\nChange-Id: Iff80392357ef5e2dd3f2e24dbf4df299660805b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2253e2a77841134a214d68d84ca9dfc4566a87e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2253e2a77841134a214d68d84ca9dfc4566a87e4"}]},"branch":"refs/heads/master"},"5912f2c16e68b4b581cb3b2d7e974e438b9333f5":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-03-13 18:29:44.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/70/710970/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/70/710970/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/710970/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/710970/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/710970/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/70/710970/3"}}},"commit":{"parents":[{"commit":"79c0dd49cbb89f918778f85b83b250f15eb64c1f","subject":"elasticsearch-curator: Add python3-wheel for Ubuntu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/79c0dd49cbb89f918778f85b83b250f15eb64c1f"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-03-03 10:53:01.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-03-13 18:29:44.000000000","tz":0},"subject":"CI: Use network mode \u003d host","message":"CI: Use network mode \u003d host\n\nThis avoids issues with IPv6 not enabled in docker.\n\nChange-Id: Iff80392357ef5e2dd3f2e24dbf4df299660805b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5912f2c16e68b4b581cb3b2d7e974e438b9333f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5912f2c16e68b4b581cb3b2d7e974e438b9333f5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
