)]}'
{"id":"openstack%2Fkolla~711003","triplet_id":"openstack%2Fkolla~master~I68aa6f4b7fb1c8d63c84f3436a34399cc9644060","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I68aa6f4b7fb1c8d63c84f3436a34399cc9644060","subject":"nova-libvirt: use dmidecode and xen-utils on archs which have them","status":"MERGED","created":"2020-03-03 13:24:31.000000000","updated":"2020-03-06 09:38:51.000000000","submitted":"2020-03-06 09:37:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"711003-1583487425941-4545ba28","meta_rev_id":"ce21664d29a902838fb49631aaffd9ea8f2d4e10","_number":711003,"virtual_id_number":711003,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-06 09:37:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-03-05 22:24:38.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"all":[{"value":1,"date":"2020-03-05 21:56:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":2,"date":"2020-03-05 21:56:08.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-03-05 21:56:08.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-03-06 08:12:37.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-03 14:02:07.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-03-03 15:29:46.000000000","updated_by":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"reviewer":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"state":"REVIEWER"},{"updated":"2020-03-05 22:24:38.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-06 08:12:37.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-03-06 09:37:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"448905d65fc252d3f28ef54aa31efd42d9ab1f46","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-03-03 13:24:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9065cd742f5f6140404015709b3d76592f356cbd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 13:33:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7ba54294aad44167be3bebb7cd21252f : FAILURE in 7m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5d0bb67d3c298d3a04dd2a53866b8147684eadd9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-03-03 13:35:54.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/057bf2cb887e4f7eb339f585b7fbedc5 : FAILURE in 9m 16s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/56a81ff7210442308ea5cb0b820500f9 : FAILURE in 9m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bdc1dd289ee0f56c933f52d63c4e3260f870e638","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-03 13:37:29.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"c20b4264d5ad8c959a30e9f6f3827962a11182f0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-03 13:37:59.000000000","message":"Patch Set 2: Code-Review+2\n\nmakes adding any easier (I fixed missing paren)","accounts_in_message":[],"_revision_number":2},{"id":"ec69fc7655d6d7fcb43c3f3375e653840730ef22","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-03 13:44:34.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"2bf208e35d1ae720ca79976a567e0dedc85ff368","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-03 13:44:49.000000000","message":"Patch Set 3:\n\nrebased on host to test aarch64 build","accounts_in_message":[],"_revision_number":3},{"id":"0fee0cb7a8a7e1dd019a2a6de3cc56040639c8b1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-03 14:02:07.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ef1059672129a5bb0f6fe8f3fa970c0a65c5ea84","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-03 14:02:24.000000000","message":"Patch Set 3: -Workflow","accounts_in_message":[],"_revision_number":3},{"id":"abf4f17172d550c94f9a8e0048d450f16e7c4180","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-03-03 14:17:14.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/99e986dfb3894f51bc5f76a95cf105c5 : SUCCESS in 30m 49s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/79734618d1f5415c896b8a3b0436b0aa : SUCCESS in 27m 39s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"304c0553ce269d9698993db56a68c635a39cf510","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 14:35:10.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/80ba4b643d2c4728a905c555e0e54e79 : SUCCESS in 49m 00s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5dcd322bc017f9a9a7812af236e4a923eaf1ca4c","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2020-03-03 15:29:46.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"cf36aaabaabc0a174fc8c440a77be522b3d6d0ac","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-03 17:11:08.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"5dd3dd56f9a292eeb519e47f27e369a7917a4fe8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 17:41:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/e1a053cb6b2740bf867bad33f0c6e978 : SUCCESS in 40m 06s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/f139b34d7e774866a560fb0231d2667d : SUCCESS in 48m 04s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/8941de71f9f94625bc7f68c3f361a5fa : SUCCESS in 34m 48s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/c946ad4575564cd38bc9ddf1e646f73e : SUCCESS in 36m 14s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/cf0cbee26e8444ca9b34de1edaa3980a : SUCCESS in 1h 39m 54s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/7b2c76fe2df5453a94c4f46b2087f1d5 : SUCCESS in 1h 25m 49s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/7f1dc04900b243bdbe4e21d787a984ca : SUCCESS in 1h 37m 54s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/42efaad7ae994578bd1bc1308eea3465 : SUCCESS in 1h 30m 14s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/a14a3906869d459e8721840d7aa9f737 : SUCCESS in 1h 50m 21s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/a395e3b9c929439e90e380a12de81123 : SUCCESS in 55m 45s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/345b03795cc74c108e33632a05128daa : SUCCESS in 33m 22s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/4f55b3c6a67b4c778f1e6de051602c5a : SUCCESS in 37m 14s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e3854dcc9e2b4ac99b1697886f7b3a20 : SUCCESS in 1h 17m 21s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/310cf50303714939a14967bf36db6275 : SUCCESS in 7m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa3b6327ec3a4ff486ffc891be4f5b61 : SUCCESS in 6m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ec45d81fbbfb4592b8635604752cb7d3 : SUCCESS in 7m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/74c1b985ff364c979fe17aacf7a1f509 : SUCCESS in 7m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e0090d8ad102419fb77d124b23e64220 : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c30d99f454594156826820a7c9f367f1 : SUCCESS in 7m 47s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e0e365c3f9b045e79426ee22ba3c4a0a : SUCCESS in 35m 34s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4d87c6a7b4154f549f422550a310b042 : SUCCESS in 30m 11s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3f905dcaf6154bf8a110acfe234aab4b : SUCCESS in 1h 20m 02s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f11bd4653f4a46059bf50a065fc2abe2 : SUCCESS in 1h 04m 07s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d9f4696b9bda414bb9cee6be92632198 : SUCCESS in 1h 29m 06s","accounts_in_message":[],"_revision_number":3},{"id":"916176451274a61b5fb90ed5d1217937e513e7b2","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-03 17:43:05.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"1df451f512f5e9c311163092ad21a522c03d8f73","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-03 17:43:34.000000000","message":"Patch Set 4: Workflow+1\n\nok, it builds, let\u0027s untangle it while we debate on IPv6...","accounts_in_message":[],"_revision_number":4},{"id":"567c7988105a8921e5cd59c961a536b122655cc6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-03-03 18:20:18.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/50dfd33829804ed097a831a37c5933ab : SUCCESS in 25m 09s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/9d656a8c7b2445f9b7e588225d7e515f : SUCCESS in 35m 37s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"53011cf0351428225753944cedc9001d99918fd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 19:25:25.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/f40f7253ff4e450c84850ad430ea1c10 : SUCCESS in 33m 35s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/713de5cae2334df487b95a2eedca1aba : FAILURE in 41m 43s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/dbf83b5bb126470fa986a0b98b116b33 : SUCCESS in 32m 17s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/c250199638f3449fbe55696cb31e3dd4 : SUCCESS in 32m 54s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/07e7ceb0bb554b9e992d2061feb83796 : SUCCESS in 1h 20m 39s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/52d4750172704f9ca91ac545cb7d4249 : SUCCESS in 1h 15m 54s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/e74045cf3d2b4d598d22aeab7c1cb8a9 : SUCCESS in 1h 24m 37s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/023cdff89878426d9b9110cc817195d9 : SUCCESS in 1h 13m 49s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/9a4fc8d38ace4bb2b26e5035701e412d : SUCCESS in 1h 40m 00s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/e70168e2a2aa4ab7823c24999f22b7c5 : SUCCESS in 45m 11s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/463c5904eee945b8911dfd2006b4c9c2 : SUCCESS in 32m 17s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/36ff22cc59ba4951907e9d8ce5cbdbe0 : SUCCESS in 32m 42s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7b013e68bb0e4f0bae15ee619db0f2d7 : SUCCESS in 1h 09m 08s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2456d25d965344a6bd09c10f272e41db : SUCCESS in 7m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81954e30b7ed4cc4851a22946e545dde : SUCCESS in 6m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b180c3b198f44d0b8dc5f6f9330a7275 : SUCCESS in 4m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6f95a37cd04d4bee8ba793691ea67d03 : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f4f2927056945d39106b5f80bf70429 : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea0a6ed228434007921813ab0e834d36 : SUCCESS in 5m 53s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8f485daa7f4542ce97d747c3699522af : SUCCESS in 25m 36s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e98a76c9da76496db4dbfb32feeeda2f : SUCCESS in 28m 52s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/dbf2ba9f28184d7a9bb481487c84c83c : SUCCESS in 1h 02m 19s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2bfc9fa3bd384706a78825a111c4f0b5 : SUCCESS in 1h 03m 08s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9ead573dff234c638900d23860abc900 : SUCCESS in 1h 18m 23s","accounts_in_message":[],"_revision_number":4},{"id":"19873dfbc8df0a98598a0fd392f5f86683ff349b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-03 19:28:44.000000000","message":"Patch Set 4: Workflow-1\n\nsheee... patrole dropped py2 support and tempest image blocks gate","accounts_in_message":[],"_revision_number":4},{"id":"d2a7ac467414b4bf4a92da13eaccf954bb95c89a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 20:48:30.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/82686062c6a54143b6a39fb9c1ace626 : TIMED_OUT in 3h 01m 33s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d7f6af2559df18fa71f572918fd4ee71fd29a167","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-03-05 21:56:08.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"61573b529506c1ae90918f1a1e561ad994c75afc","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-03-05 22:24:38.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/cee7e536258c447e97ccbe196d3611b1 : SUCCESS in 26m 46s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/c95ac06cefe44b20b4f35cf3dc5ee175 : SUCCESS in 25m 44s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"b67ebeef0670f4ed1895f59ff3c23b4176f57c53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 23:45:19.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/8bdf23bd525547a1bf20636e987f9445 : SUCCESS in 44m 31s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/9798e0fe1adb4a6d96836f20eff031e7 : SUCCESS in 31m 53s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/0d6e23936b594153b2f0cd79872229b4 : SUCCESS in 1h 20m 04s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/4dfc0e869f264d97ba1136dbfb6b436d : SUCCESS in 1h 14m 12s (non-voting)\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/3b51bd40fff34408810801208c8a1c5f : SUCCESS in 39m 38s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d40b64416bc4459a8899f7885b86c27a : SUCCESS in 20m 41s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/85d64883f79141fca461d34716e43eb2 : SUCCESS in 27m 02s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f0673e46388d46739385d78410acfe3e : SUCCESS in 1h 06m 43s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/491b85d8295f4fc3b5a47ce933997dcc : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/920fb608fd4f4b45baaa33ca0c3a2094 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b1bea0815796446fa8be595a4eea8d87 : SUCCESS in 3m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f736450a52b0435aa282c88d2f9f5d80 : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/13cd67c91474466c9b0e48c6caa9c337 : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b04d52b9f624927b971b668709bc3f8 : SUCCESS in 5m 43s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f56dcb0344f84a4b92e4ad093bb28f3e : SUCCESS in 35m 17s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5675c478df1141a28cfa873f41a391c1 : SUCCESS in 40m 15s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ea725da47a0d445f978385886429f9b9 : SUCCESS in 57m 03s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ebac7fe2647641cf88d7aef01af62765 : SUCCESS in 1h 05m 33s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b9a34641448b489c8a3d77608d101438 : SUCCESS in 1h 08m 35s","accounts_in_message":[],"_revision_number":5},{"id":"68d9ef2c6acbdf44170ccf038d0bfb2b1839ff6d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 00:59:28.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/496079a42d0c4db49be0cfa6f7022f99 : TIMED_OUT in 3h 01m 32s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"80bf8849356e766ee3cb40ebdb36eed0fc4c38fb","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-06 08:12:37.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"aa7dfce2ee13e4a950b706a66db7be751b086b7a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 08:12:49.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"fd353b0ab9a2da157595af564506996fb6dd77fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 09:37:05.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/66f0ce5eff994295b850512fb4fb77dd : SUCCESS in 29m 57s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/26ae4d90385e4c7da2cd511b3870dea6 : SUCCESS in 1h 22m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/21e9ddca295d47d2a66bc5c64fc40439 : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d834f14604dd4959b14e9acc10c7968c : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a0a2211a76f74aef9a7e59fd8d27f4a0 : SUCCESS in 3m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e4906da724104a5b8f45f55d9db343ef : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2895d121f4764b29a694f31eb62cb769 : SUCCESS in 5m 15s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a46729d856c24bd8adf7262caf2fca72 : SUCCESS in 23m 54s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8c98dd933e98469a9b09a0373ecb476a : SUCCESS in 1h 04m 12s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/307604709ff144788a2a382ffba21e02 : SUCCESS in 1h 14m 50s","accounts_in_message":[],"_revision_number":5},{"id":"d7076a7c73e36f6a9329ed4f12138d7f5b726f1b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 09:37:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"ce21664d29a902838fb49631aaffd9ea8f2d4e10","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 09:38:51.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56effade5c634444a1a17c6e68624046 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"1c8c9b42f295663fd641eadab89a82d4b93d18bf","revisions":{"d4bfdb56c1b57964ec7cbdc07fbb35ab4c6f6c6e":{"kind":"REWORK","_number":1,"created":"2020-03-03 13:24:31.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/03/711003/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/03/711003/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/711003/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/711003/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/711003/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/03/711003/1"}}},"commit":{"parents":[{"commit":"96d4e1ad3afec83de6e019a051686e05c6611c25","subject":"Merge \"Add Apache httpd setup to heat-api-cfn\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/96d4e1ad3afec83de6e019a051686e05c6611c25"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-03-03 13:23:06.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-03-03 13:24:17.000000000","tz":60},"subject":"nova-libvirt: use dmidecode and xen-utils on archs which have them","message":"nova-libvirt: use dmidecode and xen-utils on archs which have them\n\nInstead of listing architectures which do not have those packages we\nshould list those which have. Makes adding s390x easier.\n\nChange-Id: I68aa6f4b7fb1c8d63c84f3436a34399cc9644060\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d4bfdb56c1b57964ec7cbdc07fbb35ab4c6f6c6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d4bfdb56c1b57964ec7cbdc07fbb35ab4c6f6c6e"}]},"branch":"refs/heads/master"},"6c5bc16e22fad0b6a531683def53d4bbbef582e3":{"kind":"REWORK","_number":2,"created":"2020-03-03 13:37:29.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/03/711003/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/03/711003/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/711003/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/711003/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/711003/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/03/711003/2"}}},"commit":{"parents":[{"commit":"96d4e1ad3afec83de6e019a051686e05c6611c25","subject":"Merge \"Add Apache httpd setup to heat-api-cfn\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/96d4e1ad3afec83de6e019a051686e05c6611c25"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-03-03 13:23:06.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-03-03 13:37:19.000000000","tz":0},"subject":"nova-libvirt: use dmidecode and xen-utils on archs which have them","message":"nova-libvirt: use dmidecode and xen-utils on archs which have them\n\nInstead of listing architectures which do not have those packages we\nshould list those which have. Makes adding s390x easier.\n\nChange-Id: I68aa6f4b7fb1c8d63c84f3436a34399cc9644060\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6c5bc16e22fad0b6a531683def53d4bbbef582e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6c5bc16e22fad0b6a531683def53d4bbbef582e3"}]},"branch":"refs/heads/master"},"ca7021974506cc8640ffa6994da0eebff49614fc":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-03-03 13:44:34.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/03/711003/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/03/711003/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/711003/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/711003/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/711003/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/03/711003/3"}}},"commit":{"parents":[{"commit":"2253e2a77841134a214d68d84ca9dfc4566a87e4","subject":"CI: Use network mode \u003d host","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2253e2a77841134a214d68d84ca9dfc4566a87e4"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-03-03 13:23:06.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-03-03 13:44:34.000000000","tz":0},"subject":"nova-libvirt: use dmidecode and xen-utils on archs which have them","message":"nova-libvirt: use dmidecode and xen-utils on archs which have them\n\nInstead of listing architectures which do not have those packages we\nshould list those which have. Makes adding s390x easier.\n\nChange-Id: I68aa6f4b7fb1c8d63c84f3436a34399cc9644060\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ca7021974506cc8640ffa6994da0eebff49614fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ca7021974506cc8640ffa6994da0eebff49614fc"}]},"branch":"refs/heads/master"},"79e994292036ab7b59bf03baa1ef2f7e3705ecbd":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-03-03 17:43:05.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/03/711003/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/03/711003/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/711003/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/711003/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/711003/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/03/711003/4"}}},"commit":{"parents":[{"commit":"0c6312968214e403b5a408d69df17bd0a9267095","subject":"Use StorageSIG repos for Ceph in CentOS8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0c6312968214e403b5a408d69df17bd0a9267095"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-03-03 13:23:06.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-03-03 17:43:05.000000000","tz":0},"subject":"nova-libvirt: use dmidecode and xen-utils on archs which have them","message":"nova-libvirt: use dmidecode and xen-utils on archs which have them\n\nInstead of listing architectures which do not have those packages we\nshould list those which have. Makes adding s390x easier.\n\nChange-Id: I68aa6f4b7fb1c8d63c84f3436a34399cc9644060\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/79e994292036ab7b59bf03baa1ef2f7e3705ecbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/79e994292036ab7b59bf03baa1ef2f7e3705ecbd"}]},"branch":"refs/heads/master"},"1c8c9b42f295663fd641eadab89a82d4b93d18bf":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-03-05 21:56:08.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/03/711003/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/03/711003/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/711003/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/711003/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/711003/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/03/711003/5"}}},"commit":{"parents":[{"commit":"5f21b322d01d61f7be2542b36d4e2eee40ae1dc7","subject":"Merge \"Multiple fixes to unblock CI\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5f21b322d01d61f7be2542b36d4e2eee40ae1dc7"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-03-03 13:23:06.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-03-05 21:56:08.000000000","tz":0},"subject":"nova-libvirt: use dmidecode and xen-utils on archs which have them","message":"nova-libvirt: use dmidecode and xen-utils on archs which have them\n\nInstead of listing architectures which do not have those packages we\nshould list those which have. Makes adding s390x easier.\n\nChange-Id: I68aa6f4b7fb1c8d63c84f3436a34399cc9644060\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1c8c9b42f295663fd641eadab89a82d4b93d18bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1c8c9b42f295663fd641eadab89a82d4b93d18bf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
