)]}'
{"id":"openstack%2Fkolla~711186","triplet_id":"openstack%2Fkolla~master~Ifae18c407210c12745d29fc4c95dca69aeafe6a8","project":"openstack/kolla","branch":"master","topic":"bug/1866017","hashtags":[],"change_id":"Ifae18c407210c12745d29fc4c95dca69aeafe6a8","subject":"Ensure proper JSON in keystone bootstrap","status":"MERGED","created":"2020-03-04 09:26:58.000000000","updated":"2020-04-14 12:44:12.000000000","submitted":"2020-04-14 12:35:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"711186-1586867730791-a88869dc","meta_rev_id":"61957a4366aca00c3bf8bcebb51a058bff45894a","_number":711186,"virtual_id_number":711186,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"date":"2020-04-04 19:39:10.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-14 12:35:30.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-04-04 20:45:55.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2020-04-14 10:07:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2020-03-04 13:37:37.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2020-04-14 10:07:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-04 13:37:37.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-04-04 20:45:55.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-14 10:07:04.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-04-14 12:35:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9794034fac3f89f0dcbe830a5b21c0a2b40da0fe","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-04 09:26:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"06ef16ecab33abadc4da5c506bb165a76ca99208","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-03-04 09:57:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/a485393d26b741a497bf31fa0c232f20 : SUCCESS in 28m 42s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/dd837e069ba84baf8685acd38ec12a72 : SUCCESS in 28m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e368481536770524850c13c9c0a303c1e4a87795","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-04 10:50:57.000000000","message":"Patch Set 1:\n\nyay, helped","accounts_in_message":[],"_revision_number":1},{"id":"74c8152c839ec97c36a62e1ea34a1a2916e4477b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 11:15:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/7a3b68b8e9d6465089e23ca59b9cb234 : SUCCESS in 33m 29s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/3eb377d20e0b46a7975168fc2bcacbb8 : FAILURE in 42m 57s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/ff69f02c2aa8473588594dd091229fa9 : SUCCESS in 24m 30s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/066c4a8c6dc84e31a225603de8e58462 : SUCCESS in 30m 12s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/6240ddd4664b470aa313dab14fd8713a : SUCCESS in 1h 31m 57s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/d2308ec061324280b5000646b61b83b5 : FAILURE in 1h 18m 54s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/fc664163c7cb495c9aff39a450d62e11 : SUCCESS in 1h 19m 53s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/461ba69c72d343c8b8626d4037b6a398 : SUCCESS in 1h 10m 53s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/ebfe43007810442891cfde24c5baef32 : SUCCESS in 1h 35m 10s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/6df8e257a2de4e81a79d62c50da57c5d : SUCCESS in 57m 56s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/4905a1b7cb334d24ada25f0b9a624708 : SUCCESS in 40m 54s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/7b7a11a880694a8597b9fef8718d6863 : SUCCESS in 40m 15s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/782d88b79e744b45a15dcb51bd8361f3 : FAILURE in 31m 27s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c9de0147399c45ac97a86eefd425a663 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/064b303d7ee34642be3ac603931c2ba3 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2fa4adcf8b91485fa91321dc7b5589fd : SUCCESS in 4m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3aae936753fd4e52941cc60d02319390 : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c25fc8b512840168a794f9823b5041d : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1120e3353434457ea19814d8b5e06c54 : SUCCESS in 7m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b8bd76a5b016403fb8a1f2929fffe908 : SUCCESS in 6m 05s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/319f2b44ab5b465eb41d273abdadfb9a : SUCCESS in 48m 22s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/afadce026e6c46269cfd0de3843ea313 : SUCCESS in 30m 54s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/25db443ee8e54d52966a87714c55d5c6 : FAILURE in 28m 59s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/49aa7874296d4fea8f1fb55458ea087d : SUCCESS in 1h 07m 14s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3f51258617d5493fb66b5c308aa0d6fe : FAILURE in 55m 07s","accounts_in_message":[],"_revision_number":1},{"id":"b09a80eef2a69b399d7caa13d2cb962c745923f7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 12:28:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4b19dce503e0442bbc76040aa9c8fd45 : TIMED_OUT in 3h 01m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7540ce84e084821f3850b99bc6591e83201e9890","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-04 13:37:37.000000000","message":"Patch Set 1: Code-Review+2\n\nVerified that it allows the error to be displayed from logs.","accounts_in_message":[],"_revision_number":1},{"id":"94a9618d1660cddfb2438c5012d2d82729330cba","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-04 16:07:34.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"114d2b7e0a5232b9270e62c7c1af20cc377e534d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-04 16:58:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8acdb1f7a5f244eb882ec55797262b14 : FAILURE in 3m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9aa890e990706f177d0c9e24d52a770c0f37dccd","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-04 18:05:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/c1c4c09b9ceb4eeba53f93d1767bddbc : SUCCESS in 34m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0127927be41284a6b43cff3207d6d2dd836a8dfc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-04 19:09:57.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/f70770cdb8644e1fa60b1bb227a77e3d : SUCCESS in 25m 18s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/7376b582a16b4cb38b94734f0746790f : FAILURE in 6m 25s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/bd38d3a4e5fb4d39b7d656f08a6938b7 : SUCCESS in 1h 08m 35s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/892a3402635b46e3b288d182b3c67bd9 : SUCCESS in 1h 17m 52s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/04e0f242525946c1831f80b661fffd58 : SUCCESS in 1h 14m 34s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/ddc5bed9eb974bbca3e0ef6dfee30866 : SUCCESS in 43m 43s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6f8945468df042809e3b0e20e251eedc : SUCCESS in 19m 28s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/aab2ab4fa8174e0a968873dedae3e0ea : SUCCESS in 18m 21s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f2e8063391e547e6a8ad40df54d596d6 : SUCCESS in 1h 04m 27s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f41f5ae4127349068ed1fc3c300052ff : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37b8005dbcce48848ca6e41ab0faa5b3 : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d2253db0f8f24b79bcb624274a676f5c : SUCCESS in 3m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9f2c26582a44473b9271107e1ebaf218 : SUCCESS in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/667bf742b8134d75b011a74f18e15328 : SUCCESS in 4m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c115781ebda47d3a2b78f35fbef5859 : SUCCESS in 5m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/069a94b796fc4776b246e75d58a291aa : SUCCESS in 4m 53s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/277d40708fb14e5fb4e24743db83bd20 : SUCCESS in 21m 46s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c4a4675f5d724264abd33d238d9916e1 : SUCCESS in 25m 56s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/eef09d6a51e740a392a730f8062b852b : SUCCESS in 1h 22m 30s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2da5efb090b542d9be18cd17ed0f058c : SUCCESS in 57m 19s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3db07dcf343b4b66a21cd32796116cfb : SUCCESS in 1h 07m 11s\n\nWarning:\n  Job openstack-tox-py38: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path .tox/py38/lib/python3.8/site-packages/netaddr/strategy/__init__.py in the commit\u0027\n  Comments left for invalid file .tox/py38/lib/python3.8/site-packages/netaddr/strategy/__init__.py","accounts_in_message":[],"_revision_number":1},{"id":"25461b8d8ef125368d0712ca1c5d76049e8eaac9","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-04 19:39:10.000000000","message":"Patch Set 1:\n\nrecheck\n\nweird, looks like partially older code...","accounts_in_message":[],"_revision_number":1},{"id":"9f21620f124aec3cc91e011d1fb93a645fcf0ca5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-04 20:27:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/89106b27c4cc4bebaade73ef14750b79 : SUCCESS in 43m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d2a05dc52e5b55af27d648bf8542a319e21d7d31","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-04 20:45:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/8d48ade2e91c4f00a10e72e12ed72050 : SUCCESS in 1h 05m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f584847c5d5ae0572ef5c8f5ea3a366036a7ceeb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-04 21:24:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/513c6202c8414cd19a99224235200bec : SUCCESS in 22m 41s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/81bd26a9a3d34339aa8e55f4ba96a8da : SUCCESS in 26m 48s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/f9c311872bde48869617aa8080bc4f8b : SUCCESS in 1h 21m 37s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/0582ba79ede942c6a42205ef4cc53e60 : SUCCESS in 1h 10m 09s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/e5e6d180334c46f3b1839e1da685857d : SUCCESS in 1h 22m 56s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/ae58694d502c44459206cd811dd187d9 : SUCCESS in 46m 07s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/07ec96c6e3ef447e93a1896849e89437 : SUCCESS in 18m 59s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/62dfd52309334d3586e571acb3143de2 : SUCCESS in 1h 20m 26s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/55e2439c0dfc4cd197b0fa42b8c54c74 : SUCCESS in 1h 43m 31s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e44e341264fa4c4bb8a2282cb93a917e : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d1daa03ae144727841527ba0eaae2fd : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de5614c23aa84b5eb4bbb7b267a01143 : SUCCESS in 3m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/977ae94eb02949bbb1709b9c18e31fce : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4073d92daae64703b14cb0587686a0a4 : SUCCESS in 4m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c162269dbad94562bcb713d93cba492b : SUCCESS in 8m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c2a39f488d9449f0aa8aa7dbf075fb0b : SUCCESS in 4m 48s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/11b0b28231984bef8f4958ba12157f35 : SUCCESS in 1h 25m 12s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/94e1ef3e146c47e1a0a1ca63ed787ee7 : SUCCESS in 1h 25m 39s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c261212caffd43ebbf4ee5724844ccee : SUCCESS in 57m 59s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5189e9d3d78a4c1887c6540061c43af1 : SUCCESS in 52m 13s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e1560e5a0154416c9fadfccda3937e2c : SUCCESS in 1h 02m 23s","accounts_in_message":[],"_revision_number":1},{"id":"6c41f7d024e5dcd7a455ae6c3e9db792e87ae4b0","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-04-14 10:07:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a51b7ca5eff2c336fe6b3a2e9c261895558e17fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 10:07:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2d362c4d693155e652cfbf3f3a48aafa2cfbdb16","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 12:35:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a9645cf0901dd6589141c5f3bb4b80fd0114a36c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 12:35:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/e65d98b610034f199bdaf848e5bc8bd2 : SUCCESS in 50m 24s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/576ac088fb2c480ba4ea7298c3691082 : SUCCESS in 1h 26m 21s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/1369a82c12514262af661da027e3c4f0 : SUCCESS in 1h 35m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3ca481c944224786b2cd381befd4c0e2 : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29dd5e7b4ddb4125964c303047020da6 : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c131b57dff694fcf926dfeaa39985f78 : SUCCESS in 7m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/823b98c2b1204e7e902c5e51a26bf719 : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc036a15797c4249a2ee0b22f938effa : SUCCESS in 14m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f29598a05e8c42b288876cb8dbb18fd1 : SUCCESS in 6m 58s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d387edc3a58948ecaffb979976303b1e : SUCCESS in 1h 23m 29s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e4e11eb95038464cafcad1ddc050b735 : SUCCESS in 1h 47m 03s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f7c7c3ee2822487397d67b76b542b555 : SUCCESS in 2h 00m 27s","accounts_in_message":[],"_revision_number":1},{"id":"e519736b1d5e82e7eb7b4dffefb48a73ffbc99e2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 12:38:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/281571c88b734ef6aa124c3d059282bc : SUCCESS in 2m 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6eff57e94cae4ca9b661177b55fc5c3d : SUCCESS in 1m 25s","accounts_in_message":[],"_revision_number":1},{"id":"6676aff688364870675aca564ca53788b1cf6ec7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-14 12:44:06.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 44ef591464d03960d0bc8c322cacbb97a44968fe","accounts_in_message":[],"_revision_number":1},{"id":"7a16e24937dc7ad843961ca1294f53db86fa83bf","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-14 12:44:12.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit ab8bcf5a4f4b1a6a8ae698eda80f2a93d00aa84e","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"33d184a731458bb45d1206bdae4fb172629a76e7","revisions":{"33d184a731458bb45d1206bdae4fb172629a76e7":{"kind":"REWORK","_number":1,"created":"2020-03-04 09:26:58.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/86/711186/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/86/711186/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/711186/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/711186/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/711186/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/86/711186/1"}}},"commit":{"parents":[{"commit":"e58653c436af6c3cc9f0dfd6019c43be8e5844da","subject":"Merge \"Throw TypeError in repository enable/disable macros\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e58653c436af6c3cc9f0dfd6019c43be8e5844da"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-03-04 09:22:19.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-03-04 09:25:54.000000000","tz":60},"subject":"Ensure proper JSON in keystone bootstrap","message":"Ensure proper JSON in keystone bootstrap\n\nThis fixes the issue when keystone-manage output included\nbackslashes and/or double quotation marks which broke JSON string.\n\nChange-Id: Ifae18c407210c12745d29fc4c95dca69aeafe6a8\nCloses-bug: #1866017\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/33d184a731458bb45d1206bdae4fb172629a76e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/33d184a731458bb45d1206bdae4fb172629a76e7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
