)]}'
{"id":"openstack%2Fkolla~711227","triplet_id":"openstack%2Fkolla~master~I70e9a49eab9ce5523e09e2c7824fd44a0ec18704","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I70e9a49eab9ce5523e09e2c7824fd44a0ec18704","subject":"DNM: Test oslo.cache fix","status":"ABANDONED","created":"2020-03-04 13:33:12.000000000","updated":"2020-03-06 08:24:26.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e625a499af5817456058fc2c0c5dc45d71455737","_number":711227,"virtual_id_number":711227,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-03-05 09:22:55.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-03-04 17:12:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-03-05 09:52:52.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"date":"2020-03-05 09:39:23.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-04 14:49:49.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-03-04 17:12:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-05 09:39:23.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-03-05 09:52:52.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"416a72ef8abcb6f5a2e1b3b97d64a42d9f4c1157","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-04 13:33:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"322bf89038a30a1cc9cb07b3441492404c98039d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-03-04 13:45:46.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/d71bf68b784445e3be29d84662058765 : FAILURE in 9m 14s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/107bebf871b84a8f9986d46eb56190a0 : RETRY_LIMIT in 2m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5ee2a0c57701129d8e9340d31c59c59bc6c26114","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-04 13:50:28.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"a652d1695ec287ef0e8a9391ced5f223cc394f27","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-03-04 13:52:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"935e3838436c193e2e36720cd6b0c0e251e6c4f2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-03-04 14:20:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/0eecc0a23af341dc86c60a4e4e69aa13 : SUCCESS in 27m 51s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/94b1c93ad7614fdf860436ecb7cdd0f1 : RETRY_LIMIT in 2m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c46d8bdded92c7ba9367a545916b444e6176a7a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 14:22:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/0e1d5406544244f480bdb7f13e5f8e6c : FAILURE in 20m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b7a1f375940c409e82299b942935b922 : SUCCESS in 7m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78f4c05067444ec3a2e3118884f3b5b5 : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5dde72f46e6c460e9393e57ffe7c04a6 : SUCCESS in 4m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/db26f00dd5fb468aaca3e9f8edfeb78c : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7dcad9aefbb84a3f836f5a922bd20c0e : SUCCESS in 4m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b59fa797a5da48bdad43459e606c5b9a : SUCCESS in 8m 09s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4d60d20a285c44e6aa24c228c0640fe1 : FAILURE in 14m 41s","accounts_in_message":[],"_revision_number":2},{"id":"2ec3a87e7c2b6381d82593354c403f162b72e1c2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-04 14:48:20.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9373daa0f5f4baba1c0b0cbbb376ed7bf0d553fa","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-04 14:48:47.000000000","message":"Patch Set 3:\n\ndoh, this is oslo.cache, not requirements.","accounts_in_message":[],"_revision_number":3},{"id":"533d0d2c9eacf6267a05d5fc0b4c3ee3bcbf8926","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-03-04 14:49:49.000000000","message":"Patch Set 3:\n\n\u003e doh, this is oslo.cache, not requirements.\n\n:)","accounts_in_message":[],"_revision_number":3},{"id":"a73f36bf166908c83e6945e380c5499009a05887","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-03-04 15:37:13.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/db308c12f0db44f28fd6b19c3262daab : SUCCESS in 29m 09s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/db92f45b494a4c82bfba7c9c7fbb8020 : SUCCESS in 44m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"91933c1da82167766fb4e63303e1f1c165a06507","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 15:51:53.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/c22922afb08b49a68111eff048eee6e8 : FAILURE in 43m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/32b5377f98ba43d5b50a3897a95fb4be : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/183f01f1823c4bdaa009bee9f7d323b4 : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/78c1b7714e374faa9d6b393c5acc5448 : SUCCESS in 5m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/92210ebddec148eea35d4d5b8b75e63d : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b80221a9704e4f12a309402cb16a3732 : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57a830a1ff6a4573bab8c1a672cb17b7 : SUCCESS in 5m 55s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f833ad44a7df4dc5a8849716fd443c28 : FAILURE in 35m 41s","accounts_in_message":[],"_revision_number":3},{"id":"d19c4ffc0154ba453c893eae3027669e81966191","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-04 16:11:48.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b7faffa91bf214e7c608e660f25f382877eebbc6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-03-04 16:55:08.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/91cfc742812e4e599d44dd81209327d3 : SUCCESS in 29m 31s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/e05734372b264e81bc80cb8b7e76ef8f : SUCCESS in 37m 09s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d967d2f2768157f21fcb75cf4eca99bd85f6ead4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 17:12:13.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/2f1fcda21e144c54be5612bea75a7080 : FAILURE in 44m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/42b2493b001f47e58780028f3666c4d0 : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb81513b0381441b99c31d04c6367b0f : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c3377764fa04b8a99b4cf7e7c62444b : SUCCESS in 4m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0a774ff51b06498a9d584fe1c8d6e8e3 : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/06b33de3c2ad4564bb70a0e09167817b : SUCCESS in 4m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/656c23f4df2745c1a2b8ceede3853e8c : SUCCESS in 7m 37s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/26b13e0cc8b2421e93432a4e81e0314c : FAILURE in 30m 30s","accounts_in_message":[],"_revision_number":4},{"id":"78e09087f1ff48d23fce1738fa920862a4134b55","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-04 18:09:54.000000000","message":"Patch Set 4:\n\nStill fails:\n\nTypeError: __init__() got an unexpected keyword argument \u0027server_max_value_length\u0027","accounts_in_message":[],"_revision_number":4},{"id":"473158e285a08dbdef503b135cd0e14d08c20e7f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-05 09:22:55.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"beb81e13e5739b2385a4c5ab1e28484966c03a2a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-03-05 09:52:52.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/e8716b5b9c174191b407174493c40f4d : SUCCESS in 28m 43s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/4931f0c785b24975a20267c41e183144 : SUCCESS in 24m 56s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"39b559160a180585dc1309885cc3ee08d86b24e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 10:03:01.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/4448393f508644aeba0a6b5393409cc4 : FAILURE in 37m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/77366851943643dc95760646fafeac1f : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/287735d0c1814f40874b6881de39b252 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b7e8b213c7c44997bd7a72f24b612d78 : SUCCESS in 4m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bf85676cae7340da8d8861bb68cc351b : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/70bff129037f4a928ccc76dfc7c74c5b : SUCCESS in 4m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cb410547c47418297b9128988806951 : SUCCESS in 4m 44s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c63cafbc366b49c1a73e336746cb4e0a : FAILURE in 26m 24s","accounts_in_message":[],"_revision_number":4},{"id":"e625a499af5817456058fc2c0c5dc45d71455737","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-06 08:24:26.000000000","message":"Abandoned\n\nnot needed","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"3f07a1e0dbb33ec117ca2cfc6175f7dc665258d1","revisions":{"32d139293ac8a4e1c3220425f3c40f0cdbb6a083":{"kind":"REWORK","_number":1,"created":"2020-03-04 13:33:12.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/27/711227/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/27/711227/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/711227/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/711227/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/711227/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/27/711227/1"}}},"commit":{"parents":[{"commit":"e58653c436af6c3cc9f0dfd6019c43be8e5844da","subject":"Merge \"Throw TypeError in repository enable/disable macros\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e58653c436af6c3cc9f0dfd6019c43be8e5844da"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-04 13:33:02.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-04 13:33:02.000000000","tz":0},"subject":"DNM: Test oslo.cache fix","message":"DNM: Test oslo.cache fix\n\nTests https://review.opendev.org/#/c/711220/4.\n\nChange-Id: I70e9a49eab9ce5523e09e2c7824fd44a0ec18704\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/32d139293ac8a4e1c3220425f3c40f0cdbb6a083"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/32d139293ac8a4e1c3220425f3c40f0cdbb6a083"}]},"branch":"refs/heads/master"},"314c894d8168d246bfa8079264c42da31739f34e":{"kind":"REWORK","_number":2,"created":"2020-03-04 13:50:28.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/27/711227/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/27/711227/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/711227/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/711227/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/711227/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/27/711227/2"}}},"commit":{"parents":[{"commit":"e58653c436af6c3cc9f0dfd6019c43be8e5844da","subject":"Merge \"Throw TypeError in repository enable/disable macros\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e58653c436af6c3cc9f0dfd6019c43be8e5844da"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-04 13:33:02.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-04 13:50:24.000000000","tz":0},"subject":"DNM: Test oslo.cache fix","message":"DNM: Test oslo.cache fix\n\nTests https://review.opendev.org/#/c/711220/4.\n\nChange-Id: I70e9a49eab9ce5523e09e2c7824fd44a0ec18704\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/314c894d8168d246bfa8079264c42da31739f34e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/314c894d8168d246bfa8079264c42da31739f34e"}]},"branch":"refs/heads/master"},"80837e4de8b35bf9cc6b1a1c5d605906002414df":{"kind":"REWORK","_number":3,"created":"2020-03-04 14:48:20.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/27/711227/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/27/711227/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/711227/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/711227/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/711227/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/27/711227/3"}}},"commit":{"parents":[{"commit":"e58653c436af6c3cc9f0dfd6019c43be8e5844da","subject":"Merge \"Throw TypeError in repository enable/disable macros\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e58653c436af6c3cc9f0dfd6019c43be8e5844da"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-04 13:33:02.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-04 14:48:10.000000000","tz":0},"subject":"DNM: Test oslo.cache fix","message":"DNM: Test oslo.cache fix\n\nTests https://review.opendev.org/#/c/711220/4.\n\nChange-Id: I70e9a49eab9ce5523e09e2c7824fd44a0ec18704\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/80837e4de8b35bf9cc6b1a1c5d605906002414df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/80837e4de8b35bf9cc6b1a1c5d605906002414df"}]},"branch":"refs/heads/master"},"3f07a1e0dbb33ec117ca2cfc6175f7dc665258d1":{"kind":"REWORK","_number":4,"created":"2020-03-04 16:11:48.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/27/711227/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/27/711227/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/711227/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/711227/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/711227/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/27/711227/4"}}},"commit":{"parents":[{"commit":"e58653c436af6c3cc9f0dfd6019c43be8e5844da","subject":"Merge \"Throw TypeError in repository enable/disable macros\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e58653c436af6c3cc9f0dfd6019c43be8e5844da"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-04 13:33:02.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-04 16:11:37.000000000","tz":0},"subject":"DNM: Test oslo.cache fix","message":"DNM: Test oslo.cache fix\n\nTests https://review.opendev.org/#/c/711220/4.\n\nChange-Id: I70e9a49eab9ce5523e09e2c7824fd44a0ec18704\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3f07a1e0dbb33ec117ca2cfc6175f7dc665258d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3f07a1e0dbb33ec117ca2cfc6175f7dc665258d1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
