)]}'
{"id":"openstack%2Fkolla~711635","triplet_id":"openstack%2Fkolla~stable%2Ftrain~If73c419bf13866507a8258dca6d4ae30c5177ba9","project":"openstack/kolla","branch":"stable/train","topic":"bug/1866360","hashtags":[],"change_id":"If73c419bf13866507a8258dca6d4ae30c5177ba9","subject":"Fix pip2 install without upper constraints","status":"ABANDONED","created":"2020-03-06 14:49:36.000000000","updated":"2020-03-12 15:45:10.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"101d1e70a70f6ba038cdb31e6555858293f7d3a3","_number":711635,"virtual_id_number":711635,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-03-06 16:58:45.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"date":"2020-03-12 15:33:53.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-03-06 16:33:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-06 16:33:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-12 15:33:53.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"479d3ca25ce89218b92f6b5215c1b371c3b65b2e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-06 14:49:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"55b103b22865379a96fd31d37fcd59195b6f450d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 16:33:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/c45a971bd97749a2b198a6475ff6e200 : SUCCESS in 39m 16s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/f153279c6d05415388644877349395f5 : SUCCESS in 52m 19s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/01433eae6e9d49c18ff900c1c45f61d2 : SUCCESS in 33m 11s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/ecc6c238eeb1485a95cbde8cc76e3739 : SUCCESS in 39m 02s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/7a13804fe79a40b688e0aabd0015f554 : SUCCESS in 1h 19m 28s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/a098bde81e9b4b85ade9eac9d828b27e : SUCCESS in 1h 16m 03s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/5f2385b919ac48efadedc2394f7b73bc : SUCCESS in 1h 27m 16s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/d183e99530164c35ae0b6e38b6ae2743 : FAILURE in 1h 01m 38s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/a7337432e3844c548f905ba1c542739e : SUCCESS in 1h 36m 22s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/e9e9fe578ecd4379b33f771d5f5139a9 : SUCCESS in 53m 22s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/aa62600c14df406a993c7eb124a96897 : SUCCESS in 29m 11s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/05a2d33fa5d24e91b3acb93ebb52d02e : SUCCESS in 28m 26s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7ec1c474a61542dab281b27e1b1b2025 : SUCCESS in 1h 17m 39s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/36525643145f4fe39acba06b0118d441 : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96c1d77b259d4469a017cb31174c7071 : SUCCESS in 5m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e2fadd07feb14aa1b692c204c26c5a1b : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9669d72d84414280864144da7c741974 : SUCCESS in 4m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/98f51a71148c4e1e900f2d0069fc4c09 : SUCCESS in 5m 58s\n- requirements-check https://zuul.opendev.org/t/openstack/build/460cc35c6a174a29949ac60bc78acfa8 : FAILURE in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7735911c935a47769402c61825acd25f : SUCCESS in 7m 30s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f8ea4d5cd7284f7f8257ec2cf11aa337 : SUCCESS in 45m 26s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b0a81d538e51433ea09aceeee4244e79 : SUCCESS in 39m 07s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c520f2529e7d4978a2a03b4097eefbc2 : SUCCESS in 58m 16s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4ce31469daa342f5a8a4af3de19a3710 : SUCCESS in 56m 03s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/53b4e6dfa8204d83a80ece5973166e6e : SUCCESS in 1h 05m 23s","accounts_in_message":[],"_revision_number":1},{"id":"12c3c49ad4bd3c3faed29bd2bbb86b5fa38e6c9e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-06 16:58:45.000000000","message":"Patch Set 1:\n\nHmm, requirements says no.","accounts_in_message":[],"_revision_number":1},{"id":"a45d484e5410128155a4176053b35f31e4d17e2e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-12 15:33:53.000000000","message":"Patch Set 1:\n\nhmm, abandon?","accounts_in_message":[],"_revision_number":1},{"id":"cbad6576f5e71defa700e38c6e7c48e8452936e3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-12 15:45:10.000000000","message":"Abandoned\n\nNot working currently. gitdb2 not in GR, so can\u0027t be restricted by us.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"49caaa4033cde9b1b301bf0509baf189a7435276","revisions":{"49caaa4033cde9b1b301bf0509baf189a7435276":{"kind":"REWORK","_number":1,"created":"2020-03-06 14:49:36.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/35/711635/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/35/711635/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/711635/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/711635/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/711635/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/35/711635/1"}}},"commit":{"parents":[{"commit":"64208cdb5e4483bbf6a46a8716597a66a4d1269f","subject":"Merge \"Bump train versions\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/64208cdb5e4483bbf6a46a8716597a66a4d1269f"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-06 14:48:19.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-06 14:48:21.000000000","tz":0},"subject":"Fix pip2 install without upper constraints","message":"Fix pip2 install without upper constraints\n\nWithout this we get the following error:\n\nERROR: No matching distribution found for gitdb\u003e\u003d4.0.1\n(from gitdb2\u003e\u003d2.0.0-\u003eGitPython\u003c2.1.12,\u003e\u003d1.0.1-\u003ekolla\u003d\u003d9.0.1)\n\nThis is because gitdb has dropped support for Python 2.\n\nAdding an upper limit to requirements.txt for gitdb2 matching train\nupper constraints fixes the issue.\n\nChange-Id: If73c419bf13866507a8258dca6d4ae30c5177ba9\nCloses-Bug: #1866360\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/49caaa4033cde9b1b301bf0509baf189a7435276"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/49caaa4033cde9b1b301bf0509baf189a7435276"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
