)]}'
{"id":"openstack%2Fkolla~711636","triplet_id":"openstack%2Fkolla~master~I4d7ef79df435b540137d1eb4058f8222baf17529","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I4d7ef79df435b540137d1eb4058f8222baf17529","subject":"Deprecate Sensu images","status":"MERGED","created":"2020-03-06 15:02:32.000000000","updated":"2020-03-07 10:10:23.000000000","submitted":"2020-03-07 10:08:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"711636-1583575732456-97979107","meta_rev_id":"97370b706474fff754aaac46236c1a303caa3f78","_number":711636,"virtual_id_number":711636,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"date":"2020-03-06 17:28:40.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-07 10:08:52.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-03-06 17:57:43.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-03-06 17:32:35.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-03-07 08:44:39.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-06 15:04:37.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-03-06 17:57:43.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-07 08:44:39.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-03-07 10:08:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"794c1e275501543dc6089850989831907dff4bc9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-06 15:02:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"352ce350a2f5de567698d729dee0f6571e7f0516","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-03-06 15:04:37.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"568f2517e242e2321d4fe0b0db40f91a3962584d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-03-06 15:37:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/fdf7deb017cb4249bd1ca211fa2822d7 : SUCCESS in 33m 07s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/082fa08bc05740d69e6d1dbb2e22fe77 : SUCCESS in 33m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cfa90ee0d5eeda89290018dc7e3f2aed7e3ff262","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 16:28:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/65d6eb555717496eb29466c499332db3 : SUCCESS in 33m 40s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/eed5fbf157bf4e65b0a75abcb14bf2aa : SUCCESS in 39m 01s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/893f47d69b5f412193d570d38e2995dc : SUCCESS in 1h 20m 42s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/f26ca83379b84ccb9a9be294e5af5661 : SUCCESS in 1h 21m 14s (non-voting)\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/4c0c6f8bc55645158ec24eec6231953e : SUCCESS in 51m 17s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/303de0eeb915498fb825380dca5d8433 : SUCCESS in 19m 43s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/d77199d6e3314389b6db6f3fd2cc49cd : SUCCESS in 18m 52s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7f919aa17aac4c92af0438dcaf1ad0cb : SUCCESS in 1h 01m 46s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d33d9cb1d84b4e9a829fb0e3744b557b : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cce4cd65f23540509f26e78599c47f0e : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d9c6e65c84b4b918013f9e920ca1cc6 : SUCCESS in 4m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a207220e825643dcb419c717287c40ba : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5bf9119ae7db4c93a538d58c4b9b9ec4 : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1fa73830d854b8aae9c9928c5793455 : SUCCESS in 5m 32s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9763221a3a1043cc9fb8e792cafdff5b : SUCCESS in 22m 38s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/529750bb75c54e63a84cc607fa19e021 : SUCCESS in 27m 01s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/930dfe002f8d46c1857957ec1363f006 : SUCCESS in 1h 06m 30s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/69732b4878f34ad7b906c21673dc7762 : SUCCESS in 1h 06m 15s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e07cd43e96dc4c1ebc5a37cb43734e0d : SUCCESS in 1h 20m 53s","accounts_in_message":[],"_revision_number":1},{"id":"b49985630e26dcc2d7ae5aa81a1eed2240e44961","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-06 17:20:48.000000000","message":"Patch Set 1: Code-Review-1 Workflow-1\n\nno reno; and we still have sensu in ubuntu...","accounts_in_message":[],"_revision_number":1},{"id":"03c61e9cb60357cb3785690fc1480e57a409c07d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-06 17:28:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"73dec2d36c729fdecb6067f02158875620ab40d4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-06 17:28:40.000000000","message":"Patch Set 2:\n\n\u003e no reno; and we still have sensu in ubuntu...\n\nForgot to add. We don\u0027t deploy it in k-a, and tripleo dropped it.","accounts_in_message":[],"_revision_number":2},{"id":"a53364a7e8d3fd5dcafb2e2098ccc362918a79ac","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-06 17:32:35.000000000","message":"Patch Set 2: Code-Review+2\n\nNow it makes sense, thank you.","accounts_in_message":[],"_revision_number":2},{"id":"e8609b7d95fc1f69dc721c32e9f0ed0f707f1782","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-03-06 17:57:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/45bcf507d88f42fabf4b3495c971e55b : SUCCESS in 27m 59s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/dd058debca5d42829af2b7fa365cf3d2 : SUCCESS in 25m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fc8549ecfada830ff3376bed52318ee4b3dc138e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 18:55:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/c33c1bd34bb6430a900581e4c7c7446b : SUCCESS in 35m 27s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/2f6cf267c0c64b30a4a0e0182fedfa88 : SUCCESS in 37m 58s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/bdc3a6b66f4f4a6e967de8f974e61b56 : SUCCESS in 1h 23m 44s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/da83b2b85ed34efca0cdd951a02ad02b : SUCCESS in 1h 19m 13s (non-voting)\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/a6e7abe8a82947a2a90ab7c7c4c50538 : SUCCESS in 44m 51s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/49a53effbb68445994b149468ca5ecff : SUCCESS in 22m 27s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/d23ec7c8f92049c7a39850b165ea0e5f : SUCCESS in 19m 47s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/98a5e0e7736f4bfc9fa26f84dae2b9dc : SUCCESS in 1h 07m 49s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e20ccad77b26444393c9697e4dc56f03 : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e9eed966af2427799c43110d65831ff : SUCCESS in 7m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76a016228f7c448da86eed59e08be386 : SUCCESS in 3m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3b3fba8a78484b92aed44b57aa7c1295 : SUCCESS in 6m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7ac94f5824e341f38d25e43d171fc055 : SUCCESS in 8m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d41aabab40aa4d26b4ea4447ccdb59b1 : SUCCESS in 6m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8d068d963bba4bdf8687c42df727c463 : SUCCESS in 4m 54s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1e2beb6ca7824f71a4f885663ca9eaea : SUCCESS in 24m 54s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d0f4d33e603d4ec59218351a11b64d9d : SUCCESS in 23m 35s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ea1b5df83aa04419864c0a023dbc6da6 : SUCCESS in 1h 02m 48s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/501c43d105e748dbb36edb3e8d5ab784 : SUCCESS in 56m 08s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0001e2e127af459795d8dfaf71d046b2 : SUCCESS in 1h 12m 26s","accounts_in_message":[],"_revision_number":2},{"id":"da24876f23c07feb04189549b1861b35ebb19f03","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 20:31:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/bb4baf5d51c5402eb64cfc390560c9aa : TIMED_OUT in 3h 02m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"614849683c602d77d21f18d0e6289cdd22c7fb41","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-07 08:44:39.000000000","message":"Patch Set 2: Workflow+1\n\nalready had two +2s","accounts_in_message":[],"_revision_number":2},{"id":"58be2ac64250b3ec1f16a4d911618ac36b8e4f9a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-07 08:44:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0a28bc3411c1f48aa953cb3f140dd5768519a318","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-07 10:08:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"e1c64d988874696f5d32fe4ed54e68293750a250","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-07 10:08:52.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/b67cdb8bafdd4cc9ba0858479a2b0d45 : SUCCESS in 25m 08s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/11d46a970ccf4f258a050028413b2f99 : SUCCESS in 1h 22m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/10be44f49ada4cb297c7591a3e254353 : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc02be7d20cb4b04aa37ee8e854c7c22 : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/928a44b6f65b40e69b13f4f8de873f14 : SUCCESS in 3m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/28db6f7a9014488b844b4521d28e5b24 : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65d71a3f332244c4b7a4aaea95feaf6c : SUCCESS in 5m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0bf45e17369a42eab6a5e9938380400a : SUCCESS in 4m 43s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c7f7e041d15647d08d6ea60e968151e1 : SUCCESS in 27m 43s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/aace41de3af045f4b4ea8bb6da392d0e : SUCCESS in 1h 05m 21s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/52da06f31d584ec4abd8d2f64453994b : SUCCESS in 1h 10m 27s","accounts_in_message":[],"_revision_number":2},{"id":"97370b706474fff754aaac46236c1a303caa3f78","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-07 10:10:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8946c3d2120441b0a29013c95ae87bd6 : SUCCESS in 1m 10s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/34d40be96aa643b49c354450156cab39 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"569807738bb96a872a1d014f1272063b4f3d3cc8","revisions":{"6500ff5adc343241bd643e0d3305c27f0800bc1b":{"kind":"REWORK","_number":1,"created":"2020-03-06 15:02:32.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/36/711636/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/36/711636/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/711636/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/711636/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/711636/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/36/711636/1"}}},"commit":{"parents":[{"commit":"d4b4a4ba251ac0537067970acff15e3a0645baa9","subject":"Merge \"Unpin openstack/requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d4b4a4ba251ac0537067970acff15e3a0645baa9"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-06 15:01:20.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-06 15:01:20.000000000","tz":0},"subject":"Deprecate Sensu images","message":"Deprecate Sensu images\n\nSupport for Sensu has been dropped by CentOS/RHEL, and this was the main\nconsumer of these images.\n\nThis patch also marks opendaylight and mongodb images as deprecated on\nthe support matrix, as they have also been deprecated.\n\nChange-Id: I4d7ef79df435b540137d1eb4058f8222baf17529\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6500ff5adc343241bd643e0d3305c27f0800bc1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6500ff5adc343241bd643e0d3305c27f0800bc1b"}]},"branch":"refs/heads/master"},"569807738bb96a872a1d014f1272063b4f3d3cc8":{"kind":"REWORK","_number":2,"created":"2020-03-06 17:28:08.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/36/711636/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/36/711636/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/711636/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/711636/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/711636/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/36/711636/2"}}},"commit":{"parents":[{"commit":"d4b4a4ba251ac0537067970acff15e3a0645baa9","subject":"Merge \"Unpin openstack/requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d4b4a4ba251ac0537067970acff15e3a0645baa9"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-06 15:01:20.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-06 17:27:59.000000000","tz":0},"subject":"Deprecate Sensu images","message":"Deprecate Sensu images\n\nSupport for Sensu has been dropped by CentOS/RHEL, and this was the main\nconsumer of these images.\n\nThis patch also marks opendaylight and mongodb images as deprecated on\nthe support matrix, as they have also been deprecated.\n\nChange-Id: I4d7ef79df435b540137d1eb4058f8222baf17529\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/569807738bb96a872a1d014f1272063b4f3d3cc8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/569807738bb96a872a1d014f1272063b4f3d3cc8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
