)]}'
{"id":"openstack%2Fkolla~711637","triplet_id":"openstack%2Fkolla~master~I3e12ab51e9f01d383268d75ce72903686564ada6","project":"openstack/kolla","branch":"master","topic":"bp/centos-rhel-8","hashtags":[],"change_id":"I3e12ab51e9f01d383268d75ce72903686564ada6","subject":"CentOS 8: Drop trickle install to build freezer","status":"MERGED","created":"2020-03-06 15:09:10.000000000","updated":"2020-03-25 13:39:09.000000000","submitted":"2020-03-25 12:30:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"711637-1585139458198-a663f06a","meta_rev_id":"ce5841cae8d2b47bcd4c14196b2c47ae4f7bbc33","_number":711637,"virtual_id_number":711637,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"date":"2020-03-23 12:44:32.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-25 12:30:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-03-23 13:10:06.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2020-03-25 10:55:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2020-03-23 12:51:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-03-23 15:34:49.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2020-03-25 10:55:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-23 12:51:19.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2020-03-23 13:10:06.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-23 15:34:49.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-03-25 10:55:55.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-03-25 12:30:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"35af68b66f75be85a334c1f1e6a8376b82844e56","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-06 15:09:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0dd7713312eef4f985a208e1b5522cd29f1bb318","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-03-06 15:43:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/7b973068967548d2b16d8da90ffb5fd4 : SUCCESS in 27m 48s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/720c7c5193d5487daaa45cf0ac3b2d5c : SUCCESS in 33m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"96112e406056bbbbd957ae7f4fc38c957cb3cbdc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 16:42:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/be8c7e9bfc0043f9bb68e5cde66c0f62 : SUCCESS in 31m 53s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/50627132e9dd4f09984c765336b70c1b : SUCCESS in 35m 15s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/4a1ee27d20904d6d8a0d611c168347cd : SUCCESS in 1h 25m 08s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/19c83af744e148cbbca1504fdf6915ae : SUCCESS in 1h 30m 41s (non-voting)\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/0d751108bafe4d37a430566c51b11d18 : SUCCESS in 52m 31s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/730e8567068e49158c99d8e0308bb36c : SUCCESS in 39m 26s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a531572b558a4786bcdcc584cb97c1d0 : SUCCESS in 26m 32s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/1cb1b309d3364ae08497027d317dbe62 : SUCCESS in 1h 21m 52s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d2c01868ee484a828a23809e232a8b48 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca765f8adfc64fc7ae1f29b5ccd058d1 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc2cc3798e854379b7e3444ff64da586 : SUCCESS in 4m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d7734ac57da74342a3a69f38b2161d01 : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3bfdb108e5c14d2e887a16577bce1fe3 : SUCCESS in 4m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f0f18fd7f0b4af1adc5e45d9e213d68 : SUCCESS in 8m 30s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/32673457102548ab92c436a257fdd37d : SUCCESS in 24m 01s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9152d879820b47daae68871536eb4bff : SUCCESS in 41m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2713b399aa114db691aba705495b50ee : SUCCESS in 1h 04m 36s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4e8696e688114166a86a7ca0a5975e0d : SUCCESS in 1h 00m 35s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/de8c4d95ab08483f9d8a4813c3a6ad05 : SUCCESS in 1h 14m 50s","accounts_in_message":[],"_revision_number":1},{"id":"d1d46aeca97a4d2fffdd92559c72f3055b7170c7","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2020-03-06 16:53:30.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"230971de81b0b9b109763dcad316157e7a40f9e6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-06 17:25:40.000000000","message":"Patch Set 1: Code-Review-1\n\ntrickler is to limit bandwidth, sad to lose it, could opstools build it too? Should not be too hard. Otherwise requires a reno.","accounts_in_message":[],"_revision_number":1},{"id":"a06886071ec6e8f0107501aebdd20b440e5ddb2b","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2020-03-06 17:53:10.000000000","message":"Patch Set 1:\n\nI\u0027ve filed a bug/feature request to build it in epel8. https://bugzilla.redhat.com/show_bug.cgi?id\u003d1811150\n\nIn worst case, we can build it in opstools, yes.","accounts_in_message":[],"_revision_number":1},{"id":"21c437b49d90e9f6da340e858780a92a4c7d26e0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-06 17:57:13.000000000","message":"Patch Set 1:\n\nThanks, Matthias.","accounts_in_message":[],"_revision_number":1},{"id":"436f342cc9fd8e0150a4b926b624444cdbe23453","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 18:14:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9a9ed07e644243caa833d1fc359aec85 : TIMED_OUT in 3h 01m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"632a7222ec0c5ba0069d9683110728daf3bb0187","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-23 12:44:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ff6825ff1731ff60bdd477bbad7d57c81dffb0e1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-23 12:44:32.000000000","message":"Patch Set 2:\n\nNo movement on the trickle ticket. Added a reno, we can drop until a package is available.","accounts_in_message":[],"_revision_number":2},{"id":"d59e8bc6064551f4a9a17e60532f553f43b89dc1","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2020-03-23 12:51:19.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"c8542f9b98bcc23bb905b347656c89a8020712b9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-03-23 13:10:06.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/f82d06066c164442b94f18b618b8b94f : FAILURE in 14m 23s\n- tripleo-build-containers-centos-8-master https://review.rdoproject.org/zuul/build/dc7705b663c14207854a6c80ad11b2eb : SUCCESS in 25m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"260f9fc78b4013ed86ef65d40f933b3dd93793db","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 13:29:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/887586c24e8d46688c160ac90710e288 : SUCCESS in 44m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e612f2719263308049252ff7db133cfc06540772","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 14:10:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/9922c7adf11e4fcd9177201445bc5c9a : SUCCESS in 33m 05s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/37b01ce7fd52410a9fcf476f14555494 : SUCCESS in 27m 44s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/14ca04563812480985db75dfcbcf6619 : SUCCESS in 1h 20m 58s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/55f087e6a76545a392c80db974e16c9d : FAILURE in 53m 14s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/52d4eaeba09642afa9137d6f3e9710d3 : SUCCESS in 1h 21m 22s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/c6537708cb6a403ba18a78488f40c2da : SUCCESS in 46m 22s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/4509f04287684e74a64ee068674ca171 : SUCCESS in 20m 16s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/9b05b2d7fdf74c079675e25ec6fa4d7d : SUCCESS in 17m 03s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5657160ff46c45b291ca94976442ebc4 : SUCCESS in 1h 02m 52s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7ea756e010cc45ae80dac68659b6950f : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c38dc99d7df247e4801b358b624d744b : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b8a183f17d7470790b578dbd2579104 : SUCCESS in 3m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/15044a7988954bbe8c29f1875ee08146 : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d3a46dd8eefd499bbd199a1531cb40e7 : SUCCESS in 4m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d1572b38f374e448b44d607d2f8aca3 : SUCCESS in 7m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/48fbc36e9ac44bc8bc21b740b879b41d : SUCCESS in 5m 02s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bbc023cedb3f4396bf096c1a57f54667 : SUCCESS in 23m 14s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/78c08573c9124b56b80220a672af6650 : SUCCESS in 26m 29s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/078829f9e0b34729b753e1ffe3bdaf4c : SUCCESS in 1h 00m 52s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/07588f47265640ba884d419750d5d3d3 : SUCCESS in 1h 02m 07s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c920029b73bb4903b8473f7a38f339ed : SUCCESS in 1h 08m 17s","accounts_in_message":[],"_revision_number":2},{"id":"1c6d286335b63cc2e6d363a169d406b8113728ed","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-23 15:34:49.000000000","message":"Patch Set 2: Code-Review+2\n\nno crying","accounts_in_message":[],"_revision_number":2},{"id":"0f1a772b45944043ea53ea73771425790eac1fbe","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-03-25 10:55:55.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f361dc3ae67714505afe626b65fb3d8a29e83a80","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 10:56:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cc0bdb326e4d3b938fd5ceb9d11009674682210c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 12:30:57.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/58065470966a41b2a2bc15206f58e1bd : SUCCESS in 27m 30s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/312c5236996e4e00ab0acfe9ab6a3f53 : SUCCESS in 1h 15m 27s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/614598ce9b9548fdac4a797e8cf2a12c : SUCCESS in 1h 15m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8a81b3e2d3e84f77aa377ea8b9bb7426 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1dbc136bbb7422f916a1a54496144b4 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ba51514314a4a90bf37b236db79e60a : SUCCESS in 3m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/27456be91b424876b2cfeb31c6a2fac0 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/098fb99b4654427a9541ccacb9a8e88b : SUCCESS in 5m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/06904f4ed2484cff862cc14d3ccd4aed : SUCCESS in 5m 13s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4e5624391bd2406d85854c560b21f0e5 : SUCCESS in 27m 00s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9141b3ea7530419ba2b884e6b6ad3c1c : SUCCESS in 1h 02m 35s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/90bc9a1965974a609c052a8068e219c8 : SUCCESS in 1h 10m 57s","accounts_in_message":[],"_revision_number":2},{"id":"a8b5001b4a5d3e9dace2a61712088e9e8b0be99a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 12:30:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"43e7673a3d219020d3cd5ff6130711c74f93785f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 12:35:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28be737e7605467db3387dbab5cd2d30 : SUCCESS in 1m 18s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/98e6a55cc2f54bfbbe272c85ea1f941b : SUCCESS in 1m 36s","accounts_in_message":[],"_revision_number":2},{"id":"9b49b54333207b550449afea7f8387377b5fd563","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-25 13:39:09.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 9d5a8fcd73e2b177609be64840757848f57a39cd","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"548144fc3c2564c9c44a3a42372a1f3b8f2fef34","revisions":{"223f30ef1bfe2fbf0472b21c3474882d42446b7c":{"kind":"REWORK","_number":1,"created":"2020-03-06 15:09:10.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/37/711637/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/37/711637/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/711637/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/711637/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/711637/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/37/711637/1"}}},"commit":{"parents":[{"commit":"d4b4a4ba251ac0537067970acff15e3a0645baa9","subject":"Merge \"Unpin openstack/requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d4b4a4ba251ac0537067970acff15e3a0645baa9"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-06 15:07:26.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-06 15:07:26.000000000","tz":0},"subject":"CentOS 8: Drop trickle install to build freezer","message":"CentOS 8: Drop trickle install to build freezer\n\nTrickle was previously installed from EPEL7, but is not available in\nEPEL8. It does not appear critical to the functioning of freezer, so\ndrop it to allow freezer images to build for CentOS 8.\n\nChange-Id: I3e12ab51e9f01d383268d75ce72903686564ada6\nPartially-Implements: blueprint centos-rhel-8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/223f30ef1bfe2fbf0472b21c3474882d42446b7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/223f30ef1bfe2fbf0472b21c3474882d42446b7c"}]},"branch":"refs/heads/master"},"548144fc3c2564c9c44a3a42372a1f3b8f2fef34":{"kind":"REWORK","_number":2,"created":"2020-03-23 12:44:01.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/37/711637/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/37/711637/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/711637/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/711637/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/711637/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/37/711637/2"}}},"commit":{"parents":[{"commit":"39820c4abb83c55e66e7760c5844754fc07e4599","subject":"Zun: add zun-cni-daemon image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/39820c4abb83c55e66e7760c5844754fc07e4599"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-06 15:07:26.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-23 12:43:47.000000000","tz":0},"subject":"CentOS 8: Drop trickle install to build freezer","message":"CentOS 8: Drop trickle install to build freezer\n\nTrickle was previously installed from EPEL7, but is not available in\nEPEL8. It does not appear critical to the functioning of freezer, so\ndrop it to allow freezer images to build for CentOS 8.\n\nChange-Id: I3e12ab51e9f01d383268d75ce72903686564ada6\nPartially-Implements: blueprint centos-rhel-8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/548144fc3c2564c9c44a3a42372a1f3b8f2fef34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/548144fc3c2564c9c44a3a42372a1f3b8f2fef34"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
