)]}'
{"id":"openstack%2Fkolla~711683","triplet_id":"openstack%2Fkolla~stable%2Fstein~I9032e43bffa3a3de6c33a4ab59f65bafb0129dd5","project":"openstack/kolla","branch":"stable/stein","topic":"fetch_fernet_python3-stable/train-stable/stein","hashtags":[],"change_id":"I9032e43bffa3a3de6c33a4ab59f65bafb0129dd5","subject":"Adapt fetch-fernet-tokens script to Python 3","status":"MERGED","created":"2020-03-06 17:47:34.000000000","updated":"2020-03-10 12:15:12.000000000","submitted":"2020-03-10 12:12:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"711683-1583842376719-9c168db8","meta_rev_id":"90a47589737d1e6fa469bf0e2e3a8f4d93b3f943","_number":711683,"virtual_id_number":711683,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-10 12:12:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2020-03-09 11:03:19.000000000","_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":2,"date":"2020-03-10 11:06:41.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":1,"date":"2020-03-10 11:06:41.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2020-03-06 17:47:34.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-06 17:47:34.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-03-09 11:03:19.000000000","updated_by":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"reviewer":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"state":"REVIEWER"},{"updated":"2020-03-10 11:06:41.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-03-10 12:12:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8bc86064a778d6de2181e329485cc5f1ed9f75f7","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-06 17:47:34.000000000","message":"Patch Set 1: Cherry Picked from branch stable/train.","accounts_in_message":[],"_revision_number":1},{"id":"afcefe3b0b073e8e1135e80ca65002b1a17818b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 19:22:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/ebd27d2077f3419d89fd1ddc9835a8a9 : SUCCESS in 51m 40s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/6e008e636ab442bc83ef6604632c3dd3 : SUCCESS in 59m 18s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/94ddba4d713d433ea41289526d2782c3 : SUCCESS in 1h 26m 56s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/19c3d18041664ffb8dac79e862bcbbf1 : SUCCESS in 1h 15m 10s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/c64a958fcda646fb8150e4fcb14d6648 : SUCCESS in 1h 27m 24s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/5a3379e2b29f4855b4d88666fc6a8d74 : SUCCESS in 1h 01m 12s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/addf78354ffc48708c6ff878e6065f26 : FAILURE in 48m 42s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/6cd9c5a28a7b433b8ff56d24b34a5fc7 : SUCCESS in 59m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6a222a869f1d4879a2cb505fc6f67e30 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85f205e2f8f74fc3a28fa13230f6c8eb : SUCCESS in 4m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bb475f02fcea4b06a26c827b7676864e : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b1853216c83a4de78fc4407fff71bb6b : SUCCESS in 6m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ead5c6a04a8f4dd2abe3104ff1629445 : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/969d765eefee43dab8e6f7df1a0ced44 : SUCCESS in 4m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bf1e25f3a9df4b6d90cf9996707dcbed : SUCCESS in 4m 57s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/219a110731ea425bac8f33c5865bf917 : SUCCESS in 27m 33s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/abb4a8df307d435eb51e85d7a8ac8685 : SUCCESS in 32m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/10289c3b333e4da18acb0d81625a62ae : SUCCESS in 56m 55s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/64c90afbe5864f1fabe7335bdffc4fa5 : FAILURE in 42m 28s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a6b1a24de06c4ce6a79523faeaadd489 : SUCCESS in 1h 07m 55s","accounts_in_message":[],"_revision_number":1},{"id":"b4ceb4762273cf20c48009a07dfae2f3dad6c75a","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2020-03-09 11:03:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e8373621720f7416d20a3654009e6e03756cde01","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-10 11:06:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"26308ff6daee7588f31c245e942db269d4702482","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 11:07:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"abdfc66e23da8ef1a70943a8416fe3fd3706afc8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 12:12:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/3b5999c25a2a43f38d86450671df6fef : SUCCESS in 51m 33s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/458290f5097e4b9ba8655b5906684d84 : SUCCESS in 50m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e095246e03a94c95b9a7ae2faa22b3e8 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10d4dacebf1c4ee58f527871fa88853c : SUCCESS in 3m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1db3a07dace24699a881709d0b2bab71 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f77b3ef753944537b81136183c2aa462 : SUCCESS in 6m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b9b4f85aa3124ca9b56a9e7673c6cb0c : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4dbec0f751249d8942069da4c1fad36 : SUCCESS in 3m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9216bdcaa8ab4ec1a283e5766c1b02bb : SUCCESS in 4m 57s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8d70b09731c74f6b9ebf73d932e6bf19 : SUCCESS in 27m 59s","accounts_in_message":[],"_revision_number":1},{"id":"196274d04b802dfa7c2c7c615cfd3da37542e7b0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 12:12:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3c78de05f111ad50793f08991cf5d446f9165ee9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 12:15:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60e37597bfda4d0e84fc0fd1a68a2a27 : SUCCESS in 1m 26s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fb7b1259799046be83b15d773e9af7f1 : SUCCESS in 1m 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a83173da818e8a4b14812dd111f017b34beb6abc","revisions":{"a83173da818e8a4b14812dd111f017b34beb6abc":{"kind":"REWORK","_number":1,"created":"2020-03-06 17:47:34.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/83/711683/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/83/711683/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/711683/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/711683/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/711683/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/83/711683/1"}}},"commit":{"parents":[{"commit":"d53ead9421a288137fba8e20b22844d1cdce6506","subject":"Fix kolla_toolbox_pip_virtualenv_packages override","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d53ead9421a288137fba8e20b22844d1cdce6506"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-01-09 17:32:32.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-03-06 17:47:34.000000000","tz":0},"subject":"Adapt fetch-fernet-tokens script to Python 3","message":"Adapt fetch-fernet-tokens script to Python 3\n\nBackport: train\nCloses-Bug: #1859047\nChange-Id: I9032e43bffa3a3de6c33a4ab59f65bafb0129dd5\n(cherry picked from commit 902f16cb1a8efc44da8669a6707d93d47160ae9d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a83173da818e8a4b14812dd111f017b34beb6abc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a83173da818e8a4b14812dd111f017b34beb6abc"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
