)]}'
{"id":"openstack%2Fkolla~711891","triplet_id":"openstack%2Fkolla~stable%2Ftrain~I46d567f7ddadf4a40a4af9652248b50b3ac7be80","project":"openstack/kolla","branch":"stable/train","hashtags":[],"change_id":"I46d567f7ddadf4a40a4af9652248b50b3ac7be80","subject":"Clean up collectd dockerfile","status":"MERGED","created":"2020-03-09 10:13:00.000000000","updated":"2020-03-12 19:25:29.000000000","submitted":"2020-03-12 19:25:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"711891-1584041129164-9c168db8","meta_rev_id":"794f028eec686797b93a46a2167fbebe298c1bc1","_number":711891,"virtual_id_number":711891,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-12 19:25:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2020-03-12 15:34:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2020-03-12 15:34:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-03-12 15:34:34.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-03-12 15:34:39.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-09 13:07:00.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2020-03-10 13:32:50.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-03-12 15:34:39.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-03-12 19:25:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"553cc80cb406bf79310f8edb6800645e13d3af02","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-09 10:13:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2e40cb3af5151e652c15bd423eb78f1f2f341962","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-09 10:26:54.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"c0df2842f53ead512222850e3d8cebfc04e75ba5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-09 11:15:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a7b92f5c8051b42d721d8ecd5b8f035b1e165c66","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-09 11:18:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"78447567d5445522b13044aa305214d13c0c4760","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-09 13:00:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/ad01719e52cc4b59bb2357944926934a : SUCCESS in 1h 01m 52s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/85861c46c1124c6e8a12684ab1c9d843 : SUCCESS in 56m 11s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/603e2d492f4a4452bf5606f68e3b2082 : SUCCESS in 23m 57s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/b551ed509c874e679cdf2970609de1f1 : SUCCESS in 29m 38s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/76bf3f1d3bd94dc9895b17d089203767 : SUCCESS in 1h 31m 59s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/c11282befe2548da84d2edc9c123cd1e : SUCCESS in 1h 10m 56s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/e347152afa984068943419ba175821bc : SUCCESS in 1h 30m 19s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/d74e1f167a3c4d7796ea8974ea481033 : FAILURE in 49m 53s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/053b96a4ec5c45b9a305be31634639c0 : SUCCESS in 1h 37m 09s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/dc96bbdb85a34e33820b54b8a4936c51 : SUCCESS in 57m 33s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6f69c06dd405447fb7674f4c3e722982 : SUCCESS in 23m 29s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/9bf3ecf7487c4501b1302391a5fa9136 : SUCCESS in 18m 25s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d8bc3bce39724b9a88f75810e5dd2baf : SUCCESS in 1h 12m 58s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eac170b6fe3940c9b2de1138830018a6 : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edbddf92536842b0976f4350faf7a0d7 : SUCCESS in 4m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2eb37577f5184042ac2f31ab19c5273d : SUCCESS in 3m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/839dc60247c342209a4d37c7283780f2 : SUCCESS in 3m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/539f919648fa401cb57804ba6603ef99 : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84971bbeaec94d54b9f028ce4870b6af : SUCCESS in 4m 54s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/02bf983943a643f29063194eba144ace : SUCCESS in 22m 02s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/fec615d57e1b4570a6ca5820e05a3f54 : SUCCESS in 33m 14s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/55f7e4e3712f4a659a071656ef649b43 : SUCCESS in 53m 24s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/968ba6cb0c4d4b67ba1d4d17c0b557d9 : SUCCESS in 54m 14s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ba613c9932ed4266bcb63c70b3a32d7f : SUCCESS in 1h 02m 02s","accounts_in_message":[],"_revision_number":2},{"id":"012bb67afbc5cb1298fcb54fa24196b10fbed0e2","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2020-03-09 13:06:12.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nThank you, I\u0027d say it is safe for now to backport this. Since I can not accept collectd in downstream to have regressions (e.g in terms of missing functionality), I\u0027d be putting that back asap.","accounts_in_message":[],"_revision_number":1},{"id":"3533ba3591e87868c833f6ac5a8393681b0a4881","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2020-03-09 13:07:00.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"5ad5fdda4a8917f8e19d3665a7d1d5665089bf29","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-09 20:39:04.000000000","message":"Patch Set 2: Code-Review+2\n\nless full but safe","accounts_in_message":[],"_revision_number":2},{"id":"6bc83db843baf19d95ba157741aa78263951e104","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-03-10 13:32:50.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"213922658e5bb238f3c43e2b70efab12ed5af59b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 13:33:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"eb512626d27d118da2e34dcf70256362551b2984","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-10 15:10:39.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/f41baa819fb7483aa489bcf9d79fc277 : SUCCESS in 58m 14s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/69e81c88e7934f679a73a240a7977591 : FAILURE in 33m 48s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/762b2dc4f11b474491dac8e6f9afa488 : SUCCESS in 1h 33m 26s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/733065e7ef424df185686eec0a9f7173 : FAILURE in 31m 58s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/d91e1814877a4636a741807d14174d03 : SUCCESS in 1h 24m 21s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/ce0ac9ed152d4a40aac82f48f7d2dedf : SUCCESS in 57m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bf5dc97d9a054ce4b2cb275558845240 : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17f59cc9bd3e4eca97b435c71144fb61 : SUCCESS in 6m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e72c6227a1fe4283a3b5718448c735a0 : SUCCESS in 3m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2af69917c1ae4dbfafbb9d861dcfde31 : SUCCESS in 4m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c9031a1ba1254d2fb205910ff565081e : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72837e272d094db9b6c7aed9d5080a78 : SUCCESS in 5m 33s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0a8d7dadd5ed49bca0d1b7021a92017b : FAILURE in 35m 33s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c187473f339e4b629a68f10f3f36ec21 : FAILURE in 27m 23s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/fbae8128197447d2882aad3429e15e72 : FAILURE in 22m 49s","accounts_in_message":[],"_revision_number":2},{"id":"b9991431353fbfe3b56260fd518c01a19ebe1a92","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-12 15:34:34.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"b4559cfac8c38514fc8b4689d4a4a7286356bcc1","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-12 15:34:39.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"408e894064f732d844240db4361cc5dfa16a380b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 17:22:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/93dab9aec07f4cf58f532fe286adcd84 : SUCCESS in 54m 37s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/e8d31f9c8e7f4201b491595d857ad14e : SUCCESS in 1h 12m 40s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/9a16031af273419db48ea21d02dc45bd : SUCCESS in 25m 09s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/b1d514eadd6c428b9266b4d029b438ad : SUCCESS in 30m 37s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/593507053db54fe7967d8e465f2f73b2 : SUCCESS in 1h 22m 31s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/669ad54992404fc38d9c1cbdd81b38b1 : SUCCESS in 1h 07m 04s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/7f8158d7562945b183b1efec9f07d6b4 : SUCCESS in 1h 25m 22s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/73c7bf638b254610b32b0aa566a6b6e2 : FAILURE in 51m 10s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/2ef8197328d540d886c24e0848b135e2 : SUCCESS in 1h 23m 47s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/30f91892e38843e49a341573cfe0155c : SUCCESS in 45m 29s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/a0e222320ace4f29997f78c7f6d2bf3a : SUCCESS in 28m 58s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b36d4ac846e14fbaa948e4f960951af6 : SUCCESS in 21m 41s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3580e1bcdfa94665a08b8fbb08a05c61 : SUCCESS in 1h 01m 45s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b50b76c5512b4c7bbdf5bacfd977726d : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73d7d15ae6934dd3b022252a2dcfffa7 : SUCCESS in 7m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/77ea9c71ea0341beb134cce50d629d80 : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f074555ffd441f2bcfe7a279b78e293 : SUCCESS in 7m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/958da73561184f33b2d05afb00183867 : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07cdb47276694403966b07fd9dbafa76 : SUCCESS in 5m 45s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ca3ac5a69bd340c7bde2b9d47471a3c7 : SUCCESS in 26m 00s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8d97e66f3e234330b961a0d7a4314689 : SUCCESS in 48m 59s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7cf7ccfd140445f3904e6f87ec069a20 : SUCCESS in 1h 07m 43s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/88e264a72e2e41bdb7a4366b77a4f54c : SUCCESS in 58m 26s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/505441c47d8d4a5b820d19877758d5a0 : SUCCESS in 1h 05m 00s","accounts_in_message":[],"_revision_number":3},{"id":"ef3b3b401248f659471ad592845814475bb25c98","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 17:41:47.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"31782fb73ebc7dff33d07f950a6a36b6553f6760","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 19:25:28.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/cc7768cc124a46ccaf32898521125ae8 : SUCCESS in 57m 45s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/d2ac0ad70d5248879f749aa7d11a031b : SUCCESS in 38m 41s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/2a02cff68db54ce5af306ab646ae8763 : SUCCESS in 1h 29m 13s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/25fea21fa0024bc68940b338bcf7614a : SUCCESS in 1h 14m 33s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/0e938c3bb6b64ce48339102a555e21fc : SUCCESS in 1h 23m 10s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/24097d99e1fa4b08af7b2c95f4996fa6 : SUCCESS in 49m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3437cef747204920978845ce1be324ad : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5b61e5f3d3c4bbcb3d9f31757612dd0 : SUCCESS in 5m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/11c7aacc4b814192bcc0bc8df2de486f : SUCCESS in 6m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/842c47314c7b4227ac80a43e50e5ed09 : SUCCESS in 3m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9801c6a070d245cea9d0b225de43b857 : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00f80bb652404e6d92e36db30cc6371c : SUCCESS in 9m 08s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c56b1f836a454efeaeb599cf4a5ec3fd : SUCCESS in 32m 11s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6e2d9cfcefd04cb0b6ef6744b62e80ea : SUCCESS in 1h 00m 16s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b09cccff86e74c0ba5281489f2a88215 : SUCCESS in 1h 06m 50s","accounts_in_message":[],"_revision_number":3},{"id":"794f028eec686797b93a46a2167fbebe298c1bc1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 19:25:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c55fbc3265ab15bb71973c8cd7be08f5353b1fc1","revisions":{"7b91ffca05f05186a74cc90064837714c02ac3bc":{"kind":"REWORK","_number":1,"created":"2020-03-09 10:13:00.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/91/711891/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/711891/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/711891/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/711891/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/711891/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/711891/1"}}},"commit":{"parents":[{"commit":"98d1f244873bec0f77df2b20492b80ac4f263c52","subject":"Merge \"Use StorageSIG repos for Ceph in CentOS8\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/98d1f244873bec0f77df2b20492b80ac4f263c52"}]}],"author":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2020-03-06 07:32:10.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-09 10:12:45.000000000","tz":0},"subject":"Clean up collectd dockerfile","message":"Clean up collectd dockerfile\n\nfor CentOS 7 and CentOS 8.\n\nChange-Id: I46d567f7ddadf4a40a4af9652248b50b3ac7be80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7b91ffca05f05186a74cc90064837714c02ac3bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7b91ffca05f05186a74cc90064837714c02ac3bc"}]},"branch":"refs/heads/stable/train"},"152f1bfaa3fc2d059ca71928459b1db10bf58222":{"kind":"REWORK","_number":2,"created":"2020-03-09 11:18:56.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/91/711891/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/711891/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/711891/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/711891/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/711891/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/711891/2"}}},"commit":{"parents":[{"commit":"98d1f244873bec0f77df2b20492b80ac4f263c52","subject":"Merge \"Use StorageSIG repos for Ceph in CentOS8\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/98d1f244873bec0f77df2b20492b80ac4f263c52"}]}],"author":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2020-03-06 07:32:10.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-09 11:18:06.000000000","tz":0},"subject":"Clean up collectd dockerfile","message":"Clean up collectd dockerfile\n\nfor CentOS 7 and CentOS 8.\n\nChange-Id: I46d567f7ddadf4a40a4af9652248b50b3ac7be80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/152f1bfaa3fc2d059ca71928459b1db10bf58222"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/152f1bfaa3fc2d059ca71928459b1db10bf58222"}]},"branch":"refs/heads/stable/train"},"c55fbc3265ab15bb71973c8cd7be08f5353b1fc1":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-03-12 15:34:34.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/91/711891/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/711891/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/711891/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/711891/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/711891/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/711891/3"}}},"commit":{"parents":[{"commit":"ca8b61516401d61d90e2940fa130c8f55a8b6e45","subject":"openstack-base: pin setuptools \u003c 46 to get horizon working","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ca8b61516401d61d90e2940fa130c8f55a8b6e45"}]}],"author":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2020-03-06 07:32:10.000000000","tz":60},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-03-12 15:34:34.000000000","tz":0},"subject":"Clean up collectd dockerfile","message":"Clean up collectd dockerfile\n\nfor CentOS 7 and CentOS 8.\n\nChange-Id: I46d567f7ddadf4a40a4af9652248b50b3ac7be80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c55fbc3265ab15bb71973c8cd7be08f5353b1fc1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c55fbc3265ab15bb71973c8cd7be08f5353b1fc1"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
